)]}'
{"id":"openstack%2Fironic-ui~852702","triplet_id":"openstack%2Fironic-ui~master~I40857bea1c83704106277bbb858eb35ee39f360f","project":"openstack/ironic-ui","branch":"master","topic":"angular_updates","attention_set":{},"removed_from_attention_set":{"29313":{"account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"last_update":"2022-08-19 08:53:45.000000000","reason":"Change was submitted"},"6914":{"account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"last_update":"2022-08-18 20:36:30.000000000","reason":"\u003cGERRIT_ACCOUNT_6914\u003e replied on the change","reason_account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"}},"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2022-08-19 08:53:45.000000000","reason":"Change was submitted"},"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2022-08-19 08:35:26.000000000","reason":"\u003cGERRIT_ACCOUNT_23851\u003e replied on the change","reason_account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2022-08-19 08:53:45.000000000","reason":"Change was submitted"},"15519":{"account":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"last_update":"2022-08-18 18:25:25.000000000","reason":"\u003cGERRIT_ACCOUNT_15519\u003e replied on the change","reason_account":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}}},"hashtags":[],"change_id":"I40857bea1c83704106277bbb858eb35ee39f360f","subject":"Migrate to AngularJS v1.8.2","status":"MERGED","created":"2022-08-10 07:06:05.000000000","updated":"2022-08-19 08:54:40.000000000","submitted":"2022-08-19 08:53:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852702-angular_updates","meta_rev_id":"852cb2b29ebbee3d1267d3f122f67652319f2d44","_number":852702,"virtual_id_number":852702,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-19 08:53:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"recommended":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":1,"date":"2022-08-18 20:36:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":1,"date":"2022-08-18 20:03:44.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":2,"date":"2022-08-19 08:35:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-18 18:25:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":1,"date":"2022-08-19 08:35:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"reviewers":{"REVIEWER":[{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-10 07:42:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-10 11:50:23.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"state":"REVIEWER"},{"updated":"2022-08-18 17:48:04.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2022-08-18 17:48:59.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2022-08-18 17:48:59.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2022-08-18 17:54:02.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2022-08-18 20:03:44.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"}],"messages":[{"id":"f8a03ce6c8c4d963379e1e15de699951ad2d43f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-08-10 07:06:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"83e3395b3ff3acbe13877349ba80e10ef22bad00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 07:42:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8c928b526e8c4aaf81ffdab9a51fe109\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be55c775f4ce46cb883e34ae718fb153 : SUCCESS in 4m 08s\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/ec8ab6cb629b4ad8ad906f1c276197a5 : SUCCESS in 6m 28s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/a0eb2994dee640ee81e03e78f5bd39c9 : FAILURE in 8m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dde95e867b384b0a8c65e6e9a18d49ba : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/844ceff1ad6f4f77b7efeedc971d0327 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0dfcd08848594458bc26a44efb3889c6 : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0f30e44b4eab441db39ec6f9d798385b : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa9cc38696e4472fa50ed71342d2f570 : SUCCESS in 10m 40s\n- ironic-ui-integration-tests https://zuul.opendev.org/t/openstack/build/2107df9530ba4bb197299deadebba7ee : SUCCESS in 35m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ad9f2e5784a06937c47a385b7dd18fbc3090a4b1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-08-10 11:50:28.000000000","message":"Topic set to angular_updates","accounts_in_message":[],"_revision_number":1},{"id":"6725a565f64c251ed3bc8cac2b138e1d91ff2aae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-08-18 10:07:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1708e06de2dd5a10dc72a424f10ca2ee75969916","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 10:54:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dee1b6c3582f4329bb72759d48e8f7f3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7cf0a40aa667455c88415495c7b0a093 : SUCCESS in 4m 52s\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/a3517de23b504b3b9c96921bd1beb5de : SUCCESS in 5m 21s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/f59459d6c45f4b0797b416afa1a8d681 : FAILURE in 11m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b33e3e95b3b540149505a40181abb7d7 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/943d8d24079447a89b0ee7f7e39dc526 : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb9fa6a9024b4b8fa976b138603168fa : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f4f3a4e3578b45b4a02f8f2036c9033c : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dea24158e28e4bb485d4a37106909cd6 : SUCCESS in 6m 51s\n- ironic-ui-integration-tests https://zuul.opendev.org/t/openstack/build/bd6fcc66d6be440b855d44b6847e15f3 : SUCCESS in 44m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e382ca25758ab3bf6bf4b97615a3fa3d3085236b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-08-18 17:41:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d8c086055b6c1808208924e72858d1377cc43b07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 18:24:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/552512ecb9fa4241bafd9f8dc847d421\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/91b88e9fbf564fe9a14dcdbae9325bf4 : SUCCESS in 6m 18s\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/818e6640ce1546b2af43eb1935fa40aa : SUCCESS in 5m 56s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/f0382b5fc688475daa3269840ebf7280 : SUCCESS in 7m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f4828b246ad413ba3a17aa3c4a08639 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/570450d119d54614be315332307dea13 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3356e2be52f3441daeca04c4f4793ff4 : SUCCESS in 3m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b6cb67afba4b4c1a905e44939274114e : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/211d46f43cf541f4be81ead562dca1a7 : SUCCESS in 6m 47s\n- ironic-ui-integration-tests https://zuul.opendev.org/t/openstack/build/f3168e1d0ace4abb873334484a343d10 : SUCCESS in 42m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c354ae57896911087f4222a688ec6b9277195a20","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2022-08-18 18:25:25.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"96f5359ab19f2e13c53e3e305c594af173fc3084","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2022-08-18 20:03:44.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bc501ef4fe1af5d8736f6fe7b558f6e66d8f349d","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2022-08-18 20:36:30.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"7a4cf45e6931ffdff96c120f70acf1b2f32f3dec","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-08-19 08:35:26.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"919e78acd39302103c35fd2625e888321f085763","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 08:35:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1ff1100db7b5c129a0dcdf50d4c48a837a675130","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 08:53:45.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3f42d31001d4edb95f2a24dab1f9b7e\n\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/b99fdbf9c4024079999bf87f830a0380 : SUCCESS in 8m 42s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/224bf0b68cb046a0be6d21972c74e8a8 : SUCCESS in 7m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/935b4eb2b92b415fa467b9826eac8d3d : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5e7d87796c04809a89facaa9ecb3dc6 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11732937ba3e44faa87e5e3cec1c8b03 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7668cbbedce24843af9f21b79bc12e37 : SUCCESS in 7m 23s","accounts_in_message":[],"_revision_number":3},{"id":"b3f08a1c70ceb9dce482724039ec64d2c2c884bd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 08:53:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"852cb2b29ebbee3d1267d3f122f67652319f2d44","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 08:54:40.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fadc999f66d4ff99049ea5d27bec25c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c847fc18171b4015a95e65922c479c90 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"23914cc56cab845e29bc22bd23de48ab648ebaef","revisions":{"e148a5d77010192989cf34b6d9a9cfd9515651f4":{"kind":"REWORK","_number":1,"created":"2022-08-10 07:06:05.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/02/852702/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/02/852702/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/02/852702/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/02/852702/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/02/852702/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/02/852702/1"}}},"commit":{"parents":[{"commit":"8974cb6771a663d5be146fabaec2abd5982426e8","subject":"Merge \"Use Horizon project template for nodejs jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/8974cb6771a663d5be146fabaec2abd5982426e8"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2022-08-10 07:04:42.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2022-08-10 07:06:02.000000000","tz":330},"subject":"[DNM] Migrate to Angularjs v1.8.2","message":"[DNM] Migrate to Angularjs v1.8.2\n\nChange-Id: I40857bea1c83704106277bbb858eb35ee39f360f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/e148a5d77010192989cf34b6d9a9cfd9515651f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/e148a5d77010192989cf34b6d9a9cfd9515651f4"}]},"branch":"refs/heads/master"},"530780a38beb4db0b0f842efb97a721b0791aca2":{"kind":"REWORK","_number":2,"created":"2022-08-18 10:07:39.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/02/852702/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/02/852702/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/02/852702/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/02/852702/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/02/852702/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/02/852702/2"}}},"commit":{"parents":[{"commit":"8974cb6771a663d5be146fabaec2abd5982426e8","subject":"Merge \"Use Horizon project template for nodejs jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/8974cb6771a663d5be146fabaec2abd5982426e8"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2022-08-10 07:04:42.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2022-08-18 10:07:34.000000000","tz":330},"subject":"Migrate to AngularJS v1.8.2","message":"Migrate to AngularJS v1.8.2\n\nThis patch aligns current code with AngularJS v1.8.2 requirements.\nIt also updates zun-ui to use the Firefox browser instead of phantomjs\nbrowser to fix \"nodjs-run-test\" job because phantomjs browser has some\nissue with backticks.\n\nChange-Id: I40857bea1c83704106277bbb858eb35ee39f360f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/530780a38beb4db0b0f842efb97a721b0791aca2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/530780a38beb4db0b0f842efb97a721b0791aca2"}]},"branch":"refs/heads/master"},"23914cc56cab845e29bc22bd23de48ab648ebaef":{"kind":"REWORK","_number":3,"created":"2022-08-18 17:41:28.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/02/852702/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/02/852702/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/02/852702/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/02/852702/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/02/852702/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/02/852702/3"}}},"commit":{"parents":[{"commit":"8974cb6771a663d5be146fabaec2abd5982426e8","subject":"Merge \"Use Horizon project template for nodejs jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/8974cb6771a663d5be146fabaec2abd5982426e8"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2022-08-10 07:04:42.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2022-08-18 17:38:18.000000000","tz":330},"subject":"Migrate to AngularJS v1.8.2","message":"Migrate to AngularJS v1.8.2\n\nThis patch aligns current code with AngularJS v1.8.2 requirements.\nIt also updates zun-ui to use the Firefox browser instead of phantomjs\nbrowser to fix \"nodjs-run-test\" job because phantomjs browser has some\nissue with backticks.\n\nChange-Id: I40857bea1c83704106277bbb858eb35ee39f360f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/23914cc56cab845e29bc22bd23de48ab648ebaef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/23914cc56cab845e29bc22bd23de48ab648ebaef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}}]}],"submit_requirements":[]}
