)]}'
{"id":"openstack%2Fironic-ui~856700","triplet_id":"openstack%2Fironic-ui~master~Ib31c5f4f2d848bffda2154db0fa235fb53e8b3ee","project":"openstack/ironic-ui","branch":"master","topic":"add-antelope-python-jobtemplates","hashtags":[],"change_id":"Ib31c5f4f2d848bffda2154db0fa235fb53e8b3ee","subject":"Add Python3 antelope unit tests","status":"MERGED","created":"2022-09-09 10:27:13.000000000","updated":"2022-09-09 11:27:43.000000000","submitted":"2022-09-09 11:25:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"856700-add-antelope-python-jobtemplates","meta_rev_id":"80ad76302abc5083f19be648619de7efb875aea2","_number":856700,"virtual_id_number":856700,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-09 11:25:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-09 11:03:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-09 11:03:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-09 11:03:08.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2022-09-09 11:14:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4f81c703d5882220f32a2dd7e018360af65d2e6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2022-09-09 10:27:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"666ca0e6604f299d869a390a8993670fc2919724","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-09-09 11:03:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2d50ff3caa4e57e3c1e3183d3bf73e2a2f75e6a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 11:14:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c24645a61a3e42d3a5ae7018d1572b3c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb8125ce21124310b50dd60b74d6d112 : SUCCESS in 4m 29s\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/3f4b55c26b8d40d5b1fc951561c034bf : SUCCESS in 4m 36s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/425be223f73b417f87c662b3b590ac16 : SUCCESS in 11m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b60dc52df974c56bfc521b46056fef5 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e79dcbb0465445884c7b61b133eb43a : SUCCESS in 6m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cebb7ffd61c948838b6686b967e61df0 : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5698097f758c4a798e364134d6e56e87 : SUCCESS in 10m 14s\n- ironic-ui-integration-tests https://zuul.opendev.org/t/openstack/build/c414b5867ecb4840b6184c029adb6ed9 : SUCCESS in 40m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e25fc36ebd074638bedafd5b440e6803449bb9f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 11:14:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1e71b78012905849a9cc378a9619690967b36bbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 11:25:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c38343ac3524c00a703379de98d0002\n\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/0d5b3de0e38a4b7496445398fa6fe96b : SUCCESS in 6m 16s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/e6542e137c7e4348994593af19d1da1f : SUCCESS in 9m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/237c5211d1674ea097eca08da138d7c1 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55fbac4e13864ffbb5cc04ae7c0358c2 : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8416d1a45c074abebcec67836e3844d3 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5acf8153129c46e5b78e0f61278fde35 : SUCCESS in 7m 26s","accounts_in_message":[],"_revision_number":1},{"id":"3f30b854d44c409edf0a560f7e814ecfcf3bccac","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 11:25:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"80ad76302abc5083f19be648619de7efb875aea2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 11:27:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ca78de72dde4eb0ad39d9e95d8581c9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d0bb4fed3f741cfb84a44ae30769e70 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"708e850b43f7257e3f34409321bccb5855f426cc","revisions":{"708e850b43f7257e3f34409321bccb5855f426cc":{"kind":"REWORK","_number":1,"created":"2022-09-09 10:27:13.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/00/856700/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/00/856700/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/00/856700/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/00/856700/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/00/856700/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/00/856700/1"}}},"commit":{"parents":[{"commit":"d4b43161678025406ef6b60b4763bf5d6bd6ddbd","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/d4b43161678025406ef6b60b4763bf5d6bd6ddbd"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-09-09 10:27:12.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-09-09 10:27:12.000000000","tz":0},"subject":"Add Python3 antelope unit tests","message":"Add Python3 antelope unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for antelope.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: Ib31c5f4f2d848bffda2154db0fa235fb53e8b3ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/708e850b43f7257e3f34409321bccb5855f426cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/708e850b43f7257e3f34409321bccb5855f426cc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}}]}],"submit_requirements":[]}
