)]}'
{"id":"openstack%2Fironic-ui~906783","triplet_id":"openstack%2Fironic-ui~master~I1efadb54995b0f5a2a8b672f017831a3009cb6d2","project":"openstack/ironic-ui","branch":"master","topic":"ironic-codespell","attention_set":{},"removed_from_attention_set":{"36102":{"account":{"_account_id":36102,"name":"Adam McArthur","display_name":"Adam McArthur","email":"adam@mcaq.me","username":"adamcarthur","status":"PhD Candidate in Computational Imaging. Open Source Software Engineer at G-Research."},"last_update":"2024-01-30 17:22:53.000000000","reason":"Change was submitted"},"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2024-01-30 17:22:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1efadb54995b0f5a2a8b672f017831a3009cb6d2","subject":"[codespell] Fixing Spelling Mistakes","status":"MERGED","created":"2024-01-26 04:22:53.000000000","updated":"2024-01-30 17:23:58.000000000","submitted":"2024-01-30 17:22:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"906783-ironic-codespell","meta_rev_id":"8faf4044e8260f309aefa8fb769d2f6c7456daff","_number":906783,"virtual_id_number":906783,"owner":{"_account_id":36102,"name":"Adam McArthur","display_name":"Adam McArthur","email":"adam@mcaq.me","username":"adamcarthur","status":"PhD Candidate in Computational Imaging. Open Source Software Engineer at G-Research."},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-30 17:22:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":2,"date":"2024-01-30 08:28:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-30 17:10:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-01-30 17:10:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-26 04:23:20.000000000","updated_by":{"_account_id":36102,"name":"Adam McArthur","display_name":"Adam McArthur","email":"adam@mcaq.me","username":"adamcarthur","status":"PhD Candidate in Computational Imaging. Open Source Software Engineer at G-Research."},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2024-01-26 05:49:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-30 08:28:34.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2024-01-30 17:10:16.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"39b4a5e567403907092ac3e6e959acb1561c6a4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36102,"name":"Adam McArthur","display_name":"Adam McArthur","email":"adam@mcaq.me","username":"adamcarthur","status":"PhD Candidate in Computational Imaging. Open Source Software Engineer at G-Research."},"date":"2024-01-26 04:22:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"772e2c95b38c12c3d10d8388b64aed8228eaba56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 05:49:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee2805c131e441d4abb316bdf910c3c3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f4cae2911f94ffcbc9b1ff19273f0b1 : SUCCESS in 5m 24s\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/0a025500af734fc58bdcfd047966125c : SUCCESS in 5m 00s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/d0accddb86e8403f8a88c77c749fb433 : SUCCESS in 6m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/acf0eeae48cd402b8c1cc7505e61190f : SUCCESS in 3m 55s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/e604c44b64db48ce94eab30c66575b29 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0025323d9fd74eaf875e4fc8ba8c101e : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b36346e28b4d48f9be468f16bf4c9c52 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6aa12a462ccb46a4944d2e9cbffe050f : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66f102db8cf54c6b80c29149946c1fed : SUCCESS in 5m 16s\n- ironic-ui-integration-tests https://zuul.opendev.org/t/openstack/build/de297be92dbc44d097ea3c121a181feb : RETRY_LIMIT in 12m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"48b0b67f689d2ec5ac18fb72cca5bfd14e4aca98","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-01-26 23:32:07.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aaf9074bf93740eb5ae5b683b55f325dd808d434","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":36102,"name":"Adam McArthur","display_name":"Adam McArthur","email":"adam@mcaq.me","username":"adamcarthur","status":"PhD Candidate in Computational Imaging. Open Source Software Engineer at G-Research."},"date":"2024-01-27 06:31:24.000000000","message":"Topic set to ironic-codespell","accounts_in_message":[],"_revision_number":1},{"id":"e3c19dafd9ef2fc0799896732ecf1de8bb8f9bc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36102,"name":"Adam McArthur","display_name":"Adam McArthur","email":"adam@mcaq.me","username":"adamcarthur","status":"PhD Candidate in Computational Imaging. Open Source Software Engineer at G-Research."},"date":"2024-01-29 21:08:52.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"a5a8a65e309930750458bea19e0e8ab4a195acd0","author":{"_account_id":36102,"name":"Adam McArthur","display_name":"Adam McArthur","email":"adam@mcaq.me","username":"adamcarthur","status":"PhD Candidate in Computational Imaging. Open Source Software Engineer at G-Research."},"date":"2024-01-29 21:09:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0486a631daab227718aed466491ba1540e1ff778","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-29 22:38:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7872a3dd1d94bb7be04c6e9d91ddc14\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4e47b13ac25402cb90496ce849c09bf : SUCCESS in 3m 27s\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/5f3b824e2d624243b6302bde183a1e76 : SUCCESS in 5m 01s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/6ffde25ef4844acc85096bb8e15e9194 : SUCCESS in 6m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e5d184501046437ea41cac073f59bd38 : SUCCESS in 4m 07s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/b41a9fc19f0f4dcb85db80310be2c288 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98a17c37975e413bbd3add13efc7a128 : SUCCESS in 2m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a24a0fccea64c0796415281afc1692d : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84b7af57ba9f433b95bac3740cd7a322 : SUCCESS in 7m 06s\n- ironic-ui-integration-tests https://zuul.opendev.org/t/openstack/build/59af0a9a6f1e4f40b5851ae45c32f66b : RETRY_LIMIT in 19m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0f56855a937f2909caa48e3f2d60c54831103dc9","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2024-01-30 08:28:34.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0383aba6767232511bdcc75d36c2ec1b6772a50c","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2024-01-30 17:10:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d2f2c1f6ec0f768483478d9952cb681c4c9d12ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-30 17:10:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a0b235abfa06b5406589670046669bd9ef8d3bc5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-30 17:22:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afd98d738a384eddb24f6c972dee47d7\n\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/9c1af3ad444b40159b2b495006ed1dc2 : SUCCESS in 4m 59s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/bcea4abd19ee460b95912ee27c00761a : SUCCESS in 6m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a979c59b32384afa9aa65212b66dbc44 : SUCCESS in 4m 00s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/62e16336e1954bafa47dd97465b39e86 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9a5ec80bd2347c099de20ab213df26f : SUCCESS in 2m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f9ab4642d6a4be3af5762842b2ce2df : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbfca31cc83247b99b901b5bc1a27b9b : SUCCESS in 6m 15s","accounts_in_message":[],"_revision_number":2},{"id":"a9be05541c8dc06e138313278e299dcd8cb3d893","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-30 17:22:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8faf4044e8260f309aefa8fb769d2f6c7456daff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-30 17:23:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c391dee138374c81ba5c3786d2088206\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04f9bc1989844a9b8f059e3d1a4a92da : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c5cf60633991f25e779066130db3c5e209b3d3a7","revisions":{"25c13e5dd674b4bb33892c7f00dab3002917b4ec":{"kind":"REWORK","_number":1,"created":"2024-01-26 04:22:53.000000000","uploader":{"_account_id":36102,"name":"Adam McArthur","display_name":"Adam McArthur","email":"adam@mcaq.me","username":"adamcarthur","status":"PhD Candidate in Computational Imaging. Open Source Software Engineer at G-Research."},"ref":"refs/changes/83/906783/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/83/906783/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/83/906783/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/83/906783/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/83/906783/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/83/906783/1"}}},"commit":{"parents":[{"commit":"23aea8b8fb029151395d881467fa95aba295033b","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/23aea8b8fb029151395d881467fa95aba295033b"}]}],"author":{"name":"Sharpz7","email":"adam.mcarthur62@gmail.com","date":"2024-01-26 04:22:15.000000000","tz":0},"committer":{"name":"Sharpz7","email":"adam.mcarthur62@gmail.com","date":"2024-01-26 04:22:15.000000000","tz":0},"subject":"[codespell] Fixing Spelling Mistakes","message":"[codespell] Fixing Spelling Mistakes\n\nThis is the first in a series of commits to add support for codespell. This is continuning the process completed in ironic-python-agent.\n\nFuture Commits will add a Tox Target, CI support and potentially a git-blame-ignore-revs file if their are lots of spelling mistakes that could clutter git blame.\n\nChange-Id: I1efadb54995b0f5a2a8b672f017831a3009cb6d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/25c13e5dd674b4bb33892c7f00dab3002917b4ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/25c13e5dd674b4bb33892c7f00dab3002917b4ec"}]},"branch":"refs/heads/master"},"c5cf60633991f25e779066130db3c5e209b3d3a7":{"kind":"REWORK","_number":2,"created":"2024-01-29 21:08:52.000000000","uploader":{"_account_id":36102,"name":"Adam McArthur","display_name":"Adam McArthur","email":"adam@mcaq.me","username":"adamcarthur","status":"PhD Candidate in Computational Imaging. Open Source Software Engineer at G-Research."},"ref":"refs/changes/83/906783/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/83/906783/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/83/906783/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/83/906783/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/83/906783/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/83/906783/2"}}},"commit":{"parents":[{"commit":"23aea8b8fb029151395d881467fa95aba295033b","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/23aea8b8fb029151395d881467fa95aba295033b"}]}],"author":{"name":"Sharpz7","email":"adam.mcarthur62@gmail.com","date":"2024-01-29 21:08:48.000000000","tz":0},"committer":{"name":"Sharpz7","email":"adam.mcarthur62@gmail.com","date":"2024-01-29 21:08:48.000000000","tz":0},"subject":"[codespell] Fixing Spelling Mistakes","message":"[codespell] Fixing Spelling Mistakes\n\nThis is the first in a series of commits to add support for codespell. This is continuning the process completed in ironic-python-agent.\n\nFuture Commits will add a Tox Target, CI support and potentially a git-blame-ignore-revs file if their are lots of spelling mistakes that could clutter git blame.\n\nChange-Id: I1efadb54995b0f5a2a8b672f017831a3009cb6d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/c5cf60633991f25e779066130db3c5e209b3d3a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/c5cf60633991f25e779066130db3c5e209b3d3a7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
