)]}'
{"id":"openstack%2Fironic-ui~912922","triplet_id":"openstack%2Fironic-ui~unmaintained%2Fvictoria~I004ed35c30128324207688313a9039a9f1c117d0","project":"openstack/ironic-ui","branch":"unmaintained/victoria","topic":"create-victoria","hashtags":[],"change_id":"I004ed35c30128324207688313a9039a9f1c117d0","subject":"Update .gitreview for unmaintained/victoria","status":"MERGED","created":"2024-03-14 11:18:22.000000000","updated":"2025-02-16 14:09:29.000000000","submitted":"2025-02-16 14:08:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"912922-create-victoria","meta_rev_id":"847480d7a7cc9386e4a20be26a85bff0405c3cbc","_number":912922,"virtual_id_number":912922,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-16 14:08:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2025-02-16 12:29:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2025-02-16 12:29:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-14 11:47:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-16 07:56:55.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2025-02-16 12:29:48.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"48e1e4c65abddbf56614a353eee86af1c0f4879f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2024-03-14 11:18:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ccf17e0dd2f9ce7a133f8bf34fffa5dba5db96e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 11:47:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7152dabf67fe4c09a1e58e2000af14d7\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/4d51d68391bb44f49189029d8a845ce7 : SUCCESS in 3m 38s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/7a5836b3cc024189b5f59569b3ea0a3d : SUCCESS in 7m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9fcd75e55dc4b048da0e8691634eb39 : SUCCESS in 3m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5d7078696a9456595eb4576608cb540 : FAILURE in 2m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/485473d4c85c4fbb9b75f756b39ac203 : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4cc33485b8b429a86c761b57c307c26 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8046bebce52043d0912192eb069c77ca : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":1},{"id":"227afe53b5b77aca5d1c4bde9ae8e8c7cd882b4d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-02-16 07:56:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4e4af0d0558346372dba2a3fbb7f73e52aaa0606","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-16 09:03:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d39cf0f0e01f49e7986140abac93b9ac\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/c621172b654049d0916b8d8259f79e3c : SUCCESS in 3m 09s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/156466d9f031473da8f9cbd9e9680ef2 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f37fc00cac284da4be22387979c71611 : FAILURE in 2m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dbc76e51a1a04347ac883023c54bdb7e : FAILURE in 1m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/644dbaacad28417db29baaa7d4c2f385 : FAILURE in 2m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/833e4e1392c94ab1a008a9242b8f6fa1 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2bef65fc92f4413b5eb8a99aeda5c75 : FAILURE in 3m 25s\n- ironic-ui-integration-tests https://zuul.opendev.org/t/openstack/build/73c3ea4ba5ff4c07bebb75f3aba7cd6e : RETRY_LIMIT in 11m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4a4f2d620bdd0f26e33554153784bd57daf74881","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-02-16 12:29:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ff9bbfaf79f200b779a633f500dcb5ff93e902aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-16 13:54:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec2668a8358341efa40d0832d41a8324\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/94d2c997c27849a8b8c7d02b1c005af8 : SUCCESS in 3m 31s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/4dafb5f4b14b40128084f5e4a766ac88 : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2930cc7714c48a58b84de7eba3f49f9 : SUCCESS in 2m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2db1ee784fe14d2a884acdf21c383d44 : SUCCESS in 8m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dfea99eddce546b18a2c9b25d3a3666b : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c821eaf0c584042848354ed2f4f8387 : SUCCESS in 8m 51s\n- ironic-ui-integration-tests https://zuul.opendev.org/t/openstack/build/6735592022a148468294b79a1e657c02 : RETRY_LIMIT in 24m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e6718b4ba7f23c0750932e390c8b919b948450b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-16 13:54:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d78994dc1f57874e32a40d6fb68c6cba52db5afd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-16 14:08:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eeb4c06841664a2f91d3cf97c3e18648\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/b10c995bcfca4712b69aeb955f8a2ce0 : SUCCESS in 4m 24s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/ba79fb46445846829ee4f82aa530e21c : SUCCESS in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b33804c181740ba90f67164aa781f34 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ead337318c0d479d84cbaff4c81d504a : SUCCESS in 2m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cdc10f6ed08d4e1e8b2c3039afd56282 : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/766ef335a6784989b9b74bdb99aeb219 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":1},{"id":"1a6ab816e73f52097a2cac75ce15ac5226c4ed19","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-16 14:08:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"847480d7a7cc9386e4a20be26a85bff0405c3cbc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-16 14:09:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e9ba1e51dd94f1a8b5575cf70e121ff\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eca6b587852e49d29d240bd36306c502 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d1c7ef3c2f47aa67e8437c0ff91452f9044911d2","revisions":{"d1c7ef3c2f47aa67e8437c0ff91452f9044911d2":{"kind":"REWORK","_number":1,"created":"2024-03-14 11:18:22.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/22/912922/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/22/912922/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/22/912922/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/22/912922/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/22/912922/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/22/912922/1"}}},"commit":{"parents":[{"commit":"268629d80ebbdb00668a6902625911b45fb0ab5a","subject":"Merge \"Update TOX_CONSTRAINTS_FILE for stable/victoria\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/268629d80ebbdb00668a6902625911b45fb0ab5a"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-03-14 11:18:20.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-03-14 11:18:20.000000000","tz":0},"subject":"Update .gitreview for unmaintained/victoria","message":"Update .gitreview for unmaintained/victoria\n\nChange-Id: I004ed35c30128324207688313a9039a9f1c117d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/d1c7ef3c2f47aa67e8437c0ff91452f9044911d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/d1c7ef3c2f47aa67e8437c0ff91452f9044911d2"}]},"branch":"refs/heads/unmaintained/victoria"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
