)]}'
{"id":"openstack%2Fironic-ui~953362","triplet_id":"openstack%2Fironic-ui~master~I0ac37d408a1c66d749ae313a7711bf7a612b9b26","project":"openstack/ironic-ui","branch":"master","topic":"remove-py39","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-08-06 22:43:47.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"I0ac37d408a1c66d749ae313a7711bf7a612b9b26","subject":"Remove Python 3.9 support","status":"MERGED","created":"2025-06-26 01:13:21.000000000","updated":"2025-08-06 22:44:46.000000000","submitted":"2025-08-06 22:43:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953362-remove-py39","meta_rev_id":"bb27edd6b52063c80040699010c355fb5b489bd6","_number":953362,"virtual_id_number":953362,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-06 22:43:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":2,"date":"2025-08-06 22:34:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":2,"date":"2025-08-06 11:04:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":1,"date":"2025-08-06 22:34:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"date":"2025-08-06 11:04:53.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-26 01:24:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-26 06:54:10.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2025-08-06 22:34:25.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"}],"messages":[{"id":"0743b761e59d28c671b8c305317fca358cd7f246","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-26 01:13:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d409b719d56b06e22aff002ca8f690388ce14cfe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 01:24:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3d7fdd896894f80a234b41cad210475\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/609f7a59d667422ea540ef8b233f15bb : SUCCESS in 2m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f5c70e494e14bdcb0d0fda943865340 : SUCCESS in 4m 14s\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/c742b21236d3440aba7fb245c556eb62 : SUCCESS in 3m 55s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/dab3723bf2bf468a9cb2f14af23012ea : SUCCESS in 7m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/378c9912af6a47bcbf51b51860ad1d19 : SUCCESS in 3m 27s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/53788f79691841e084b7f6eaa4da17b5 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d1bc50f34154f589a698708ad992abe : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bf97446b2f8a44018ddf3bcc9f9c8ea5 : SUCCESS in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/19e7a7863062457883e40e92672fb040 : SUCCESS in 8m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bab2c07b717a4f3fbca8b2a7c24c4965 : SUCCESS in 7m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a429cecd4ce8417084398138e20dc14e : SUCCESS in 5m 10s\n- ironic-ui-tox-codespell https://zuul.opendev.org/t/openstack/build/306af90d45f24c15a6cf026eaa4c4c44 : FAILURE in 1m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8806cb84934bd7941ea425107513f78f1d52baa2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-26 02:24:40.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"128bcd1f8ce52a7d008a444df89229f4803140e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 03:02:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b9a53bc0f3b4d7691dd63d08c513ef6\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/113dd580c507436d8ef74baed6a17fb7 : SUCCESS in 2m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c38a920e1724fd593c01cd77e8f0568 : SUCCESS in 6m 19s\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/ec61dd92ed73448093433a704b7daeb0 : SUCCESS in 2m 24s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/3e44a79966c94d2aabbf7fcfb29163d1 : SUCCESS in 4m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6f0d4b3be9bc4d44b024ee463812d4db : SUCCESS in 4m 04s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/839f3e349d2f49eab2f6c5e2dbc9dad2 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d59185947d5c4c4f8595d11d0bd8133c : SUCCESS in 1m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/420eaa08be5e4fce9bc0776b21a1951d : SUCCESS in 3m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b8e5b011b1a04f2ab807131b734475fe : SUCCESS in 3m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f66955ca1704d15a8f26c813c95799a : SUCCESS in 7m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53b7d05f1cf741ba85aae3ece8d2591d : SUCCESS in 3m 29s\n- ironic-ui-tox-codespell https://zuul.opendev.org/t/openstack/build/087a9c97a94442ac8a33521d1a293a42 : SUCCESS in 4m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2d3854a6403812f34afaab0825a877ef199795b4","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-06-26 06:54:10.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9b7796aef601dd7c0ad5cf4bc79f8c7bd76d3a53","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-08-06 11:04:53.000000000","message":"Patch Set 2: -Workflow Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4ec680900ff45cb2e1a48aa9e84ff4641dbf38bb","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-08-06 11:08:17.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":2},{"id":"928ebdd86a82110c493b00089293ea783826bff8","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-08-06 22:34:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8376a73cb9ec90ac88dccf33318912896703cb27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 22:34:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"89b7ca7b3f8f4ed40cb45ca7999803b82366399b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 22:43:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33a8e785404a4720a2b9834f83e2bb8f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/977762b2a8af46cdbb5d9bf2cf727ddb : SUCCESS in 1m 21s\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/42f68714580549cab7ad0d00cbdcf231 : SUCCESS in 4m 36s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/335464fe16334284815e6d3a96112617 : SUCCESS in 6m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/660ae2cff1874867811601deab76d2ec : SUCCESS in 3m 37s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/7ba2209729934ebc88a01fd81da333ef : SUCCESS in 2m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/793230f0409e41e6aff422df2c71d64c : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fe75ee4f754e4dc597f446d84fa43cb6 : SUCCESS in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15bb3e97772e4390a344330e955510bf : SUCCESS in 3m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad57963080bb41429b71233a83fdfb4e : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":2},{"id":"26774dc0dd9765d3d8d3c91582c4bde39843f940","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 22:43:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"bb27edd6b52063c80040699010c355fb5b489bd6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 22:44:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea2b208fcc7645a388e6ba11cf107772\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ab789a4d0fb4dddb50c5829b61d03ba : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b416ea0a4584438981d361bad583aaa : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9554edf4a201ff80097f3741647992f93d88bb8b","revisions":{"95c44ebb882059ac2c756697fbc0e2170c983cea":{"kind":"REWORK","_number":1,"created":"2025-06-26 01:13:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/62/953362/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/62/953362/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/62/953362/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/62/953362/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/62/953362/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/62/953362/1"}}},"commit":{"parents":[{"commit":"c3e13081351cd6bdf143f73ec5d7ad68f53c211f","subject":"Merge \"Revert \"Replace license classifier\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/c3e13081351cd6bdf143f73ec5d7ad68f53c211f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-26 01:13:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-26 01:13:12.000000000","tz":540},"subject":"Remove Python 3.9 support","message":"Remove Python 3.9 support\n\nPython 3.9 is no longer part of the tested runtimes[1].\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2025.2.html\n\nChange-Id: I0ac37d408a1c66d749ae313a7711bf7a612b9b26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/95c44ebb882059ac2c756697fbc0e2170c983cea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/95c44ebb882059ac2c756697fbc0e2170c983cea"}]},"branch":"refs/heads/master"},"9554edf4a201ff80097f3741647992f93d88bb8b":{"kind":"REWORK","_number":2,"created":"2025-06-26 02:24:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/62/953362/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/62/953362/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/62/953362/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/62/953362/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/62/953362/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/62/953362/2"}}},"commit":{"parents":[{"commit":"c3e13081351cd6bdf143f73ec5d7ad68f53c211f","subject":"Merge \"Revert \"Replace license classifier\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/c3e13081351cd6bdf143f73ec5d7ad68f53c211f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-26 01:13:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-26 02:24:39.000000000","tz":0},"subject":"Remove Python 3.9 support","message":"Remove Python 3.9 support\n\nPython 3.9 is no longer part of the tested runtimes[1].\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2025.2.html\n\nChange-Id: I0ac37d408a1c66d749ae313a7711bf7a612b9b26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/9554edf4a201ff80097f3741647992f93d88bb8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/9554edf4a201ff80097f3741647992f93d88bb8b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
