)]}'
{"id":"openstack%2Fironic-ui~954408","triplet_id":"openstack%2Fironic-ui~stable%2F2024.1~I60da4ebe63c916cdd78577c74b0ce8105c5a210f","project":"openstack/ironic-ui","branch":"stable/2024.1","topic":"zanata/translations","hashtags":[],"change_id":"I60da4ebe63c916cdd78577c74b0ce8105c5a210f","subject":"Imported Translations from Zanata","status":"MERGED","created":"2025-07-09 03:24:20.000000000","updated":"2025-07-12 06:02:54.000000000","submitted":"2025-07-09 12:07:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954408-zanata/translations","meta_rev_id":"a63bf790a9c946ff0f95d32e3d65cbab011d2af8","_number":954408,"virtual_id_number":954408,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-09 12:07:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2025-07-09 11:48:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-07-12 06:02:54.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2025-07-09 11:48:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-09 03:48:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-09 11:48:32.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2025-07-12 06:02:54.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"eeca154014e27fda1585a5ea9f2422f4b1fa51ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-07-09 03:24:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b1e779d860185c984a6e98186006e1407ed883a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 03:48:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e68bfa35ffb7438d8200c2de69d2e821\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4511ce20f1524d2b95d17db228271734 : SUCCESS in 3m 14s\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/dfd15f6d80a842b5ba0e6a88459ac846 : SUCCESS in 3m 50s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/a17b31820ab6499081cb305ca44ad037 : SUCCESS in 4m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/82335383c7624239acbe4b01bd31a265 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd9dff4878ef432ab980ff94358b02da : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/933fd92aa215472a9d9eb06288381b45 : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28b32bc025e942c496072e64a3c2b855 : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ab6a9345f6d4a478d5c94f5ef32c881 : SUCCESS in 4m 57s\n- ironic-ui-integration-tests https://zuul.opendev.org/t/openstack/build/26a720df21224c0abe32e4d0ec9a10cd : RETRY_LIMIT in 5m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f61cfd868ef63c86ba0564176639ca19758e1907","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-07-09 11:48:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"634bc1f22c8f9535f19f36b078c5ec4b4cbc41f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 11:48:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b09f9e127d81d3e9bc19c0fa5d403408a45fc14c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 12:07:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22b9928cfc2a4007865329658828ec66\n\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/a0b3d4e71a57412982c70cd104fe030e : SUCCESS in 4m 59s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/e662abe6999340678d3e037ba9b2f488 : SUCCESS in 4m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/763f9f9d30ea41659c584ee43cb6abb1 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efbb3814efe94bb8af3eeb2c17fe9e2b : SUCCESS in 2m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55efe9d415dd4ec99c0e1de6ff51d078 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbc24bf4a8e0461d9350debd1d395a86 : SUCCESS in 7m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63343c9b3ed347ba8e14b623f28bdb16 : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":1},{"id":"09130de895466b80a8b14cff5b7d204bf93a92ce","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 12:07:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"64c5426585c9b5fe08f446b2f9572bfdb2dc9cd0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 12:08:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb59361b68694709becfd4e50af6277a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/689d4f50c5584e24bb54caedaa9ee202 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/deeaec1cd29046138a277b450ce90a7c : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1},{"id":"a63bf790a9c946ff0f95d32e3d65cbab011d2af8","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-12 06:02:54.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4bf4e0a09a220ce892a7026e69fabf2f36b4d780","revisions":{"4bf4e0a09a220ce892a7026e69fabf2f36b4d780":{"kind":"REWORK","_number":1,"created":"2025-07-09 03:24:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/08/954408/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/08/954408/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/08/954408/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/08/954408/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/08/954408/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/08/954408/1"}}},"commit":{"parents":[{"commit":"a50a6e491c85de91bc751a621082f9dac594dac2","subject":"Update TOX_CONSTRAINTS_FILE for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/a50a6e491c85de91bc751a621082f9dac594dac2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-07-09 03:24:18.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-07-09 03:24:18.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I60da4ebe63c916cdd78577c74b0ce8105c5a210f\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/openstack-zuul-jobs:roles/prepare-zanata-client/files/common_translation_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/4bf4e0a09a220ce892a7026e69fabf2f36b4d780"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/4bf4e0a09a220ce892a7026e69fabf2f36b4d780"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
