)]}'
{"id":"openstack%2Fironic-ui~962518","triplet_id":"openstack%2Fironic-ui~master~I9147f1e1be95ee2ea93cd6a4df6f49c72a2da2b7","project":"openstack/ironic-ui","branch":"master","attention_set":{"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2025-09-29 15:58:53.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I9147f1e1be95ee2ea93cd6a4df6f49c72a2da2b7","subject":"WIP/DNM: Add conductor list view into ironic-ui","status":"NEW","created":"2025-09-29 15:43:57.000000000","updated":"2025-09-29 15:58:53.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"96ffde567033955a2887d7b58afd2a499ad89f1e","_number":962518,"virtual_id_number":962518,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-09-29 15:58:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-29 15:58:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9169223d772d309cf56cb1cd3387d57aa26486c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2025-09-29 15:43:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96ffde567033955a2887d7b58afd2a499ad89f1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 15:58:53.000000000","message":"Patch Set 1: Verified-1\n\n(30 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a624e28d7c84afaa32538f43f10213d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f8e8eded36c4eb3bd7db76829523ccd : SUCCESS in 5m 26s\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/277e0f7d0d3d4b8aaaeaa342e8676c77 : FAILURE in 4m 56s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/d6081db760c847a597c929fc298e60e2 : FAILURE in 6m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7c768c7c08c04512a5e1d93961a3fb5c : SUCCESS in 4m 02s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/c26e2d44764044018ff743f08af83bec : SUCCESS in 4m 30s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/4cb86b7b2fda41e2b1f44a16e21245c6 : FAILURE in 4m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8601a86a61d24037b1b85db855488246 : FAILURE in 4m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f2ba1d9262bc447bb502c909cf726f46 : SUCCESS in 3m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f5075eafdbe04584801e4564d7be5453 : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0f2047deb934a99be50e3ddf9f3f5eb : SUCCESS in 8m 18s\n- ironic-ui-tox-codespell https://zuul.opendev.org/t/openstack/build/4cf349dfcfe74777b8535e045e252a67 : SUCCESS in 4m 57s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cf73cdbe8f4a94c1c934cf02987e0516022d1bc6","revisions":{"cf73cdbe8f4a94c1c934cf02987e0516022d1bc6":{"kind":"REWORK","_number":1,"created":"2025-09-29 15:43:57.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/18/962518/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/18/962518/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/18/962518/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/18/962518/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/18/962518/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/18/962518/1"}}},"commit":{"parents":[{"commit":"e542f3fac4dfca6b1f17d62db7877648efdba41d","subject":"WIP/DNM: Node history support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/e542f3fac4dfca6b1f17d62db7877648efdba41d"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2025-09-29 14:05:07.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2025-09-29 14:05:07.000000000","tz":-420},"subject":"WIP/DNM: Add conductor list view into ironic-ui","message":"WIP/DNM: Add conductor list view into ironic-ui\n\nThis was an experiment, anyone may take this change set over.\n\nAssisted-By: Claude Code - Claude Sonnet 4\nChange-Id: I9147f1e1be95ee2ea93cd6a4df6f49c72a2da2b7\nSigned-off-by: Julia Kreger \u003cjuliaashleykreger@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/cf73cdbe8f4a94c1c934cf02987e0516022d1bc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/cf73cdbe8f4a94c1c934cf02987e0516022d1bc6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
