)]}'
{"id":"openstack%2Fironic-ui~962519","triplet_id":"openstack%2Fironic-ui~master~I4e218860c764da2b55457491db73d0761cfb0227","project":"openstack/ironic-ui","branch":"master","attention_set":{"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2025-09-29 15:59:38.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I4e218860c764da2b55457491db73d0761cfb0227","subject":"WIP/DNM: Add node traits support to ironic-ui","status":"NEW","created":"2025-09-29 15:43:57.000000000","updated":"2025-09-29 15:59:38.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d9e376cabe03d2cabb35bd05c64039a56c1ca1d5","_number":962519,"virtual_id_number":962519,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-09-29 15:59:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-29 15:59:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9a05c3529fce383e0de9559525ba82bcefd7e001","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2025-09-29 15:43:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d9e376cabe03d2cabb35bd05c64039a56c1ca1d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 15:59:38.000000000","message":"Patch Set 1: Verified-1\n\n(44 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/79ba6ec6a9ef46ba8ac601d15ddcb432\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf84aa6c5ecc43f8a2db26e53c76d399 : SUCCESS in 4m 23s\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/cb3901f93253485a9ba2f60212e3ee8f : FAILURE in 7m 23s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/0b2120d4fef04382a144d494e6957aea : FAILURE in 7m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6e9cd50dbfe647ddbdd8ab56c9d49a7f : SUCCESS in 4m 06s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/933fad376c254e31925b34b926320b05 : SUCCESS in 4m 56s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/49f343b6cd4b43ddb48f2f6676d318c1 : FAILURE in 5m 39s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad9b8fa21d8a4deaaee4ac26aa768a56 : FAILURE in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2ae3606cec344298a6bb78711b04ff04 : SUCCESS in 3m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/423e2e10da724ce3a537cc8b6bc17109 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b5071198c0d425297ddeee6e6fd1d1f : SUCCESS in 9m 48s\n- ironic-ui-tox-codespell https://zuul.opendev.org/t/openstack/build/e116ca75968147048ec9d70182e214c3 : SUCCESS in 3m 21s (non-voting)\n\nWarning:\n  Comments left for invalid file ironic_ui/content/ironic/urls.py","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eafd8435a00f11749a94aa14522d2e994b2c3023","revisions":{"eafd8435a00f11749a94aa14522d2e994b2c3023":{"kind":"REWORK","_number":1,"created":"2025-09-29 15:43:57.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/19/962519/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/19/962519/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/19/962519/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/19/962519/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/19/962519/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/19/962519/1"}}},"commit":{"parents":[{"commit":"cf73cdbe8f4a94c1c934cf02987e0516022d1bc6","subject":"WIP/DNM: Add conductor list view into ironic-ui","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/cf73cdbe8f4a94c1c934cf02987e0516022d1bc6"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2025-09-29 14:36:06.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2025-09-29 14:36:06.000000000","tz":-420},"subject":"WIP/DNM: Add node traits support to ironic-ui","message":"WIP/DNM: Add node traits support to ironic-ui\n\nFeel free to take over this change and move it along, I generated\nit as an experiment.\n\nAssisted-By: Claude Code - Claude Sonnet 4\nChange-Id: I4e218860c764da2b55457491db73d0761cfb0227\nSigned-off-by: Julia Kreger \u003cjuliaashleykreger@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/eafd8435a00f11749a94aa14522d2e994b2c3023"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/eafd8435a00f11749a94aa14522d2e994b2c3023"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
