)]}'
{"id":"openstack%2Fironic-ui~972615","triplet_id":"openstack%2Fironic-ui~master~I259252858f2d7d46a19d2a33976efe1d9c57b87f","project":"openstack/ironic-ui","branch":"master","topic":"zanata/translations","attention_set":{},"removed_from_attention_set":{"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2026-01-14 12:51:33.000000000","reason":"\u003cGERRIT_ACCOUNT_23851\u003e replied on the change","reason_account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}}},"hashtags":[],"change_id":"I259252858f2d7d46a19d2a33976efe1d9c57b87f","subject":"Imported Translations from Zanata","status":"MERGED","created":"2026-01-08 02:18:34.000000000","updated":"2026-01-14 13:02:31.000000000","submitted":"2026-01-14 13:01:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"972615-zanata/translations","meta_rev_id":"a25312ac51df02cd76a02cd110e2aeda7583ca06","_number":972615,"virtual_id_number":972615,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-14 13:01:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2026-01-14 12:51:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-08 15:27:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2026-01-14 12:51:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-08 03:08:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-08 15:27:30.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-01-08 15:27:30.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"7545ef855d17b82ac90a307bdf0ba39329d187ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-01-08 02:18:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"66bf344e890f7b3769d3f3ba1feaff64fe303b9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 03:08:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04708ee98f3249d9a92fd8a8d331201f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8410b8168cd241f0850616f86bd5076a : SUCCESS in 4m 07s\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/e6be0986dcac4a988e767c1ba70307e5 : SUCCESS in 4m 15s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/ac5d758f4c92489ca30280d96e063cac : SUCCESS in 3m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bf0342b4365e447a8d6d5c0f4fa98117 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9c9a33cdf82489ab8723e4b8a3ae2cc : SUCCESS in 2m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/27235eff391f4402b3792b9198ffcd71 : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/971e6bb34136418d8353965645329502 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f40841fc54a44483bbf884c38e8a0b0b : SUCCESS in 7m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e659713eb043448798eb357f28541422 : SUCCESS in 4m 30s\n- ironic-ui-tox-codespell https://zuul.opendev.org/t/openstack/build/8b73d83c0a0046778a7dd1c69f8b5221 : SUCCESS in 3m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0293f10402cb7bc2029822bf8c1b6c6a9dcd9127","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-08 15:27:30.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"61404ce9417a8f58b9d51f885598c127b57b3913","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2026-01-14 12:51:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"94bc68d15cc947c7679d59279969056d1f5f66d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 12:51:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8730c0a72cea3ac72b94c5dbd46f807121935a28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 13:01:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af1dabd58b8042bdb0420a3303e56378\n\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/ed6a9ab049a84dc0ab9c1dbf036606e6 : SUCCESS in 4m 04s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/ee634ac84f8f48d99927aca1e35d1fdd : SUCCESS in 5m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/43ed1ef3836145f0958d55e28cf5ebee : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc15c015272743fc836b467a422baa6f : SUCCESS in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/15607078046b429dbcd354904686ab8e : SUCCESS in 2m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/abb3354a6b964e9b9688daff09824b93 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f98c94898d24556a94325683ff52a82 : SUCCESS in 6m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/177729842bfb4ad9af919226b0daca76 : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":1},{"id":"51ec388abb6253a66b5d07b0cce0b3f73d86b4cc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 13:01:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a25312ac51df02cd76a02cd110e2aeda7583ca06","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 13:02:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6160ee786bb479799cef5dcc39cc148\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34fda9a919674b26a61b0db572858274 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60dd3430b97540ca839d43fb83be4cb9 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"acb691bf705b11dbb0b3182a18d44deeff2ee3a1","revisions":{"acb691bf705b11dbb0b3182a18d44deeff2ee3a1":{"kind":"REWORK","_number":1,"created":"2026-01-08 02:18:34.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/15/972615/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/15/972615/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/15/972615/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/15/972615/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/15/972615/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/15/972615/1"}}},"commit":{"parents":[{"commit":"fc986ecf74dcea1d913200b41606523fa883e218","subject":"reno: Update master for unmaintained/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/fc986ecf74dcea1d913200b41606523fa883e218"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-08 02:18:33.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-08 02:18:33.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I259252858f2d7d46a19d2a33976efe1d9c57b87f\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/openstack-zuul-jobs:roles/prepare-zanata-client/files/common_translation_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/acb691bf705b11dbb0b3182a18d44deeff2ee3a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/acb691bf705b11dbb0b3182a18d44deeff2ee3a1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
