)]}'
{"id":"openstack%2Fironic-ui~987281","triplet_id":"openstack%2Fironic-ui~master~Ib9ad8b667fc8f70feee32a625333c270ec1cde97","project":"openstack/ironic-ui","branch":"master","attention_set":{},"removed_from_attention_set":{"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-05-05 13:52:06.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"Ib9ad8b667fc8f70feee32a625333c270ec1cde97","subject":"Update packaging configuration","status":"MERGED","created":"2026-05-04 20:38:51.000000000","updated":"2026-05-05 13:53:28.000000000","submitted":"2026-05-05 13:52:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"987281","meta_rev_id":"8a3bd12e4f0aab7b58df693ee9b7c01cc68d2631","_number":987281,"virtual_id_number":987281,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-05 13:52:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":2,"date":"2026-05-05 13:39:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":2,"date":"2026-05-04 21:14:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":1,"date":"2026-05-05 13:39:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-04 20:57:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-04 21:14:09.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2026-05-05 13:39:13.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"}],"messages":[{"id":"292b63f833d85ade6ef8a2525205c29693fd7a3e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-04 20:38:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa3a6ab6bd1e1907806419502eb8c54e5fd55e65","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-04 20:41:18.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"1f477e0572bd78c0ef24a82d579ae0ffeec14855","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-04 20:41:29.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"7644bed66239c2523cb7580bf18d114c970ab162","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-04 20:41:31.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"aebca210e7d67f5dc0624d61ad1d6799ced8b529","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 20:57:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0ab01bdcfb0440cbeb8a08ab7ae2d52\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/439fe46c1d1149f3bc9c763f0b715c06 : SUCCESS in 2m 52s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3a6b436ded7446ec90d7cef9792a49a9 : SUCCESS in 2m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8441c1f2122a4cbc8f997b808d336120 : SUCCESS in 3m 39s\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/f00a5f2ac9f8452cb20d3444bdcd442e : SUCCESS in 5m 56s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/9dc90878a73e4f618a427f397d5b5260 : SUCCESS in 7m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ac2621ed49e94432a513530e09362f1a : SUCCESS in 3m 32s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/5785aec15bc54b03868e87e074101d8a : SUCCESS in 4m 32s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/9c432d426cf64e24b0ce9e0d18b5a748 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/259e45232fd74fc3ae6fee8fbbda37b7 : SUCCESS in 2m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2b249c0afa27402caf8ac81ff7adfb9b : SUCCESS in 3m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/80e25a4481ea4da08e7c40cc0dedf6a1 : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a622ad104644823ac3a074645be157d : SUCCESS in 4m 13s\n- ironic-ui-tox-codespell https://zuul.opendev.org/t/openstack/build/68e4f92190444cf98682bf53b2859660 : SUCCESS in 4m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"69039a7d1d1457a67c643d54f0958cd1f96f0e61","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2026-05-04 21:14:09.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"58b264119f63f469d7047043b0a7ae1e01872b4e","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2026-05-04 21:14:16.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":3},{"id":"21cf7c5b72c45d021be9adfbae2027cba8ffa3b8","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-05-05 13:39:13.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"bf8e3de5efd0408484298c6657dd127ba1618bbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 13:39:26.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2dde9617e0d02e5dddda71e773e3fd33aabacbb5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 13:52:06.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b99294a470f8422581236ed0ad98b36f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3a6249dd3cb3409ea3ec25cee238f8ef : SUCCESS in 2m 46s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a58eb5f0da804f3492ea44820e4e29d5 : SUCCESS in 3m 16s\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/c5253e9f9d774dcd930f6026b7517b2b : SUCCESS in 5m 48s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/cafebf811d864b5f81a6192595f9d2ff : SUCCESS in 6m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ac1282d2190b4e128dd7e5c13ea655bb : SUCCESS in 3m 07s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/b3cf36888d8c484195748873fa9c66aa : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21c2819bbde848298db1b9b65fc181fb : SUCCESS in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4a3af750e80c47d98dba993e2866c24d : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8449190eb0574982abc8686929d13e33 : SUCCESS in 8m 54s","accounts_in_message":[],"_revision_number":3},{"id":"dc54278da80cc7e60196d7c8ffcd6aa1d2a41d14","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 13:52:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"8a3bd12e4f0aab7b58df693ee9b7c01cc68d2631","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 13:53:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fba314981b7743b081da41c97b417e49\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06dbfdd6008848ae845d0d2853984925 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"01b3f8319599b1bcb13c6e5135e4cb39fe704f8a","revisions":{"4fac00edf3bf638cd72b1cd4bffd9b43a28a721b":{"kind":"REWORK","_number":1,"created":"2026-05-04 20:38:51.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/81/987281/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/81/987281/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/81/987281/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/81/987281/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/81/987281/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/81/987281/1"}}},"commit":{"parents":[{"commit":"273780538b046eca1007bd5ea61b56ed52429a87","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/273780538b046eca1007bd5ea61b56ed52429a87"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-04 20:38:51.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-04 20:38:51.000000000","tz":0},"subject":"Update packaging configuration","message":"Update packaging configuration\n\nEnable package discovery and resolve setuptools deprecation warnings\naround license configuration.\n\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Ib9ad8b667fc8f70feee32a625333c270ec1cde97\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/4fac00edf3bf638cd72b1cd4bffd9b43a28a721b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/4fac00edf3bf638cd72b1cd4bffd9b43a28a721b"}]},"branch":"refs/heads/master","conflicts":{"contains_conflicts":false}},"0efafb5bea137d8f3acba041687284718a8ebf76":{"kind":"REWORK","_number":2,"created":"2026-05-04 20:41:18.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/81/987281/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/81/987281/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/81/987281/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/81/987281/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/81/987281/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/81/987281/2"}}},"commit":{"parents":[{"commit":"273780538b046eca1007bd5ea61b56ed52429a87","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/273780538b046eca1007bd5ea61b56ed52429a87"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-04 20:38:51.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-04 20:41:18.000000000","tz":0},"subject":"Update packaging configuration","message":"Update packaging configuration\n\nEnable package discovery and resolve setuptools deprecation warnings\naround license configuration.\n\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Ib9ad8b667fc8f70feee32a625333c270ec1cde97\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/0efafb5bea137d8f3acba041687284718a8ebf76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/0efafb5bea137d8f3acba041687284718a8ebf76"}]},"branch":"refs/heads/master"},"01b3f8319599b1bcb13c6e5135e4cb39fe704f8a":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-05-04 20:41:29.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/81/987281/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-ui","ref":"refs/changes/81/987281/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/81/987281/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/81/987281/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-ui refs/changes/81/987281/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-ui refs/changes/81/987281/3"}}},"commit":{"parents":[{"commit":"273780538b046eca1007bd5ea61b56ed52429a87","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/273780538b046eca1007bd5ea61b56ed52429a87"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-04 20:38:51.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-04 20:41:29.000000000","tz":0},"subject":"Update packaging configuration","message":"Update packaging configuration\n\nEnable package discovery and resolve setuptools deprecation warnings\naround license configuration.\n\nChange-Id: Ib9ad8b667fc8f70feee32a625333c270ec1cde97\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/01b3f8319599b1bcb13c6e5135e4cb39fe704f8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-ui/commit/01b3f8319599b1bcb13c6e5135e4cb39fe704f8a"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
