)]}'
{"id":"openstack%2Fironic~1005083","triplet_id":"openstack%2Fironic~master~Ic0ae24b17fe20df08fca1cfa0ea116ab26abbae4","project":"openstack/ironic","branch":"master","attention_set":{},"removed_from_attention_set":{"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2026-09-11 15:49:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic0ae24b17fe20df08fca1cfa0ea116ab26abbae4","subject":"Fix device class detection for PCI devices during Redfish inspection","status":"MERGED","created":"2026-09-10 16:25:02.000000000","updated":"2026-09-11 15:50:43.000000000","submitted":"2026-09-11 15:49:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005083","meta_rev_id":"a5df7f588b56cedec437bfaff64670942cbf4c49","_number":1005083,"virtual_id_number":1005083,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-11 15:49:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":2,"date":"2026-09-11 14:29:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-11 13:32:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":1,"date":"2026-09-11 14:29:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-10 16:48:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-10 18:25:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-11 13:32:25.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2026-09-11 14:29:37.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"}],"messages":[{"id":"78381545c6fd47feb0e6e61b41ab47d3e2148609","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2026-09-10 16:25:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9c9a4909497004307d752b702ee380fff8982f97","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 16:48:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e624257fa894cf5be6537241edd008e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/61f673ec29164fa3a269aae31e821953 : SUCCESS in 9m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d8ca7c08ef3b42c9be26c09cfb10b1a1 : SUCCESS in 8m 11s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f98d3b0da8a940cabc310209b590f1a6 : SUCCESS in 17m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a4d54609832d2ea166689612f9b92a167df7fe0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 18:25:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2e7cac5c4b94a99953f0b2f9fe3e0e9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d81b523def34487e924299e35e0ab185 : SUCCESS in 11m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2529dd96d474f44b520704b2d79ca3d : SUCCESS in 7m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/33e7bcb28f094eeda73c9e9b805d3e50 : SUCCESS in 8m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/539c97d159e843099d51f295c9445b97 : SUCCESS in 4m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d7a544cc69ff45d69016aab00b6da4df : SUCCESS in 13m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44c1e707401d4a8698c905230e326dba : SUCCESS in 16m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7100377f31b4cce87186ae6d4cc4d9c : SUCCESS in 22m 06s\n- ironic-tox-mypy https://zuul.opendev.org/t/openstack/build/819846abefd145379d2acb5d7119ae34 : SUCCESS in 5m 21s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/88c144f737be42dc88e20e2fb29c11ee : SUCCESS in 5m 17s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/07872999c3dc4f55bb5a7c0356e2d7a5 : SUCCESS in 12m 15s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/7f3ac9cd9ea04507a889b49aabb9039d : SUCCESS in 9m 33s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/0f8ee8756808497e8687aee58f0ce244 : SUCCESS in 26m 01s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/a233744a7fb549278fd9ce1002c9b6ff : SUCCESS in 54m 21s (non-voting)\n- ironic-tempest-vnc-container https://zuul.opendev.org/t/openstack/build/0f014b1aaddc4d0a81439623222543ee : SUCCESS in 31m 08s (non-voting)\n- ironic-tempest-standalone-advanced https://zuul.opendev.org/t/openstack/build/a2a524a0a37b4c3382945edd1d5ee69e : SUCCESS in 55m 49s (non-voting)\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/2df20e9b95564624a114ce7ffcbae905 : SUCCESS in 41m 51s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/804216ea9dc047b5a75f8c975792745b : SUCCESS in 45m 09s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/c1233082bf2247249a4ba35dcf20da83 : SUCCESS in 30m 00s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/8fe8c60664a24316ac4f34e61a65bbb3 : SUCCESS in 49m 03s (non-voting)\n- ironic-tempest-ovn-uefi-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/2f654af5d8794eef9606e82244b63d84 : SUCCESS in 58m 08s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-multinode https://zuul.opendev.org/t/openstack/build/092560e0b267490da00ff83349fe65e3 : SUCCESS in 1h 43m 21s (non-voting)\n- ironic-tempest-ipa-wholedisk-direct-multinode-shard https://zuul.opendev.org/t/openstack/build/812c0bbdc77f4e53a9cc6a131883ed6b : SUCCESS in 1h 15m 26s (non-voting)\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/f3bdef44578740e39ffcffd19f0b501a : SUCCESS in 1h 11m 07s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/09955cd7637d4261976389915c8028d0 : SUCCESS in 30m 19s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/ecac0684d0c74b778db4e9b2bda40847 : SUCCESS in 47m 35s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/8005a4106c624002bcc2a4298c056001 : SUCCESS in 1h 17m 06s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/be27f19476824b5096851aceb9fb6c26 : SUCCESS in 1h 08m 30s (non-voting)\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/1641c5ca5a05403f898d678cca1a4a42 : SUCCESS in 1h 14m 59s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/02a1c9b9b9574276b4cc92a0363d00bc : FAILURE in 30m 58s (non-voting)\n- metal3-baremetal-operator-functional https://zuul.opendev.org/t/openstack/build/1872274c1d664e9cb2d7b63784701dab : FAILURE in 51m 19s (non-voting)\n- metal3-ironic-standalone-operator-functional https://zuul.opendev.org/t/openstack/build/ff25ed942cac40ac8e3de95cda17f57d : SUCCESS in 29m 40s (non-voting)\n- ironic-standalone-aarch64 https://zuul.opendev.org/t/openstack/build/078a0390f6c24fafaa9bcd07a8a9883f : SUCCESS in 51m 12s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/250b8bbe487a4ff7ad0c5ff34061111d : SUCCESS in 30m 40s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/bcd5e932ac8b4db1a9dc226cc1053784 : FAILURE in 28m 48s (non-voting)\n- bifrost-integration-on-ubuntu-noble https://zuul.opendev.org/t/openstack/build/79ee325e93ac49a7a6cad8175e8ada2e : SUCCESS in 58m 47s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c51f1c143a934409bf82272a98c67743 : SUCCESS in 36m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"23830a183fc0f6c3916ec0c27e96a9926cd9ad51","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2026-09-11 13:32:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c9a76c44a606e977bd25756062020bef6b0b8bc7","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-09-11 14:29:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2ca152ccbc8a45fed158d3c16083dc87ab0b20a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 14:30:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5b7f99272fbc4d86ec5ba3910228c8bb3e64c147","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 15:49:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6e1a747bec64747b15f3604648c9efd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffc7e492847f4ebf8107c6271ec7f995 : SUCCESS in 6m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0f8e8a4cea9245ed83e9778ba9251d59 : SUCCESS in 8m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fcbbf6a627e84e3f9f404a83240d760a : SUCCESS in 8m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0729861218cc4356add2f4a73b3723a6 : SUCCESS in 16m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb2e5d05205f47caa645a8737c2cd375 : SUCCESS in 20m 13s\n- ironic-tox-mypy https://zuul.opendev.org/t/openstack/build/5a142414763b4969838bbb18c0f592b3 : SUCCESS in 4m 55s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/c1e5994c8702456990ea712bcc88123f : SUCCESS in 6m 00s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/43ae3a9e78924776afa0c0fc03e89c4d : SUCCESS in 7m 52s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/99939144d449411da36a4cdfff345ee3 : SUCCESS in 38m 32s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/8e919b3a97bf4d448d1b8407c6940c4c : SUCCESS in 34m 42s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/06d97a500cfe4e52892a250822e38bf3 : SUCCESS in 35m 11s\n- ironic-tempest-uefi-redfish-https https://zuul.opendev.org/t/openstack/build/659389bc34db4cf68fd5da4c1c1fe6d5 : SUCCESS in 37m 55s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/70e34d4563c446ad984b65356130049a : SUCCESS in 33m 12s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/4d5f6e5d4998478590d1f532ae7f983c : SUCCESS in 41m 08s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/c7eebded6c414fadb8757932d9bb9ada : SUCCESS in 47m 01s\n- ironic-tempest-uefi-redfish-vmedia-4k https://zuul.opendev.org/t/openstack/build/202d48dc4a304aae97a2469883d3bce8 : SUCCESS in 1h 13m 07s","accounts_in_message":[],"_revision_number":1},{"id":"47ba2cdb5c4ca8ca0f2faa253aaa7d12e010479a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 15:49:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a5df7f588b56cedec437bfaff64670942cbf4c49","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 15:50:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aecccf7a6b504907b5b346eecd73b1ae\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d35a5fb9250f4eecb21145a8882ec54c : SUCCESS in 1m 26s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9c2e228ead74c5c9a7a6c1e7e2db23d : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cfe1af450e1c3afac6509bf18b3af24b2d72bf31","revisions":{"cfe1af450e1c3afac6509bf18b3af24b2d72bf31":{"kind":"REWORK","_number":1,"created":"2026-09-10 16:25:02.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/83/1005083/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/83/1005083/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/83/1005083/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/83/1005083/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/83/1005083/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/83/1005083/1"}}},"commit":{"parents":[{"commit":"de09a8bbb315d83dbb425c75b973ff402e32566b","subject":"Merge \"Replace deprecated pyparsing interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/de09a8bbb315d83dbb425c75b973ff402e32566b"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2026-09-10 16:20:40.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2026-09-10 16:23:50.000000000","tz":120},"subject":"Fix device class detection for PCI devices during Redfish inspection","message":"Fix device class detection for PCI devices during Redfish inspection\n\nSushy uses Python enums. Quite annoyingly, str(enum) returns something\nlike \"DeviceClass.NETWORK_CONTROLLER\" instead of expected value.\n\nChange-Id: Ic0ae24b17fe20df08fca1cfa0ea116ab26abbae4\nSigned-off-by: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/cfe1af450e1c3afac6509bf18b3af24b2d72bf31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/cfe1af450e1c3afac6509bf18b3af24b2d72bf31"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
