)]}'
{"id":"openstack%2Fironic~655596","triplet_id":"openstack%2Fironic~master~I7f23cd29fe7b1deea3b14cf5d7e8ad21ff824c74","project":"openstack/ironic","branch":"master","topic":"fix_spelling","hashtags":[],"change_id":"I7f23cd29fe7b1deea3b14cf5d7e8ad21ff824c74","subject":"Correct spelling errors","status":"MERGED","created":"2019-04-25 01:44:26.000000000","updated":"2019-04-25 16:54:54.000000000","submitted":"2019-04-25 16:54:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655596-1556211294168-7785bd57","meta_rev_id":"89b776c99a2e5556f74d4a53daa72f5ad7348c09","_number":655596,"virtual_id_number":655596,"owner":{"_account_id":29139,"name":"shenxindi","email":"shenxindi_yewu@cmss.chinamobile.com","username":"shenxindi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-04-25 12:34:00.000000000","_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":2,"date":"2019-04-25 16:54:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"recommended":{"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},"all":[{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-04-25 06:48:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},{"value":2,"date":"2019-04-25 15:30:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},{"value":1,"date":"2019-04-25 15:30:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-25 06:48:24.000000000","updated_by":{"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},"reviewer":{"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},"state":"REVIEWER"},{"updated":"2019-04-25 12:34:00.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"REVIEWER"},{"updated":"2019-04-25 15:30:57.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2019-04-25 16:54:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"140553ce93dedadf7c69112be3aedf4d2abf038f","author":{"_account_id":29139,"name":"shenxindi","email":"shenxindi_yewu@cmss.chinamobile.com","username":"shenxindi"},"date":"2019-04-25 01:44:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9d4b89a0641b4dbdda72c0fff6a98ab129e2c693","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 03:13:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/96/655596/1/check/openstack-tox-cover/17400b4/cover/ : SUCCESS in 11m 14s\n- openstack-tox-lower-constraints http://logs.openstack.org/96/655596/1/check/openstack-tox-lower-constraints/e359cdf/ : SUCCESS in 6m 13s\n- openstack-tox-pep8 http://logs.openstack.org/96/655596/1/check/openstack-tox-pep8/21c256e/ : SUCCESS in 4m 47s\n- openstack-tox-py27 http://logs.openstack.org/96/655596/1/check/openstack-tox-py27/7ca6c64/ : SUCCESS in 6m 31s\n- openstack-tox-py36 http://logs.openstack.org/96/655596/1/check/openstack-tox-py36/7f9b24a/ : SUCCESS in 7m 56s\n- openstack-tox-docs http://logs.openstack.org/96/655596/1/check/openstack-tox-docs/a1ff025/html/ : SUCCESS in 6m 40s\n- ironic-tox-unit-with-driver-libs http://logs.openstack.org/96/655596/1/check/ironic-tox-unit-with-driver-libs/35d51de/ : SUCCESS in 7m 37s\n- ironic-tox-unit-with-driver-libs-python3 http://logs.openstack.org/96/655596/1/check/ironic-tox-unit-with-driver-libs-python3/9f68904/ : SUCCESS in 7m 30s\n- bifrost-integration-tinyipa-ubuntu-xenial http://logs.openstack.org/96/655596/1/check/bifrost-integration-tinyipa-ubuntu-xenial/b3a64ee/ : SUCCESS in 16m 56s (non-voting)\n- metalsmith-integration-glance-localboot-centos7 http://logs.openstack.org/96/655596/1/check/metalsmith-integration-glance-localboot-centos7/06d1b00/ : SUCCESS in 1h 20m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b34748e1a81765b7013e38288e0e6dc6bd8aea83","author":{"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},"date":"2019-04-25 06:48:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2c041f5bea783f9fdbeca9e9517e5ee78e933139","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2019-04-25 12:34:00.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-R640-UEFI-py2-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/96/655596/1/check/dellemc-poweredge-R740-R640-UEFI-py2-tempest-dsvm-ironic-idrac/c7f720b/ : FAILURE in 13m 01s\n- dellemc-DSS9000-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/96/655596/1/check/dellemc-DSS9000-tempest-dsvm-ironic-idrac/aba8066/ : SUCCESS in 47m 54s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/96/655596/1/check/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac/0879c47/ : SUCCESS in 43m 29s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/96/655596/1/check/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac/3039508/ : SUCCESS in 40m 16s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/96/655596/1/check/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac/eba8806/ : SUCCESS in 43m 54s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/96/655596/1/check/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish/8de462c/ : SUCCESS in 39m 54s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/96/655596/1/check/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish/61a52a1/ : SUCCESS in 40m 15s\n- dellemc-DSS9000-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/96/655596/1/check/dellemc-DSS9000-tempest-dsvm-ironic-redfish/2127244/ : SUCCESS in 42m 58s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/96/655596/1/check/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi/433444c/ : SUCCESS in 38m 37s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/96/655596/1/check/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi/434454a/ : SUCCESS in 1h 06m 40s","accounts_in_message":[],"_revision_number":1},{"id":"cf565ea510a5ce630a94fc1b30dfa0cb76ce26c7","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-04-25 15:30:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nTrivial test fix, approving","accounts_in_message":[],"_revision_number":1},{"id":"b4891f993112ee2532b35bddd62e7eed14252e82","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 15:31:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bb8499cf22872d847cb27ffa72bf5f8aa776e2e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 16:54:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/96/655596/1/gate/openstack-tox-lower-constraints/541f043/ : SUCCESS in 6m 09s\n- openstack-tox-pep8 http://logs.openstack.org/96/655596/1/gate/openstack-tox-pep8/82adaab/ : SUCCESS in 5m 19s\n- openstack-tox-py27 http://logs.openstack.org/96/655596/1/gate/openstack-tox-py27/3abde2e/ : SUCCESS in 7m 06s\n- openstack-tox-py36 http://logs.openstack.org/96/655596/1/gate/openstack-tox-py36/c7119f5/ : SUCCESS in 7m 51s\n- openstack-tox-docs http://logs.openstack.org/96/655596/1/gate/openstack-tox-docs/a110553/html/ : SUCCESS in 6m 29s\n- ironic-tox-unit-with-driver-libs http://logs.openstack.org/96/655596/1/gate/ironic-tox-unit-with-driver-libs/fcd8d71/ : SUCCESS in 7m 22s\n- ironic-tox-unit-with-driver-libs-python3 http://logs.openstack.org/96/655596/1/gate/ironic-tox-unit-with-driver-libs-python3/b9367a7/ : SUCCESS in 7m 27s","accounts_in_message":[],"_revision_number":1},{"id":"b4ac2afefd716aca9e14ad4fa0ff05d5e41f561a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 16:54:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"810670cf134a28adecada566895e685af6c67ef2","revisions":{"810670cf134a28adecada566895e685af6c67ef2":{"kind":"REWORK","_number":1,"created":"2019-04-25 01:44:26.000000000","uploader":{"_account_id":29139,"name":"shenxindi","email":"shenxindi_yewu@cmss.chinamobile.com","username":"shenxindi"},"ref":"refs/changes/96/655596/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/96/655596/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/655596/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/655596/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/655596/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/96/655596/1"}}},"commit":{"parents":[{"commit":"95abbde644ed1ff6606409e9ad3c7e85452a15a3","subject":"Merge \"Jsonschema 3.0.1: Binding the schema to draft-04\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/95abbde644ed1ff6606409e9ad3c7e85452a15a3"}]}],"author":{"name":"shenxindi","email":"shenxindi@cmss.chinamobile.com","date":"2019-04-25 01:36:32.000000000","tz":480},"committer":{"name":"shenxindi","email":"shenxindi@cmss.chinamobile.com","date":"2019-04-25 01:36:32.000000000","tz":480},"subject":"Correct spelling errors","message":"Correct spelling errors\n\nChange-Id: I7f23cd29fe7b1deea3b14cf5d7e8ad21ff824c74\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/810670cf134a28adecada566895e685af6c67ef2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/810670cf134a28adecada566895e685af6c67ef2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
