)]}'
{"id":"openstack%2Fironic~662963","triplet_id":"openstack%2Fironic~stable%2Fstein~Ia0dd6883b4c15b8a0646ee765689343c116b2bcd","project":"openstack/ironic","branch":"stable/stein","topic":"update-sphinx-req-stable/stein","hashtags":[],"change_id":"Ia0dd6883b4c15b8a0646ee765689343c116b2bcd","subject":"Update sphinx requirements","status":"MERGED","created":"2019-06-04 08:04:54.000000000","updated":"2019-06-04 11:50:34.000000000","submitted":"2019-06-04 11:50:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662963-1559649034393-03aab1a2","meta_rev_id":"5cf1d875dcd6cf9118537948f1d8c539bfceac94","_number":662963,"virtual_id_number":662963,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-06-04 09:40:47.000000000","_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":2,"date":"2019-06-04 11:50:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-04 11:23:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-04 11:23:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-04 09:40:47.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"REVIEWER"},{"updated":"2019-06-04 11:23:21.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2019-06-04 11:50:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a8c68b7b45055801bdd81f1f9982bd115e5defcb","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-06-04 08:04:54.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"135a200106ce8f9741d0840c4c3bdab7c5c3a37a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 08:19:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/63/662963/1/check/requirements-check/7c5a53b/ : SUCCESS in 3m 01s\n- openstack-tox-pep8 http://logs.openstack.org/63/662963/1/check/openstack-tox-pep8/f094354/ : SUCCESS in 4m 50s\n- openstack-tox-docs http://logs.openstack.org/63/662963/1/check/openstack-tox-docs/3f081b6/html/ : SUCCESS in 7m 30s\n- build-openstack-releasenotes http://logs.openstack.org/63/662963/1/check/build-openstack-releasenotes/ea68f71/html/ : SUCCESS in 8m 38s\n- ironic-tox-unit-with-driver-libs http://logs.openstack.org/63/662963/1/check/ironic-tox-unit-with-driver-libs/9a31350/ : SUCCESS in 7m 07s\n- ironic-tox-unit-with-driver-libs-python3 http://logs.openstack.org/63/662963/1/check/ironic-tox-unit-with-driver-libs-python3/ec16014/ : SUCCESS in 7m 00s","accounts_in_message":[],"_revision_number":1},{"id":"ba6e90c0992aed08fd6370d82d170eb7b593dd59","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2019-06-04 09:40:47.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-R640-UEFI-py2-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/63/662963/1/check/dellemc-poweredge-R740-R640-UEFI-py2-tempest-dsvm-ironic-idrac/aff4366/ : SUCCESS in 46m 32s\n- dellemc-DSS9000-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/63/662963/1/check/dellemc-DSS9000-tempest-dsvm-ironic-idrac/43e67de/ : SUCCESS in 40m 57s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/63/662963/1/check/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac/8a70ed4/ : SUCCESS in 46m 13s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/63/662963/1/check/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac/6421cf9/ : FAILURE in 49m 22s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/63/662963/1/check/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac/0c027c9/ : FAILURE in 35m 00s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/63/662963/1/check/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish/2d25312/ : SUCCESS in 38m 22s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/63/662963/1/check/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish/2093d3c/ : FAILURE in 46m 52s\n- dellemc-DSS9000-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/63/662963/1/check/dellemc-DSS9000-tempest-dsvm-ironic-redfish/bd77878/ : SUCCESS in 38m 10s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/63/662963/1/check/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi/c39cc0d/ : FAILURE in 47m 19s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/63/662963/1/check/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi/84ce77d/ : FAILURE in 42m 04s","accounts_in_message":[],"_revision_number":1},{"id":"17855be85779b9d0cf1a40b7015fddca09b030db","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-06-04 11:23:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nReflects global-requirements, approving","accounts_in_message":[],"_revision_number":1},{"id":"8bf1aae8323751a12e3afd6f92022374352b306e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 11:23:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a29bc9cbbc63dff790782c6ed34e123a75449668","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 11:50:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/63/662963/1/gate/requirements-check/7473271/ : SUCCESS in 2m 41s\n- openstack-tox-pep8 http://logs.openstack.org/63/662963/1/gate/openstack-tox-pep8/0413955/ : SUCCESS in 5m 57s\n- openstack-tox-docs http://logs.openstack.org/63/662963/1/gate/openstack-tox-docs/d669b0c/html/ : SUCCESS in 5m 33s\n- build-openstack-releasenotes http://logs.openstack.org/63/662963/1/gate/build-openstack-releasenotes/51cadec/html/ : SUCCESS in 8m 47s\n- ironic-tox-unit-with-driver-libs http://logs.openstack.org/63/662963/1/gate/ironic-tox-unit-with-driver-libs/ed35590/ : SUCCESS in 8m 28s\n- ironic-tox-unit-with-driver-libs-python3 http://logs.openstack.org/63/662963/1/gate/ironic-tox-unit-with-driver-libs-python3/ce4d447/ : SUCCESS in 7m 31s","accounts_in_message":[],"_revision_number":1},{"id":"af4530b60d1f3a60f6b64ed97aa852eb8e378733","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 11:50:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e48f1c4386595189678f99defb4351ad69c202a8","revisions":{"e48f1c4386595189678f99defb4351ad69c202a8":{"kind":"REWORK","_number":1,"created":"2019-06-04 08:04:54.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/63/662963/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/63/662963/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/63/662963/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/63/662963/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/63/662963/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/63/662963/1"}}},"commit":{"parents":[{"commit":"596e82baab4953cf5b4a372aad9f8731bb6b1b58","subject":"devstack: configure rabbit outside of API configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/596e82baab4953cf5b4a372aad9f8731bb6b1b58"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-05-22 13:03:52.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-06-04 08:04:54.000000000","tz":0},"subject":"Update sphinx requirements","message":"Update sphinx requirements\n\nFollowing recent changes to global requirements [1], we need\nto update sphinx requirements locally.\n\n[1] If558f184c959e4b63b56dec3ca1571d1034cfe5c\n\nChange-Id: Ia0dd6883b4c15b8a0646ee765689343c116b2bcd\n(cherry picked from commit 9ed91013944544559f4011ecf07817e929a41985)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/e48f1c4386595189678f99defb4351ad69c202a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/e48f1c4386595189678f99defb4351ad69c202a8"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
