)]}'
{"id":"openstack%2Fironic~662964","triplet_id":"openstack%2Fironic~stable%2Frocky~Ia0dd6883b4c15b8a0646ee765689343c116b2bcd","project":"openstack/ironic","branch":"stable/rocky","topic":"update-sphinx-req-stable/rocky","hashtags":[],"change_id":"Ia0dd6883b4c15b8a0646ee765689343c116b2bcd","subject":"Update sphinx requirements","status":"MERGED","created":"2019-06-04 08:05:06.000000000","updated":"2019-06-04 11:50:35.000000000","submitted":"2019-06-04 11:50:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662964-1559649035815-d9d96933","meta_rev_id":"dc914348ba19d0034e2280c58ecbd2933cc06042","_number":662964,"virtual_id_number":662964,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-06-04 11:10:13.000000000","_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":2,"date":"2019-06-04 11:50:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-04 11:23:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-04 11:23:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-04 11:10:13.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"REVIEWER"},{"updated":"2019-06-04 11:23:41.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2019-06-04 11:50:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"801ca6aec8f476da7b6811a26c0e88e726f2cc42","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-06-04 08:05:06.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d3180471071442cf10ce8444d47d2d59c3c71ce6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 08:21:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/64/662964/1/check/openstack-tox-pep8/54c63c4/ : SUCCESS in 4m 36s\n- build-openstack-sphinx-docs http://logs.openstack.org/64/662964/1/check/build-openstack-sphinx-docs/17393b7/html/ : SUCCESS in 6m 31s\n- requirements-check http://logs.openstack.org/64/662964/1/check/requirements-check/a6aa014/ : SUCCESS in 2m 51s\n- build-openstack-releasenotes http://logs.openstack.org/64/662964/1/check/build-openstack-releasenotes/9761a3f/html/ : SUCCESS in 9m 06s","accounts_in_message":[],"_revision_number":1},{"id":"edd95d6602262b47c20ede1591119343bbc0472a","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2019-06-04 11:10:13.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-R640-UEFI-py2-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/64/662964/1/check/dellemc-poweredge-R740-R640-UEFI-py2-tempest-dsvm-ironic-idrac/1c0ecdb/ : FAILURE in 39m 06s\n- dellemc-DSS9000-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/64/662964/1/check/dellemc-DSS9000-tempest-dsvm-ironic-idrac/0dce9e9/ : SUCCESS in 40m 40s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/64/662964/1/check/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac/9e82d56/ : SUCCESS in 45m 22s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/64/662964/1/check/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac/a2bae89/ : FAILURE in 47m 20s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/64/662964/1/check/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac/039d3e1/ : SUCCESS in 42m 59s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/64/662964/1/check/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish/d420036/ : FAILURE in 6s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/64/662964/1/check/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish/95ac3d9/ : FAILURE in 42m 17s\n- dellemc-DSS9000-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/64/662964/1/check/dellemc-DSS9000-tempest-dsvm-ironic-redfish/805d9ff/ : SUCCESS in 41m 09s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/64/662964/1/check/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi/580e72b/ : FAILURE in 43m 45s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/64/662964/1/check/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi/d590fef/ : FAILURE in 42m 58s","accounts_in_message":[],"_revision_number":1},{"id":"78b303f97f7b796d94473b4fde76d487f0faabed","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-06-04 11:23:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nReflects global-requirements, approving","accounts_in_message":[],"_revision_number":1},{"id":"c44164e9e4bec2bf0226d49dd5f9cb64ec1d6bb5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 11:23:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e4ddb7eb179c3454b98fbe96d9546b460c563d4d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 11:50:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6ea31fd2f4ab3e04f0b32bc05851898e344da01c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 11:50:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/64/662964/1/gate/openstack-tox-pep8/d603c39/ : SUCCESS in 5m 24s\n- build-openstack-sphinx-docs http://logs.openstack.org/64/662964/1/gate/build-openstack-sphinx-docs/c14c57d/html/ : SUCCESS in 5m 49s\n- requirements-check http://logs.openstack.org/64/662964/1/gate/requirements-check/13fb295/ : SUCCESS in 3m 49s\n- build-openstack-releasenotes http://logs.openstack.org/64/662964/1/gate/build-openstack-releasenotes/5784004/html/ : SUCCESS in 8m 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1e1f1723ac970fa909b15b7dbbfa1a17d3e206ba","revisions":{"1e1f1723ac970fa909b15b7dbbfa1a17d3e206ba":{"kind":"REWORK","_number":1,"created":"2019-06-04 08:05:06.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/64/662964/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/64/662964/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/64/662964/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/64/662964/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/64/662964/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/64/662964/1"}}},"commit":{"parents":[{"commit":"a2dea908c2156145402bcd3e040b4b7f3198bf5d","subject":"Merge \"Ansible module fix: stream_url\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/a2dea908c2156145402bcd3e040b4b7f3198bf5d"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-05-22 13:03:52.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-06-04 08:05:06.000000000","tz":0},"subject":"Update sphinx requirements","message":"Update sphinx requirements\n\nFollowing recent changes to global requirements [1], we need\nto update sphinx requirements locally.\n\n[1] If558f184c959e4b63b56dec3ca1571d1034cfe5c\n\nChange-Id: Ia0dd6883b4c15b8a0646ee765689343c116b2bcd\n(cherry picked from commit 9ed91013944544559f4011ecf07817e929a41985)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1e1f1723ac970fa909b15b7dbbfa1a17d3e206ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1e1f1723ac970fa909b15b7dbbfa1a17d3e206ba"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
