)]}'
{"id":"openstack%2Fironic~679403","triplet_id":"openstack%2Fironic~master~I80ba4c5575feae9b03bb8fa30eba68312d4c9ce2","project":"openstack/ironic","branch":"master","topic":"ironic_specs_link","hashtags":[],"change_id":"I80ba4c5575feae9b03bb8fa30eba68312d4c9ce2","subject":"Add ironic-specs link to readme.rst","status":"MERGED","created":"2019-08-30 09:20:30.000000000","updated":"2019-08-31 12:26:10.000000000","submitted":"2019-08-31 12:22:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679403-1567254165289-1d8aaa17","meta_rev_id":"03adfc8178ae8ae6e38788f2c86bbe2af887847b","_number":679403,"virtual_id_number":679403,"owner":{"_account_id":27383,"name":"Shuo Liu","email":"liushuobj@inspur.com","username":"liushuobj"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"date":"2019-08-30 13:43:09.000000000","_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":2,"date":"2019-08-31 12:22:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":2,"date":"2019-08-31 12:01:21.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-30 09:26:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},{"value":2,"date":"2019-08-30 09:21:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":1,"date":"2019-08-31 12:01:21.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-30 09:21:48.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2019-08-30 09:26:13.000000000","updated_by":{"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},"reviewer":{"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},"state":"REVIEWER"},{"updated":"2019-08-30 13:43:09.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"REVIEWER"},{"updated":"2019-08-31 12:01:21.000000000","updated_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"reviewer":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"state":"REVIEWER"},{"updated":"2019-08-31 12:22:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d81942b5d2e9e215a67a6aa4e1a4f408bf05e784","author":{"_account_id":27383,"name":"Shuo Liu","email":"liushuobj@inspur.com","username":"liushuobj"},"date":"2019-08-30 09:20:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7fc528d9707650945ed3bc96d59340a19387c678","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-08-30 09:21:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"192be6935ee4d6587430032fe8099e402ca9aaa7","author":{"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},"date":"2019-08-30 09:26:13.000000000","message":"Patch Set 1: Code-Review+1\n\nGood addition, thanks!","accounts_in_message":[],"_revision_number":1},{"id":"0ba754bd8bfcd4683be8a333fca983ae101d564b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 10:16:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/10cce3cc73e846c6b014ff5211351e27 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3083c30090c34b3494639038dc2d6931 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eadbe2fa882844f1b6087d40ff00dd29 : SUCCESS in 13m 54s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/ae16718061bb4ef9ba25580c9fc15f90 : SUCCESS in 8m 50s\n- ironic-tox-unit-with-driver-libs-python3 https://zuul.opendev.org/t/openstack/build/08555afb80bd462183ea3947c14d4eb3 : SUCCESS in 8m 41s","accounts_in_message":[],"_revision_number":1},{"id":"6194c3ded8d6a07f7c87700575151f87616fefef","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2019-08-30 13:43:09.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-R640-UEFI-py2-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-py2-tempest-dsvm-ironic-idrac/2019-08-30:04:20/679403/1/ : SUCCESS in 46m 13s\n- dellemc-DSS9000-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-DSS9000-tempest-dsvm-ironic-idrac/2019-08-30:04:20/679403/1/ : SUCCESS in 42m 13s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac/2019-08-30:04:20/679403/1/ : SUCCESS in 43m 17s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac/2019-08-30:04:20/679403/1/ : SUCCESS in 46m 29s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac/2019-08-30:04:20/679403/1/ : FAILURE in 31m 57s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish/2019-08-30:04:20/679403/1/ : SUCCESS in 42m 59s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish/2019-08-30:04:20/679403/1/ : SUCCESS in 44m 45s\n- dellemc-DSS9000-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-DSS9000-tempest-dsvm-ironic-redfish/2019-08-30:04:20/679403/1/ : SUCCESS in 41m 22s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi/2019-08-30:04:20/679403/1/ : SUCCESS in 41m 49s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi/2019-08-30:04:20/679403/1/ : FAILURE in 44m 23s","accounts_in_message":[],"_revision_number":1},{"id":"f08e2edbdbdd2d183356f8f4e024653a3effbd7b","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2019-08-31 12:01:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThank you Shuo.","accounts_in_message":[],"_revision_number":1},{"id":"d7c5ed8c5fa219b58db927784ac071275d7babe1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-31 12:01:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"19c3811582b1bb1cdca487f65328ae23b2fed664","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-31 12:22:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0019eb076d224041a4c781feda13136d : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8efeb7e0150c4762b4bb846e814aeb68 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ea24a3eaab548b48768d579027eaad8 : SUCCESS in 8m 21s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/868b54198d4440779ba2f78eb86e0fa3 : SUCCESS in 7m 24s\n- ironic-tox-unit-with-driver-libs-python3 https://zuul.opendev.org/t/openstack/build/0339615cf1814469881e35c518d78f2f : SUCCESS in 7m 49s","accounts_in_message":[],"_revision_number":1},{"id":"ff19eea350f5556a1a1f0484a05f99520221b2c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-31 12:22:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fc516491450909c7c07055379e9b2c421eb00763","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-31 12:26:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34f867d8cc00459289e7c8687168b9ce : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3747e14189bbf703eb1e9b85b660a345116d5b51","revisions":{"3747e14189bbf703eb1e9b85b660a345116d5b51":{"kind":"REWORK","_number":1,"created":"2019-08-30 09:20:30.000000000","uploader":{"_account_id":27383,"name":"Shuo Liu","email":"liushuobj@inspur.com","username":"liushuobj"},"ref":"refs/changes/03/679403/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/03/679403/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/03/679403/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/03/679403/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/03/679403/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/03/679403/1"}}},"commit":{"parents":[{"commit":"3c9f2a834e9256d961e0c622d47bc7699f0cd070","subject":"Merge \"Adding file uri support for ipa image location\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/3c9f2a834e9256d961e0c622d47bc7699f0cd070"}]}],"author":{"name":"liushuobj","email":"liushuobj@inspur.com","date":"2019-08-30 09:16:09.000000000","tz":480},"committer":{"name":"liushuobj","email":"liushuobj@inspur.com","date":"2019-08-30 09:16:09.000000000","tz":480},"subject":"Add ironic-specs link to readme.rst","message":"Add ironic-specs link to readme.rst\n\nChange-Id: I80ba4c5575feae9b03bb8fa30eba68312d4c9ce2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/3747e14189bbf703eb1e9b85b660a345116d5b51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/3747e14189bbf703eb1e9b85b660a345116d5b51"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
