)]}'
{"id":"openstack%2Fironic~680643","triplet_id":"openstack%2Fironic~master~I8a4dfa27d0d90ac7287d90aa0174c5fd91870708","project":"openstack/ironic","branch":"master","topic":"test","hashtags":[],"change_id":"I8a4dfa27d0d90ac7287d90aa0174c5fd91870708","subject":"DNM Test nova fix","status":"ABANDONED","created":"2019-09-06 09:48:05.000000000","updated":"2019-09-09 13:09:00.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a4755f12c10d8afa7cc7fddf734dbb397130d592","_number":680643,"virtual_id_number":680643,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-09-06 17:05:59.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"date":"2019-09-06 14:40:13.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"date":"2019-09-07 11:57:18.000000000","_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":-1,"date":"2019-09-06 13:31:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-09-06 14:23:24.000000000","_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-06 13:31:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-06 14:40:13.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-09-06 17:05:59.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-07 11:57:18.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"REVIEWER"}],"messages":[{"id":"fc5e48a32e66745592cca3d65f0eed7cdd221d38","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-09-06 09:48:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c8476a300bc1ea9bed4e5953977404593857b440","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-09-06 09:48:59.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-ironic-agent_ipmitool https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/ironic/43/680643/1/check-ironic/tempest-dsvm-ironic-agent_ipmitool/233f6d3/ : FAILURE in 35s","accounts_in_message":[],"_revision_number":1},{"id":"c3c29868b26ed0d6b8721525f48b2938170de738","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-09-06 13:19:49.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"3c0e9db1ac282b8d9184fb3a91726c808b1b2026","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 13:31:48.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"29ff46046362c1b049d1f9f311431ccf7b73f2a3","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-09-06 13:35:00.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"67c7fb30d874842e6a36ccc42e394b92fc53e78e","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-06 13:43:34.000000000","message":"Patch Set 2:\n\n(8:41:48 AM) mriedem: dtantsur: is this the job we care about? ironic-grenade-dsvm-multinode-multitenant\n(8:41:58 AM) dtantsur: lemme double-check\n(8:42:27 AM) mriedem: that\u0027s the only multinode grenade job i see in zuul on that change\n(8:42:32 AM) dtantsur: mriedem: oh, it seems to be a normal (non-grenade) multinode job: ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode\n(8:43:04 AM) mriedem: ack\n(8:43:05 AM) dtantsur: although I did see something similar on one or two grenade runs, but they don\u0027t seem consistently broken\n(8:43:17 AM) mriedem: maybe randomly picking the correct node\n(8:43:21 AM) dtantsur: yep","accounts_in_message":[],"_revision_number":2},{"id":"0b976b80bb129781720736205801c30d88c61a01","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 14:23:00.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"985ed800c66bece3f2f3d289e8a67f572c68c6b0","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-09-06 14:23:24.000000000","message":"Patch Set 2:\n\nrecheck\n\ndeps updated","accounts_in_message":[],"_revision_number":2},{"id":"d0420b0d4ab0d691cc44f20701478a4a60c7ffa5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 14:39:57.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"34e0e0bf1a3bad63b6f68be4167539880c47b455","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-06 14:40:13.000000000","message":"Patch Set 2:\n\nrecheck dep updated","accounts_in_message":[],"_revision_number":2},{"id":"da54f4f2794b2212ed3b7976e1ed4f80797ef5b9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-09-06 17:05:59.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-ironic-agent_ipmitool https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/ironic/43/680643/2/check-ironic/tempest-dsvm-ironic-agent_ipmitool/7370cd7/ : FAILURE in 24m 58s","accounts_in_message":[],"_revision_number":2},{"id":"ef53c32198d540e1a7f35001cfd5991fb34d4412","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 00:23:53.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9df6eaaccd0445c68c3e02d76e5ab5d4 : FAILURE in 7m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5d5dcf8f3d94f878a4c6246138655b4 : FAILURE in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2959e3c6736d44e9880c461a8a662999 : SUCCESS in 6m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/984ce1b1391c4fe0b1387ae99ffe1582 : FAILURE in 8m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c41b196bd534dbbbea64af046d4b10e : FAILURE in 8m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/416f5b2ed8994793a4c20559f4cd1005 : FAILURE in 9m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27f0011880a240c4af009b4845f169c5 : SUCCESS in 9m 46s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/6c72f8945925463a9214175683b19422 : FAILURE in 8m 56s\n- ironic-tox-unit-with-driver-libs-python3 https://zuul.opendev.org/t/openstack/build/6e7a570c2be04207b74fee76e6fdb100 : FAILURE in 9m 16s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/a73ec7bbfcab48f7979022c3b53813c6 : RETRY_LIMIT in 3h 01m 09s\n- ironic-tempest-functional-python2 https://zuul.opendev.org/t/openstack/build/a5c3954e47e74153aa14ec17785571eb : SUCCESS in 24m 56s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/65b78478b3484d4b81aa6caf9024190c : SUCCESS in 38m 54s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/e6077c56b98f46b5bf2951b8374ee5b4 : SUCCESS in 1h 39m 48s\n- ironic-grenade-dsvm-multinode-multitenant https://zuul.opendev.org/t/openstack/build/b6207bdd22e3485d9414c543808cf94a : SUCCESS in 1h 38m 10s\n- ironic-tempest-ipa-partition-pxe_ipmitool-tinyipa-python3 https://zuul.opendev.org/t/openstack/build/bef12a7c031a485ca5dc1d21b5c672a4 : SUCCESS in 53m 32s\n- ironic-tempest-ipa-partition-redfish-tinyipa https://zuul.opendev.org/t/openstack/build/6e41a140b4e541be8fbf8f6a46a299b8 : SUCCESS in 51m 01s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/459c3c28f6f5403b8c4b8a80992c0300 : SUCCESS in 50m 11s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/ddfc609319074da4826d7183fb058722 : RETRY_LIMIT in 3h 03m 29s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/63120ff6cfe8414d855121a4680b07f3 : SUCCESS in 1h 03m 00s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/d925887abafc40b4a5e31aba5a1500f5 : SUCCESS in 51m 20s\n- ironic-tempest-ipa-partition-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/65390ec07f6642ea917aac9525920caf : SUCCESS in 41m 42s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/ce60ff43b1da403884e23fec48d2783e : SUCCESS in 33m 06s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/4634d6a72fd948db8e942b80bf690e26 : SUCCESS in 51m 30s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/c6a387bde2b6495cb56934c1b88e2238 : SUCCESS in 5m 18s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-pxe_snmp-tinyipa https://zuul.opendev.org/t/openstack/build/b03c3c79eaf94bcea88ccedff7aa4f4a : SUCCESS in 46m 06s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/686193f3722f409bbf87e352e71fd5c6 : SUCCESS in 45m 34s (non-voting)\n- ironic-inspector-tempest-discovery-fast-track https://zuul.opendev.org/t/openstack/build/599f9b312ed242fd8a153b641c5c8060 : FAILURE in 38m 52s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/6ae0c14032fd413791009f513ce26af7 : SUCCESS in 22m 30s (non-voting)\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/c95535f3734541efbc65ac0dca40b18f : SUCCESS in 1h 23m 05s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/cd57cc0f8acf4246b4bb9d25dc0dac5b : SUCCESS in 47m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"09722ae9d59b39bb064839fbc41deb8ee05b2cfa","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2019-09-07 11:57:18.000000000","message":"Patch Set 2:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-R640-UEFI-py2-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-py2-tempest-dsvm-ironic-idrac/2019-09-06:08:19/680643/2/ : FAILURE in 38m 10s\n- dellemc-DSS9000-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-DSS9000-tempest-dsvm-ironic-idrac/2019-09-06:08:19/680643/2/ : SUCCESS in 44m 32s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac/2019-09-06:08:19/680643/2/ : FAILURE in 51m 39s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac/2019-09-06:08:19/680643/2/ : SUCCESS in 45m 24s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac/2019-09-06:08:19/680643/2/ : SUCCESS in 46m 32s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish/2019-09-06:08:19/680643/2/ : SUCCESS in 55m 48s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish/2019-09-06:08:19/680643/2/ : FAILURE in 50m 07s\n- dellemc-DSS9000-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-DSS9000-tempest-dsvm-ironic-redfish/2019-09-06:08:19/680643/2/ : SUCCESS in 42m 31s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi/2019-09-06:08:19/680643/2/ : SUCCESS in 47m 07s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi/2019-09-06:08:19/680643/2/ : SUCCESS in 43m 32s","accounts_in_message":[],"_revision_number":2},{"id":"237748ad9c570256ba0ec1fe194c79eb462175a2","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-09-09 13:09:00.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bb1d572629e7a9aff6f53e5ea1d5e8489356d48c","revisions":{"53b40a049c9ca527007938522ad92c7e44b7826b":{"kind":"REWORK","_number":1,"created":"2019-09-06 09:48:05.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/43/680643/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/43/680643/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/43/680643/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/43/680643/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/43/680643/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/43/680643/1"}}},"commit":{"parents":[{"commit":"1f68fb9073d79f2471d2aa83a902b3e01cec6a90","subject":"Merge \"add table of available cleaning steps to documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1f68fb9073d79f2471d2aa83a902b3e01cec6a90"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-09-06 09:47:33.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-09-06 09:47:33.000000000","tz":120},"subject":"DNM Test nova fix","message":"DNM Test nova fix\n\nChange-Id: I8a4dfa27d0d90ac7287d90aa0174c5fd91870708\nDepends-On: https://review.opendev.org/#/c/680542/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/53b40a049c9ca527007938522ad92c7e44b7826b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/53b40a049c9ca527007938522ad92c7e44b7826b"}]},"branch":"refs/heads/master"},"bb1d572629e7a9aff6f53e5ea1d5e8489356d48c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-09-06 13:19:49.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/43/680643/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/43/680643/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/43/680643/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/43/680643/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/43/680643/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/43/680643/2"}}},"commit":{"parents":[{"commit":"1f68fb9073d79f2471d2aa83a902b3e01cec6a90","subject":"Merge \"add table of available cleaning steps to documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1f68fb9073d79f2471d2aa83a902b3e01cec6a90"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-09-06 09:47:33.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-09-06 13:19:23.000000000","tz":120},"subject":"DNM Test nova fix","message":"DNM Test nova fix\n\nChange-Id: I8a4dfa27d0d90ac7287d90aa0174c5fd91870708\nDepends-On: https://review.opendev.org/#/c/680542/\nDepends-On: https://review.opendev.org/#/c/680684/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/bb1d572629e7a9aff6f53e5ea1d5e8489356d48c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/bb1d572629e7a9aff6f53e5ea1d5e8489356d48c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
