)]}'
{"id":"openstack%2Fironic~680660","triplet_id":"openstack%2Fironic~master~I18aa79774c0b562bf1e8b973db3d7860a01367e7","project":"openstack/ironic","branch":"master","topic":"sort_controllers","hashtags":[],"change_id":"I18aa79774c0b562bf1e8b973db3d7860a01367e7","subject":"DRAC: Fix a bug for delete_config with multiple controllers","status":"MERGED","created":"2019-09-06 11:07:15.000000000","updated":"2019-10-14 12:19:02.000000000","submitted":"2019-10-11 20:21:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":62,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680660-1570825275834-d762e54f","meta_rev_id":"d13564633f026690c89e521cb706503b2a45e2b3","_number":680660,"virtual_id_number":680660,"owner":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8580,"name":"Digambar","email":"digambarpatil15@yahoo.co.in","username":"digambar"},{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"date":"2019-10-03 18:18:06.000000000","_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"value":0,"date":"2019-10-05 11:06:52.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},{"value":0,"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":0,"date":"2019-10-04 16:57:48.000000000","_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-11 20:21:15.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"date":"2019-10-04 05:04:47.000000000","_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"recommended":{"_account_id":8580,"name":"Digambar","email":"digambarpatil15@yahoo.co.in","username":"digambar"},"all":[{"value":1,"date":"2019-10-04 10:20:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8580,"name":"Digambar","email":"digambarpatil15@yahoo.co.in","username":"digambar"},{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},{"value":2,"date":"2019-10-11 14:12:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":1,"date":"2019-10-03 14:30:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},{"value":1,"date":"2019-10-07 17:14:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":2,"date":"2019-10-08 09:23:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":0,"_account_id":8580,"name":"Digambar","email":"digambarpatil15@yahoo.co.in","username":"digambar"},{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},{"value":1,"date":"2019-10-11 14:55:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},{"value":0,"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":8580,"name":"Digambar","email":"digambarpatil15@yahoo.co.in","username":"digambar"},{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},{"value":0,"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"_account_id":8580,"name":"Digambar","email":"digambarpatil15@yahoo.co.in","username":"digambar"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-06 11:36:09.000000000","updated_by":{"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},"reviewer":{"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},"state":"REVIEWER"},{"updated":"2019-09-10 05:39:43.000000000","updated_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"reviewer":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"state":"REVIEWER"},{"updated":"2019-09-17 21:37:02.000000000","updated_by":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"reviewer":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"state":"REVIEWER"},{"updated":"2019-09-30 10:52:22.000000000","updated_by":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"reviewer":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"state":"REVIEWER"},{"updated":"2019-09-30 10:52:41.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2019-10-03 14:30:59.000000000","updated_by":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"reviewer":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"state":"REVIEWER"},{"updated":"2019-10-03 18:18:06.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"REVIEWER"},{"updated":"2019-10-04 10:20:22.000000000","updated_by":{"_account_id":8580,"name":"Digambar","email":"digambarpatil15@yahoo.co.in","username":"digambar"},"reviewer":{"_account_id":8580,"name":"Digambar","email":"digambarpatil15@yahoo.co.in","username":"digambar"},"state":"REVIEWER"},{"updated":"2019-10-04 16:57:48.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"REVIEWER"},{"updated":"2019-10-05 11:06:52.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-07 17:14:12.000000000","updated_by":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"reviewer":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"state":"REVIEWER"},{"updated":"2019-10-08 09:23:08.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2019-10-11 14:55:13.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2019-10-11 20:21:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"941fd9a86051e2b476e7790f945020edcddf0c97","author":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"date":"2019-09-06 11:07:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea8f83824abf01840e31c632c80fc4df254a4307","author":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"date":"2019-09-06 11:18:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5f1d915ad1c17625b198e9cae52c1437e261c18a","author":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"date":"2019-09-06 14:17:24.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"e60259637bd75db6124b4196a4d1e19b5a9ea7e9","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2019-09-06 14:23:59.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"23667283185f5c1a10966454733cc36f56b10040","author":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"date":"2019-09-06 14:47:22.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"fa92b53c5ddd83368b594746594bb2edb2dd1ed2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-09-06 17:32:27.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-ironic-agent_ipmitool https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/ironic/60/680660/4/check-ironic/tempest-dsvm-ironic-agent_ipmitool/4ff964e/ : FAILURE in 24m 09s","accounts_in_message":[],"_revision_number":4},{"id":"d8aea160558c49f1d17314baeb0df820f4ec80da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 00:24:50.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64b27949d52d46039444432c45b30266 : SUCCESS in 7m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/338d35dd8a574889a4f3b0c35e3aaec2 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e857f10ee6b340d9a4ceaf18a97ce2ef : SUCCESS in 5m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3b06aad4828742988b6163581c94be0d : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/926605927d014e7199fb7a5c26211945 : SUCCESS in 6m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/29690289881d4beabf613f3388ce7483 : SUCCESS in 9m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/441a6b147b1e40768ae12198ef7a508c : SUCCESS in 9m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/592ea2f81499434aa5259df5d2f14fae : SUCCESS in 9m 27s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/63f25995691d48c18938097c985e4d8d : SUCCESS in 7m 16s\n- ironic-tox-unit-with-driver-libs-python3 https://zuul.opendev.org/t/openstack/build/20587bcb98c6415e815de22a4d695e7e : SUCCESS in 9m 46s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/3f55060972724791b2998a65bd761486 : RETRY_LIMIT in 3h 01m 17s\n- ironic-tempest-functional-python2 https://zuul.opendev.org/t/openstack/build/79a6e8f8b6854843ba531b95be9eed88 : SUCCESS in 29m 55s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/d6c95d6400e8490b901be589567033b3 : SUCCESS in 27m 18s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/137a675a9c3f495c89312ddd05a8e993 : FAILURE in 1h 55m 17s\n- ironic-grenade-dsvm-multinode-multitenant https://zuul.opendev.org/t/openstack/build/7a350383a8ac498ba360410960eb09bb : SUCCESS in 1h 41m 39s\n- ironic-tempest-ipa-partition-pxe_ipmitool-tinyipa-python3 https://zuul.opendev.org/t/openstack/build/1351a3a17f184c8d80af77d4381530a1 : SUCCESS in 1h 02m 56s\n- ironic-tempest-ipa-partition-redfish-tinyipa https://zuul.opendev.org/t/openstack/build/6c9e137edf02416dbff6fe21617e91a7 : SUCCESS in 57m 36s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/789f6a35c74846cbbe7dd9baaba10adb : SUCCESS in 46m 30s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/a673b4af2fe64fd8a0f1cff3da5b0a13 : FAILURE in 2h 21m 04s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7fe2f33c74ff459e995d3537e458fad9 : SUCCESS in 48m 49s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/996919cc4fba4e44bb37467240999fd1 : SUCCESS in 51m 30s\n- ironic-tempest-ipa-partition-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/c59136d564b44274b2f0adb70c6d1606 : SUCCESS in 49m 12s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/0d450652c801491c8e8b279a7188a169 : FAILURE in 46m 06s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/976d498d7d854235a114279fffe7173b : SUCCESS in 40m 17s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/259ded2d72ca4324915e6b839f8ccd3a : SUCCESS in 4m 13s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-pxe_snmp-tinyipa https://zuul.opendev.org/t/openstack/build/96e1b914d6a242e59ee1406aeffea7b0 : SUCCESS in 48m 39s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/35b3231d14ff4752ac7f40339dba3450 : SUCCESS in 49m 41s (non-voting)\n- ironic-inspector-tempest-discovery-fast-track https://zuul.opendev.org/t/openstack/build/94e9d158f52249e2ad83430c79ec9637 : FAILURE in 43m 55s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/0f651a5d1a094d90bc953617b6573660 : SUCCESS in 22m 34s (non-voting)\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/10b89207f7fd4e4d94c1f12a341abeda : SUCCESS in 1h 19m 16s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/e9a2a73024754ba0aecc77c28379719e : SUCCESS in 42m 40s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"262c0396ae98968d6e270a52afa94b0dae72be8a","author":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"date":"2019-09-07 09:29:17.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"50ee95e80a3d0f6d9733463d424eae96075989c4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-09-07 09:46:38.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-ironic-agent_ipmitool https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/ironic/60/680660/4/check-ironic/tempest-dsvm-ironic-agent_ipmitool/b7f74fc/ : FAILURE in 17m 02s","accounts_in_message":[],"_revision_number":4},{"id":"e4b8b47418d31f6cf219185d723f3e2c0d590710","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 12:21:59.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5fd5a257ecd346699519b2616fe61074 : SUCCESS in 11m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e12b6408b4584428a5de38c6a03e703b : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eaca1dd9166b4973957fc5c14477955d : SUCCESS in 7m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f109a531d6fd4e968848284580ac390e : SUCCESS in 8m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85c432db26544c1fafe722144073acef : POST_FAILURE in 8m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ea71b6a6d8714fd08343651a8771bc28 : SUCCESS in 9m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dd5349a4601480fb8b081e6703e069e : SUCCESS in 11m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f5c3fd278e4421fac2176e69c6c6b3b : SUCCESS in 10m 23s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/cb532455c8b748d4aea75f9fdb3b7915 : SUCCESS in 8m 28s\n- ironic-tox-unit-with-driver-libs-python3 https://zuul.opendev.org/t/openstack/build/0b787be862da45fb996d19a740e55b7b : SUCCESS in 7m 51s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/765fcf2d7e094f0b8c4ca84d02789425 : SUCCESS in 1h 11m 26s\n- ironic-tempest-functional-python2 https://zuul.opendev.org/t/openstack/build/d62eb77affc642eb921b01290a729edd : SUCCESS in 26m 29s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/05e20c7a5c5349499b13e83b8b9e5c49 : SUCCESS in 35m 06s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/57afd263c93d4958859bee781ddba178 : SUCCESS in 1h 50m 04s\n- ironic-grenade-dsvm-multinode-multitenant https://zuul.opendev.org/t/openstack/build/1d950729f5f844f19f01e98e08a92f6e : SUCCESS in 1h 25m 12s\n- ironic-tempest-ipa-partition-pxe_ipmitool-tinyipa-python3 https://zuul.opendev.org/t/openstack/build/9fd762de91d84e618edd869d6d45b71f : SUCCESS in 46m 16s\n- ironic-tempest-ipa-partition-redfish-tinyipa https://zuul.opendev.org/t/openstack/build/99173d39470e4f39a7121183fe1d15e3 : POST_FAILURE in 45m 13s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/facff1d5cd394883bebe0974e5e46594 : SUCCESS in 49m 01s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/58d922285f7b4c0080436bc73cedf6a4 : SUCCESS in 1h 46m 39s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ade1e7181c664c048287266693b1c278 : POST_FAILURE in 52m 42s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/dbee7eb45da744189f6dd5cd2c5785f4 : POST_FAILURE in 42m 35s\n- ironic-tempest-ipa-partition-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/c64eae6ebdc24ce1b8ebe6db86258065 : SUCCESS in 43m 24s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/7f9a9bedf2cb47788d7b83c4378b6cfc : SUCCESS in 37m 48s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/a05d31a6179847f3a0c8b05967219edc : SUCCESS in 44m 36s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/0edd846128c74871b2f2b62104d437a6 : SUCCESS in 7m 15s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-pxe_snmp-tinyipa https://zuul.opendev.org/t/openstack/build/cc5d5c976501461d9f75034069a08651 : SUCCESS in 42m 05s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/45b4411671eb4d088cdac08fc675a468 : SUCCESS in 51m 53s (non-voting)\n- ironic-inspector-tempest-discovery-fast-track https://zuul.opendev.org/t/openstack/build/dd45d2461bb44ad2b835c6878768c44c : POST_FAILURE in 1h 12m 31s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/8cfa4187de694bb6897bf8a1fa730c09 : SUCCESS in 31m 15s (non-voting)\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/58defc22f0084c819ea019335a101b49 : POST_FAILURE in 1h 15m 31s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/192071ef23f54322b76b4c8a8e6a6e07 : SUCCESS in 44m 31s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4a356c769604586ac927b42d401d5f269cc192fd","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2019-09-07 13:31:19.000000000","message":"Patch Set 4:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-R640-UEFI-py2-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-py2-tempest-dsvm-ironic-idrac/2019-09-06:09:47/680660/4/ : FAILURE in 34m 50s\n- dellemc-DSS9000-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-DSS9000-tempest-dsvm-ironic-idrac/2019-09-06:09:47/680660/4/ : FAILURE in 14m 58s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac/2019-09-06:09:47/680660/4/ : SUCCESS in 52m 43s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac/2019-09-06:09:47/680660/4/ : SUCCESS in 45m 31s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac/2019-09-06:09:47/680660/4/ : SUCCESS in 46m 12s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish/2019-09-06:09:47/680660/4/ : SUCCESS in 1h 18m 46s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish/2019-09-06:09:47/680660/4/ : FAILURE in 46m 47s\n- dellemc-DSS9000-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-DSS9000-tempest-dsvm-ironic-redfish/2019-09-06:09:47/680660/4/ : SUCCESS in 47m 02s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi/2019-09-06:09:47/680660/4/ : SUCCESS in 43m 21s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi/2019-09-06:09:47/680660/4/ : SUCCESS in 43m 31s","accounts_in_message":[],"_revision_number":4},{"id":"d0835904e8ae3abbc118d47f1c12d7ea8df33c5d","author":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"date":"2019-09-09 05:18:50.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"88cc23b292e740a66224fd2058843d5f52192a89","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-09-09 06:14:18.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-ironic-agent_ipmitool https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/ironic/60/680660/4/check-ironic/tempest-dsvm-ironic-agent_ipmitool/f6c10ae/ : FAILURE in 17m 42s","accounts_in_message":[],"_revision_number":4},{"id":"bce2a381e9a3d2102319daa810539ae53a2bfd76","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2019-09-09 13:03:26.000000000","message":"Patch Set 4:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-R640-UEFI-py2-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-py2-tempest-dsvm-ironic-idrac/2019-09-09:00:18/680660/4/ : FAILURE in 35m 19s\n- dellemc-DSS9000-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-DSS9000-tempest-dsvm-ironic-idrac/2019-09-09:00:18/680660/4/ : SUCCESS in 44m 43s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac/2019-09-09:00:18/680660/4/ : SUCCESS in 52m 19s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac/2019-09-09:00:18/680660/4/ : SUCCESS in 49m 28s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac/2019-09-09:00:18/680660/4/ : SUCCESS in 46m 10s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish/2019-09-09:00:18/680660/4/ : SUCCESS in 1h 14m 34s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish/2019-09-09:00:18/680660/4/ : FAILURE in 47m 40s\n- dellemc-DSS9000-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-DSS9000-tempest-dsvm-ironic-redfish/2019-09-09:00:18/680660/4/ : FAILURE in 48m 01s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi/2019-09-09:00:18/680660/4/ : FAILURE in 53m 18s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi/2019-09-09:00:18/680660/4/ : SUCCESS in 52m 29s","accounts_in_message":[],"_revision_number":4},{"id":"0283470e31defb0f365dc01958109e1a48dee854","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 14:36:35.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f24a6d0d282430c849389099bba4a20 : FAILURE in 6m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d444b9f164a486a901d9856fef0c8a8 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d5d29e18e464f78a21612e371f53acf : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9010cd6851874485b162feb681940531 : FAILURE in 7m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7562ea7dfdc7413a8665164698f19dad : FAILURE in 6m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c0eb13a86ca848dc8508a43819e7d9d7 : FAILURE in 9m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/522586ae133d4b5290e491c517241e51 : SUCCESS in 9m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d15c193f889491b8ee18025285e4aef : SUCCESS in 11m 09s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/5d55c6d58adc46cca3ed7284664e3d6f : FAILURE in 7m 07s\n- ironic-tox-unit-with-driver-libs-python3 https://zuul.opendev.org/t/openstack/build/1a027821d61d481182334e32c45bf7e8 : FAILURE in 7m 08s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/f4e3c9320b8b4b0e8467941b57b4e60a : RETRY_LIMIT in 3h 01m 20s\n- ironic-tempest-functional-python2 https://zuul.opendev.org/t/openstack/build/52b7cbf95c2a41ca910aa44bf0a6c4cb : SUCCESS in 25m 36s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/a9bf8de079f94410a5e77932cadaedd6 : SUCCESS in 26m 36s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/1da07cf4c22f4215b726a3c16d5a3069 : SUCCESS in 1h 43m 46s\n- ironic-grenade-dsvm-multinode-multitenant https://zuul.opendev.org/t/openstack/build/171703ecb722484e931e6bc2a0b4c1cd : SUCCESS in 1h 54m 01s\n- ironic-tempest-ipa-partition-pxe_ipmitool-tinyipa-python3 https://zuul.opendev.org/t/openstack/build/5d5a92aba47b499497fd208c6eff94ff : SUCCESS in 54m 34s\n- ironic-tempest-ipa-partition-redfish-tinyipa https://zuul.opendev.org/t/openstack/build/adf05e119e844cf59d5dbea4502d0f80 : SUCCESS in 47m 39s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7402ceb2d7cd487385ee2a877317e62a : SUCCESS in 44m 18s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/ed6bc6ffdfa4426db8f2b763ab9ca848 : SUCCESS in 2h 01m 47s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e4595d42ba164f34bb969f04015ceca2 : SUCCESS in 52m 17s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/1b47cdef18954f6684aa9fc56fc8bcb3 : SUCCESS in 45m 26s\n- ironic-tempest-ipa-partition-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/bc8596a6c12444d98ae8ae941caa65b5 : SUCCESS in 45m 11s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/c34d99a930394fcdad0cb0a80e82c583 : SUCCESS in 37m 02s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/57ce46b62cfb42c58a6f6df5e5fdea37 : SUCCESS in 44m 01s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/189ec2ea47a24b12adc37290578413cc : SUCCESS in 4m 01s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-pxe_snmp-tinyipa https://zuul.opendev.org/t/openstack/build/3ef46e9293384791a9752ee18edffb0d : SUCCESS in 44m 48s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/e0acfadb66e14cb18c233291d602dc94 : SUCCESS in 52m 45s (non-voting)\n- ironic-inspector-tempest-discovery-fast-track https://zuul.opendev.org/t/openstack/build/ecacc29f8f5345819ae0cbd70ab5f757 : FAILURE in 36m 09s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/1bf6264e5b414b93abd8f6a16df44983 : SUCCESS in 19m 23s (non-voting)\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/4ff1aab35cad49c58ecf2620c502177a : SUCCESS in 1h 17m 10s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/a7411e2b8096414fa6d63b5d0fd142ff : SUCCESS in 49m 34s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"75e9f535947a754b1cef6920f5fd245209455f13","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2019-09-10 02:28:28.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"b45b6f24d1779d323e5ec9bd7c95acb521197074","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 04:28:45.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9c6aff046724ab1b5aac563b85c77db : SUCCESS in 10m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/232cad14f4d240cf868745e4a234bfb2 : SUCCESS in 7m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b483cbbc667f4aa4ae4dbc8f964e6102 : SUCCESS in 5m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/68bc0c48d4d646f6b2d9f5c8da284234 : SUCCESS in 8m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1b645ec63194749a99176102f8b18b1 : SUCCESS in 9m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8386ca2c484c4815b81bbcc4c7365bbe : SUCCESS in 8m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91e720ed466443e9a392bedebca21285 : SUCCESS in 10m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b34ee3333b5c4607a10b5caf75e83ace : SUCCESS in 12m 25s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/d3e3c86331b540d4b4cb25cc4419e1fe : SUCCESS in 9m 36s\n- ironic-tox-unit-with-driver-libs-python3 https://zuul.opendev.org/t/openstack/build/0075f18a387c4d5f9cc755cc945c1766 : SUCCESS in 8m 39s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/9c712697ddad4c4592042ff3e38d9a64 : SUCCESS in 1h 14m 06s\n- ironic-tempest-functional-python2 https://zuul.opendev.org/t/openstack/build/c1985e0d126b4b18966ce7a34268a7f7 : SUCCESS in 29m 07s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/bae911b432524eae8955519db461b3c5 : FAILURE in 37m 00s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/31e15eced2be4bffad58c72f1ece381f : SUCCESS in 1h 30m 28s\n- ironic-grenade-dsvm-multinode-multitenant https://zuul.opendev.org/t/openstack/build/36dc7b53a26a4b6990c9b0b6db044c29 : SUCCESS in 1h 44m 55s\n- ironic-tempest-ipa-partition-pxe_ipmitool-tinyipa-python3 https://zuul.opendev.org/t/openstack/build/be7e197d9c524edc9d63b46a61be5656 : SUCCESS in 49m 02s\n- ironic-tempest-ipa-partition-redfish-tinyipa https://zuul.opendev.org/t/openstack/build/a0fd861aa9c9428ab4f372892f01099c : SUCCESS in 49m 46s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/51a973a5674f41b3b34f03c9f64c8fe0 : FAILURE in 24m 50s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/7a5e83af05804535ad105572a59e4677 : SUCCESS in 1h 32m 48s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/812e4c778c3e44a09a092241dddb795c : SUCCESS in 57m 35s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/49f5ac0e46a14871af64079f9e0744be : SUCCESS in 1h 00m 15s\n- ironic-tempest-ipa-partition-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/ffde162fe69a460292e91f8da917ad3d : SUCCESS in 43m 22s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/83e6236057e94b2596a0939fc7a1db23 : SUCCESS in 39m 09s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/575ac33eecc7489494acc2df8641caa5 : SUCCESS in 57m 37s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/82327bd2db41474a939a7f7cd572f6d4 : SUCCESS in 5m 00s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-pxe_snmp-tinyipa https://zuul.opendev.org/t/openstack/build/b184f1110e1540de84be1dcd7dd47f0c : SUCCESS in 56m 35s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/3181e40f5a3540e7b3951d1e4f50857c : SUCCESS in 51m 35s (non-voting)\n- ironic-inspector-tempest-discovery-fast-track https://zuul.opendev.org/t/openstack/build/2c24ff920e9b405f9491475b75f95584 : FAILURE in 39m 32s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/7ef2ceabdaf44d30a5208f705c3db506 : SUCCESS in 20m 55s (non-voting)\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/a70238319a3f465eb6e9ed40d1d1be5b : SUCCESS in 1h 16m 59s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/38d262e53069430d995dfe2772ce50be : SUCCESS in 52m 48s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6ec9c0a7d20da8306a97f94f5421a4b4d2b0888e","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2019-09-10 05:39:43.000000000","message":"Patch Set 4: Code-Review-1\n\n(6 comments)\n\nThank you Rachit. Few comments inline.","accounts_in_message":[],"_revision_number":4},{"id":"85d58907faf5464b10f709f4acbc00fea723b9c7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-09-10 06:03:33.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-ironic-agent_ipmitool https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/ironic/60/680660/4/check-ironic/tempest-dsvm-ironic-agent_ipmitool/f3ca610/ : FAILURE in 21m 18s","accounts_in_message":[],"_revision_number":4},{"id":"fd5529be0c01bc52dc251fb032033356de543ce4","author":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"date":"2019-09-10 07:04:20.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"9f5f57bcb11f4bd5da3f78722c3a2771f6f735f0","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2019-09-10 07:04:55.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"1a87f28729d3293b77b7ed0b7f98a78c51473236","author":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"date":"2019-09-10 07:19:23.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"7537303cc7728bfdf4d0e4ce40bc70179b5923f8","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-09-10 08:18:07.000000000","message":"Patch Set 6:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-ironic-agent_ipmitool https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/ironic/60/680660/6/check-ironic/tempest-dsvm-ironic-agent_ipmitool/cb2b8fb/ : FAILURE in 19m 36s","accounts_in_message":[],"_revision_number":6},{"id":"800970c7640194db755fa3b642a3a6bf2a448210","author":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"date":"2019-09-10 08:22:36.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"815757a65f7b5b87910d809284c6a998f0612344","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2019-09-10 08:46:40.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 http://15.184.0.52/60/680660/7/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/7c48485/ : POST_FAILURE in 22m 55s","accounts_in_message":[],"_revision_number":7},{"id":"5f3953dfba57e9e0e5d98025e9b57f0967d520ab","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-09-10 09:40:20.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-ironic-agent_ipmitool https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/ironic/60/680660/7/check-ironic/tempest-dsvm-ironic-agent_ipmitool/7e515a9/ : FAILURE in 20m 41s","accounts_in_message":[],"_revision_number":7},{"id":"4597b01e9c76bf3b0eb676f9093bb16242989c4b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 11:18:15.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29c6c6c1eb254e32856cb3a1f9aee91a : SUCCESS in 9m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/79632ccc39f943f8860abf62bfbcf81a : FAILURE in 16m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26303a4db1ee4318a05c5aa9d1e489a4 : SUCCESS in 5m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/821513ff197a46b09222b17657035cb7 : SUCCESS in 7m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0017884669284b17bdbb9302964a46ea : SUCCESS in 8m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c5ed49879a334d7bb10509388f30708f : SUCCESS in 7m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0d30869d1ee474db66f108d36255ede : SUCCESS in 10m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af90f4c9526b4b7aa7c5419c96e8fe5a : SUCCESS in 13m 15s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/e49db39efd3241c1b98c9f694f8a5c1e : SUCCESS in 8m 33s\n- ironic-tox-unit-with-driver-libs-python3 https://zuul.opendev.org/t/openstack/build/245e834cdf4c4802a3ee3594be5519b0 : SUCCESS in 10m 43s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/03f7de7bda2040e4a953841ddff8275c : SUCCESS in 1h 06m 03s\n- ironic-tempest-functional-python2 https://zuul.opendev.org/t/openstack/build/cb317220133a41b4b040743b9801c6df : SUCCESS in 24m 17s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/8dbf24f2611748c98bb1add992584959 : SUCCESS in 29m 46s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/0fb63cdf06f44b6681a5270ccd9ef81e : SUCCESS in 1h 30m 53s\n- ironic-grenade-dsvm-multinode-multitenant https://zuul.opendev.org/t/openstack/build/6dc207f6a15643b287b9f2f7ea507a78 : SUCCESS in 1h 44m 45s\n- ironic-tempest-ipa-partition-pxe_ipmitool-tinyipa-python3 https://zuul.opendev.org/t/openstack/build/eb3b0a66e15d47a6bf489b77152de138 : SUCCESS in 47m 36s\n- ironic-tempest-ipa-partition-redfish-tinyipa https://zuul.opendev.org/t/openstack/build/bb1ec1369a4e4d668ca01fd784fc59c5 : SUCCESS in 55m 03s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/baa35843206c4845a4880bc4528e1d3c : SUCCESS in 40m 17s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/0d81596767884d758ec1962d74e11ae1 : FAILURE in 1h 51m 10s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6d3a37a5c7214f21a43a6f9315b7e78d : SUCCESS in 54m 23s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/f1acd7baa7b54672a5d32f64f060d9de : SUCCESS in 48m 24s\n- ironic-tempest-ipa-partition-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/4722cf2aed6747a994517e5a899eb2e7 : SUCCESS in 42m 55s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/2c3142bfc0f64d53b15123b48efbddf7 : SUCCESS in 40m 34s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/b68813e485944fac9d91132ab93e0532 : SUCCESS in 46m 46s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/2db2ddd693384bf1afa579417b0d5977 : SUCCESS in 4m 16s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-pxe_snmp-tinyipa https://zuul.opendev.org/t/openstack/build/9fe82d9e554f466d942f816b41922bbd : SUCCESS in 56m 04s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/1bafb811813e4c1a9541e3f758a48a37 : SUCCESS in 50m 47s (non-voting)\n- ironic-inspector-tempest-discovery-fast-track https://zuul.opendev.org/t/openstack/build/ef1a7eeb62804634be78e710c255db91 : FAILURE in 40m 12s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/397034f38db04e33929827999700f3c5 : SUCCESS in 20m 29s (non-voting)\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/d5b9772b86ef427384a77b6a3174ce2e : SUCCESS in 1h 13m 34s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/27140808259643fc9e92d54266d23ef0 : SUCCESS in 49m 08s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"bf179459aa2dc8f758c8c4ca6f7cf2cc0c2cdd8b","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2019-09-11 08:59:05.000000000","message":"Patch Set 7:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-R640-UEFI-py2-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-py2-tempest-dsvm-ironic-idrac/2019-09-10:03:22/680660/7/ : FAILURE in 36m 53s\n- dellemc-DSS9000-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-DSS9000-tempest-dsvm-ironic-idrac/2019-09-10:03:22/680660/7/ : SUCCESS in 1h 08m 00s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac/2019-09-10:03:22/680660/7/ : FAILURE in 54m 59s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac/2019-09-10:03:22/680660/7/ : SUCCESS in 44m 58s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac/2019-09-10:03:22/680660/7/ : SUCCESS in 46m 15s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish/2019-09-10:03:22/680660/7/ : SUCCESS in 47m 56s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish/2019-09-10:03:22/680660/7/ : FAILURE in 46m 44s\n- dellemc-DSS9000-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-DSS9000-tempest-dsvm-ironic-redfish/2019-09-10:03:22/680660/7/ : SUCCESS in 40m 42s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi/2019-09-10:03:22/680660/7/ : SUCCESS in 41m 33s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi/2019-09-10:03:22/680660/7/ : SUCCESS in 41m 48s","accounts_in_message":[],"_revision_number":7},{"id":"f1fda530d7f3ab25d05a3f2ffcf55d0478ad8b17","author":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"date":"2019-09-11 20:03:03.000000000","message":"Patch Set 7: Code-Review-1\n\n(3 comments)\n\nIs it really necessary to change the current logic that controls when a reboot is requested?  I\u0027m not sure that the new logic will work if there are multiple controllers that don\u0027t support realtime mode.","accounts_in_message":[],"_revision_number":7},{"id":"a13a1ad306930bf34367c35570773106de29f756","author":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"date":"2019-09-12 08:57:09.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"1517c8163e8cd835122eb98cb05d95742433cf2a","author":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"date":"2019-09-12 18:26:24.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"74bf92febb9675b6c2820d6d45f6b35dd7d0ff84","author":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"date":"2019-09-12 18:32:01.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"9c2f04c4c1c2fa15a23856fb6099abd7dad88003","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2019-09-12 22:32:26.000000000","message":"Patch Set 9:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-R640-UEFI-py2-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-py2-tempest-dsvm-ironic-idrac/2019-09-12:13:32/680660/9/ : SUCCESS in 44m 46s\n- dellemc-DSS9000-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-DSS9000-tempest-dsvm-ironic-idrac/2019-09-12:13:32/680660/9/ : FAILURE in 15m 08s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac/2019-09-12:13:32/680660/9/ : SUCCESS in 49m 47s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac/2019-09-12:13:32/680660/9/ : SUCCESS in 48m 57s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac/2019-09-12:13:32/680660/9/ : FAILURE in 36m 12s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish/2019-09-12:13:32/680660/9/ : SUCCESS in 44m 16s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish/2019-09-12:13:32/680660/9/ : SUCCESS in 43m 33s\n- dellemc-DSS9000-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-DSS9000-tempest-dsvm-ironic-redfish/2019-09-12:13:32/680660/9/ : SUCCESS in 42m 51s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi/2019-09-12:13:32/680660/9/ : SUCCESS in 45m 53s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi/2019-09-12:13:32/680660/9/ : FAILURE in 46m 17s","accounts_in_message":[],"_revision_number":9},{"id":"c1a055749472a1c8c04bab4c5de5584ac3aa0910","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 23:25:37.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/665c316a853b419c9b9ac4f99f165c62 : SUCCESS in 8m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f9cd9ce59ffe4bcd938d25578d1bec39 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/260c9ed4f7924bbf8f80b76f687bc035 : FAILURE in 5m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/215457bbef084504a1646361498be57a : SUCCESS in 7m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e063b637f3bd44d0bf0fed105e340a5d : SUCCESS in 6m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f19d3e35bd9d4d3d8ca4c32f19dc5c3a : SUCCESS in 8m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac74fbd5adf64ea5991972bbd241cd51 : SUCCESS in 9m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/68de765e6ef44e039dcc00d2dd384d2b : SUCCESS in 11m 21s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/7faab31b66b348379054aee9ad914e41 : SUCCESS in 6m 17s\n- ironic-tox-unit-with-driver-libs-python3 https://zuul.opendev.org/t/openstack/build/becb46cbeeef4d4fba99254e8f618493 : SUCCESS in 7m 29s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/9547e250549343b995ea57ef243e72b8 : SUCCESS in 1h 10m 11s\n- ironic-tempest-functional-python2 https://zuul.opendev.org/t/openstack/build/d50d61c1b67d40c1b2153c20b7b6ccdf : SUCCESS in 28m 14s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/ff997e09e1f245c4885b0eec814bde6e : SUCCESS in 30m 35s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/54b662e4b07d48b8854b5ed2651a6178 : SUCCESS in 1h 37m 38s\n- ironic-grenade-dsvm-multinode-multitenant https://zuul.opendev.org/t/openstack/build/4d8db96eb74d45ebaf71f51eb571505a : SUCCESS in 1h 36m 00s\n- ironic-tempest-ipa-partition-pxe_ipmitool-tinyipa-python3 https://zuul.opendev.org/t/openstack/build/11dc318b10844874a958389d27281d17 : SUCCESS in 50m 22s\n- ironic-tempest-ipa-partition-redfish-tinyipa https://zuul.opendev.org/t/openstack/build/5957ea1a075541e09cca3190e211db96 : SUCCESS in 46m 57s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a99d1d1d9c674054b344abe231dc80e7 : SUCCESS in 45m 39s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/e977283852724bb3b31683290407d26f : SUCCESS in 1h 36m 00s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/798ff71f86a94f83a680630a9de301ff : SUCCESS in 50m 02s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/4c8f2dc466b34ceda8765ffc0d7d8df7 : SUCCESS in 56m 29s\n- ironic-tempest-ipa-partition-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/9da6a46555b94d128d0b67842f44708f : SUCCESS in 49m 55s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/986b764b56b349a2bedbd91a5dbe3e1b : SUCCESS in 40m 40s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/6ef16cfee3a14f8cb650da707599b864 : SUCCESS in 40m 07s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/38d0ec167d01404aab4fe8212b4d1f6f : SUCCESS in 5m 02s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-pxe_snmp-tinyipa https://zuul.opendev.org/t/openstack/build/6451dce443d84fe2b39c9f70e5cbbfd2 : SUCCESS in 44m 07s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/1476d52821c34f4a8215d50ea4e16251 : SUCCESS in 48m 11s (non-voting)\n- ironic-inspector-tempest-discovery-fast-track https://zuul.opendev.org/t/openstack/build/f0e34a1c928f42de8d84a0cfc271a703 : FAILURE in 40m 32s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/8b3f543e86f248eb801e3f92d31aafe9 : SUCCESS in 21m 55s (non-voting)\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/d722762268574e958ddf62bd03bc43a2 : SUCCESS in 1h 19m 21s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/5bf590a354a24a88948e85a2017d9ae0 : SUCCESS in 53m 21s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"83e698bbd40fa359cf0b61aed29e4ceceeb65334","author":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"date":"2019-09-13 00:44:26.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"b88d8755f3b580cf2499ec55765c2b0d8464e9db","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2019-09-13 02:39:21.000000000","message":"Patch Set 10:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-R640-UEFI-py2-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-py2-tempest-dsvm-ironic-idrac/2019-09-12:19:44/680660/10/ : SUCCESS in 54m 48s\n- dellemc-DSS9000-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-DSS9000-tempest-dsvm-ironic-idrac/2019-09-12:19:44/680660/10/ : SUCCESS in 54m 49s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac/2019-09-12:19:44/680660/10/ : SUCCESS in 54m 32s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac/2019-09-12:19:44/680660/10/ : SUCCESS in 54m 37s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac/2019-09-12:19:44/680660/10/ : FAILURE in 43m 55s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish/2019-09-12:19:44/680660/10/ : SUCCESS in 51m 09s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish/2019-09-12:19:44/680660/10/ : SUCCESS in 51m 09s\n- dellemc-DSS9000-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-DSS9000-tempest-dsvm-ironic-redfish/2019-09-12:19:44/680660/10/ : FAILURE in 5s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi/2019-09-12:19:44/680660/10/ : SUCCESS in 48m 43s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi/2019-09-12:19:44/680660/10/ : FAILURE in 57m 31s","accounts_in_message":[],"_revision_number":10},{"id":"d1c0c200b1093e9fe47d92466cad931d73c31dc9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 02:48:24.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/823ced5148664e18beba3ea407381a95 : SUCCESS in 7m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ab1c79b4c7342dc992085190a88c475 : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69b7862e475c42179ef53b30b801a6e6 : SUCCESS in 6m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5fa2846621864f90b1c4135d7c78eaf0 : SUCCESS in 7m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9fe5726d9b204e21b0b17fe856cf8ea6 : SUCCESS in 12m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/54e19cd2456e41b68be8842aea93a2a4 : SUCCESS in 8m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6381b11cbee84d5caab45492b9c3365a : SUCCESS in 8m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb76f18fdf5642159c3f527135d878c7 : SUCCESS in 11m 30s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/ae4d1931ad714416aaae6c7810e2f7ec : SUCCESS in 7m 18s\n- ironic-tox-unit-with-driver-libs-python3 https://zuul.opendev.org/t/openstack/build/cf2d116c92ea4d6b928fc05b3491859f : SUCCESS in 9m 40s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/7ab6668e536f483ab20de794c706b557 : SUCCESS in 1h 21m 17s\n- ironic-tempest-functional-python2 https://zuul.opendev.org/t/openstack/build/427b7a71d6244705870f3b2c476ae01d : SUCCESS in 32m 28s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/3fef310216f646209e1271f8a2c49be3 : SUCCESS in 27m 54s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/a02465e223214f63b1d15bba3061c108 : SUCCESS in 1h 40m 46s\n- ironic-grenade-dsvm-multinode-multitenant https://zuul.opendev.org/t/openstack/build/df872e7077054499a4ba4c35f6fd66f0 : SUCCESS in 1h 40m 24s\n- ironic-tempest-ipa-partition-pxe_ipmitool-tinyipa-python3 https://zuul.opendev.org/t/openstack/build/197d803c043941e98711f116872b9335 : SUCCESS in 54m 10s\n- ironic-tempest-ipa-partition-redfish-tinyipa https://zuul.opendev.org/t/openstack/build/7eeb916c0dfe4038a92552cc29fb867a : SUCCESS in 42m 11s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/d8e34311dca447dfb9663d725bf6e822 : SUCCESS in 49m 15s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/5ee84398a9ab484f974fa63067e939f7 : SUCCESS in 1h 46m 14s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/df038ac8a9544e509dfbe4a9088b37e1 : SUCCESS in 53m 26s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/aac6a82741f345e2a8086eae1549f3ef : SUCCESS in 50m 16s\n- ironic-tempest-ipa-partition-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/0bf0d5779eaf43a183957715d2ae4abe : SUCCESS in 50m 17s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/ba250c632dda4a28ac5e07ff71544c36 : SUCCESS in 43m 50s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/42cbb2fa17844cf6b8f0e9093c11be33 : SUCCESS in 46m 51s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/d646f12084d847a1a8224ef89b376d97 : SUCCESS in 6m 46s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-pxe_snmp-tinyipa https://zuul.opendev.org/t/openstack/build/991b5872153540da9f26c8a0ad9d100f : SUCCESS in 51m 42s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/859c1475eb0f406db2a721eb63d92d54 : SUCCESS in 1h 19m 52s (non-voting)\n- ironic-inspector-tempest-discovery-fast-track https://zuul.opendev.org/t/openstack/build/3188dc6df23144139969015873487100 : FAILURE in 39m 37s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/321ccf2ef0eb4c8da89324ff533b2c7b : SUCCESS in 20m 59s (non-voting)\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/0a4ccc02a8204f48b603bfac629d8838 : SUCCESS in 1h 14m 39s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/c3ba7587f09d4f79bfe67cdc4d60d0c0 : SUCCESS in 50m 51s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"d6bbeadd097ecb122f993a32633869a60880e5b1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-09-13 09:31:29.000000000","message":"Patch Set 10:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-ironic-agent_ipmitool https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/ironic/60/680660/10/check-ironic/tempest-dsvm-ironic-agent_ipmitool/31bdcc3/ : FAILURE in 17m 44s","accounts_in_message":[],"_revision_number":10},{"id":"9f845147dd893042fea10b484b7c003f581f9295","author":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"date":"2019-09-13 15:23:39.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"01dd98ab30855f20b7df12e0efebeec6c61bb2cb","author":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"date":"2019-09-13 15:24:08.000000000","message":"Patch Set 10: Workflow-1","accounts_in_message":[],"_revision_number":10},{"id":"f1c7c2ffcfbec8923957a17640b861e4b3db95c7","author":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"date":"2019-09-13 15:26:01.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"b23df82f78688624c252c0788eeb9ea8fcec353e","author":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"date":"2019-09-18 10:56:28.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"2c3a18387e3b524fb230c0154dd44fcc1647b334","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-09-18 12:27:40.000000000","message":"Patch Set 11:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-ironic-agent_ipmitool https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/ironic/60/680660/11/check-ironic/tempest-dsvm-ironic-agent_ipmitool/914b920/ : FAILURE in 24m 55s","accounts_in_message":[],"_revision_number":11},{"id":"fcc10524fd12036f5c84cc1c5ad9f81cd2ca1d81","author":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"date":"2019-09-18 12:36:21.000000000","message":"Uploaded patch set 12: Commit message was updated.","accounts_in_message":[],"_revision_number":12},{"id":"f9d1b1747e5c6b5f125c70729bfa6d8ead12efd1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-09-18 13:16:13.000000000","message":"Patch Set 12:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-ironic-agent_ipmitool https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/ironic/60/680660/12/check-ironic/tempest-dsvm-ironic-agent_ipmitool/8dd4ee8/ : FAILURE in 21m 39s","accounts_in_message":[],"_revision_number":12},{"id":"c2c6ffe32b70d565b8b8aadbc41b3ac4acbaafbf","author":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"date":"2019-09-18 15:51:31.000000000","message":"Patch Set 12: Code-Review-1\n\n(7 comments)\n\nHey Rachit, I like the approach.  I believe I found a bug in the code.  Also I think we need to add a test case that tests deleting 2 virtual disks where not all controllers support realtime.  I think the existing test_delete_configuration test case handles the other case where all controllers support realtime.","accounts_in_message":[],"_revision_number":12},{"id":"306c589cfd9fad665a3f19deb0953db1804efd8a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 16:49:00.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de9e8f1d87fc4995ac7f4df716b3a22f : SUCCESS in 8m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c223e294ec2d493f8213d057b438461f : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b21c9e8a63b434d9c1b93fa98821914 : SUCCESS in 5m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b26e4834c9e54e27b3226d24146d81ce : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/deac1c46c5a442be8a8b44473cdc41be : SUCCESS in 7m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c111336ed7cb48fe8c767add04aafc85 : SUCCESS in 8m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e67221d9e476484fa516b49782cd0ec9 : SUCCESS in 11m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3073dd55b934eaba5265ce31dd7f44b : SUCCESS in 9m 36s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/564340e74c66415a8ff7bda8b4bb4e34 : SUCCESS in 6m 13s\n- ironic-tox-unit-with-driver-libs-python3 https://zuul.opendev.org/t/openstack/build/c3036db7cd8b436094d822f718daa5b3 : SUCCESS in 8m 54s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/236c79d7c23e4f6b984dde96a515f094 : SUCCESS in 1h 11m 29s\n- ironic-tempest-functional-python2 https://zuul.opendev.org/t/openstack/build/98d341c186614e19a8d0cfcbcfcbd296 : FAILURE in 28m 08s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/946961e4db16436cb08dc12695234409 : SUCCESS in 31m 13s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/be831f7989934072999eb3c01ce21e34 : SUCCESS in 1h 37m 48s\n- ironic-grenade-dsvm-multinode-multitenant https://zuul.opendev.org/t/openstack/build/532bdd14f5384222858c1a2a206d221a : SUCCESS in 1h 45m 27s\n- ironic-tempest-ipa-partition-pxe_ipmitool-tinyipa-python3 https://zuul.opendev.org/t/openstack/build/4ce3e02e0ae04351a1a66f0d00596f25 : SUCCESS in 49m 48s\n- ironic-tempest-ipa-partition-redfish-tinyipa https://zuul.opendev.org/t/openstack/build/32351178f3a240cb90f824c1070d6918 : SUCCESS in 48m 06s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/1dfbae6436d94aef8602342fc3e3eb51 : SUCCESS in 49m 10s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/c6db05ae72ad40829560e8a5ca81d8b7 : SUCCESS in 1h 46m 16s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/0974aac8376146b287fb1b1486092453 : SUCCESS in 51m 52s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/7e240a5e89aa45c78ef634128fb7ce9f : SUCCESS in 47m 23s\n- ironic-tempest-ipa-partition-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/31098e663e7d44a18d995ff6b73f84d7 : SUCCESS in 56m 19s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/ad02471108694d0eb5a7e0e2236dd166 : SUCCESS in 41m 20s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/c4a266fc061745d9b7b649d2e6afadcb : FAILURE in 44m 34s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/6b40ad2d383448eeb0373ed6f25379e8 : SUCCESS in 5m 17s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-pxe_snmp-tinyipa https://zuul.opendev.org/t/openstack/build/a82037854460484da3fe440bd287c935 : SUCCESS in 59m 30s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/c9bff191e1ac4227affc402481a4dee9 : SUCCESS in 58m 40s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/6decfaa004f148158d068aeda28cd427 : SUCCESS in 30m 53s (non-voting)\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/6296d548570c477abab303529ab7f176 : SUCCESS in 1h 20m 38s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/fc1607e1d0734a3ba8883cc3535c68e9 : SUCCESS in 49m 27s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"f4abcfd35dfd38d055011d7877bafa58a14c1df3","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2019-09-19 01:44:40.000000000","message":"Patch Set 12:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-R640-UEFI-py2-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-py2-tempest-dsvm-ironic-idrac/2019-09-18:07:36/680660/12/ : FAILURE in 15m 21s\n- dellemc-DSS9000-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-DSS9000-tempest-dsvm-ironic-idrac/2019-09-18:07:36/680660/12/ : FAILURE in 15m 54s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac/2019-09-18:07:36/680660/12/ : FAILURE in 17m 19s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac/2019-09-18:07:36/680660/12/ : FAILURE in 15m 53s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac/2019-09-18:07:36/680660/12/ : FAILURE in 17m 20s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish/2019-09-18:07:36/680660/12/ : FAILURE in 16m 01s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish/2019-09-18:07:36/680660/12/ : FAILURE in 16m 06s\n- dellemc-DSS9000-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-DSS9000-tempest-dsvm-ironic-redfish/2019-09-18:07:36/680660/12/ : FAILURE in 15m 54s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi/2019-09-18:07:36/680660/12/ : FAILURE in 17m 20s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi/2019-09-18:07:36/680660/12/ : FAILURE in 16m 09s","accounts_in_message":[],"_revision_number":12},{"id":"1e57f8447ec1335d64cd12fcbe2129ad5dcb8a0e","author":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"date":"2019-09-19 19:05:25.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"5cba3b4a4753b51a42d188d96b0f20a300dfa689","author":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"date":"2019-09-19 19:14:15.000000000","message":"Patch Set 13:\n\n(7 comments)\n\nHi Chris,\nThanks for the review, i have added new testcase for delete_configuration with multiple controllers that supports realtime as well as non-realtime mode.","accounts_in_message":[],"_revision_number":13},{"id":"5c1e3c7377aa0abe3ddf00e6199bb07c15f7896f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-09-19 19:29:45.000000000","message":"Patch Set 13:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-ironic-agent_ipmitool https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/ironic/60/680660/13/check-ironic/tempest-dsvm-ironic-agent_ipmitool/494d71a/ : FAILURE in 24m 00s","accounts_in_message":[],"_revision_number":13},{"id":"2bc8f553dc60b9e97cd41c07918d1b76a01e498f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 21:26:39.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a84c961f62f469bb9fec0bdce1c1984 : SUCCESS in 8m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/09f846f2b4404865bce9c9fc8cf2ee8f : SUCCESS in 8m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27c719e12a434ebc853b56f370d27b20 : SUCCESS in 5m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/612e90ada2524d06a6d88357c6f05d77 : SUCCESS in 7m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/10375009c2c94e269669e796805b108b : SUCCESS in 7m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5a0026146ee0464ebf008c995c913c2a : SUCCESS in 8m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0d5be238b0442f1a88120d8186d4906 : SUCCESS in 9m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43d3d56c32e34601a6ccbe71563cf04c : SUCCESS in 11m 23s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/54b0ed8e225443b08a13dbc54cadde67 : SUCCESS in 7m 26s\n- ironic-tox-unit-with-driver-libs-python3 https://zuul.opendev.org/t/openstack/build/83e45c43daf04c75a737b80a2ce947f6 : SUCCESS in 8m 28s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/f61d82c4a5bd46c3b0637ef4d62f1d9e : SUCCESS in 1h 07m 27s\n- ironic-tempest-functional-python2 https://zuul.opendev.org/t/openstack/build/8043786caf2a40d89fc6c33df88fc9d0 : SUCCESS in 26m 55s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/d73c0a3a305d485db7972519ff5e5268 : SUCCESS in 28m 19s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/09c65fe5469b4afc82daab65fdf9a248 : SUCCESS in 2h 02m 03s\n- ironic-grenade-dsvm-multinode-multitenant https://zuul.opendev.org/t/openstack/build/bbf3e5d768c1456389d165241daab691 : SUCCESS in 1h 40m 54s\n- ironic-tempest-ipa-partition-pxe_ipmitool-tinyipa-python3 https://zuul.opendev.org/t/openstack/build/3e4bde1fc1c4453d82df3cc63b6cc02e : SUCCESS in 1h 04m 00s\n- ironic-tempest-ipa-partition-redfish-tinyipa https://zuul.opendev.org/t/openstack/build/601aaef8933b4339af5fec9f17c52d44 : SUCCESS in 57m 52s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e83e1450c8e84311bd1497e662cf03f4 : SUCCESS in 44m 33s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/f29629d0726744ebbb300e2f7cf1f08e : SUCCESS in 1h 27m 21s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ed62f8b0fe7641799bcd084730d206cb : FAILURE in 36m 28s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/746e5d68f8b04878bedce518e5a5716c : SUCCESS in 50m 48s\n- ironic-tempest-ipa-partition-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/77bb561cd2db46dda5e59dc6e0c35fa9 : SUCCESS in 55m 40s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/6defce1908184c90836ff14fd4193465 : SUCCESS in 40m 53s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/280288eb5cea4f4283d16a14a957f739 : FAILURE in 46m 00s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/86b147ee5f4443beb953c9e84830cd16 : SUCCESS in 4m 31s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-pxe_snmp-tinyipa https://zuul.opendev.org/t/openstack/build/811ead44af024a8b8714bf6e6e8c1769 : SUCCESS in 46m 31s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/656373a0ecaa45d6b8849630d71c9044 : SUCCESS in 48m 32s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/94ff751433b34e28beff3f44c9facc41 : FAILURE in 11m 08s (non-voting)\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/345520a99ad34c369f54bf313ef1d4e4 : SUCCESS in 1h 17m 21s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/02596de0ef814336a86f54128079b925 : SUCCESS in 50m 58s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"c603fe480e6c7f70366188c740f56b840620835b","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2019-09-19 22:56:50.000000000","message":"Patch Set 13:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-R640-UEFI-py2-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-py2-tempest-dsvm-ironic-idrac/2019-09-19:14:05/680660/13/ : FAILURE in 1h 28m 47s\n- dellemc-DSS9000-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-DSS9000-tempest-dsvm-ironic-idrac/2019-09-19:14:05/680660/13/ : FAILURE in 41m 15s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac/2019-09-19:14:05/680660/13/ : FAILURE in 1h 00m 18s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac/2019-09-19:14:05/680660/13/ : FAILURE in 42m 56s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac/2019-09-19:14:05/680660/13/ : FAILURE in 40m 13s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish/2019-09-19:14:05/680660/13/ : FAILURE in 40m 36s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish/2019-09-19:14:05/680660/13/ : FAILURE in 57m 24s\n- dellemc-DSS9000-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-DSS9000-tempest-dsvm-ironic-redfish/2019-09-19:14:05/680660/13/ : FAILURE in 37m 51s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi/2019-09-19:14:05/680660/13/ : FAILURE in 57m 42s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi/2019-09-19:14:05/680660/13/ : FAILURE in 54m 22s","accounts_in_message":[],"_revision_number":13},{"id":"ac6ad5d4564825071f83af904567ac4382a07fb4","author":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"date":"2019-09-20 14:13:41.000000000","message":"Patch Set 13: Code-Review-1\n\n(7 comments)\n\nThanks for the updates Rachit!  I have just a few more comments.","accounts_in_message":[],"_revision_number":13},{"id":"14d1b10adbcdcf3623dc60c47c03952b99e5db05","author":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"date":"2019-09-23 12:59:06.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"c1617c78ca776fcdd444ea95e3f5ecd5fc0372ba","author":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"date":"2019-09-23 13:05:51.000000000","message":"Patch Set 14:\n\n(7 comments)","accounts_in_message":[],"_revision_number":14},{"id":"4398d16728a31a381ddc44ab5580ecc058b8c0d5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-09-23 13:48:54.000000000","message":"Patch Set 14:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-ironic-agent_ipmitool https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/ironic/60/680660/14/check-ironic/tempest-dsvm-ironic-agent_ipmitool/6514237/ : FAILURE in 18m 08s","accounts_in_message":[],"_revision_number":14},{"id":"8db7f00245e51608110e3765f18185fdffb6fbe9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 20:55:50.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ce8761a31da49b79ec8b966f2fff942 : SUCCESS in 7m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d43c48f362be4125bdbf10cb2b3339fd : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22c9d3c667344f01842d062daa842004 : SUCCESS in 4m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9b7e822b1e1342708b4d53e145b5136c : SUCCESS in 7m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2ca8a02a3234242851e5197f2ed3b98 : SUCCESS in 6m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1950b964280f4f5b9729b1ffee63448f : SUCCESS in 8m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82d396bf85c943489e99bb170ec9f329 : SUCCESS in 21m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04ee16f0d9e24f448684599b97f1f060 : SUCCESS in 10m 35s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/3dd1068b9ac64974a9556fafb20edfea : SUCCESS in 7m 17s\n- ironic-tox-unit-with-driver-libs-python3 https://zuul.opendev.org/t/openstack/build/35292124924f4dc2b48210e19fef10aa : SUCCESS in 9m 23s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/82918992b6df47dd9225737c167c2db5 : SUCCESS in 1h 14m 59s\n- ironic-tempest-functional-python2 https://zuul.opendev.org/t/openstack/build/1819b9170b8c47aba5c42eedcd7a158b : SUCCESS in 25m 35s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/174776e517734d08a7038812c45a7319 : SUCCESS in 25m 34s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/8a345a9866f04adca19f6d3c51621f4e : SUCCESS in 1h 38m 11s\n- ironic-grenade-dsvm-multinode-multitenant https://zuul.opendev.org/t/openstack/build/209979af465547bb969cc6cada6750cf : FAILURE in 1h 33m 39s (non-voting)\n- ironic-tempest-ipa-partition-pxe_ipmitool-tinyipa-python3 https://zuul.opendev.org/t/openstack/build/9e2ccde4ea664454bcdb2f3bb63ac581 : SUCCESS in 49m 43s\n- ironic-tempest-ipa-partition-redfish-tinyipa https://zuul.opendev.org/t/openstack/build/0ae531bf6bba4bf7a451903c0b0f6e32 : SUCCESS in 44m 36s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7ac7eda8180a44d7bc1cc60b98508c56 : SUCCESS in 43m 04s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/0f673ab1ff8444e1821640c67e44bb26 : RETRY_LIMIT in 1h 40m 23s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ec4b0904416b4e0fa3ad0410393b8420 : SUCCESS in 54m 38s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/2c4a5305eb14433393edbb7243f6e348 : SUCCESS in 53m 00s\n- ironic-tempest-ipa-partition-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/6d311f4fee66411985f84ca6190b2be2 : SUCCESS in 51m 35s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/539b35b36d204d5e92acadc9af81754a : SUCCESS in 48m 40s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/4c859f9abc7844069f7b48ba34096ad3 : SUCCESS in 46m 21s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/40ec37e1bba545e5aa02573e9c2eb522 : SUCCESS in 4m 22s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-pxe_snmp-tinyipa https://zuul.opendev.org/t/openstack/build/4fdaea9f7f0c483cb4356d620a69d18e : SUCCESS in 53m 48s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/1296c0d64176474da682e0253f3a99ad : SUCCESS in 45m 31s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/0d619a2f9dcf44c9a84ce4f00f3b5fb7 : FAILURE in 16m 49s (non-voting)\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/68609c821c0f44d28d4f588dd63bea91 : SUCCESS in 1h 21m 11s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/c09e6505602c4f51a5f611be18caaf59 : SUCCESS in 51m 39s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"7d0fc91b2faae6e8e34767b82345d6995258d5af","author":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"date":"2019-09-23 21:32:05.000000000","message":"Patch Set 14: Code-Review+1\n\n(5 comments)\n\nLooks good!  Had a few nits if you would like to address them.","accounts_in_message":[],"_revision_number":14},{"id":"7cec407f93315bab05cee4fdd52072885b2a599a","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2019-09-24 08:53:45.000000000","message":"Patch Set 14:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-R640-UEFI-py2-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-py2-tempest-dsvm-ironic-idrac/2019-09-23:07:59/680660/14/ : FAILURE in 1h 19m 04s\n- dellemc-DSS9000-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-DSS9000-tempest-dsvm-ironic-idrac/2019-09-23:07:59/680660/14/ : SUCCESS in 49m 08s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac/2019-09-23:07:59/680660/14/ : SUCCESS in 45m 36s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac/2019-09-23:07:59/680660/14/ : SUCCESS in 45m 37s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac/2019-09-23:07:59/680660/14/ : SUCCESS in 47m 42s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish/2019-09-23:07:59/680660/14/ : SUCCESS in 48m 06s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish/2019-09-23:07:59/680660/14/ : SUCCESS in 1h 19m 37s\n- dellemc-DSS9000-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-DSS9000-tempest-dsvm-ironic-redfish/2019-09-23:07:59/680660/14/ : SUCCESS in 44m 08s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi/2019-09-23:07:59/680660/14/ : SUCCESS in 42m 38s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi/2019-09-23:07:59/680660/14/ : FAILURE in 46m 28s","accounts_in_message":[],"_revision_number":14},{"id":"321c02cb5e45f3ad43b9501c7cb2691212ca5392","author":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"date":"2019-09-24 10:34:27.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"a4e909c06447c00cd6824ee543e7f0399112cff1","author":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"date":"2019-09-24 10:35:47.000000000","message":"Patch Set 15:\n\n(5 comments)","accounts_in_message":[],"_revision_number":15},{"id":"91604b4efea1bebd35431e204e8fa9d3485a1f02","author":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"date":"2019-09-24 11:10:34.000000000","message":"Patch Set 15: Code-Review+1\n\nLooks great!  Thanks for the updates Rachit!","accounts_in_message":[],"_revision_number":15},{"id":"47940d0aa0d9842157c215cc0acebc0fc454c379","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-09-24 11:34:12.000000000","message":"Patch Set 15:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-ironic-agent_ipmitool https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/ironic/60/680660/15/check-ironic/tempest-dsvm-ironic-agent_ipmitool/0ee8b70/ : FAILURE in 17m 23s","accounts_in_message":[],"_revision_number":15},{"id":"3ba1041b96d5afa1270f8f7769f728cecebd805d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 13:13:09.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94fb15ecb3bb4715aaef9fbfe31d06bb : SUCCESS in 7m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa316883e3cd4a31adc11f3a0f8d8a84 : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f47a5e4b46fc44e8a827af6c38c42ef5 : SUCCESS in 7m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/208828779a2f45a99e75c1ddb945cddd : SUCCESS in 9m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4034968e12484b6191ce454243a339dd : SUCCESS in 10m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6757022dcb9241bc948bc4030c60deb6 : SUCCESS in 8m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ada3c1b22d144cf49522096fbd0a2b58 : SUCCESS in 20m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58d05de035544f08a6d78336a1cb5f22 : SUCCESS in 11m 18s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/a5bf253da84b4bc89d9fbdc158491971 : SUCCESS in 8m 10s\n- ironic-tox-unit-with-driver-libs-python3 https://zuul.opendev.org/t/openstack/build/74dc6e04a1094a3f8bc5149690bd25dc : SUCCESS in 8m 08s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/e1c9ccd34d5442139240c91a8ade2662 : SUCCESS in 1h 11m 41s\n- ironic-tempest-functional-python2 https://zuul.opendev.org/t/openstack/build/e32d093b131e4c71b56756f2979d8d01 : SUCCESS in 27m 59s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/8ebcbc86d0ba4c4c8bd3c15cd148ea39 : SUCCESS in 27m 49s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/84041cd6b06946abbc6ee79681fb0a24 : POST_FAILURE in 1h 43m 59s\n- ironic-grenade-dsvm-multinode-multitenant https://zuul.opendev.org/t/openstack/build/61e6f9fff198412f9ff15554ac5524e0 : FAILURE in 1h 28m 41s (non-voting)\n- ironic-tempest-ipa-partition-pxe_ipmitool-tinyipa-python3 https://zuul.opendev.org/t/openstack/build/fa19ccf59ea14818a3e9edb3c2936c06 : SUCCESS in 49m 14s\n- ironic-tempest-ipa-partition-redfish-tinyipa https://zuul.opendev.org/t/openstack/build/118b58ce47d248b1a06d004248719e2c : SUCCESS in 50m 00s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/49ba9b9d9be84871aebb77655fb6f49d : SUCCESS in 55m 10s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/c75ff8db8ec940e595c8ccb8fb05c49f : SUCCESS in 1h 49m 18s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a1921a4c87444ad28e13936392c4c17d : SUCCESS in 50m 40s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/b20b1d2687a84ed398a3d8907308ea91 : SUCCESS in 44m 29s\n- ironic-tempest-ipa-partition-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/000955af20ef4d49ac5e7eb472d82dd7 : SUCCESS in 48m 57s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/e8b8648978a54b6ca91b82320f1945b0 : SUCCESS in 56m 48s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/93599f96f0bb42d098f6bc12a2d09270 : SUCCESS in 55m 21s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/f4df2fc5b6704c6490ca6afc43d8e728 : SUCCESS in 5m 22s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-pxe_snmp-tinyipa https://zuul.opendev.org/t/openstack/build/5d71668facb04065830a20f64fba6e2b : SUCCESS in 59m 57s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/90d9adc64c2542fcad458f0f84eaedb7 : SUCCESS in 51m 33s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/2023da4cb4184f30b7e093670fde3752 : FAILURE in 16m 35s (non-voting)\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/73e8e360656f4bf0b9393a283b3862f3 : SUCCESS in 1h 41m 56s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/ed406e7ddad3427e8946771dd4563cfd : SUCCESS in 58m 09s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"962b219b4386c25af266d2ebd97a6d09de992366","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2019-09-25 10:08:47.000000000","message":"Patch Set 15:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-R640-UEFI-py2-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-py2-tempest-dsvm-ironic-idrac/2019-09-24:05:34/680660/15/ : FAILURE in 1h 46m 53s\n- dellemc-DSS9000-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-DSS9000-tempest-dsvm-ironic-idrac/2019-09-24:05:34/680660/15/ : SUCCESS in 1h 15m 53s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac/2019-09-24:05:34/680660/15/ : SUCCESS in 1h 10m 49s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac/2019-09-24:05:34/680660/15/ : SUCCESS in 1h 49m 53s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac/2019-09-24:05:34/680660/15/ : SUCCESS in 51m 54s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish/2019-09-24:05:34/680660/15/ : SUCCESS in 1h 03m 48s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish/2019-09-24:05:34/680660/15/ : SUCCESS in 1h 30m 09s\n- dellemc-DSS9000-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-DSS9000-tempest-dsvm-ironic-redfish/2019-09-24:05:34/680660/15/ : SUCCESS in 55m 59s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi/2019-09-24:05:34/680660/15/ : SUCCESS in 1h 17m 49s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi/2019-09-24:05:34/680660/15/ : FAILURE in 56m 49s","accounts_in_message":[],"_revision_number":15},{"id":"6eb56beb054408fe6304ec98fe065d078a8fbf40","author":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"date":"2019-09-25 12:53:29.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"7ad1d163ba47e72b035849d548239089c470dd7c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-09-25 13:26:42.000000000","message":"Patch Set 16:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-ironic-agent_ipmitool https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/ironic/60/680660/16/check-ironic/tempest-dsvm-ironic-agent_ipmitool/16b3c63/ : FAILURE in 21m 55s","accounts_in_message":[],"_revision_number":16},{"id":"2f2152bb188c25f5c97806d07bf04034f50dd6e8","author":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"date":"2019-09-25 14:04:35.000000000","message":"Patch Set 16: Code-Review+1\n\nLooks good.","accounts_in_message":[],"_revision_number":16},{"id":"1ae5661373c1c419433d1b924d42337f4d8b5df2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 14:50:25.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f3159e2f1a14fceab67ba9554fa1f69 : SUCCESS in 7m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4f1a501b35d42d49b13217335692730 : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fab9456f669243ccb41ecb8ecb1c5e95 : SUCCESS in 5m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/caa4a9ea529e42ebbbcde53a80dad6cd : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36e4840564d249488789c12bf30c3dff : SUCCESS in 7m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6de256538fa74607bde04f61d7e48810 : SUCCESS in 7m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abf97a96d232488dacf90eee7ba5c2ee : SUCCESS in 20m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d341b62b26f94e17b59160d6a7ac45f8 : SUCCESS in 11m 16s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/b6dda5aad3e74b5cbe60c5100b35d2d6 : SUCCESS in 8m 18s\n- ironic-tox-unit-with-driver-libs-python3 https://zuul.opendev.org/t/openstack/build/6acc17946caf4c0a9c1612625d93e8a4 : SUCCESS in 12m 19s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/c97f114e368b426fbf4a01683208f0d8 : FAILURE in 1h 46m 03s\n- ironic-tempest-functional-python2 https://zuul.opendev.org/t/openstack/build/48df6389db9c4386967b36d3a7daedc7 : SUCCESS in 28m 57s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/42011266ad3b456e8fa5ee4cf02e6a78 : SUCCESS in 39m 56s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/ff5507fa8f51438bb9d0a42047467bfe : SUCCESS in 1h 37m 58s\n- ironic-grenade-dsvm-multinode-multitenant https://zuul.opendev.org/t/openstack/build/b1866a4157e94f89a5fad5d811318282 : FAILURE in 1h 21m 01s (non-voting)\n- ironic-tempest-ipa-partition-pxe_ipmitool-tinyipa-python3 https://zuul.opendev.org/t/openstack/build/68f7a5d43de545cbb6749a6a7050a295 : FAILURE in 1h 01m 59s\n- ironic-tempest-ipa-partition-redfish-tinyipa https://zuul.opendev.org/t/openstack/build/62d140cfa6cf4bef96a1a218a8592a84 : FAILURE in 1h 01m 09s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a8b35a3da23a471eac0c5c464e4b5ea3 : FAILURE in 44m 04s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/5829c188398449ccbf3252daa9055621 : FAILURE in 1h 44m 32s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ec9c8d5d3fd34d8892a64010969423a9 : FAILURE in 1h 02m 28s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/205b78dba0eb4951aa15704577499c0b : FAILURE in 51m 01s\n- ironic-tempest-ipa-partition-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/13e6fb8d3a7f419eb6c6f2f9db386666 : FAILURE in 48m 31s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/243264f16c904075ae886b707ef9d187 : SUCCESS in 53m 09s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/da335455745541c5946f06a264e91848 : FAILURE in 1h 02m 33s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/b44e947e0cc54e42b842143757d77d93 : SUCCESS in 6m 29s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-pxe_snmp-tinyipa https://zuul.opendev.org/t/openstack/build/ea74ab0a0d7a4bfcaebb48f6be3b2832 : FAILURE in 47m 08s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/49135eddeee84ada8b3a57e5d5dea078 : FAILURE in 51m 01s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/88ef6391030f4e17b10b1a8fca98418c : FAILURE in 11m 21s (non-voting)\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/f26748b69daf4d4c9208f10b4b407211 : FAILURE in 1h 34m 34s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/a0e9c1a489f54bf29d4a79d5665ebe0f : FAILURE in 1h 04m 53s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"f50065d89e01175666c29f6959b9a09c7c455b6d","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2019-09-26 04:40:34.000000000","message":"Patch Set 16:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-R640-UEFI-py2-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-py2-tempest-dsvm-ironic-idrac/2019-09-25:07:53/680660/16/ : FAILURE in 34m 12s\n- dellemc-DSS9000-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-DSS9000-tempest-dsvm-ironic-idrac/2019-09-25:07:53/680660/16/ : SUCCESS in 45m 11s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac/2019-09-25:07:53/680660/16/ : SUCCESS in 53m 07s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac/2019-09-25:07:53/680660/16/ : SUCCESS in 45m 14s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac/2019-09-25:07:53/680660/16/ : SUCCESS in 50m 31s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish/2019-09-25:07:53/680660/16/ : SUCCESS in 51m 17s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish/2019-09-25:07:53/680660/16/ : SUCCESS in 1h 07m 36s\n- dellemc-DSS9000-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-DSS9000-tempest-dsvm-ironic-redfish/2019-09-25:07:53/680660/16/ : FAILURE in 32m 15s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi/2019-09-25:07:53/680660/16/ : SUCCESS in 47m 01s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi/2019-09-25:07:53/680660/16/ : FAILURE in 50m 11s","accounts_in_message":[],"_revision_number":16},{"id":"a9f3688203d7c955ac843b21ab86fd1693c25d1a","author":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"date":"2019-09-30 08:39:14.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"43403d85ca8d8070ce7b6cc5d7133e43ca9d4f8f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-09-30 08:52:11.000000000","message":"Patch Set 16:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-ironic-agent_ipmitool https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/ironic/60/680660/16/check-ironic/tempest-dsvm-ironic-agent_ipmitool/6f62606/ : FAILURE in 12m 36s","accounts_in_message":[],"_revision_number":16},{"id":"7c21e14b45a349ae882370a9e89d867e255e6aa5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 10:35:16.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d43a96fcc7645bd89045be93118a039 : SUCCESS in 7m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/677a80812c24462b858198bb7c96d51c : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1d87bdd13ff4d4b8d859fff8c226442 : SUCCESS in 5m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/abc5f5161aa147f198d746c6c21d8f38 : SUCCESS in 7m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e6d71ca7d1a4dd8a1138abeeb602bfa : SUCCESS in 7m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/81101e137cf749e287dd5b7d8a7aa3d8 : SUCCESS in 8m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b12d1516dc234f1dba33cc39e015966e : SUCCESS in 16m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4e97974ad3c4ddaa2cd1fa0a4a97f10 : SUCCESS in 12m 51s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/eda9e5fa6b2847588b63f7bb6f007271 : SUCCESS in 8m 02s\n- ironic-tox-unit-with-driver-libs-python3 https://zuul.opendev.org/t/openstack/build/b4b8507a96894ad9a2db0c75d65ee1c2 : SUCCESS in 8m 10s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/fc8e8fcf7a9949d88368bb0fb84fbe06 : SUCCESS in 1h 14m 00s\n- ironic-tempest-functional-python2 https://zuul.opendev.org/t/openstack/build/065afc0140ed43aa81cf782d5c431931 : SUCCESS in 25m 16s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/520bfcbd839d4b1dbca15cfeade0ae42 : SUCCESS in 27m 47s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/eac91b1715fa44aaae67dc0fb6addef1 : SUCCESS in 1h 51m 08s\n- ironic-grenade-dsvm-multinode-multitenant https://zuul.opendev.org/t/openstack/build/3c0334c7f8dc42879a8d456773ed9b5a : FAILURE in 1h 24m 38s (non-voting)\n- ironic-tempest-ipa-partition-pxe_ipmitool-tinyipa-python3 https://zuul.opendev.org/t/openstack/build/30e7428d5b8d44189ed1ecfa8026cfeb : SUCCESS in 51m 51s\n- ironic-tempest-ipa-partition-redfish-tinyipa https://zuul.opendev.org/t/openstack/build/06e18314adfc4f398b0860acd8849eaf : SUCCESS in 49m 11s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7585d4c224f842a2a9a653bb2d0ed8c8 : SUCCESS in 44m 44s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/8ec61feb5a80404ca935be44bc2e82c0 : SUCCESS in 1h 31m 20s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3f3c75bd04a64d608aaa0e4dd8b3b2d2 : SUCCESS in 55m 24s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/b56a6ba8b9f941499181cd96088990f7 : SUCCESS in 46m 26s\n- ironic-tempest-ipa-partition-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/6eb41cbb7ad649abb953d73cd3a24a7d : SUCCESS in 45m 39s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/53cb7d9f2561441d8d1f8b976bbd88e5 : SUCCESS in 38m 31s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/58840c8374b047179f766143d99b9bea : SUCCESS in 47m 10s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/92881351e85040dcb112130455c2193a : SUCCESS in 4m 40s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-pxe_snmp-tinyipa https://zuul.opendev.org/t/openstack/build/568e936597be400e9b5e0bad7ce893b9 : SUCCESS in 44m 49s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/3e5ea73a110841e189e13aad46c12781 : SUCCESS in 48m 44s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/dbfd9dbbf5d8468b8f4dd9dc0faaf825 : SUCCESS in 20m 34s (non-voting)\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/25ad37203b8346a1bbe0f471cab6580d : FAILURE in 1h 10m 52s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/2b4b49bf36ce458a8ff0e0a58659fa33 : SUCCESS in 47m 40s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"0a32db02013f1ef69db9b55045d6c548c7a8f956","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2019-09-30 11:30:12.000000000","message":"Patch Set 16:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-R640-UEFI-py2-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-py2-tempest-dsvm-ironic-idrac/2019-09-30:03:39/680660/16/ : FAILURE in 51m 17s\n- dellemc-DSS9000-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-DSS9000-tempest-dsvm-ironic-idrac/2019-09-30:03:39/680660/16/ : SUCCESS in 44m 28s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac/2019-09-30:03:39/680660/16/ : SUCCESS in 53m 47s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac/2019-09-30:03:39/680660/16/ : SUCCESS in 49m 29s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac/2019-09-30:03:39/680660/16/ : SUCCESS in 56m 07s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish/2019-09-30:03:39/680660/16/ : SUCCESS in 44m 24s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish/2019-09-30:03:39/680660/16/ : SUCCESS in 1h 12m 05s\n- dellemc-DSS9000-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-DSS9000-tempest-dsvm-ironic-redfish/2019-09-30:03:39/680660/16/ : SUCCESS in 47m 30s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi/2019-09-30:03:39/680660/16/ : FAILURE in 49m 46s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi/2019-09-30:03:39/680660/16/ : FAILURE in 45m 48s","accounts_in_message":[],"_revision_number":16},{"id":"c9decb3f13cc529d5535e452570f0fea5890ea4f","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-09-30 15:02:10.000000000","message":"Patch Set 16:\n\n(7 comments)\n\nlooks good, thanks!\nJust a couple of comments, but nothing strictly code or functional related","accounts_in_message":[],"_revision_number":16},{"id":"041ffed6d3b9d4d81b773993e73ba74d5c0c2862","author":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"date":"2019-10-01 13:23:02.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"8313d167641cd11c7e97b49fd7e276014c876fa6","author":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"date":"2019-10-01 15:04:24.000000000","message":"Patch Set 17: Code-Review+1","accounts_in_message":[],"_revision_number":17},{"id":"96bff0d3891c71eee9e1b25c299532590e7eb396","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 15:11:14.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d9951314a8547e28b7a595993accb0d : SUCCESS in 8m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/408fbd4648664b64ba3e72e2b38855b1 : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2844ea1847e43f28c9fb0f8cb59e49e : SUCCESS in 4m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1b0189fa0e294705b18555b3049de6c1 : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b934d764d00243a69a6151d2136bc0b4 : SUCCESS in 7m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/475a056f4a1e4dec93f1d742e67d7431 : SUCCESS in 7m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da6ee3fdf2004d0bbd702156ed9c3654 : SUCCESS in 18m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ef7799d74c242d7b36c6fe31e80a329 : SUCCESS in 11m 15s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/bfe9aaf1d2704d6ba587f26ba6f21762 : SUCCESS in 8m 06s\n- ironic-tox-unit-with-driver-libs-python3 https://zuul.opendev.org/t/openstack/build/7387f9e726214cf9a543aefa3d6aff2f : SUCCESS in 9m 16s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/494059b4410d4ebba48a46522af8df20 : SUCCESS in 1h 14m 11s\n- ironic-tempest-functional-python2 https://zuul.opendev.org/t/openstack/build/2ab62c00e48d44bf844aa585254f0eaf : SUCCESS in 25m 22s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/e4ea124747c8477e9c4667265acf63c7 : SUCCESS in 27m 03s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/74aa5430ebf445dab24550ff6fe62324 : SUCCESS in 1h 37m 01s\n- ironic-grenade-dsvm-multinode-multitenant https://zuul.opendev.org/t/openstack/build/65addc0eff0c4f4182da59766bb13360 : FAILURE in 1h 24m 59s (non-voting)\n- ironic-tempest-ipa-partition-pxe_ipmitool-tinyipa-python3 https://zuul.opendev.org/t/openstack/build/5fec206cc9334a339a867099b3ee37fc : SUCCESS in 50m 49s\n- ironic-tempest-ipa-partition-redfish-tinyipa https://zuul.opendev.org/t/openstack/build/397ce00124294ea18ac065e3b58fe9f8 : SUCCESS in 42m 01s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/d20ee328a19d477a98fcb360f38f5919 : SUCCESS in 47m 13s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/a87827063b0b4c4fb9fdc651a1543da7 : SUCCESS in 1h 41m 04s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/df2837f8b19d482c89f3a8387cb05488 : SUCCESS in 1h 01m 52s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/540c7a50e0e44f75a0be5906c4160e8d : SUCCESS in 56m 31s\n- ironic-tempest-ipa-partition-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/cb570c5934ca45958c00569a0eccd4a0 : SUCCESS in 45m 40s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/b5a1b4a39774462c84320f28a9b0b47a : SUCCESS in 47m 05s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/abc882b4b7664191af19e901d280ea41 : SUCCESS in 50m 20s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/901785d432734a0e8bbc466a12b62441 : SUCCESS in 4m 44s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-pxe_snmp-tinyipa https://zuul.opendev.org/t/openstack/build/208175c55e484d2eb9a850f40b09c8a9 : SUCCESS in 46m 51s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/0f660610f03b4f81b527283aab88277a : SUCCESS in 58m 19s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/8d465d3a9a014eb0bff0d72d52e656a9 : SUCCESS in 24m 04s (non-voting)\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/32f15433fdf342c7b69374244c32524e : FAILURE in 1h 18m 06s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/67c581df47324919a4329e987e7add64 : SUCCESS in 50m 38s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"0bcdff6a20c9977ac4e6ff1d4012f01bd421f2fd","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2019-10-01 16:24:47.000000000","message":"Patch Set 17:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-R640-UEFI-py2-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-py2-tempest-dsvm-ironic-idrac/2019-10-01:08:23/680660/17/ : FAILURE in 48m 58s\n- dellemc-DSS9000-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-DSS9000-tempest-dsvm-ironic-idrac/2019-10-01:08:23/680660/17/ : SUCCESS in 44m 56s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac/2019-10-01:08:23/680660/17/ : FAILURE in 51m 25s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac/2019-10-01:08:23/680660/17/ : SUCCESS in 46m 01s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac/2019-10-01:08:23/680660/17/ : SUCCESS in 50m 42s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish/2019-10-01:08:23/680660/17/ : SUCCESS in 45m 05s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish/2019-10-01:08:23/680660/17/ : SUCCESS in 56m 54s\n- dellemc-DSS9000-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-DSS9000-tempest-dsvm-ironic-redfish/2019-10-01:08:23/680660/17/ : SUCCESS in 44m 33s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi/2019-10-01:08:23/680660/17/ : FAILURE in 49m 54s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi/2019-10-01:08:23/680660/17/ : FAILURE in 47m 36s","accounts_in_message":[],"_revision_number":17},{"id":"214b55beebb0e53074fd4c26d3d8ffbd5bea8988","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-10-02 02:43:15.000000000","message":"Patch Set 17:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-ironic-agent_ipmitool https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/ironic/60/680660/17/check-ironic/tempest-dsvm-ironic-agent_ipmitool/e69f5f2/ : FAILURE in 19m 30s","accounts_in_message":[],"_revision_number":17},{"id":"41a8cf08b6703ba70928ac465e14f6cc75e897e8","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-10-02 08:50:10.000000000","message":"Patch Set 17:\n\n\u003e Uploaded patch set 17.\n\nhi Rachit, did you see my notes on the release note ?","accounts_in_message":[],"_revision_number":17},{"id":"3c89c267e2a83be66f494ef5c63bd4d355c6adfe","author":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"date":"2019-10-02 12:05:52.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"537b0004c3271b46ac701f7b72789041bd0732e5","author":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"date":"2019-10-02 12:19:21.000000000","message":"Patch Set 18: Workflow-1","accounts_in_message":[],"_revision_number":18},{"id":"def491c353eecef6032b3397e033c5e84296789f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-10-02 12:25:38.000000000","message":"Patch Set 18:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-ironic-agent_ipmitool https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/ironic/60/680660/18/check-ironic/tempest-dsvm-ironic-agent_ipmitool/5dd37b2/ : FAILURE in 19m 26s","accounts_in_message":[],"_revision_number":18},{"id":"fdf82879b1da667c45e7e219741f831a9cd98b68","author":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"date":"2019-10-02 13:06:30.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"d0ab85ba100bd87f9008e3284f477619d493647b","author":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"date":"2019-10-02 13:16:10.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"8d96d153e12a30bcb976a0f831805ff927204aa3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-10-02 13:36:45.000000000","message":"Patch Set 20:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-ironic-agent_ipmitool https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/ironic/60/680660/20/check-ironic/tempest-dsvm-ironic-agent_ipmitool/2980e33/ : FAILURE in 18m 11s","accounts_in_message":[],"_revision_number":20},{"id":"a5a8d7acba756ee67ed62785d76ab1dd0244231a","author":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"date":"2019-10-02 14:55:44.000000000","message":"Patch Set 20: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":20},{"id":"c8b2a251fdf89a0b26abf77cfcd1a3eb73e3f08c","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2019-10-02 17:03:37.000000000","message":"Patch Set 20:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-R640-UEFI-py2-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-py2-tempest-dsvm-ironic-idrac/2019-10-02:08:16/680660/20/ : FAILURE in 1h 50m 44s\n- dellemc-DSS9000-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-DSS9000-tempest-dsvm-ironic-idrac/2019-10-02:08:16/680660/20/ : SUCCESS in 56m 17s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac/2019-10-02:08:16/680660/20/ : SUCCESS in 1h 07m 37s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac/2019-10-02:08:16/680660/20/ : SUCCESS in 59m 13s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac/2019-10-02:08:16/680660/20/ : FAILURE in 1h 51m 26s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish/2019-10-02:08:16/680660/20/ : SUCCESS in 44m 04s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish/2019-10-02:08:16/680660/20/ : FAILURE in 1h 46m 12s\n- dellemc-DSS9000-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-DSS9000-tempest-dsvm-ironic-redfish/2019-10-02:08:16/680660/20/ : SUCCESS in 1h 15m 24s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi/2019-10-02:08:16/680660/20/ : SUCCESS in 1h 11m 53s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi/2019-10-02:08:16/680660/20/ : FAILURE in 1h 51m 29s","accounts_in_message":[],"_revision_number":20},{"id":"8fc56eeb1fb1e4022d477c0136bd35a48d70cc03","author":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"date":"2019-10-02 17:16:45.000000000","message":"Patch Set 20:\n\n(5 comments)","accounts_in_message":[],"_revision_number":20},{"id":"caa0402fe09e3674ee440b848167c0fc10fd7b11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 17:19:38.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/883a1ce9eac44d269d34cd927b3bd5d7 : SUCCESS in 8m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf3e26a728504f9dba16ec3bc6817894 : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f0ddce7873642c3a6889b8bbd6e43e3 : SUCCESS in 5m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f5df01101553437ebe25ad619bd973ce : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/338c1f0fd03d47b2b208b03251fb5b36 : SUCCESS in 6m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dff0e593d9f2436a8eaded5b74bd4a3c : SUCCESS in 7m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ad97f2a4fa2457da948c6a942da4cf4 : SUCCESS in 18m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/415e9b2ef1844e44b1f40aa5a5e9b74b : SUCCESS in 14m 14s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/4efcc179947241fcb95cec79e79b2466 : SUCCESS in 11m 54s\n- ironic-tox-unit-with-driver-libs-python3 https://zuul.opendev.org/t/openstack/build/e9b6cb0526434f6d951911ee5d239589 : SUCCESS in 8m 02s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/aa9cf934bf994888b16d9f478d4440b0 : SUCCESS in 1h 18m 57s\n- ironic-tempest-functional-python2 https://zuul.opendev.org/t/openstack/build/4843d98b5d0045aeaa31a4033919c653 : SUCCESS in 46m 22s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/d3ef069cf7324d7cb41c24af580d48de : SUCCESS in 36m 49s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/8a362d6904b34ea992514a506e823739 : SUCCESS in 1h 35m 55s\n- ironic-grenade-dsvm-multinode-multitenant https://zuul.opendev.org/t/openstack/build/0c85fb28238f4d46aeb71c425a55113c : FAILURE in 1h 14m 02s (non-voting)\n- ironic-tempest-ipa-partition-pxe_ipmitool-tinyipa-python3 https://zuul.opendev.org/t/openstack/build/bd585b892fb04e8ca78a063a46b31e42 : SUCCESS in 50m 33s\n- ironic-tempest-ipa-partition-redfish-tinyipa https://zuul.opendev.org/t/openstack/build/7f6fd019e8aa49188e340f6d6d47b117 : SUCCESS in 55m 42s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/56ec43b2ed344b3e83e27152e14bc7c8 : SUCCESS in 44m 08s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/0d6ae2b96a2e4520aa249311fa0f6a61 : FAILURE in 1h 56m 56s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/33f4bc8f67234ff4b49016b5e99ecc65 : SUCCESS in 57m 24s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/153a3c3951ec445f8d47ee11d445211f : SUCCESS in 42m 33s\n- ironic-tempest-ipa-partition-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/d8a25e2071d04c2a87b84801e3af74d3 : SUCCESS in 48m 28s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/4dccf480393248379ce8c67c5279df28 : SUCCESS in 39m 43s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/76b343c6998541af90446916ae29bc67 : SUCCESS in 50m 36s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/fb0011a42fdf46069a63c7a55740f893 : SUCCESS in 4m 12s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-pxe_snmp-tinyipa https://zuul.opendev.org/t/openstack/build/248e454e36c247179ecbbded497c1e00 : SUCCESS in 47m 35s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/ffba0c02a4dd4caa8803c5e5ef82c950 : SUCCESS in 49m 57s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/23478ed93482476ea8c79d048f09c228 : SUCCESS in 24m 34s (non-voting)\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/e05b62010ac8410e8a4964381c5d59d7 : FAILURE in 1h 14m 02s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/4da3672817d5443fb9f30cd40819a976 : SUCCESS in 59m 23s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"dba788d2ce9738a18da00674ad2a58549e56ee60","author":{"_account_id":8580,"name":"Digambar","email":"digambarpatil15@yahoo.co.in","username":"digambar"},"date":"2019-10-03 04:28:25.000000000","message":"Patch Set 20: Code-Review+1\n\nGreat work Rachit!","accounts_in_message":[],"_revision_number":20},{"id":"c60376061ba7947bb65240c0850c3c1c1a5375c9","author":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"date":"2019-10-03 13:38:19.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"5dab79038c46d0377b9816f47eb52a338300b14a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-10-03 14:01:23.000000000","message":"Patch Set 21:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-ironic-agent_ipmitool https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/ironic/60/680660/21/check-ironic/tempest-dsvm-ironic-agent_ipmitool/3f27b21/ : FAILURE in 22m 43s","accounts_in_message":[],"_revision_number":21},{"id":"363108a16e02f66b251f01239098716f3e1bbe5b","author":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"date":"2019-10-03 14:30:59.000000000","message":"Patch Set 21: Code-Review+1","accounts_in_message":[],"_revision_number":21},{"id":"4419b0e4bf362531ce2df76c223b5782f5519faf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 15:48:40.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e2d057a6f9643699e7655781d5ecf0b : SUCCESS in 8m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b064d68ea9f4552a482b3977b4975f6 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f81bc3e2625140648cecaa7cd9d042ee : SUCCESS in 5m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/abd56ba3c6f04e54878910898f92f75a : SUCCESS in 7m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f5eee509db944ea8d53d35c612317a1 : SUCCESS in 7m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/78135a1c14fe4058a3932921f3e8e7eb : SUCCESS in 7m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd33ef1812944a199dc530ecbfb26d90 : SUCCESS in 16m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d16082708fd45eda9893219ef6605a6 : SUCCESS in 20m 02s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/c8637c2587534404b0f1f3a3d07e859f : SUCCESS in 10m 42s\n- ironic-tox-unit-with-driver-libs-python3 https://zuul.opendev.org/t/openstack/build/c45a61d774b14a11817114eb10a6ff0e : SUCCESS in 8m 27s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/20ef285037584537b1448ae7e5c1b4bd : FAILURE in 2h 06m 12s\n- ironic-tempest-functional-python2 https://zuul.opendev.org/t/openstack/build/e1301f8e31894e9085d9398e30ac5952 : SUCCESS in 29m 03s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/96dfbf1df9f24035acdd200b14886292 : SUCCESS in 33m 29s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/c4bf03fc22c944c5b17ac7ff8bc0e165 : SUCCESS in 1h 49m 43s\n- ironic-grenade-dsvm-multinode-multitenant https://zuul.opendev.org/t/openstack/build/31f939e88e0a41fba7b20b64ff1b668c : FAILURE in 1h 11m 56s (non-voting)\n- ironic-tempest-ipa-partition-pxe_ipmitool-tinyipa-python3 https://zuul.opendev.org/t/openstack/build/22d023f7c854492397140c9afa366fd4 : SUCCESS in 55m 17s\n- ironic-tempest-ipa-partition-redfish-tinyipa https://zuul.opendev.org/t/openstack/build/0d4ce640f8974755859f22715ab02cb3 : SUCCESS in 50m 05s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/d866119b30444760b6b66efbe744b167 : SUCCESS in 46m 22s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/b1a8acbcb5ff4745bb698b6213855732 : SUCCESS in 1h 46m 59s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f59fc8995c9247eaa347ad3c83cb7a61 : SUCCESS in 57m 23s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/44e3cc839dcc4b909b6ced407ee8e207 : SUCCESS in 55m 16s\n- ironic-tempest-ipa-partition-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/38b72a4cd8f0477c8b8737c01aa49a80 : SUCCESS in 52m 19s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/619a430685e34a6f9784b7f4a80dcec4 : SUCCESS in 38m 27s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/28e6da20f9cd48a98bba3b9f452f8189 : SUCCESS in 49m 27s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/523afd78ce6f494da20e00548b36bc5a : SUCCESS in 4m 01s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-pxe_snmp-tinyipa https://zuul.opendev.org/t/openstack/build/b420bdec08c14d2792089dd62917bc9e : SUCCESS in 44m 56s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/22da0eae7fd04d2f841a9200a5d9388b : SUCCESS in 57m 24s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib-centos7 https://zuul.opendev.org/t/openstack/build/878b46b3811b4ed3b0f45cc43d2b298f : SUCCESS in 1h 11m 21s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/fe73001dce0b47eba0a7b6c39768d5e2 : SUCCESS in 20m 33s (non-voting)\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/e0b0f4aab7fd4a18ab5ba2d32d3907a5 : FAILURE in 1h 04m 12s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/d33c1e6e10e44c8592854f09e69eebbf : SUCCESS in 51m 56s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"959dc2095e8af52ef88f7ede5759bff0760776e1","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2019-10-03 18:18:06.000000000","message":"Patch Set 21:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 finger://HPE-CI/a98aee19f2044e87b99c3f3506b0982f : RETRY_LIMIT in 58m 57s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python2 finger://HPE-CI/6fa541a5c3b743bab804c6a3b8717534 : RETRY_LIMIT in 48m 57s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 finger://HPE-CI/fda61603b7c849489146e7bb74f817f8 : RETRY_LIMIT in 51m 02s","accounts_in_message":[],"_revision_number":21},{"id":"a2b09efe1d20a0c4566f9f49e5b5500728b37572","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2019-10-03 20:43:19.000000000","message":"Patch Set 21:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-R640-UEFI-py2-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-py2-tempest-dsvm-ironic-idrac/2019-10-03:08:38/680660/21/ : FAILURE in 1h 51m 24s\n- dellemc-DSS9000-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-DSS9000-tempest-dsvm-ironic-idrac/2019-10-03:08:38/680660/21/ : SUCCESS in 43m 33s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac/2019-10-03:08:38/680660/21/ : SUCCESS in 45m 03s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac/2019-10-03:08:38/680660/21/ : FAILURE in 1h 11m 25s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac/2019-10-03:08:38/680660/21/ : FAILURE in 1h 51m 27s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish/2019-10-03:08:38/680660/21/ : SUCCESS in 1h 28m 12s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish/2019-10-03:08:38/680660/21/ : FAILURE in 1h 51m 22s\n- dellemc-DSS9000-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-DSS9000-tempest-dsvm-ironic-redfish/2019-10-03:08:38/680660/21/ : SUCCESS in 46m 52s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi/2019-10-03:08:38/680660/21/ : SUCCESS in 42m 00s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi/2019-10-03:08:38/680660/21/ : FAILURE in 1h 46m 05s","accounts_in_message":[],"_revision_number":21},{"id":"25bc0dd3b5a6f0fe7157e869f3a0442ab4cb8fc3","author":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"date":"2019-10-04 05:04:47.000000000","message":"Patch Set 21:\n\nrecheck","accounts_in_message":[],"_revision_number":21},{"id":"cacd2d391f9fd2e4dcce7d1d4fc0e3060494e49f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 06:46:12.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8661bf0139714debb3613d141396e0d2 : SUCCESS in 10m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a8a757467e3f47d79ab0bc6aa9b9aaab : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93e7f7b238f94f1c848cc292f1337a27 : SUCCESS in 4m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/57be29914d7848f6823ce145e08e1db0 : SUCCESS in 9m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c98468c12c04a03a3e83edb9557b143 : SUCCESS in 9m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7b880f0f91434813800f5e0370fc6987 : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de71b301541843efa5c2dfe05c0c4631 : SUCCESS in 22m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/031d2723ca374dbfb4888d68b7cdf0aa : SUCCESS in 12m 32s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/303598d63b374a6a9c678923055f25a6 : SUCCESS in 11m 41s\n- ironic-tox-unit-with-driver-libs-python3 https://zuul.opendev.org/t/openstack/build/df436d759b2f4d038a30f51b26c284eb : SUCCESS in 11m 06s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/94854b9be97f4123a6b527b373673add : SUCCESS in 1h 17m 59s\n- ironic-tempest-functional-python2 https://zuul.opendev.org/t/openstack/build/1c062f4881fa440d9a65eb2dadcf8748 : SUCCESS in 28m 49s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/ff08427fd8b5470281dd33ba8df1195c : SUCCESS in 30m 53s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/49aad71fadb148bea408e419fb0db721 : SUCCESS in 1h 30m 52s\n- ironic-grenade-dsvm-multinode-multitenant https://zuul.opendev.org/t/openstack/build/6ba8d292e30c4128af8b8142cc10d0fb : FAILURE in 1h 22m 18s (non-voting)\n- ironic-tempest-ipa-partition-pxe_ipmitool-tinyipa-python3 https://zuul.opendev.org/t/openstack/build/31bc3a20e31e41e097c4496d069880ee : SUCCESS in 49m 35s\n- ironic-tempest-ipa-partition-redfish-tinyipa https://zuul.opendev.org/t/openstack/build/b8a41e88f8f141ec9760f6d85c44316a : SUCCESS in 48m 08s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/2a7376d0425e45ea83533567f7125bb9 : SUCCESS in 50m 47s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/554fd17ee64849be845b0c4df5b6164f : SUCCESS in 1h 35m 34s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/85913b472feb45cb9f7bb7482692e9ff : SUCCESS in 54m 39s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/cb3c8a6c946246579a73e9e9da5c1c15 : SUCCESS in 54m 00s\n- ironic-tempest-ipa-partition-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/6bce6e0e3bd14d6d96c20ae3417b3930 : SUCCESS in 50m 44s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/cf31119f6dc64ac6b9e4ef07415c1664 : SUCCESS in 38m 39s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/ba088abef6a84e86813917ca1e1f726c : SUCCESS in 47m 16s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/4c54140b20654a3e95e18ee569228455 : SUCCESS in 3m 57s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-pxe_snmp-tinyipa https://zuul.opendev.org/t/openstack/build/5e57800ef62c445eb664972c5649f1d0 : SUCCESS in 1h 02m 26s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/6ef6b9cf93c44dd685b34e96b0292dd8 : SUCCESS in 52m 10s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib-centos7 https://zuul.opendev.org/t/openstack/build/754ec6013979467183b1670028be8b97 : SUCCESS in 1h 11m 00s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/d4d2af628e364cbfbbe20c77c72ed770 : SUCCESS in 18m 48s (non-voting)\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/c5fc356ae9ff4a6e9d7bf5b791c86481 : FAILURE in 1h 10m 06s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/6e734e708c6b4ddd9a4ccb0dc20a590f : SUCCESS in 53m 21s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"16d4c636377657334ef79fedc50bbb6b2b3ed806","author":{"_account_id":8580,"name":"Digambar","email":"digambarpatil15@yahoo.co.in","username":"digambar"},"date":"2019-10-04 10:20:22.000000000","message":"Patch Set 21: Code-Review+1","accounts_in_message":[],"_revision_number":21},{"id":"556f70b09265105ef8f07758585369e1911e3268","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2019-10-04 16:57:48.000000000","message":"Patch Set 21:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-R640-UEFI-py2-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-py2-tempest-dsvm-ironic-idrac/2019-10-04:00:04/680660/21/ : FAILURE in 1h 51m 23s\n- dellemc-DSS9000-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-DSS9000-tempest-dsvm-ironic-idrac/2019-10-04:00:04/680660/21/ : SUCCESS in 43m 01s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac/2019-10-04:00:04/680660/21/ : SUCCESS in 54m 15s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac/2019-10-04:00:04/680660/21/ : SUCCESS in 45m 57s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac/2019-10-04:00:04/680660/21/ : FAILURE in 1h 51m 32s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish/2019-10-04:00:04/680660/21/ : SUCCESS in 1h 02m 46s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish/2019-10-04:00:04/680660/21/ : FAILURE in 1h 46m 24s\n- dellemc-DSS9000-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-DSS9000-tempest-dsvm-ironic-redfish/2019-10-04:00:04/680660/21/ : SUCCESS in 42m 24s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi/2019-10-04:00:04/680660/21/ : SUCCESS in 41m 42s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi/2019-10-04:00:04/680660/21/ : FAILURE in 1h 51m 39s","accounts_in_message":[],"_revision_number":21},{"id":"64c21445717a1d142a31284b601f21cdc67be02d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-10-05 11:06:52.000000000","message":"Patch Set 21:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-ironic-agent_ipmitool https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/ironic/60/680660/21/check-ironic/tempest-dsvm-ironic-agent_ipmitool/a04be7b/ : FAILURE in 17m 48s","accounts_in_message":[],"_revision_number":21},{"id":"41b593fe624838902dfae2c4c242109e4bbba5f1","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2019-10-07 17:14:12.000000000","message":"Patch Set 21: Code-Review+1","accounts_in_message":[],"_revision_number":21},{"id":"5f0c767d759f9090d1c73a3b1b5fc9ebcfeecfca","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-10-08 09:23:08.000000000","message":"Patch Set 21: Code-Review+2","accounts_in_message":[],"_revision_number":21},{"id":"fe49eff363e07a48716ba747f3525cf03d8ce5c8","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2019-10-11 14:12:05.000000000","message":"Patch Set 21: Code-Review+2\n\nLGTM, Thanks!","accounts_in_message":[],"_revision_number":21},{"id":"52d6adc113a116d5569af0197af1e9db7d73edd0","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2019-10-11 14:55:13.000000000","message":"Patch Set 21: Workflow+1","accounts_in_message":[],"_revision_number":21},{"id":"c82b4145ac9ceb92bdea25954bc783e95428d739","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 14:55:32.000000000","message":"Patch Set 21: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":21},{"id":"3d0f4ede71703cdaff2934559a06448b4980f80c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 20:21:15.000000000","message":"Patch Set 21: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d8a557e41624f8083ea76c15b833616 : SUCCESS in 9m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a9ef4ca93114b959e2ed9650cbc0b7a : SUCCESS in 8m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8801367469234802a714724d3f478fd6 : SUCCESS in 9m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d325ce4fd0e243fd8ed2dcf44db2708e : SUCCESS in 6m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cfda669a567f47b7b8f2609f14f71290 : SUCCESS in 9m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32c079b66fa140a88153295c94a7662f : SUCCESS in 18m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/537b141eceba45b590ac6e986d4a118b : SUCCESS in 14m 45s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/85ab98c9cf65431e880ae2735f558edf : SUCCESS in 7m 32s\n- ironic-tox-unit-with-driver-libs-python3 https://zuul.opendev.org/t/openstack/build/326d28da8a714676bc02681e104d75f0 : SUCCESS in 8m 56s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/0719426ada1347898de2312a2f413793 : SUCCESS in 1h 19m 03s\n- ironic-tempest-functional-python2 https://zuul.opendev.org/t/openstack/build/48baf72c5c7f40ccab845416342f393d : SUCCESS in 32m 07s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/8b436dd091ff4867974d574bab6828fb : SUCCESS in 33m 10s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/faba5db7877845d8b2a5414d233c595e : SUCCESS in 1h 43m 07s\n- ironic-tempest-ipa-partition-pxe_ipmitool-tinyipa-python3 https://zuul.opendev.org/t/openstack/build/e2a44ab22b804446a58dfdb0687c4f06 : SUCCESS in 46m 22s\n- ironic-tempest-ipa-partition-redfish-tinyipa https://zuul.opendev.org/t/openstack/build/330b35267a824fcbabf1a287558697c5 : SUCCESS in 54m 43s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ec2c96f8b78d4b47bcbb05fe4f9c5af9 : SUCCESS in 51m 49s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/24d582e3bec447c2b189996d39549e36 : SUCCESS in 1h 01m 59s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/1df63e3d34224f758ba83d3c4f9f3136 : SUCCESS in 51m 34s\n- ironic-tempest-ipa-partition-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/0e9649e78f894e158fffe8216e6d270c : SUCCESS in 49m 23s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/1a9696e40cf049618ee7994aecb5e4dc : SUCCESS in 39m 46s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/9cd841b0077c48b08d5428a67eb88e18 : SUCCESS in 57m 25s","accounts_in_message":[],"_revision_number":21},{"id":"547c7a59e921311494fc3adea7bfea96ac8b7433","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 20:21:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":21},{"id":"ee88d3913d078bb2961559ac4e6b295772bf4214","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 20:22:59.000000000","message":"Patch Set 21:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7463c167f64d4e99b34ea98600b3e451 : SUCCESS in 1m 15s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3791e0bd37894dd6b3819a5b7c721b2b : SUCCESS in 46s","accounts_in_message":[],"_revision_number":21},{"id":"67497e83aa761807c449c4587686f3f00b3fa3a6","author":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"date":"2019-10-14 12:19:02.000000000","message":"Patch Set 21: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 4568cc625040cda2dd28edce0f1acb3a595158a1","accounts_in_message":[],"_revision_number":21}],"current_revision_number":21,"current_revision":"e4f3b7b813007b90a579b7d3f6c4990a6affeb39","revisions":{"4da3c6a891f033580f80673c7f32ac08c09d49a8":{"kind":"REWORK","_number":1,"created":"2019-09-06 11:07:15.000000000","uploader":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"ref":"refs/changes/60/680660/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/60/680660/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/60/680660/1"}}},"commit":{"parents":[{"commit":"1f68fb9073d79f2471d2aa83a902b3e01cec6a90","subject":"Merge \"add table of available cleaning steps to documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1f68fb9073d79f2471d2aa83a902b3e01cec6a90"}]}],"author":{"name":"Rachit7194","email":"kapadiarachit007@gmail.com","date":"2019-09-06 11:01:07.000000000","tz":-240},"committer":{"name":"Rachit7194","email":"kapadiarachit007@gmail.com","date":"2019-09-06 11:07:14.000000000","tz":-240},"subject":"DRAC: Fix a bug for delete_configuration with multiple controllers","message":"DRAC: Fix a bug for delete_configuration with multiple controllers\n\nNeed to fix a bug where host has two controllers and executes\ndelete_configuration cleaning step in a such a way it should\nfirst create configuration job for the controller which supports realtime\nand then for the other controllers\n\nStory: 2006502\nTask : 36480\n\nChange-Id: I18aa79774c0b562bf1e8b973db3d7860a01367e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4da3c6a891f033580f80673c7f32ac08c09d49a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4da3c6a891f033580f80673c7f32ac08c09d49a8"}]},"branch":"refs/heads/master"},"92e3f1c12669ff55810e0b3af37352ddaec3b58d":{"kind":"REWORK","_number":2,"created":"2019-09-06 11:18:16.000000000","uploader":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"ref":"refs/changes/60/680660/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/60/680660/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/60/680660/2"}}},"commit":{"parents":[{"commit":"1f68fb9073d79f2471d2aa83a902b3e01cec6a90","subject":"Merge \"add table of available cleaning steps to documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1f68fb9073d79f2471d2aa83a902b3e01cec6a90"}]}],"author":{"name":"Rachit7194","email":"kapadiarachit007@gmail.com","date":"2019-09-06 11:01:07.000000000","tz":-240},"committer":{"name":"Rachit7194","email":"kapadiarachit007@gmail.com","date":"2019-09-06 11:18:02.000000000","tz":-240},"subject":"DRAC: Fix a bug for delete_configuration with multiple controllers","message":"DRAC: Fix a bug for delete_configuration with multiple controllers\n\nNeed to fix a bug where host has two controllers and executes\ndelete_configuration cleaning step in a such a way it should\nfirst create configuration job for the controller which supports realtime\nand then for the other controllers\n\nStory: 2006502\nTask : 36480\n\nChange-Id: I18aa79774c0b562bf1e8b973db3d7860a01367e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/92e3f1c12669ff55810e0b3af37352ddaec3b58d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/92e3f1c12669ff55810e0b3af37352ddaec3b58d"}]},"branch":"refs/heads/master"},"306da9be088d7882720bab558a758ab6a0168605":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-09-06 14:17:24.000000000","uploader":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"ref":"refs/changes/60/680660/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/60/680660/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/60/680660/3"}}},"commit":{"parents":[{"commit":"1f68fb9073d79f2471d2aa83a902b3e01cec6a90","subject":"Merge \"add table of available cleaning steps to documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1f68fb9073d79f2471d2aa83a902b3e01cec6a90"}]}],"author":{"name":"Rachit7194","email":"kapadiarachit007@gmail.com","date":"2019-09-06 11:01:07.000000000","tz":-240},"committer":{"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","date":"2019-09-06 14:17:06.000000000","tz":0},"subject":"DRAC: Fix a bug for delete_configuration with multiple controllers","message":"DRAC: Fix a bug for delete_configuration with multiple controllers\n\nNeed to fix a bug where host has multiple controllers and executes\ndelete_configuration cleaning step in a such a way it should\nfirst create configuration job for the controller which supports realtime\nand then for the other controllers\n\nStory: 2006502\nTask : 36480\n\nChange-Id: I18aa79774c0b562bf1e8b973db3d7860a01367e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/306da9be088d7882720bab558a758ab6a0168605"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/306da9be088d7882720bab558a758ab6a0168605"}]},"branch":"refs/heads/master"},"ee6f7f5c8c0e9ce79238136b2285faa43356324d":{"kind":"REWORK","_number":4,"created":"2019-09-06 14:47:22.000000000","uploader":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"ref":"refs/changes/60/680660/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/60/680660/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/60/680660/4"}}},"commit":{"parents":[{"commit":"1f68fb9073d79f2471d2aa83a902b3e01cec6a90","subject":"Merge \"add table of available cleaning steps to documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1f68fb9073d79f2471d2aa83a902b3e01cec6a90"}]}],"author":{"name":"Rachit7194","email":"kapadiarachit007@gmail.com","date":"2019-09-06 11:01:07.000000000","tz":-240},"committer":{"name":"Rachit7194","email":"kapadiarachit007@gmail.com","date":"2019-09-06 14:47:06.000000000","tz":-240},"subject":"DRAC: Fix a bug for delete_configuration with multiple controllers","message":"DRAC: Fix a bug for delete_configuration with multiple controllers\n\nNeed to fix a bug where host has two controllers and executes\ndelete_configuration cleaning step in a such a way it should\nfirst create configuration job for the controller which supports realtime\nand then for the other controllers\n\nStory: 2006502\nTask : 36480\n\nChange-Id: I18aa79774c0b562bf1e8b973db3d7860a01367e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/ee6f7f5c8c0e9ce79238136b2285faa43356324d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/ee6f7f5c8c0e9ce79238136b2285faa43356324d"}]},"branch":"refs/heads/master"},"3d191aef26a6677be427bc77d9f96dd795707568":{"kind":"REWORK","_number":5,"created":"2019-09-10 07:04:20.000000000","uploader":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"ref":"refs/changes/60/680660/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/60/680660/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/60/680660/5"}}},"commit":{"parents":[{"commit":"1f68fb9073d79f2471d2aa83a902b3e01cec6a90","subject":"Merge \"add table of available cleaning steps to documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1f68fb9073d79f2471d2aa83a902b3e01cec6a90"}]}],"author":{"name":"Rachit7194","email":"kapadiarachit007@gmail.com","date":"2019-09-06 11:01:07.000000000","tz":-240},"committer":{"name":"Rachit7194","email":"kapadiarachit007@gmail.com","date":"2019-09-10 07:03:25.000000000","tz":-240},"subject":"DRAC: Fix a bug for delete_configuration with multiple controllers","message":"DRAC: Fix a bug for delete_configuration with multiple controllers\n\nNeed to fix a bug where host has two controllers and executes\ndelete_configuration cleaning step in a such a way it should\nfirst create configuration job for the controller which supports realtime\nand then for the other controllers\n\nStory: #2006502\nTask:  #36480\n\nChange-Id: I18aa79774c0b562bf1e8b973db3d7860a01367e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/3d191aef26a6677be427bc77d9f96dd795707568"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/3d191aef26a6677be427bc77d9f96dd795707568"}]},"branch":"refs/heads/master"},"f389a8701aa80f6261fbd0265699570a5ede60ec":{"kind":"REWORK","_number":6,"created":"2019-09-10 07:19:23.000000000","uploader":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"ref":"refs/changes/60/680660/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/60/680660/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/60/680660/6"}}},"commit":{"parents":[{"commit":"1f68fb9073d79f2471d2aa83a902b3e01cec6a90","subject":"Merge \"add table of available cleaning steps to documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1f68fb9073d79f2471d2aa83a902b3e01cec6a90"}]}],"author":{"name":"Rachit7194","email":"kapadiarachit007@gmail.com","date":"2019-09-06 11:01:07.000000000","tz":-240},"committer":{"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","date":"2019-09-10 07:19:15.000000000","tz":0},"subject":"DRAC: Fix a bug for delete_configuration with multiple controllers","message":"DRAC: Fix a bug for delete_configuration with multiple controllers\n\nNeed to fix a bug where baremetal server has multiple controllers and\nexecutes delete_configuration clean step in a such a way it should\nfirst create configuration job for the RAID controller which supports\nrealtime and then for the other controllers\n\nStory:#2006502\nTask: #36480\n\nChange-Id: I18aa79774c0b562bf1e8b973db3d7860a01367e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/f389a8701aa80f6261fbd0265699570a5ede60ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/f389a8701aa80f6261fbd0265699570a5ede60ec"}]},"branch":"refs/heads/master"},"6e559ec16be16cf526de2358f826b7d3fd39eb83":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2019-09-10 08:22:36.000000000","uploader":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"ref":"refs/changes/60/680660/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/60/680660/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/60/680660/7"}}},"commit":{"parents":[{"commit":"1f68fb9073d79f2471d2aa83a902b3e01cec6a90","subject":"Merge \"add table of available cleaning steps to documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1f68fb9073d79f2471d2aa83a902b3e01cec6a90"}]}],"author":{"name":"Rachit7194","email":"kapadiarachit007@gmail.com","date":"2019-09-06 11:01:07.000000000","tz":-240},"committer":{"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","date":"2019-09-10 08:22:27.000000000","tz":0},"subject":"DRAC: Fix a bug for delete_configuration with multiple controllers","message":"DRAC: Fix a bug for delete_configuration with multiple controllers\n\nNeed to fix a bug where baremetal server has multiple controllers and\nexecutes delete_configuration clean step in a such a way it should\nfirst create configuration job for the RAID controller which supports\nrealtime and then for the other controllers\n\nStory: #2006502\nTask: #36480\n\nChange-Id: I18aa79774c0b562bf1e8b973db3d7860a01367e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/6e559ec16be16cf526de2358f826b7d3fd39eb83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/6e559ec16be16cf526de2358f826b7d3fd39eb83"}]},"branch":"refs/heads/master"},"14d3a07c99228a8867b5702dd6ba8b437b0c1134":{"kind":"REWORK","_number":8,"created":"2019-09-12 18:26:24.000000000","uploader":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"ref":"refs/changes/60/680660/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/60/680660/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/60/680660/8"}}},"commit":{"parents":[{"commit":"1f68fb9073d79f2471d2aa83a902b3e01cec6a90","subject":"Merge \"add table of available cleaning steps to documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1f68fb9073d79f2471d2aa83a902b3e01cec6a90"}]}],"author":{"name":"Rachit7194","email":"kapadiarachit007@gmail.com","date":"2019-09-06 11:01:07.000000000","tz":-240},"committer":{"name":"Rachit7194","email":"kapadiarachit007@gmail.com","date":"2019-09-12 18:26:15.000000000","tz":-240},"subject":"DRAC: Fix a bug for delete_configuration with multiple controllers","message":"DRAC: Fix a bug for delete_configuration with multiple controllers\n\nNeed to fix a bug where host has two controllers and executes\ndelete_configuration cleaning step in a such a way it should\nfirst create configuration job for the controller which supports realtime\nand then for the other controllers\n\nStory: #2006502\nTask:  #36480\n\nChange-Id: I18aa79774c0b562bf1e8b973db3d7860a01367e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/14d3a07c99228a8867b5702dd6ba8b437b0c1134"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/14d3a07c99228a8867b5702dd6ba8b437b0c1134"}]},"branch":"refs/heads/master"},"1f57b71c02d805ffd2fa4a3e86e8c9879167eb57":{"kind":"REWORK","_number":9,"created":"2019-09-12 18:32:01.000000000","uploader":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"ref":"refs/changes/60/680660/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/60/680660/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/60/680660/9"}}},"commit":{"parents":[{"commit":"1f68fb9073d79f2471d2aa83a902b3e01cec6a90","subject":"Merge \"add table of available cleaning steps to documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1f68fb9073d79f2471d2aa83a902b3e01cec6a90"}]}],"author":{"name":"Rachit7194","email":"kapadiarachit007@gmail.com","date":"2019-09-06 11:01:07.000000000","tz":-240},"committer":{"name":"Rachit7194","email":"kapadiarachit007@gmail.com","date":"2019-09-12 18:30:24.000000000","tz":-240},"subject":"DRAC: Fix a bug for delete_configuration with multiple controllers","message":"DRAC: Fix a bug for delete_configuration with multiple controllers\n\nNeed to fix a bug where host has two controllers and executes\ndelete_configuration cleaning step in a such a way it should\nfirst create configuration job for the controller which supports\nrealtime and then for the other controllers.\n\nStory: #2006502\nTask: #36480\n\nChange-Id: I18aa79774c0b562bf1e8b973db3d7860a01367e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1f57b71c02d805ffd2fa4a3e86e8c9879167eb57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1f57b71c02d805ffd2fa4a3e86e8c9879167eb57"}]},"branch":"refs/heads/master"},"743fce848e049e322d24ddcbdad792ab81a6e52f":{"kind":"REWORK","_number":10,"created":"2019-09-13 00:44:26.000000000","uploader":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"ref":"refs/changes/60/680660/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/60/680660/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/60/680660/10"}}},"commit":{"parents":[{"commit":"1f68fb9073d79f2471d2aa83a902b3e01cec6a90","subject":"Merge \"add table of available cleaning steps to documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1f68fb9073d79f2471d2aa83a902b3e01cec6a90"}]}],"author":{"name":"Rachit7194","email":"kapadiarachit007@gmail.com","date":"2019-09-06 11:01:07.000000000","tz":-240},"committer":{"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","date":"2019-09-13 00:43:55.000000000","tz":0},"subject":"DRAC: Fix a bug for delete_configuration with multiple controllers","message":"DRAC: Fix a bug for delete_configuration with multiple controllers\n\nNeed to fix a bug where host has two controllers and executes\ndelete_configuration cleaning step in a such a way it should\nfirst create configuration job for the controller which supports\nrealtime and then for the other controllers.\n\nStory: #2006502\nTask: #36480\n\nChange-Id: I18aa79774c0b562bf1e8b973db3d7860a01367e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/743fce848e049e322d24ddcbdad792ab81a6e52f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/743fce848e049e322d24ddcbdad792ab81a6e52f"}]},"branch":"refs/heads/master"},"ee3d902302225a112c89e285e33467fbef695f86":{"kind":"REWORK","_number":11,"created":"2019-09-18 10:56:28.000000000","uploader":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"ref":"refs/changes/60/680660/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/60/680660/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/60/680660/11"}}},"commit":{"parents":[{"commit":"1f68fb9073d79f2471d2aa83a902b3e01cec6a90","subject":"Merge \"add table of available cleaning steps to documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1f68fb9073d79f2471d2aa83a902b3e01cec6a90"}]}],"author":{"name":"Rachit7194","email":"kapadiarachit007@gmail.com","date":"2019-09-06 11:01:07.000000000","tz":-240},"committer":{"name":"Rachit7194","email":"kapadiarachit007@gmail.com","date":"2019-09-18 10:56:12.000000000","tz":-240},"subject":"DRAC: Fix a bug for delete_configuration with multiple controllers","message":"DRAC: Fix a bug for delete_configuration with multiple controllers\n\nNeed to fix a bug where host has two controllers and executes\ndelete_configuration cleaning step in a such a way it should\nfirst create configuration job for the controller which supports\nrealtime and then for the other controllers.\n\nStory: #2006502\nTask: #36480\n\nChange-Id: I18aa79774c0b562bf1e8b973db3d7860a01367e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/ee3d902302225a112c89e285e33467fbef695f86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/ee3d902302225a112c89e285e33467fbef695f86"}]},"branch":"refs/heads/master"},"d859fb16a33ae75d3cd028e43c250e0d532e0c32":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2019-09-18 12:36:21.000000000","uploader":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"ref":"refs/changes/60/680660/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/60/680660/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/60/680660/12"}}},"commit":{"parents":[{"commit":"1f68fb9073d79f2471d2aa83a902b3e01cec6a90","subject":"Merge \"add table of available cleaning steps to documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1f68fb9073d79f2471d2aa83a902b3e01cec6a90"}]}],"author":{"name":"Rachit7194","email":"kapadiarachit007@gmail.com","date":"2019-09-06 11:01:07.000000000","tz":-240},"committer":{"name":"Rachit7194","email":"kapadiarachit007@gmail.com","date":"2019-09-18 10:57:35.000000000","tz":-240},"subject":"DRAC: Fix a bug for delete_config with multiple controllers","message":"DRAC: Fix a bug for delete_config with multiple controllers\n\nNeed to fix a bug where host has two controllers and executes\ndelete_configuration cleaning step in a such a way it should\nfirst create configuration job for the controller which supports\nrealtime and then for the other controllers.\n\nStory: #2006502\nTask: #36480\n\nChange-Id: I18aa79774c0b562bf1e8b973db3d7860a01367e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/d859fb16a33ae75d3cd028e43c250e0d532e0c32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/d859fb16a33ae75d3cd028e43c250e0d532e0c32"}]},"branch":"refs/heads/master"},"d8ae19b5f3ea240b3964eb6ecca3e5ea672db2ab":{"kind":"REWORK","_number":13,"created":"2019-09-19 19:05:25.000000000","uploader":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"ref":"refs/changes/60/680660/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/60/680660/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/60/680660/13"}}},"commit":{"parents":[{"commit":"1f68fb9073d79f2471d2aa83a902b3e01cec6a90","subject":"Merge \"add table of available cleaning steps to documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1f68fb9073d79f2471d2aa83a902b3e01cec6a90"}]}],"author":{"name":"Rachit7194","email":"kapadiarachit007@gmail.com","date":"2019-09-06 11:01:07.000000000","tz":-240},"committer":{"name":"Rachit7194","email":"kapadiarachit007@gmail.com","date":"2019-09-19 19:03:15.000000000","tz":-240},"subject":"DRAC: Fix a bug for delete_config with multiple controllers","message":"DRAC: Fix a bug for delete_config with multiple controllers\n\nThis fix a bug where a race condition can occur on a host that\nhas a mix of controllers where some supports realtime mode and\nsome do not. The approach is to use only realtime mode if all\ncontrollers supports realtime.This removes the race condition.\n\nStory: #2006502\nTask: #36480\n\nChange-Id: I18aa79774c0b562bf1e8b973db3d7860a01367e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/d8ae19b5f3ea240b3964eb6ecca3e5ea672db2ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/d8ae19b5f3ea240b3964eb6ecca3e5ea672db2ab"}]},"branch":"refs/heads/master"},"1e224d655eb3108f7d558a393ca42d3cdd4dad1f":{"kind":"REWORK","_number":14,"created":"2019-09-23 12:59:06.000000000","uploader":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"ref":"refs/changes/60/680660/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/60/680660/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/60/680660/14"}}},"commit":{"parents":[{"commit":"1f68fb9073d79f2471d2aa83a902b3e01cec6a90","subject":"Merge \"add table of available cleaning steps to documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1f68fb9073d79f2471d2aa83a902b3e01cec6a90"}]}],"author":{"name":"Rachit7194","email":"kapadiarachit007@gmail.com","date":"2019-09-06 11:01:07.000000000","tz":-240},"committer":{"name":"Rachit7194","email":"kapadiarachit007@gmail.com","date":"2019-09-23 12:58:49.000000000","tz":-240},"subject":"DRAC: Fix a bug for delete_config with multiple controllers","message":"DRAC: Fix a bug for delete_config with multiple controllers\n\nThis fix a bug where a race condition can occur on a host that\nhas a mix of controllers where some supports realtime mode and\nsome do not. The approach is to use only realtime mode if all\ncontrollers supports realtime.This removes the race condition.\n\nStory: #2006502\nTask: #36480\n\nChange-Id: I18aa79774c0b562bf1e8b973db3d7860a01367e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1e224d655eb3108f7d558a393ca42d3cdd4dad1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1e224d655eb3108f7d558a393ca42d3cdd4dad1f"}]},"branch":"refs/heads/master"},"72916a08cf455f6ea9bf80914c8e3f1b4696cbf2":{"kind":"REWORK","_number":15,"created":"2019-09-24 10:34:27.000000000","uploader":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"ref":"refs/changes/60/680660/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/60/680660/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/60/680660/15"}}},"commit":{"parents":[{"commit":"1f68fb9073d79f2471d2aa83a902b3e01cec6a90","subject":"Merge \"add table of available cleaning steps to documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1f68fb9073d79f2471d2aa83a902b3e01cec6a90"}]}],"author":{"name":"Rachit7194","email":"kapadiarachit007@gmail.com","date":"2019-09-06 11:01:07.000000000","tz":-240},"committer":{"name":"Rachit7194","email":"kapadiarachit007@gmail.com","date":"2019-09-24 10:33:55.000000000","tz":-240},"subject":"DRAC: Fix a bug for delete_config with multiple controllers","message":"DRAC: Fix a bug for delete_config with multiple controllers\n\nThis fix a bug where a race condition can occur on a host that\nhas a mix of controllers where some supports realtime mode and\nsome do not. The approach is to use only realtime mode if all\ncontrollers supports realtime. This removes the race condition.\n\nStory: #2006502\nTask: #36480\n\nChange-Id: I18aa79774c0b562bf1e8b973db3d7860a01367e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/72916a08cf455f6ea9bf80914c8e3f1b4696cbf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/72916a08cf455f6ea9bf80914c8e3f1b4696cbf2"}]},"branch":"refs/heads/master"},"88e3c987e5270a58b661af61ecf61379bbde6b8c":{"kind":"REWORK","_number":16,"created":"2019-09-25 12:53:29.000000000","uploader":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"ref":"refs/changes/60/680660/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/60/680660/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/60/680660/16"}}},"commit":{"parents":[{"commit":"1f68fb9073d79f2471d2aa83a902b3e01cec6a90","subject":"Merge \"add table of available cleaning steps to documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1f68fb9073d79f2471d2aa83a902b3e01cec6a90"}]}],"author":{"name":"Rachit7194","email":"kapadiarachit007@gmail.com","date":"2019-09-06 11:01:07.000000000","tz":-240},"committer":{"name":"Rachit7194","email":"kapadiarachit007@gmail.com","date":"2019-09-25 12:53:12.000000000","tz":-240},"subject":"DRAC: Fix a bug for delete_config with multiple controllers","message":"DRAC: Fix a bug for delete_config with multiple controllers\n\nThis fix a bug where a race condition can occur on a host that\nhas a mix of controllers where some supports realtime mode and\nsome do not. The approach is to use only realtime mode if all\ncontrollers supports realtime. This removes the race condition.\n\nStory: #2006502\nTask: #36480\n\nChange-Id: I18aa79774c0b562bf1e8b973db3d7860a01367e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/88e3c987e5270a58b661af61ecf61379bbde6b8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/88e3c987e5270a58b661af61ecf61379bbde6b8c"}]},"branch":"refs/heads/master"},"9e7a8e71241bb56bd0b5d8ee154fe3f895df716f":{"kind":"REWORK","_number":17,"created":"2019-10-01 13:23:02.000000000","uploader":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"ref":"refs/changes/60/680660/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/60/680660/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/60/680660/17"}}},"commit":{"parents":[{"commit":"1f68fb9073d79f2471d2aa83a902b3e01cec6a90","subject":"Merge \"add table of available cleaning steps to documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1f68fb9073d79f2471d2aa83a902b3e01cec6a90"}]}],"author":{"name":"Rachit7194","email":"kapadiarachit007@gmail.com","date":"2019-09-06 11:01:07.000000000","tz":-240},"committer":{"name":"Rachit7194","email":"kapadiarachit007@gmail.com","date":"2019-10-01 13:21:54.000000000","tz":-240},"subject":"DRAC: Fix a bug for delete_config with multiple controllers","message":"DRAC: Fix a bug for delete_config with multiple controllers\n\nThis fix a bug where a race condition can occur on a host that\nhas a mix of controllers where some supports realtime mode and\nsome do not. The approach is to use only realtime mode if all\ncontrollers supports realtime. This removes the race condition.\n\nStory: #2006502\nTask: #36480\n\nChange-Id: I18aa79774c0b562bf1e8b973db3d7860a01367e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/9e7a8e71241bb56bd0b5d8ee154fe3f895df716f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/9e7a8e71241bb56bd0b5d8ee154fe3f895df716f"}]},"branch":"refs/heads/master"},"aab62b49f3ec48334afd926370f6be7168c6455d":{"kind":"REWORK","_number":18,"created":"2019-10-02 12:05:52.000000000","uploader":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"ref":"refs/changes/60/680660/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/60/680660/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/60/680660/18"}}},"commit":{"parents":[{"commit":"2408987175ce1d2516348cfef1281d42b41e25ae","subject":"Merge \"DRAC: Fix a bug for job creation when only required\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/2408987175ce1d2516348cfef1281d42b41e25ae"}]}],"author":{"name":"Rachit7194","email":"kapadiarachit007@gmail.com","date":"2019-09-06 11:01:07.000000000","tz":-240},"committer":{"name":"Rachit7194","email":"kapadiarachit007@gmail.com","date":"2019-10-02 12:05:12.000000000","tz":-240},"subject":"DRAC: Fix a bug for delete_config with multiple controllers","message":"DRAC: Fix a bug for delete_config with multiple controllers\n\nThis fix a bug where a race condition can occur on a host that\nhas a mix of controllers where some supports realtime mode and\nsome do not. The approach is to use only realtime mode if all\ncontrollers supports realtime. This removes the race condition.\n\nStory: #2006502\nTask: #36480\n\nChange-Id: I18aa79774c0b562bf1e8b973db3d7860a01367e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/aab62b49f3ec48334afd926370f6be7168c6455d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/aab62b49f3ec48334afd926370f6be7168c6455d"}]},"branch":"refs/heads/master"},"fdae89a6cbb6b423af5bd6ff52eb3b59a3aae769":{"kind":"REWORK","_number":19,"created":"2019-10-02 13:06:30.000000000","uploader":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"ref":"refs/changes/60/680660/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/60/680660/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/60/680660/19"}}},"commit":{"parents":[{"commit":"2408987175ce1d2516348cfef1281d42b41e25ae","subject":"Merge \"DRAC: Fix a bug for job creation when only required\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/2408987175ce1d2516348cfef1281d42b41e25ae"}]}],"author":{"name":"Rachit7194","email":"kapadiarachit007@gmail.com","date":"2019-09-06 11:01:07.000000000","tz":-240},"committer":{"name":"Rachit7194","email":"kapadiarachit007@gmail.com","date":"2019-10-02 13:05:11.000000000","tz":-240},"subject":"DRAC: Fix a bug for delete_config with multiple controllers","message":"DRAC: Fix a bug for delete_config with multiple controllers\n\nThis fix a bug where a race condition can occur on a host that\nhas a mix of controllers where some supports realtime mode and\nsome do not. The approach is to use only realtime mode if all\ncontrollers supports realtime. This removes the race condition.\n\nStory: #2006502\nTask: #36480\n\nChange-Id: I18aa79774c0b562bf1e8b973db3d7860a01367e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/fdae89a6cbb6b423af5bd6ff52eb3b59a3aae769"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/fdae89a6cbb6b423af5bd6ff52eb3b59a3aae769"}]},"branch":"refs/heads/master"},"76c1cab3728763059ab546393d3360fbb3714dcf":{"kind":"REWORK","_number":20,"created":"2019-10-02 13:16:10.000000000","uploader":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"ref":"refs/changes/60/680660/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/60/680660/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/60/680660/20"}}},"commit":{"parents":[{"commit":"2408987175ce1d2516348cfef1281d42b41e25ae","subject":"Merge \"DRAC: Fix a bug for job creation when only required\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/2408987175ce1d2516348cfef1281d42b41e25ae"}]}],"author":{"name":"Rachit7194","email":"kapadiarachit007@gmail.com","date":"2019-09-06 11:01:07.000000000","tz":-240},"committer":{"name":"Rachit7194","email":"kapadiarachit007@gmail.com","date":"2019-10-02 13:16:00.000000000","tz":-240},"subject":"DRAC: Fix a bug for delete_config with multiple controllers","message":"DRAC: Fix a bug for delete_config with multiple controllers\n\nThis fix a bug where a race condition can occur on a host that\nhas a mix of controllers where some supports realtime mode and\nsome do not. The approach is to use only realtime mode if all\ncontrollers supports realtime. This removes the race condition.\n\nStory: #2006502\nTask: #36480\n\nChange-Id: I18aa79774c0b562bf1e8b973db3d7860a01367e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/76c1cab3728763059ab546393d3360fbb3714dcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/76c1cab3728763059ab546393d3360fbb3714dcf"}]},"branch":"refs/heads/master"},"e4f3b7b813007b90a579b7d3f6c4990a6affeb39":{"kind":"REWORK","_number":21,"created":"2019-10-03 13:38:19.000000000","uploader":{"_account_id":29405,"name":"Rachit Kapadia","email":"kapadiarachit007@gmail.com","username":"rachit7194"},"ref":"refs/changes/60/680660/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/60/680660/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/60/680660/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/60/680660/21"}}},"commit":{"parents":[{"commit":"2408987175ce1d2516348cfef1281d42b41e25ae","subject":"Merge \"DRAC: Fix a bug for job creation when only required\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/2408987175ce1d2516348cfef1281d42b41e25ae"}]}],"author":{"name":"Rachit7194","email":"kapadiarachit007@gmail.com","date":"2019-09-06 11:01:07.000000000","tz":-240},"committer":{"name":"Rachit7194","email":"kapadiarachit007@gmail.com","date":"2019-10-03 13:37:53.000000000","tz":-240},"subject":"DRAC: Fix a bug for delete_config with multiple controllers","message":"DRAC: Fix a bug for delete_config with multiple controllers\n\nThis fix a bug where a race condition can occur on a host that\nhas a mix of controllers where some supports realtime mode and\nsome do not. The approach is to use only realtime mode if all\ncontrollers supports realtime. This removes the race condition.\n\nStory: #2006502\nTask: #36480\n\nChange-Id: I18aa79774c0b562bf1e8b973db3d7860a01367e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/e4f3b7b813007b90a579b7d3f6c4990a6affeb39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/e4f3b7b813007b90a579b7d3f6c4990a6affeb39"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
