)]}'
{"id":"openstack%2Fironic~689554","triplet_id":"openstack%2Fironic~stable%2Ftrain~I0bf38321b0135ea17439d35081b5727fd4d827c8","project":"openstack/ironic","branch":"stable/train","topic":"another-fix-to-console-stable/train","hashtags":[],"change_id":"I0bf38321b0135ea17439d35081b5727fd4d827c8","subject":"Read in non-blocking fashion when starting console","status":"MERGED","created":"2019-10-18 17:37:55.000000000","updated":"2019-10-21 10:30:56.000000000","submitted":"2019-10-21 10:28:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689554-1571653739419-70f2cce9","meta_rev_id":"d80e68a0ec23a8668a4937c5f872aa23ad9361f6","_number":689554,"virtual_id_number":689554,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},{"value":0,"date":"2019-10-20 18:20:45.000000000","_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"value":0,"date":"2019-10-18 23:01:54.000000000","_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-21 10:28:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},{"value":0,"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":2,"date":"2019-10-21 08:51:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"date":"2019-10-18 17:37:55.000000000","_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},{"value":0,"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":1,"date":"2019-10-21 08:51:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},{"value":0,"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-18 17:37:55.000000000","updated_by":{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},"reviewer":{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},"state":"REVIEWER"},{"updated":"2019-10-18 23:01:54.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"REVIEWER"},{"updated":"2019-10-20 18:20:45.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"REVIEWER"},{"updated":"2019-10-21 08:51:34.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2019-10-21 10:28:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e05ca31f5f214a8cdd112bab28faafe5af9faa9a","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2019-10-18 17:37:55.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"af3cbb97a92ec9880316c1b56d4e11e0d42c0903","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2019-10-18 17:38:20.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 899a0d3784284fb6c2b09e61fe2293eaaa7ac93f","accounts_in_message":[],"_revision_number":1},{"id":"ffa35474eeabc03453daacddae9669696e441431","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 19:38:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e0ab6e8346a4eaebb58b036a1f989f3 : SUCCESS in 7m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b9c37e81f884d08bf8ccae14c5cf9ef : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da08649d91fb49709df6d5b3c8305256 : SUCCESS in 5m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/429e0cecbfd24f99b15e433b15766a55 : SUCCESS in 8m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b455a85ccb2499084c69f9dd2a73fa7 : SUCCESS in 6m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/72fb5a22be5244d6a323b6b7db837b2f : SUCCESS in 7m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faeb4d31104949bcafd688966c2617f0 : SUCCESS in 18m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0fcec6accdd043be9d046d46b471a2fe : SUCCESS in 11m 03s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/517c1d8648e74696b7edaec589036463 : SUCCESS in 8m 13s\n- ironic-tox-unit-with-driver-libs-python3 https://zuul.opendev.org/t/openstack/build/7e5d77d7ec354e44b49fa77e47e2522e : SUCCESS in 7m 41s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/7f5e32fcfda9470fa825a24ebcc72098 : SUCCESS in 1h 14m 45s\n- ironic-tempest-functional-python2 https://zuul.opendev.org/t/openstack/build/cee5ef4c1b1941f89ba32aaca6253ba6 : SUCCESS in 24m 24s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/698ca61f02394cb5890bf93018973812 : SUCCESS in 30m 10s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/cb2953ebd99a430dadb8aeb5bb349524 : SUCCESS in 1h 40m 27s\n- ironic-grenade-dsvm-multinode-multitenant https://zuul.opendev.org/t/openstack/build/b5aaf602378645fd953f4640055e0d18 : FAILURE in 1h 48m 12s (non-voting)\n- ironic-tempest-ipa-partition-pxe_ipmitool-tinyipa-python3 https://zuul.opendev.org/t/openstack/build/43195e8085ac473493b9e47906ba7069 : SUCCESS in 1h 01m 07s\n- ironic-tempest-ipa-partition-redfish-tinyipa https://zuul.opendev.org/t/openstack/build/dc06b45cee404860ab7e88e7e5cd7ac3 : SUCCESS in 58m 06s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c901a3959beb448690c737934e36be7d : SUCCESS in 44m 21s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/bcf0020f6d37450691aac97393766882 : FAILURE in 1h 58m 08s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/cb7a76bb6d0448cab50c16aa41844db4 : SUCCESS in 1h 07m 09s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/32af624b239f466b9765f2b78b5001e4 : SUCCESS in 1h 07m 12s\n- ironic-tempest-ipa-partition-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/2560f96a726a493b958ba6852b224dc6 : SUCCESS in 57m 18s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/3085234abe4449ad9b778f791f416fc0 : SUCCESS in 37m 13s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/0bb2bf7b77ad4b90a87cecbd5737bd3b : SUCCESS in 1h 08m 16s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/bd7106d2a3c34085adffd928d01b50ea : SUCCESS in 5m 35s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-pxe_snmp-tinyipa https://zuul.opendev.org/t/openstack/build/0c0ad5733a1644b0accff50735b18217 : SUCCESS in 56m 42s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/b41e465d79e049bb9eea2799cd08e75e : SUCCESS in 1h 08m 06s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/663364c75b1d4d85962776e4ea662328 : SUCCESS in 20m 35s (non-voting)\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/4c85d083eb544da7b9ec8e56d060a53f : SUCCESS in 1h 29m 24s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/b3544007cac64c1fa025f75e459f9fa0 : SUCCESS in 1h 04m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e616bef4c6ddb7ea48524508329020eb7ada7d3f","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2019-10-18 23:01:54.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-DSS9000-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-DSS9000-tempest-dsvm-ironic-idrac/2019-10-18:12:38/689554/1/ : SUCCESS in 47m 20s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac/2019-10-18:12:38/689554/1/ : FAILURE in 37m 12s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac/2019-10-18:12:38/689554/1/ : SUCCESS in 47m 46s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac/2019-10-18:12:38/689554/1/ : SUCCESS in 44m 37s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish/2019-10-18:12:38/689554/1/ : SUCCESS in 42m 46s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish/2019-10-18:12:38/689554/1/ : SUCCESS in 1h 09m 33s\n- dellemc-DSS9000-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-DSS9000-tempest-dsvm-ironic-redfish/2019-10-18:12:38/689554/1/ : SUCCESS in 45m 20s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi/2019-10-18:12:38/689554/1/ : SUCCESS in 42m 12s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi/2019-10-18:12:38/689554/1/ : SUCCESS in 46m 30s","accounts_in_message":[],"_revision_number":1},{"id":"26d0575a3810d42aaf23f417330bf94bfdca8a5c","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2019-10-20 18:20:45.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/54/689554/1/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/29bfe73/29bfe73.html : TIMED_OUT in 3h 09m 45s\n- hpe-gen8-dsvm-ironic-standalone-wholedisk-bios-ilo-virtual-media-iscsi-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/54/689554/1/check/hpe-gen8-dsvm-ironic-standalone-wholedisk-bios-ilo-virtual-media-iscsi-python3/671aab4/671aab4.html : TIMED_OUT in 3h 09m 42s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python2 https://hpproliant.github.io/hpeproliant.github.io/logs/54/689554/1/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python2/e0b7d1f/e0b7d1f.html : TIMED_OUT in 3h 09m 44s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/54/689554/1/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/a89f568/a89f568.html : TIMED_OUT in 3h 09m 46s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/54/689554/1/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/4daab38/4daab38.html : TIMED_OUT in 3h 09m 40s","accounts_in_message":[],"_revision_number":1},{"id":"35f5980c1d256882d82a275e81a5813e917d200b","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-10-21 08:51:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f7bda6f2dee5c6d842b48fc8afafd73749f1b855","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 08:51:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0f0eaed21fec0565c395f0b5c82b5f1bf38016cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 10:28:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4dcf45148fc7406a8e37bc7a3d937476 : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19356064ef2b4e68bf14dd4de6738f34 : SUCCESS in 5m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4cd3335c6f33402496cdd9e9ccd8c031 : SUCCESS in 8m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c34cf11a89a44a498704afabafe08738 : SUCCESS in 6m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3219a8b3de1b49c5a647b0f6c7130697 : SUCCESS in 7m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da0601c9c4ec4b88a9c67e18dfc92496 : SUCCESS in 16m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5e374f6b06f41f9bdf4644a6e3d532e : SUCCESS in 9m 37s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/b0920af45a3346b29fb0e9eadc48a1b6 : SUCCESS in 7m 58s\n- ironic-tox-unit-with-driver-libs-python3 https://zuul.opendev.org/t/openstack/build/99e448c0567844c38ad8f98099bbe91f : SUCCESS in 7m 04s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/64516018daa44d81ae75551e2a7e810c : SUCCESS in 1h 10m 21s\n- ironic-tempest-functional-python2 https://zuul.opendev.org/t/openstack/build/58ad9b2d734743cbb063f639a1d8aa4c : SUCCESS in 27m 07s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/c1d9c991038b4b2aa2cbe8e6f888b78f : SUCCESS in 28m 16s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/d82ebf4c3fe64bc9b5e3bc017d41f0e6 : SUCCESS in 1h 31m 37s\n- ironic-tempest-ipa-partition-pxe_ipmitool-tinyipa-python3 https://zuul.opendev.org/t/openstack/build/e60bc1245bdf4806bed0e7d70fcd701c : SUCCESS in 51m 58s\n- ironic-tempest-ipa-partition-redfish-tinyipa https://zuul.opendev.org/t/openstack/build/c898a5d480304b568bbd579d2fbea612 : SUCCESS in 49m 34s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e5ff5a6c0d4b4c2a8e694253e7a5d8bb : SUCCESS in 43m 18s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e4ff9e92451d41399c2840a238974443 : SUCCESS in 52m 16s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/d0645275cdc9419d88e1e1bb7fc89b47 : SUCCESS in 46m 28s\n- ironic-tempest-ipa-partition-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/63a598c4b74946e48521ccfc7528c73a : SUCCESS in 54m 13s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/b9413e589c144736b07eb3afced39785 : SUCCESS in 39m 17s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/fb7714f5e09545f497546b810d578d36 : SUCCESS in 58m 06s","accounts_in_message":[],"_revision_number":1},{"id":"77f1da0e929643af48e57579a205c3c789d61183","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 10:28:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"519b56359a9fb76cc33bb3479dd0868e087775ba","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 10:30:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a81482b5ec243e6a19a1072ecc32c9d : SUCCESS in 1m 32s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6231f1fa897b4a71b2e091b07eae2631 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"680c8b7745499b916c5f7d1614b61df738f40df2","revisions":{"680c8b7745499b916c5f7d1614b61df738f40df2":{"kind":"REWORK","_number":1,"created":"2019-10-18 17:37:55.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/54/689554/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/54/689554/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/54/689554/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/54/689554/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/54/689554/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/54/689554/1"}}},"commit":{"parents":[{"commit":"ba32a5f4bfbd4b7e3ff1ee652b973f6469b57c79","subject":"Fix drive sensors collection in `redfish` mgmt interface","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/ba32a5f4bfbd4b7e3ff1ee652b973f6469b57c79"}]}],"author":{"name":"Vladyslav Drok","email":"vdrok@mirantis.com","date":"2019-08-30 12:52:42.000000000","tz":120},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2019-10-18 17:37:55.000000000","tz":0},"subject":"Read in non-blocking fashion when starting console","message":"Read in non-blocking fashion when starting console\n\nAs PID file is created by daemon, it might happen that although the\nreturn code is 0, the PID file does not exist yet. And afterwards,\ncommunicate() call might hang for some reason. As there is no way to\nput timeout on communicate call in python2, and in order to make\nthis change more backportable, this patch uses non-blocking read to\nget the values of stdout and stderr of the call.\n\nStory: 2006514\nTask: 36511\nChange-Id: I0bf38321b0135ea17439d35081b5727fd4d827c8\n(cherry picked from commit efc2d5fcd3e4c3d8e24801b45f61f7f747b7f340)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/680c8b7745499b916c5f7d1614b61df738f40df2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/680c8b7745499b916c5f7d1614b61df738f40df2"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
