)]}'
{"id":"openstack%2Fironic~698696","triplet_id":"openstack%2Fironic~master~Id849d0967ba168a6e8940d830d62e62ea1f34f29","project":"openstack/ironic","branch":"master","topic":"test-multinode","hashtags":[],"change_id":"Id849d0967ba168a6e8940d830d62e62ea1f34f29","subject":"[DNM] test multinode job","status":"ABANDONED","created":"2019-12-12 11:18:31.000000000","updated":"2020-01-30 16:24:55.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b25e1d4b92216388af5b5b60030937ded8ab8291","_number":698696,"virtual_id_number":698696,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-01-03 23:05:03.000000000","_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"date":"2020-01-03 16:34:19.000000000","_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-01-03 11:03:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-02 18:01:09.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-03 11:03:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-03 16:34:19.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"REVIEWER"},{"updated":"2020-01-03 23:05:03.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"REVIEWER"}],"messages":[{"id":"c8dbdbfbe473edcd8a4b44b772a9553caab728eb","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-12-12 11:18:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c5927ba3c8fee4aa19b495ccfd558075f3b8e242","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 13:35:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/67c326d6d872498f8b3cae157f959f40 : SUCCESS in 2h 13m 51s","accounts_in_message":[],"_revision_number":1},{"id":"f67dd68879f4434486eb99fdd89611e18b52bfdc","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-12-12 13:56:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ac6a0342811c320c5b46c147a9b0d35c92d038c8","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-12-12 14:01:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5461ae0c265ed3da02ea20c151a256fd872e4507","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 15:41:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- ironic-grenade-dsvm-multinode-multitenant https://zuul.opendev.org/t/openstack/build/f4d27980753d4d10bf09be1bd557eae1 : FAILURE in 1h 27m 52s","accounts_in_message":[],"_revision_number":3},{"id":"9554599b5fc34c6ce8553cdadf6f6ed5aaa4be0b","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-12-12 17:45:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e47e7e6cadf54898a26a95e1abb10beb00037f50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 20:31:11.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- ironic-grenade-dsvm-multinode-multitenant https://zuul.opendev.org/t/openstack/build/27b0bcdc416344de9e39d50deb5d3fdd : FAILURE in 2h 29m 40s","accounts_in_message":[],"_revision_number":4},{"id":"8544b0ed04b95cf4a5dd491e625ef31722a12262","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2019-12-13 07:00:50.000000000","message":"Patch Set 4:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac/2019-12-12:11:46/698696/4/ : FAILURE in 34m 47s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac/2019-12-12:11:46/698696/4/ : SUCCESS in 42m 41s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac/2019-12-12:11:46/698696/4/ : FAILURE in 1h 01m 55s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish/2019-12-12:11:46/698696/4/ : SUCCESS in 40m 31s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish/2019-12-12:11:46/698696/4/ : SUCCESS in 1h 07m 02s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi/2019-12-12:11:46/698696/4/ : SUCCESS in 38m 48s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi/2019-12-12:11:46/698696/4/ : SUCCESS in 1h 06m 46s","accounts_in_message":[],"_revision_number":4},{"id":"d4045fdca4392332f4bea4ff71fa2305300e3949","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-12-16 11:25:55.000000000","message":"Patch Set 4:\n\nrecheck ./safe-devstack-vm-gate-wrap.sh: line 583: cannot create temp file for here-document: No space left on device","accounts_in_message":[],"_revision_number":4},{"id":"85615543274c5035f7ee0dcc89b3d7024f552635","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 11:26:09.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"7e9c2a5b3c17c09334c71d39d09eba294797456e","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-12-16 11:28:00.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7496a0f0034b34ca0fb022043bbb5210a1ce8dc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 13:21:28.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- ironic-grenade-dsvm-multinode-multitenant https://zuul.opendev.org/t/openstack/build/975d3539d7074bd5b76f393ae38279a7 : FAILURE in 1h 44m 55s","accounts_in_message":[],"_revision_number":5},{"id":"ec645d9d1cfab89419c83214a430157c6a5e859b","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2019-12-16 18:37:30.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/96/698696/5/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/77254dc/77254dc.html : FAILURE in 1h 25m 41s\n- hpe-gen8-dsvm-ironic-standalone-wholedisk-bios-ilo-virtual-media-iscsi-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/96/698696/5/check/hpe-gen8-dsvm-ironic-standalone-wholedisk-bios-ilo-virtual-media-iscsi-python3/7cb98e4/7cb98e4.html : POST_FAILURE in 1h 30m 38s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python2 https://hpproliant.github.io/hpeproliant.github.io/logs/96/698696/5/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python2/e0d8a16/e0d8a16.html : TIMED_OUT in 3h 02m 25s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/96/698696/5/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/4214800/4214800.html : POST_FAILURE in 1h 09m 56s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/96/698696/5/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/fcaa383/fcaa383.html : FAILURE in 1h 35m 03s","accounts_in_message":[],"_revision_number":5},{"id":"2269d76c9e2630dc10f0339a1a46e364220d7518","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-12-17 17:22:42.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"9b66a014a44cb92681b77141e42994596f4de0ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 20:06:42.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- ironic-grenade-dsvm-multinode-multitenant https://zuul.opendev.org/t/openstack/build/87aad58f2f77400fb8503003ff538465 : FAILURE in 2h 28m 11s","accounts_in_message":[],"_revision_number":6},{"id":"d82404d06f6fecb1bcba14d5614068b317796272","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2019-12-17 20:51:05.000000000","message":"Patch Set 6:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac/2019-12-17:11:22/698696/6/ : FAILURE in 40m 26s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac/2019-12-17:11:22/698696/6/ : SUCCESS in 41m 22s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac/2019-12-17:11:22/698696/6/ : FAILURE in 41m 14s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish/2019-12-17:11:22/698696/6/ : FAILURE in 46m 33s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish/2019-12-17:11:22/698696/6/ : SUCCESS in 39m 40s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi/2019-12-17:11:22/698696/6/ : SUCCESS in 38m 32s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi/2019-12-17:11:22/698696/6/ : SUCCESS in 47m 05s","accounts_in_message":[],"_revision_number":6},{"id":"4ba49e6d4331ba0e482513e3d122923aac7f774d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-12-17 23:25:48.000000000","message":"Patch Set 6:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-ironic-agent_ipmitool https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/ironic/96/698696/6/check-ironic/tempest-dsvm-ironic-agent_ipmitool/9dd88c4/ : FAILURE in 33m 36s","accounts_in_message":[],"_revision_number":6},{"id":"685fa0286066b60c0a9c51e122e44a133322b008","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-12-18 09:17:07.000000000","message":"Uploaded patch set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"0b6fbfaa88df106a48d19ffde73a6b709320402e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-12-18 09:51:50.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-ironic-agent_ipmitool https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/ironic/96/698696/7/check-ironic/tempest-dsvm-ironic-agent_ipmitool/e06eb46/ : FAILURE in 34m 03s","accounts_in_message":[],"_revision_number":7},{"id":"422d2a681f5889a7f749e225690ac44c8ca4c7bd","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2019-12-18 10:42:08.000000000","message":"Patch Set 7:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac/2019-12-18:03:17/698696/7/ : FAILURE in 37m 00s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac/2019-12-18:03:17/698696/7/ : SUCCESS in 51m 30s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac/2019-12-18:03:17/698696/7/ : FAILURE in 42m 07s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish/2019-12-18:03:17/698696/7/ : SUCCESS in 47m 12s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish/2019-12-18:03:17/698696/7/ : SUCCESS in 41m 51s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi/2019-12-18:03:17/698696/7/ : SUCCESS in 40m 07s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi/2019-12-18:03:17/698696/7/ : SUCCESS in 41m 05s","accounts_in_message":[],"_revision_number":7},{"id":"4387a56ce98ac88472fe61040c65b074ff5a76a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 10:42:27.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- ironic-grenade-dsvm-multinode-multitenant https://zuul.opendev.org/t/openstack/build/1b92b6ee838b4d0995ea59eb7b0c6df3 : FAILURE in 1h 18m 02s","accounts_in_message":[],"_revision_number":7},{"id":"ce5e5fffecdf332d515de08600fff733ef258145","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-12-18 13:36:10.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"1dc2d77a5ab26a29a88f60f72090e7d98fc6f783","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-12-18 14:10:44.000000000","message":"Patch Set 8:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-ironic-agent_ipmitool https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/ironic/96/698696/8/check-ironic/tempest-dsvm-ironic-agent_ipmitool/5594dc5/ : FAILURE in 33m 53s","accounts_in_message":[],"_revision_number":8},{"id":"e6eb0dbb9b9bb2c0da63063aab13436f94e8b378","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2019-12-18 15:02:21.000000000","message":"Patch Set 8:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac/2019-12-18:07:36/698696/8/ : FAILURE in 39m 16s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac/2019-12-18:07:36/698696/8/ : SUCCESS in 49m 05s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac/2019-12-18:07:36/698696/8/ : SUCCESS in 47m 18s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish/2019-12-18:07:36/698696/8/ : SUCCESS in 47m 29s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish/2019-12-18:07:36/698696/8/ : SUCCESS in 44m 37s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi/2019-12-18:07:36/698696/8/ : SUCCESS in 38m 25s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi/2019-12-18:07:36/698696/8/ : SUCCESS in 39m 25s","accounts_in_message":[],"_revision_number":8},{"id":"c8dba49ae47579c291dc0508a1bcaad9a2753fb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 15:16:41.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- ironic-grenade-dsvm-multinode-multitenant https://zuul.opendev.org/t/openstack/build/b2200dea5eff4474b16a36ddcec336db : FAILURE in 1h 36m 53s","accounts_in_message":[],"_revision_number":8},{"id":"157e79cf79481677ba5589cb1eddb5016b4ef4c0","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-12-18 15:59:48.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"82f279d33a35be58eb720e7101f707947426fdef","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-12-18 16:06:10.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"243784e3775d1f01db8a2a329d87465e990461ee","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-12-18 16:52:06.000000000","message":"Patch Set 9:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-ironic-agent_ipmitool https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/ironic/96/698696/9/check-ironic/tempest-dsvm-ironic-agent_ipmitool/bd286c3/ : FAILURE in 34m 43s","accounts_in_message":[],"_revision_number":9},{"id":"443c90380f4f1ac3c02bf5d04ebb886160f081e4","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2019-12-18 17:32:53.000000000","message":"Patch Set 9:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac/2019-12-18:10:06/698696/9/ : FAILURE in 34m 55s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac/2019-12-18:10:06/698696/9/ : SUCCESS in 50m 57s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac/2019-12-18:10:06/698696/9/ : SUCCESS in 42m 35s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish/2019-12-18:10:06/698696/9/ : SUCCESS in 45m 57s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish/2019-12-18:10:06/698696/9/ : SUCCESS in 46m 43s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi/2019-12-18:10:06/698696/9/ : FAILURE in 44m 09s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi/2019-12-18:10:06/698696/9/ : SUCCESS in 39m 25s","accounts_in_message":[],"_revision_number":9},{"id":"780e30890c5ed2ab41e1841efd7919efc2a3834a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 17:46:05.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- ironic-grenade-dsvm-multinode-multitenant https://zuul.opendev.org/t/openstack/build/33de058a39a242968fefa92d2a3aa669 : FAILURE in 1h 36m 17s","accounts_in_message":[],"_revision_number":9},{"id":"981fd19f7c30ab7525ab7bbc61ec210c737beed6","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-12-23 13:54:14.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"f6cf6658b69f320693371c1db9b3898c4cd3baa8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-23 15:32:19.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- ironic-grenade-dsvm-multinode-multitenant https://zuul.opendev.org/t/openstack/build/a7f7b7fa58f14511ad3e335f7e9b5c6a : FAILURE in 1h 26m 02s","accounts_in_message":[],"_revision_number":10},{"id":"54b2e168f8fe78134ecb3a3e43120cdea4cc5077","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2019-12-23 16:23:34.000000000","message":"Patch Set 10:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac/2019-12-23:07:54/698696/10/ : SUCCESS in 48m 14s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac/2019-12-23:07:54/698696/10/ : SUCCESS in 46m 46s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac/2019-12-23:07:54/698696/10/ : SUCCESS in 1h 34m 39s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish/2019-12-23:07:54/698696/10/ : SUCCESS in 51m 11s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish/2019-12-23:07:54/698696/10/ : FAILURE in 1h 46m 15s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi/2019-12-23:07:54/698696/10/ : SUCCESS in 44m 01s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi/2019-12-23:07:54/698696/10/ : SUCCESS in 1h 30m 16s","accounts_in_message":[],"_revision_number":10},{"id":"c65c650586fda5f0d3051caab4538779bddf65f5","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2019-12-23 17:13:07.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/96/698696/10/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/0318fe3/0318fe3.html : FAILURE in 1h 30m 54s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python2 https://hpproliant.github.io/hpeproliant.github.io/logs/96/698696/10/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python2/ba2af93/ba2af93.html : FAILURE in 2h 23m 23s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/96/698696/10/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/a1ac2e8/a1ac2e8.html : FAILURE in 1h 16m 36s","accounts_in_message":[],"_revision_number":10},{"id":"73a829ed3632596b8a1e93c3d0a5c99bc26ee2eb","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-12-24 14:16:50.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"d5d4fc48ff5286a78d254448db2898065b5c3f92","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-12-24 14:32:28.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"18de7659728058be6dbcd72aa8f5b860bc74d770","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-24 15:49:06.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- ironic-grenade-dsvm-multinode-multitenant https://zuul.opendev.org/t/openstack/build/4a5773fffbd942e99d836f24b242f203 : FAILURE in 1h 14m 35s","accounts_in_message":[],"_revision_number":12},{"id":"629e8257bbb1e7a1401d1bb4482267287fdc39ee","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2019-12-24 17:12:46.000000000","message":"Patch Set 12:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac/2019-12-24:08:32/698696/12/ : SUCCESS in 50m 01s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac/2019-12-24:08:32/698696/12/ : SUCCESS in 55m 00s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac/2019-12-24:08:32/698696/12/ : FAILURE in 1h 46m 06s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish/2019-12-24:08:32/698696/12/ : SUCCESS in 43m 33s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish/2019-12-24:08:32/698696/12/ : FAILURE in 1h 45m 56s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi/2019-12-24:08:32/698696/12/ : FAILURE in 48m 06s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi/2019-12-24:08:32/698696/12/ : FAILURE in 1h 46m 08s","accounts_in_message":[],"_revision_number":12},{"id":"99a68fe04b76925af7c76bcf1211eed64a6bda69","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2019-12-24 17:41:13.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/96/698696/12/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/dc29f02/dc29f02.html : TIMED_OUT in 3h 03m 06s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python2 https://hpproliant.github.io/hpeproliant.github.io/logs/96/698696/12/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python2/b1d863a/b1d863a.html : TIMED_OUT in 3h 03m 21s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/96/698696/12/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/56259fa/56259fa.html : TIMED_OUT in 3h 03m 23s","accounts_in_message":[],"_revision_number":12},{"id":"dbfdc94becc4da8782564a499e6aa8e1891d08da","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-02 16:29:36.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"f62b363b44a11c0de2d6451c127ce5ffd2296d44","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-02 16:32:13.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"4d8748cf39c568d94101556b571435169a8f314d","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-02 16:33:33.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"40c1d8709c2d164874c9ac51a25bc2fa6a29a87a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-01-02 18:01:09.000000000","message":"Patch Set 15:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-ironic-agent_ipmitool https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/ironic/96/698696/15/check-ironic/tempest-dsvm-ironic-agent_ipmitool/95ca365/ : FAILURE in 34m 56s","accounts_in_message":[],"_revision_number":15},{"id":"eff5d7d1e9f73287878ca5029808e70b6c133e5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-02 18:16:37.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/f150df13782048a3a30af84ed518ece2 : SUCCESS in 1h 37m 25s","accounts_in_message":[],"_revision_number":15},{"id":"0b4d464a24f4b7ac49a5ce819699f596e309babf","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-01-03 04:03:08.000000000","message":"Patch Set 15:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac/2020-01-02:10:33/698696/15/ : SUCCESS in 55m 48s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac/2020-01-02:10:33/698696/15/ : SUCCESS in 44m 55s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac/2020-01-02:10:33/698696/15/ : FAILURE in 1h 46m 07s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish/2020-01-02:10:33/698696/15/ : SUCCESS in 50m 10s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish/2020-01-02:10:33/698696/15/ : FAILURE in 1h 46m 13s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi/2020-01-02:10:33/698696/15/ : SUCCESS in 43m 54s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi/2020-01-02:10:33/698696/15/ : FAILURE in 1h 46m 08s","accounts_in_message":[],"_revision_number":15},{"id":"eb851cddd40489fa3dea94ff2214becc40c09fbe","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-03 09:26:16.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"cfd4e7252374f50fd25f0db608f448de0b7a2b3e","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-03 09:28:11.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"1e8d4edcce8948e9bf60bb5b3a9fd033a933b0c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-03 11:03:36.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- ironic-grenade-dsvm-multinode-multitenant https://zuul.opendev.org/t/openstack/build/035ce65316ee4ed38863025a02a8728c : FAILURE in 1h 28m 56s","accounts_in_message":[],"_revision_number":17},{"id":"de8e9c4da5e86e645e34bcbfd156451a418fdd30","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-01-03 16:34:19.000000000","message":"Patch Set 17:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac/2020-01-03:03:28/698696/17/ : SUCCESS in 52m 41s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac/2020-01-03:03:28/698696/17/ : SUCCESS in 45m 19s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac/2020-01-03:03:28/698696/17/ : FAILURE in 1h 46m 06s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish/2020-01-03:03:28/698696/17/ : SUCCESS in 50m 28s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish/2020-01-03:03:28/698696/17/ : FAILURE in 1h 46m 21s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi/2020-01-03:03:28/698696/17/ : SUCCESS in 50m 35s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi/2020-01-03:03:28/698696/17/ : FAILURE in 1h 46m 06s","accounts_in_message":[],"_revision_number":17},{"id":"2c037ab96a6ce4b84340e57d9c0d5d07697e7f16","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2020-01-03 23:05:03.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/96/698696/17/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/c7522ea/c7522ea.html : FAILURE in 1h 36m 34s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python2 https://hpproliant.github.io/hpeproliant.github.io/logs/96/698696/17/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python2/d16f602/d16f602.html : TIMED_OUT in 3h 03m 16s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/96/698696/17/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/d573e03/d573e03.html : FAILURE in 11m 07s","accounts_in_message":[],"_revision_number":17},{"id":"a13e0e1a36e6ed42db92e7a7230df7d8885a995d","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-30 16:24:55.000000000","message":"Abandoned\n\nwe\u0027re migrating directly to zuulv3","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"66ac02d291173bc48c7a11cf40de5d9d1243f5f2","revisions":{"1ddf0ab98d0d470d1c3d7e53852f8d257012e9f0":{"kind":"REWORK","_number":1,"created":"2019-12-12 11:18:31.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/96/698696/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/96/698696/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/96/698696/1"}}},"commit":{"parents":[{"commit":"b293eb1af0cda474a7262787efeb67d29591616b","subject":"Remove deprecated ironic-agent element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/b293eb1af0cda474a7262787efeb67d29591616b"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-12-12 11:18:06.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-12-12 11:18:06.000000000","tz":60},"subject":"[DNM] test multinode job","message":"[DNM] test multinode job\n\nChange-Id: Id849d0967ba168a6e8940d830d62e62ea1f34f29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1ddf0ab98d0d470d1c3d7e53852f8d257012e9f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1ddf0ab98d0d470d1c3d7e53852f8d257012e9f0"}]},"branch":"refs/heads/master"},"1a2daea3ad68a3b21d5892671bff1e928d142de4":{"kind":"REWORK","_number":2,"created":"2019-12-12 13:56:06.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/96/698696/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/96/698696/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/96/698696/2"}}},"commit":{"parents":[{"commit":"b293eb1af0cda474a7262787efeb67d29591616b","subject":"Remove deprecated ironic-agent element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/b293eb1af0cda474a7262787efeb67d29591616b"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-12-12 11:18:06.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-12-12 13:55:56.000000000","tz":60},"subject":"[DNM] test multinode job","message":"[DNM] test multinode job\n\nChange-Id: Id849d0967ba168a6e8940d830d62e62ea1f34f29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1a2daea3ad68a3b21d5892671bff1e928d142de4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1a2daea3ad68a3b21d5892671bff1e928d142de4"}]},"branch":"refs/heads/master"},"4b2ee02722e483c8afe49393914e478f8c1b8084":{"kind":"REWORK","_number":3,"created":"2019-12-12 14:01:29.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/96/698696/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/96/698696/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/96/698696/3"}}},"commit":{"parents":[{"commit":"b293eb1af0cda474a7262787efeb67d29591616b","subject":"Remove deprecated ironic-agent element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/b293eb1af0cda474a7262787efeb67d29591616b"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-12-12 11:18:06.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-12-12 14:01:20.000000000","tz":60},"subject":"[DNM] test multinode job","message":"[DNM] test multinode job\n\nChange-Id: Id849d0967ba168a6e8940d830d62e62ea1f34f29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4b2ee02722e483c8afe49393914e478f8c1b8084"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4b2ee02722e483c8afe49393914e478f8c1b8084"}]},"branch":"refs/heads/master"},"11352353e94b845de5d7c3b608e05285efb58f38":{"kind":"REWORK","_number":4,"created":"2019-12-12 17:45:47.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/96/698696/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/96/698696/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/96/698696/4"}}},"commit":{"parents":[{"commit":"b293eb1af0cda474a7262787efeb67d29591616b","subject":"Remove deprecated ironic-agent element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/b293eb1af0cda474a7262787efeb67d29591616b"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-12-12 11:18:06.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-12-12 17:45:39.000000000","tz":60},"subject":"[DNM] test multinode job","message":"[DNM] test multinode job\n\nChange-Id: Id849d0967ba168a6e8940d830d62e62ea1f34f29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/11352353e94b845de5d7c3b608e05285efb58f38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/11352353e94b845de5d7c3b608e05285efb58f38"}]},"branch":"refs/heads/master"},"b9d7b8320fafcc1f96c860024921560f14cbdace":{"kind":"REWORK","_number":5,"created":"2019-12-16 11:28:00.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/96/698696/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/96/698696/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/96/698696/5"}}},"commit":{"parents":[{"commit":"2edb84db23e508f80ca6b5919db08ad96ee4735c","subject":"Merge \"Fix use of urlparse.urljoin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/2edb84db23e508f80ca6b5919db08ad96ee4735c"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-12-12 11:18:06.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-12-16 11:27:38.000000000","tz":60},"subject":"[DNM] test multinode job","message":"[DNM] test multinode job\n\nChange-Id: Id849d0967ba168a6e8940d830d62e62ea1f34f29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/b9d7b8320fafcc1f96c860024921560f14cbdace"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/b9d7b8320fafcc1f96c860024921560f14cbdace"}]},"branch":"refs/heads/master"},"a35b11671594a400b10f0ee2471fcd7227ba9c53":{"kind":"REWORK","_number":6,"created":"2019-12-17 17:22:42.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/96/698696/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/96/698696/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/96/698696/6"}}},"commit":{"parents":[{"commit":"76597ca9331ab8727d20301e42c970c302597eda","subject":"Merge \"Correct power state handling for managed in-band inspection\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/76597ca9331ab8727d20301e42c970c302597eda"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-12-12 11:18:06.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-12-17 17:22:33.000000000","tz":60},"subject":"[DNM] test multinode job","message":"[DNM] test multinode job\n\nChange-Id: Id849d0967ba168a6e8940d830d62e62ea1f34f29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/a35b11671594a400b10f0ee2471fcd7227ba9c53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/a35b11671594a400b10f0ee2471fcd7227ba9c53"}]},"branch":"refs/heads/master"},"742148aab33ce6ebb14c614a1df0b8dd3b4ee545":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2019-12-18 09:17:07.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/96/698696/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/96/698696/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/96/698696/7"}}},"commit":{"parents":[{"commit":"76597ca9331ab8727d20301e42c970c302597eda","subject":"Merge \"Correct power state handling for managed in-band inspection\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/76597ca9331ab8727d20301e42c970c302597eda"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-12-12 11:18:06.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-12-18 09:16:16.000000000","tz":60},"subject":"[DNM] test multinode job","message":"[DNM] test multinode job\n\nDepends-On: https://review.opendev.org/699338\n\nChange-Id: Id849d0967ba168a6e8940d830d62e62ea1f34f29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/742148aab33ce6ebb14c614a1df0b8dd3b4ee545"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/742148aab33ce6ebb14c614a1df0b8dd3b4ee545"}]},"branch":"refs/heads/master"},"4962801c74560fa89bc1ad9759400e7fafb1c5a6":{"kind":"REWORK","_number":8,"created":"2019-12-18 13:36:10.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/96/698696/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/96/698696/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/96/698696/8"}}},"commit":{"parents":[{"commit":"76597ca9331ab8727d20301e42c970c302597eda","subject":"Merge \"Correct power state handling for managed in-band inspection\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/76597ca9331ab8727d20301e42c970c302597eda"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-12-12 11:18:06.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-12-18 13:35:56.000000000","tz":60},"subject":"[DNM] test multinode job","message":"[DNM] test multinode job\n\nChange-Id: Id849d0967ba168a6e8940d830d62e62ea1f34f29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4962801c74560fa89bc1ad9759400e7fafb1c5a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4962801c74560fa89bc1ad9759400e7fafb1c5a6"}]},"branch":"refs/heads/master"},"257d6c903262ff9e89a4eb49471c746610110b46":{"kind":"REWORK","_number":9,"created":"2019-12-18 16:06:10.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/96/698696/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/96/698696/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/96/698696/9"}}},"commit":{"parents":[{"commit":"8d8bfc9cc5e3b1384d4df90cd998b64425514831","subject":"Merge \"Tune down multinode concurrency\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/8d8bfc9cc5e3b1384d4df90cd998b64425514831"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-12-12 11:18:06.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-12-18 16:06:01.000000000","tz":60},"subject":"[DNM] test multinode job","message":"[DNM] test multinode job\n\nChange-Id: Id849d0967ba168a6e8940d830d62e62ea1f34f29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/257d6c903262ff9e89a4eb49471c746610110b46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/257d6c903262ff9e89a4eb49471c746610110b46"}]},"branch":"refs/heads/master"},"9df4757450379ff8b0cd149a43dd8dc0714c40c1":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2019-12-23 13:54:14.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/96/698696/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/96/698696/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/96/698696/10"}}},"commit":{"parents":[{"commit":"67f5a6bf5c8922a56d70daae023a004e672a88db","subject":"Merge \"Disable automated clean on newer jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/67f5a6bf5c8922a56d70daae023a004e672a88db"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-12-12 11:18:06.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-12-23 13:54:00.000000000","tz":60},"subject":"[DNM] test multinode job","message":"[DNM] test multinode job\n\nChange-Id: Id849d0967ba168a6e8940d830d62e62ea1f34f29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/9df4757450379ff8b0cd149a43dd8dc0714c40c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/9df4757450379ff8b0cd149a43dd8dc0714c40c1"}]},"branch":"refs/heads/master"},"4a46ab8e1586d617b817113ab642c07bf70d0850":{"kind":"REWORK","_number":11,"created":"2019-12-24 14:16:50.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/96/698696/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/96/698696/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/96/698696/11"}}},"commit":{"parents":[{"commit":"67f5a6bf5c8922a56d70daae023a004e672a88db","subject":"Merge \"Disable automated clean on newer jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/67f5a6bf5c8922a56d70daae023a004e672a88db"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-12-12 11:18:06.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-12-24 14:15:56.000000000","tz":60},"subject":"[DNM] test multinode job","message":"[DNM] test multinode job\n\nChange-Id: Id849d0967ba168a6e8940d830d62e62ea1f34f29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4a46ab8e1586d617b817113ab642c07bf70d0850"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4a46ab8e1586d617b817113ab642c07bf70d0850"}]},"branch":"refs/heads/master"},"640e40e89c88ed1d5afedf228d27c3d5157a2a75":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2019-12-24 14:32:28.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/96/698696/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/96/698696/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/96/698696/12"}}},"commit":{"parents":[{"commit":"0a675e397742e28cab9754bb91604c91c5022379","subject":"Merge \"Stop using six library\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/0a675e397742e28cab9754bb91604c91c5022379"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-12-12 11:18:06.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-12-24 14:32:18.000000000","tz":60},"subject":"[DNM] test multinode job","message":"[DNM] test multinode job\n\nChange-Id: Id849d0967ba168a6e8940d830d62e62ea1f34f29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/640e40e89c88ed1d5afedf228d27c3d5157a2a75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/640e40e89c88ed1d5afedf228d27c3d5157a2a75"}]},"branch":"refs/heads/master"},"224af75687069c006e614d4d5f83b036c945445a":{"kind":"REWORK","_number":13,"created":"2020-01-02 16:29:36.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/96/698696/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/96/698696/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/96/698696/13"}}},"commit":{"parents":[{"commit":"0a675e397742e28cab9754bb91604c91c5022379","subject":"Merge \"Stop using six library\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/0a675e397742e28cab9754bb91604c91c5022379"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-12-12 11:18:06.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-02 16:29:24.000000000","tz":60},"subject":"[DNM] test multinode job","message":"[DNM] test multinode job\n\nChange-Id: Id849d0967ba168a6e8940d830d62e62ea1f34f29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/224af75687069c006e614d4d5f83b036c945445a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/224af75687069c006e614d4d5f83b036c945445a"}]},"branch":"refs/heads/master"},"9f59cee30cc31a392c5d1ea5cfa1687f503cbc46":{"kind":"REWORK","_number":14,"created":"2020-01-02 16:32:13.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/96/698696/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/96/698696/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/96/698696/14"}}},"commit":{"parents":[{"commit":"2a43e371f17e78ea749bb0d17c98df24652aeb0c","subject":"Merge \"Remove those switches for python2.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/2a43e371f17e78ea749bb0d17c98df24652aeb0c"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-12-12 11:18:06.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-02 16:32:04.000000000","tz":60},"subject":"[DNM] test multinode job","message":"[DNM] test multinode job\n\nChange-Id: Id849d0967ba168a6e8940d830d62e62ea1f34f29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/9f59cee30cc31a392c5d1ea5cfa1687f503cbc46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/9f59cee30cc31a392c5d1ea5cfa1687f503cbc46"}]},"branch":"refs/heads/master"},"f51073e96f2724b07623bd4d65722b5a3f88398b":{"kind":"REWORK","_number":15,"created":"2020-01-02 16:33:33.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/96/698696/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/96/698696/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/96/698696/15"}}},"commit":{"parents":[{"commit":"2a43e371f17e78ea749bb0d17c98df24652aeb0c","subject":"Merge \"Remove those switches for python2.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/2a43e371f17e78ea749bb0d17c98df24652aeb0c"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-12-12 11:18:06.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-02 16:33:25.000000000","tz":60},"subject":"[DNM] test multinode job","message":"[DNM] test multinode job\n\nChange-Id: Id849d0967ba168a6e8940d830d62e62ea1f34f29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/f51073e96f2724b07623bd4d65722b5a3f88398b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/f51073e96f2724b07623bd4d65722b5a3f88398b"}]},"branch":"refs/heads/master"},"6e480aa34dcb3a146ec6ab43d440eebf5849409e":{"kind":"REWORK","_number":16,"created":"2020-01-03 09:26:16.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/96/698696/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/96/698696/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/96/698696/16"}}},"commit":{"parents":[{"commit":"2a43e371f17e78ea749bb0d17c98df24652aeb0c","subject":"Merge \"Remove those switches for python2.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/2a43e371f17e78ea749bb0d17c98df24652aeb0c"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-12-12 11:18:06.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-03 09:25:17.000000000","tz":60},"subject":"[DNM] test multinode job","message":"[DNM] test multinode job\n\nChange-Id: Id849d0967ba168a6e8940d830d62e62ea1f34f29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/6e480aa34dcb3a146ec6ab43d440eebf5849409e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/6e480aa34dcb3a146ec6ab43d440eebf5849409e"}]},"branch":"refs/heads/master"},"66ac02d291173bc48c7a11cf40de5d9d1243f5f2":{"kind":"REWORK","_number":17,"created":"2020-01-03 09:28:11.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/96/698696/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/96/698696/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/96/698696/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/96/698696/17"}}},"commit":{"parents":[{"commit":"2a43e371f17e78ea749bb0d17c98df24652aeb0c","subject":"Merge \"Remove those switches for python2.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/2a43e371f17e78ea749bb0d17c98df24652aeb0c"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-12-12 11:18:06.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-01-03 09:28:02.000000000","tz":60},"subject":"[DNM] test multinode job","message":"[DNM] test multinode job\n\nChange-Id: Id849d0967ba168a6e8940d830d62e62ea1f34f29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/66ac02d291173bc48c7a11cf40de5d9d1243f5f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/66ac02d291173bc48c7a11cf40de5d9d1243f5f2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
