)]}'
{"id":"openstack%2Fironic~721256","triplet_id":"openstack%2Fironic~master~I19fdb1bf580345a72958ba161d4e1b5d651c5fbe","project":"openstack/ironic","branch":"master","topic":"02-fix-vmedia-tests","hashtags":[],"change_id":"I19fdb1bf580345a72958ba161d4e1b5d651c5fbe","subject":"Fix vmedia test methods and remove `six`","status":"ABANDONED","created":"2020-04-20 12:39:16.000000000","updated":"2020-06-18 16:40:06.000000000","total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c314fd31eb4fafeacfce5b0bffb0fce51350c729","_number":721256,"virtual_id_number":721256,"owner":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-04-21 00:09:44.000000000","_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-04-20 20:11:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-05-05 13:31:48.000000000","_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":2,"date":"2020-04-20 14:27:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":2,"date":"2020-04-20 14:27:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-20 12:42:25.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-04-20 12:46:31.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-04-20 20:11:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-21 00:09:44.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"REVIEWER"},{"updated":"2020-05-05 13:31:48.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"}],"messages":[{"id":"3ba86ba0fc0a82f002a07845403fcd6589e20a3e","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2020-04-20 12:39:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a727a1fc13a34c583beff87885a0e6ac63d5020f","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-04-20 12:42:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0a336386492e8bb97b7b91f9bfbe169028fd625a","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-04-20 12:46:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4d9276e75fc66df6949a57f245b87235896826a6","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2020-04-20 14:27:18.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"d83a7cf94c788517c779c18a2c8a793597d7c9b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 20:11:23.000000000","message":"Patch Set 2: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab7d96d2e1a1404583a5a42b1b5b9d47 : SUCCESS in 8m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be5d78708056400c895e448e5ec1228a : SUCCESS in 8m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1022d95dc5ae4e4ba5c6e35b7c7eb9d3 : FAILURE in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04d3ccf86e2f47f08377da0b6b239caa : SUCCESS in 7m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6a2f05179b4943f08a1751aaa66f9d9a : SUCCESS in 11m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1af85c2666ea4916b014b1c81237ada6 : SUCCESS in 7m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9abb0766a27743bbbce78bd4d615b66b : SUCCESS in 20m 23s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/6a502dc5c43142b089b96eaf4217344c : SUCCESS in 8m 21s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/29e8ef77809e4eb3b43cf87b995bf6c6 : SUCCESS in 1h 32m 15s\n- ironic-inspector-tempest-managed https://zuul.opendev.org/t/openstack/build/f7a19269199f4591abce4314f4098e51 : SUCCESS in 1h 02m 23s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/25a70818d8374b94b7452d47ef0a5398 : SUCCESS in 16m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0b7ed55ef0c34babd60a2e62d21a09fb3766bb9b","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-04-21 00:09:44.000000000","message":"Patch Set 2:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac/2020-04-20:09:27/721256/2/ : FAILURE in 48m 01s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac/2020-04-20:09:27/721256/2/ : FAILURE in 50m 35s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac/2020-04-20:09:27/721256/2/ : FAILURE in 45m 48s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish/2020-04-20:09:27/721256/2/ : FAILURE in 50m 40s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish/2020-04-20:09:27/721256/2/ : FAILURE in 42m 43s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi/2020-04-20:09:27/721256/2/ : FAILURE in 56m 40s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi/2020-04-20:09:27/721256/2/ : FAILURE in 49m 18s\n- dellemc-poweredge-XE-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-XE-UEFI-tempest-dsvm-ironic-redfish/2020-04-20:09:27/721256/2/ : FAILURE in 42m 37s","accounts_in_message":[],"_revision_number":2},{"id":"51bb260fcce97abfc06d0fc60f3ebc2bfc3d6d76","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-05-05 13:31:48.000000000","message":"Patch Set 2:\n\nI have the feeling we won\u0027t need this anymore since the file was removed in the revert: https://review.opendev.org/#/c/721309/","accounts_in_message":[],"_revision_number":2},{"id":"2bff167ed9bcd0d1c4dce685f47cda77801e9aac","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-06-18 16:40:06.000000000","message":"Abandoned\n\nThis is not necessary since we did a revert","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f2e892b2cf839feb28259b5e5bc988224f05481d","revisions":{"e94304cd9912625084a24dff32346a5a586f35a8":{"kind":"REWORK","_number":1,"created":"2020-04-20 12:39:16.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/56/721256/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/56/721256/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/56/721256/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/56/721256/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/56/721256/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/56/721256/1"}}},"commit":{"parents":[{"commit":"716eb5829e198c347358887b3c8e00e4381e8202","subject":"Add missing and remove non-existing parameters.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/716eb5829e198c347358887b3c8e00e4381e8202"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2020-04-20 12:38:34.000000000","tz":120},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2020-04-20 12:38:34.000000000","tz":120},"subject":"Fix vmedia test methods and remove `six`","message":"Fix vmedia test methods and remove `six`\n\nCommit [1] caused `six` to re-appear in ironic tests, this\npatch removes it.\n\nAlso, this patch renames test methods to reflect the implementation\nmethods being tested.\n\n1. https://review.opendev.org/#/c/677996/16\n\nChange-Id: I19fdb1bf580345a72958ba161d4e1b5d651c5fbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/e94304cd9912625084a24dff32346a5a586f35a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/e94304cd9912625084a24dff32346a5a586f35a8"}]},"branch":"refs/heads/master"},"f2e892b2cf839feb28259b5e5bc988224f05481d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-04-20 14:27:18.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/56/721256/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/56/721256/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/56/721256/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/56/721256/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/56/721256/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/56/721256/2"}}},"commit":{"parents":[{"commit":"36df9149bb0111a03dc04077c9b8084f627b984f","subject":"Add missing and remove non-existing parameters.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/36df9149bb0111a03dc04077c9b8084f627b984f"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2020-04-20 12:38:34.000000000","tz":120},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2020-04-20 14:26:28.000000000","tz":120},"subject":"Fix vmedia test methods and remove `six`","message":"Fix vmedia test methods and remove `six`\n\nCommit [1] caused `six` to re-appear in ironic tests, this\npatch removes it.\n\nAlso, this patch renames test methods to reflect the implementation\nmethods being tested.\n\n1. https://review.opendev.org/#/c/677996/16\n\nChange-Id: I19fdb1bf580345a72958ba161d4e1b5d651c5fbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/f2e892b2cf839feb28259b5e5bc988224f05481d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/f2e892b2cf839feb28259b5e5bc988224f05481d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
