)]}'
{"id":"openstack%2Fironic~725922","triplet_id":"openstack%2Fironic~master~Id0d3db5d4faeb317272885dd9e98527e15c33c12","project":"openstack/ironic","branch":"master","topic":"fix-3rdpdm","hashtags":[],"change_id":"Id0d3db5d4faeb317272885dd9e98527e15c33c12","subject":"Enforce autospec in some unit tests modules","status":"MERGED","created":"2020-05-06 16:23:22.000000000","updated":"2020-05-08 05:27:04.000000000","submitted":"2020-05-08 05:06:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725922-1588914373171-0280b767","meta_rev_id":"a626ddf99f05cadc016cb4de5fbf0ef711105ae0","_number":725922,"virtual_id_number":725922,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-05-07 03:53:56.000000000","_account_id":30100,"name":"Huawei Ironic CI","email":"zhengxianhui@huawei.com","username":"huawei-ironic-driver-ci"},{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"date":"2020-05-07 23:22:17.000000000","_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-08 05:06:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":0,"_account_id":30100,"name":"Huawei Ironic CI","email":"zhengxianhui@huawei.com","username":"huawei-ironic-driver-ci"},{"value":2,"date":"2020-05-07 10:22:00.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":2,"date":"2020-05-07 03:46:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":0,"_account_id":30100,"name":"Huawei Ironic CI","email":"zhengxianhui@huawei.com","username":"huawei-ironic-driver-ci"},{"value":1,"date":"2020-05-07 10:22:00.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30100,"name":"Huawei Ironic CI","email":"zhengxianhui@huawei.com","username":"huawei-ironic-driver-ci"},{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"_account_id":30100,"name":"Huawei Ironic CI","email":"zhengxianhui@huawei.com","username":"huawei-ironic-driver-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-07 03:46:24.000000000","updated_by":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"reviewer":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"state":"REVIEWER"},{"updated":"2020-05-07 03:53:56.000000000","updated_by":{"_account_id":30100,"name":"Huawei Ironic CI","email":"zhengxianhui@huawei.com","username":"huawei-ironic-driver-ci"},"reviewer":{"_account_id":30100,"name":"Huawei Ironic CI","email":"zhengxianhui@huawei.com","username":"huawei-ironic-driver-ci"},"state":"REVIEWER"},{"updated":"2020-05-07 10:22:00.000000000","updated_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"reviewer":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"state":"REVIEWER"},{"updated":"2020-05-07 23:22:17.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"REVIEWER"},{"updated":"2020-05-08 05:06:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"34487893bc38239e059402e7964adf03be461f87","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-05-06 16:23:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"97ba5ad6653c259b188f00ae1d04d3f5ccb58a6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 17:52:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/344e8072325146458aaa863319b1b3c0 : SUCCESS in 7m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9cba539cf35a489aa877ec76a7e4f7b5 : POST_FAILURE in 7m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b1c255348174657849f78318d884cda : SUCCESS in 11m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cc6737d4bff4779a3c9a127b10f1a45 : SUCCESS in 7m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3067eb1f37eb4ba5b0c3e5cea24143a1 : POST_FAILURE in 7m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/befee25d0277463eaed8836c2d1ffc5f : SUCCESS in 12m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/391c2fb80403439f83db1ce44c58a510 : SUCCESS in 12m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81935c04244e4bec849b1600114c6bb6 : SUCCESS in 22m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a5a1ac351184d9aa9bae8633bfbb1aa : SUCCESS in 13m 08s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/b4188722afe940f5b4cf19fad4f1ec52 : SUCCESS in 8m 59s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/3a8dcc95d2e34253b05292ef1699da6c : SUCCESS in 1h 27m 15s\n- ironic-inspector-tempest-managed https://zuul.opendev.org/t/openstack/build/134ff5349cbb4d4ca8769e1fec83e252 : SUCCESS in 49m 39s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e407fa223781421ab86f4bfb2f537b6d : SUCCESS in 19m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bf345cc1e9711aa13608c8ab6c617189d2ede2d9","author":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"date":"2020-05-07 03:46:24.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1436663c3bee7c9e0103abd896c0b1be9f020a95","author":{"_account_id":30100,"name":"Huawei Ironic CI","email":"zhengxianhui@huawei.com","username":"huawei-ironic-driver-ci"},"date":"2020-05-07 03:53:56.000000000","message":"Patch Set 1:\n\nSUCCESS \n\n* huawei-rack-pxe http://116.66.187.11/builds/Ironic-Unsuri-CI/119 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"f10e7c7719e2f62b12c1a6a013f26f24940424c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 05:00:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/977f14723a1848cbb5787a484cbe75c7 : SUCCESS in 7m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5bfb9345e88c42d1aa81227104b9e6b3 : SUCCESS in 12m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea99bb6c8b2348318f01b60c5e1344a9 : SUCCESS in 8m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af131745ae8744f490b37d28c8b6c63b : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/943836c8a2b046d4a8f5817f369cc677 : SUCCESS in 6m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6eb73d9139ef43b5be1725fb5b40f4eb : SUCCESS in 12m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/53899b0769094ca9b128a7f78db0cb06 : SUCCESS in 13m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cf58a4fccf44a519284d587db46969e : SUCCESS in 20m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/448d36bb80d8468199f4632825cd5c12 : SUCCESS in 12m 58s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/08c17d92f09a4512a8810c0730a86d79 : SUCCESS in 12m 55s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/729ae7b1cfbc4f97aeecfa2cd695872e : SUCCESS in 1h 13m 45s\n- ironic-inspector-tempest-managed https://zuul.opendev.org/t/openstack/build/fe8da07ada864ca19e7daf9c3685d08d : SUCCESS in 54m 43s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d2617fa80d2146c1a2535f8e743bad6d : SUCCESS in 15m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2a7bb9b618b312e985ba2976bd5be12ec2f15c6f","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2020-05-07 10:22:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThank you Riccardo.","accounts_in_message":[],"_revision_number":1},{"id":"6e4431829c9dd8eb7185fb1bc9c77489f3fe6ed6","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-05-07 23:22:17.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac/2020-05-06:11:23/725922/1/ : FAILURE in 20m 44s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac/2020-05-06:11:23/725922/1/ : FAILURE in 19m 30s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac/2020-05-06:11:23/725922/1/ : FAILURE in 21m 50s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish/2020-05-06:11:23/725922/1/ : FAILURE in 37m 24s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish/2020-05-06:11:23/725922/1/ : SUCCESS in 46m 16s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi/2020-05-06:11:23/725922/1/ : FAILURE in 45m 04s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi/2020-05-06:11:23/725922/1/ : FAILURE in 19m 57s\n- dellemc-poweredge-XE-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-XE-UEFI-tempest-dsvm-ironic-redfish/2020-05-06:11:23/725922/1/ : FAILURE in 19m 59s\n- dellemc-poweredge-R7515-R6515-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dsvm-ironic-idrac/2020-05-06:11:23/725922/1/ : FAILURE in 21m 42s\n- dellemc-poweredge-R7515-R6515-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-UEFI-tempest-dsvm-ironic-redfish/2020-05-06:11:23/725922/1/ : FAILURE in 19m 32s","accounts_in_message":[],"_revision_number":1},{"id":"df407f9269f97b3d9904c7dcca95d636a4274f5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 03:31:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"22389bd7e5b56f7e5fd0bc3f7f1bf8d5424d118d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 05:06:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d33290ee64894e16ac2d83aaad6b6b64 : SUCCESS in 5m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/89771b5c9a104cca8b9f0839c4118f8b : SUCCESS in 11m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ccc166ed44c4efdae36c1b08fd399c3 : SUCCESS in 7m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d0981f71d66410ca965e1969ddf6d3b : SUCCESS in 7m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b6abffba16c340698d20446684863d91 : SUCCESS in 7m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f958c883d75e401391750905e3b16e49 : SUCCESS in 21m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c77065c9dd094c23aabb1d303f4163b8 : SUCCESS in 13m 03s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/6bbc2a7efb3e4b1498aa6b714e83095c : SUCCESS in 7m 15s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/edf39103df184b6e84eb263f12f42de0 : SUCCESS in 1h 10m 28s","accounts_in_message":[],"_revision_number":1},{"id":"8b0c4fa195d851f455d1e0002c970351ff5da070","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 05:06:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"470f80b6eb9490f6d56c47153b04318414434fa9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 05:08:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/e128feba0dcc4870b80d14fc690c04aa : SUCCESS in 56s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00d9822c202b447fbac30e23c88153ea : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9028a584450441878cfe0c3922350962 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1},{"id":"b1906f5e69b335990844f7c08a8d00bd829cd9cb","author":{"_account_id":30100,"name":"Huawei Ironic CI","email":"zhengxianhui@huawei.com","username":"huawei-ironic-driver-ci"},"date":"2020-05-08 05:27:04.000000000","message":"Patch Set 1:\n\nSUCCESS \n\n* huawei-rack-pxe http://116.66.187.11/builds/Ironic-Unsuri-CI/146 : SUCCESS","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c5543b572b9585e76d1e1b94c1192b5018fce0a9","revisions":{"c5543b572b9585e76d1e1b94c1192b5018fce0a9":{"kind":"REWORK","_number":1,"created":"2020-05-06 16:23:22.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/22/725922/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/22/725922/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/22/725922/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/22/725922/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/22/725922/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/22/725922/1"}}},"commit":{"parents":[{"commit":"981b24255e032a44cb195f488770d3c4491d40ad","subject":"Fix 3rd party driver mocks and related tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/981b24255e032a44cb195f488770d3c4491d40ad"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-05-06 16:20:01.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-05-06 16:20:01.000000000","tz":120},"subject":"Enforce autospec in some unit tests modules","message":"Enforce autospec in some unit tests modules\n\nAlso changed filters in tox.ini to be more granular and allow\nworking in small batches.\n\nChange-Id: Id0d3db5d4faeb317272885dd9e98527e15c33c12\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/c5543b572b9585e76d1e1b94c1192b5018fce0a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/c5543b572b9585e76d1e1b94c1192b5018fce0a9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
