)]}'
{"id":"openstack%2Fironic~726978","triplet_id":"openstack%2Fironic~master~Ie7c85db6ddceba0206f95dfd23cfe30447eb7068","project":"openstack/ironic","branch":"master","hashtags":[],"change_id":"Ie7c85db6ddceba0206f95dfd23cfe30447eb7068","subject":"Add ussuri release notes version","status":"MERGED","created":"2020-05-11 23:54:08.000000000","updated":"2020-05-14 14:07:31.000000000","submitted":"2020-05-14 01:52:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726978-1589421130090-78abda74","meta_rev_id":"e0e66546db32b4f67a26342b4df3244686fe6e55","_number":726978,"virtual_id_number":726978,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-05-13 17:17:16.000000000","_account_id":30100,"name":"Huawei Ironic CI","email":"zhengxianhui@huawei.com","username":"huawei-ironic-driver-ci"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"date":"2020-05-14 14:07:31.000000000","post_submit":true,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-14 01:52:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":30100,"name":"Huawei Ironic CI","email":"zhengxianhui@huawei.com","username":"huawei-ironic-driver-ci"},{"value":2,"date":"2020-05-14 01:33:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-13 19:22:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":30100,"name":"Huawei Ironic CI","email":"zhengxianhui@huawei.com","username":"huawei-ironic-driver-ci"},{"value":1,"date":"2020-05-14 01:33:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30100,"name":"Huawei Ironic CI","email":"zhengxianhui@huawei.com","username":"huawei-ironic-driver-ci"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":30100,"name":"Huawei Ironic CI","email":"zhengxianhui@huawei.com","username":"huawei-ironic-driver-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-13 17:17:16.000000000","updated_by":{"_account_id":30100,"name":"Huawei Ironic CI","email":"zhengxianhui@huawei.com","username":"huawei-ironic-driver-ci"},"reviewer":{"_account_id":30100,"name":"Huawei Ironic CI","email":"zhengxianhui@huawei.com","username":"huawei-ironic-driver-ci"},"state":"REVIEWER"},{"updated":"2020-05-13 19:22:15.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-05-14 01:33:01.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-05-14 01:52:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-14 14:07:31.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"REVIEWER"}],"messages":[{"id":"e0a0945cbd032eea2bcdda96c1b87411c1d745e5","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-05-11 23:54:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"797f87ac8efeefb08be6413cbcd53ea7bcab8362","author":{"_account_id":30100,"name":"Huawei Ironic CI","email":"zhengxianhui@huawei.com","username":"huawei-ironic-driver-ci"},"date":"2020-05-11 23:54:42.000000000","message":"Patch Set 1:\n\nFAILURE \n\n* huawei-rack-pxe http://116.66.187.11/builds/Ironic-Unsuri-CI/192 : Build failed. \n- To recheck use \u0027run huawei-ironic-ci\u0027. \n- For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems","accounts_in_message":[],"_revision_number":1},{"id":"47f24328ab4aaf9dd4ec4f79c201c152bd50afbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 00:15:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9af3924c00649558fec37314afbb60d : FAILURE in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f381e894f2fa4ca5b10d86235441aa82 : SUCCESS in 20m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/618fb2df649a420387e9459140553616 : SUCCESS in 12m 34s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/48a308d6e773488a9f156a89f2c5b618 : SUCCESS in 7m 10s\n\nWarning:\n  Comments left for invalid file ironic/tests/unit/api/controllers/v1/test_node.py\n  Comments left for invalid file ironic/tests/unit/api/controllers/v1/test_chassis.py\n  Comments left for invalid file ironic/tests/unit/api/controllers/v1/test_port.py\n  Comments left for invalid file ironic/tests/unit/api/controllers/v1/test_driver.py\n  Comments left for invalid file ironic/tests/unit/api/controllers/v1/test_portgroup.py\n  Comments left for invalid file ironic/tests/unit/api/controllers/v1/test_conductor.py\n  Comments left for invalid file ironic/tests/unit/api/controllers/v1/test_deploy_template.py\n  Comments left for invalid file ironic/tests/unit/api/controllers/v1/test_volume.py\n  Comments left for invalid file ironic/tests/unit/api/controllers/v1/test_volume_target.py\n  Comments left for invalid file ironic/cmd/__init__.py\n  Comments left for invalid file ironic/common/pxe_utils.py\n  Comments left for invalid file ironic/tests/unit/api/controllers/v1/test_allocation.py\n  Comments left for invalid file ironic/tests/unit/api/controllers/v1/test_volume_connector.py\n  Comments left for invalid file ironic/tests/base.py\n  Comments left for invalid file ironic/tests/unit/test_base.py","accounts_in_message":[],"_revision_number":1},{"id":"29444f48193d1fd2a67ec45083f1352112cf54a6","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-05-12 04:50:51.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac/2020-05-11:18:53/726978/1/ : FAILURE in 58m 12s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac/2020-05-11:18:53/726978/1/ : SUCCESS in 53m 09s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac/2020-05-11:18:53/726978/1/ : SUCCESS in 53m 49s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish/2020-05-11:18:53/726978/1/ : SUCCESS in 49m 53s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish/2020-05-11:18:53/726978/1/ : SUCCESS in 49m 26s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi/2020-05-11:18:53/726978/1/ : FAILURE in 1h 08m 01s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi/2020-05-11:18:53/726978/1/ : SUCCESS in 52m 30s\n- dellemc-poweredge-XE-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-XE-UEFI-tempest-dsvm-ironic-redfish/2020-05-11:18:53/726978/1/ : SUCCESS in 47m 59s\n- dellemc-poweredge-R7515-R6515-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dsvm-ironic-idrac/2020-05-11:18:53/726978/1/ : FAILURE in 50m 40s\n- dellemc-poweredge-R7515-R6515-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-UEFI-tempest-dsvm-ironic-redfish/2020-05-11:18:53/726978/1/ : SUCCESS in 46m 35s","accounts_in_message":[],"_revision_number":1},{"id":"4900c6bc4d534af310435713d4f503908f217a74","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-05-12 09:53:17.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nWhile the CI is broken, one minor issue inline.","accounts_in_message":[],"_revision_number":1},{"id":"1b8341ec08beff29dd09a07d72d6d3dd276795de","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-05-13 08:00:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"949a6b88a4c35ef9d557630690e7b6530b82e77d","author":{"_account_id":30100,"name":"Huawei Ironic CI","email":"zhengxianhui@huawei.com","username":"huawei-ironic-driver-ci"},"date":"2020-05-13 08:00:56.000000000","message":"Patch Set 1:\n\nFAILURE \n\n* huawei-rack-pxe http://116.66.187.11/builds/Ironic-Unsuri-CI/217 : Build failed. \n- To recheck use \u0027run huawei-ironic-ci\u0027. \n- For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems","accounts_in_message":[],"_revision_number":1},{"id":"cfb43658414234d471c8b5207789cfe33dee4830","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 08:25:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40ae51917b9a4effac6c6a902afcb67e : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b67e83b2bac54542b63a80d3e46749a6 : SUCCESS in 21m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/369bdee91d3a4b51b16015e27920e1ae : SUCCESS in 13m 53s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/15fa035501db45b8be7ec00bede76a59 : SUCCESS in 9m 13s","accounts_in_message":[],"_revision_number":1},{"id":"59056dfbc65b3dbf735590118b7677be97c2c9c8","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-05-13 17:16:26.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"14f5913552dab2a6b7a2cf4eac5c8d3527d3d000","author":{"_account_id":30100,"name":"Huawei Ironic CI","email":"zhengxianhui@huawei.com","username":"huawei-ironic-driver-ci"},"date":"2020-05-13 17:17:16.000000000","message":"Patch Set 2:\n\nFAILURE \n\n* huawei-rack-pxe http://116.66.187.11/builds/Ironic-Unsuri-CI/235 : Build failed. \n- To recheck use \u0027run huawei-ironic-ci\u0027. \n- For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems","accounts_in_message":[],"_revision_number":2},{"id":"18868d3b966da6e11b9583746659138eb0cae506","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 17:39:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb180c49492a457698898dfb664c16a1 : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c07af1f57e314358b9ecdace37e1a71a : SUCCESS in 20m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4803b4cd65a9478ca2bd653382a1efc1 : SUCCESS in 13m 58s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/755f6d69bb124fd793869ae5cd425542 : SUCCESS in 11m 57s","accounts_in_message":[],"_revision_number":2},{"id":"0d97b03cbd66b816ba8ac0eadb68b9c295197f24","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-05-13 19:22:15.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5dc4acdc30154707fbfef88ebb3c95d64bafbbdd","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-05-14 01:33:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ed2de97e73ce06a258215ee988e998fd4b559958","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 01:33:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3c007980df09cbabf27c348bd9b0b7e21b15196b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 01:52:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/483252fade8649a8934191b556250f1c : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b04fd3588ba41deab8ffe02a5735b48 : SUCCESS in 18m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50cadd491e9b441c9aa6ca1153045c20 : SUCCESS in 11m 11s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/8b68e29ee9c3432386460f17d66b809b : SUCCESS in 9m 41s","accounts_in_message":[],"_revision_number":2},{"id":"d8639348d7ced5099d12673da2e2426cb59a04f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 01:52:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ab5caac4e530ff50abb91b9b970382b8a844a233","author":{"_account_id":30100,"name":"Huawei Ironic CI","email":"zhengxianhui@huawei.com","username":"huawei-ironic-driver-ci"},"date":"2020-05-14 01:52:54.000000000","message":"Patch Set 2:\n\nFAILURE \n\n* huawei-rack-pxe http://116.66.187.11/builds/Ironic-Unsuri-CI/238 : Build failed. \n- To recheck use \u0027run huawei-ironic-ci\u0027. \n- For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems","accounts_in_message":[],"_revision_number":2},{"id":"956fa59a68a7ebf42c61782a6c3f7d7086ed550b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 01:53:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5526de775dd94f3bad6a213ec02eca15 : SUCCESS in 1m 16s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/999e6fe1382d43e599875d1313769d9c : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2},{"id":"01d92dfc7c686bc588bede3eab6581b313135a25","author":{"_account_id":30100,"name":"Huawei Ironic CI","email":"zhengxianhui@huawei.com","username":"huawei-ironic-driver-ci"},"date":"2020-05-14 07:42:27.000000000","message":"Patch Set 2:\n\nFAILURE \n\n* huawei-rack-pxe http://116.66.187.11/builds/Ironic-Unsuri-CI/239 : Build failed. \n- To recheck use \u0027run huawei-ironic-ci\u0027. \n- For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems","accounts_in_message":[],"_revision_number":2},{"id":"860878de21241311f02e25016cc4fb62b007fd05","author":{"_account_id":30100,"name":"Huawei Ironic CI","email":"zhengxianhui@huawei.com","username":"huawei-ironic-driver-ci"},"date":"2020-05-14 10:36:29.000000000","message":"Patch Set 2:\n\nSUCCESS \n\n* huawei-rack-pxe http://116.66.187.11/builds/Ironic-Unsuri-CI/240 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"e0e66546db32b4f67a26342b4df3244686fe6e55","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-05-14 14:07:31.000000000","message":"Patch Set 2:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac/2020-05-13:12:16/726978/2/ : SUCCESS in 53m 43s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac/2020-05-13:12:16/726978/2/ : SUCCESS in 57m 09s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac/2020-05-13:12:16/726978/2/ : SUCCESS in 54m 31s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish/2020-05-13:12:16/726978/2/ : SUCCESS in 48m 06s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish/2020-05-13:12:16/726978/2/ : SUCCESS in 57m 09s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi/2020-05-13:12:16/726978/2/ : SUCCESS in 57m 53s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi/2020-05-13:12:16/726978/2/ : SUCCESS in 52m 05s\n- dellemc-poweredge-XE-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-XE-UEFI-tempest-dsvm-ironic-redfish/2020-05-13:12:16/726978/2/ : SUCCESS in 47m 21s\n- dellemc-poweredge-R7515-R6515-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dsvm-ironic-idrac/2020-05-13:12:16/726978/2/ : FAILURE in 56m 37s\n- dellemc-poweredge-R7515-R6515-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-UEFI-tempest-dsvm-ironic-redfish/2020-05-13:12:16/726978/2/ : SUCCESS in 46m 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"893029493ac4a2c018964b752ce86e741f4886dc","revisions":{"6730a628ab1b89af0565f2f56d877c3075afe329":{"kind":"REWORK","_number":1,"created":"2020-05-11 23:54:08.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/78/726978/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/78/726978/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/78/726978/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/78/726978/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/78/726978/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/78/726978/1"}}},"commit":{"parents":[{"commit":"855986c0244d31f02cc6712b4669b2069ae3bcc1","subject":"Merge \"Add timeout and retries to JSON RPC client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/855986c0244d31f02cc6712b4669b2069ae3bcc1"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-05-11 23:53:56.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-05-11 23:53:56.000000000","tz":-420},"subject":"Add ussuri release notes version","message":"Add ussuri release notes version\n\nChange-Id: Ie7c85db6ddceba0206f95dfd23cfe30447eb7068\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/6730a628ab1b89af0565f2f56d877c3075afe329"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/6730a628ab1b89af0565f2f56d877c3075afe329"}]},"branch":"refs/heads/master"},"893029493ac4a2c018964b752ce86e741f4886dc":{"kind":"REWORK","_number":2,"created":"2020-05-13 17:16:26.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/78/726978/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/78/726978/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/78/726978/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/78/726978/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/78/726978/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/78/726978/2"}}},"commit":{"parents":[{"commit":"855986c0244d31f02cc6712b4669b2069ae3bcc1","subject":"Merge \"Add timeout and retries to JSON RPC client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/855986c0244d31f02cc6712b4669b2069ae3bcc1"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-05-11 23:53:56.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-05-13 17:16:24.000000000","tz":0},"subject":"Add ussuri release notes version","message":"Add ussuri release notes version\n\nChange-Id: Ie7c85db6ddceba0206f95dfd23cfe30447eb7068\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/893029493ac4a2c018964b752ce86e741f4886dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/893029493ac4a2c018964b752ce86e741f4886dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
