)]}'
{"id":"openstack%2Fironic~732736","triplet_id":"openstack%2Fironic~master~I72990b22a8dfab7c3abe04d976ee0118df015b6e","project":"openstack/ironic","branch":"master","topic":"fix-3rdpdm","hashtags":[],"change_id":"I72990b22a8dfab7c3abe04d976ee0118df015b6e","subject":"Enforce autospec in test_agent_base module","status":"MERGED","created":"2020-06-02 16:32:40.000000000","updated":"2020-06-04 09:33:53.000000000","submitted":"2020-06-04 09:32:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"732736-1591263120144-24af4908","meta_rev_id":"7e580cf1be7a04b2bae9438746905ccf8c511c97","_number":732736,"virtual_id_number":732736,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-06-02 16:32:56.000000000","_account_id":30100,"name":"Huawei Ironic CI","email":"zhengxianhui@huawei.com","username":"huawei-ironic-driver-ci"},{"value":0,"date":"2020-06-02 21:08:48.000000000","_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-04 09:31:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":30100,"name":"Huawei Ironic CI","email":"zhengxianhui@huawei.com","username":"huawei-ironic-driver-ci"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-02 19:16:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":2,"date":"2020-06-04 09:12:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":30100,"name":"Huawei Ironic CI","email":"zhengxianhui@huawei.com","username":"huawei-ironic-driver-ci"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":1,"date":"2020-06-04 09:12:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30100,"name":"Huawei Ironic CI","email":"zhengxianhui@huawei.com","username":"huawei-ironic-driver-ci"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30100,"name":"Huawei Ironic CI","email":"zhengxianhui@huawei.com","username":"huawei-ironic-driver-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-02 16:32:56.000000000","updated_by":{"_account_id":30100,"name":"Huawei Ironic CI","email":"zhengxianhui@huawei.com","username":"huawei-ironic-driver-ci"},"reviewer":{"_account_id":30100,"name":"Huawei Ironic CI","email":"zhengxianhui@huawei.com","username":"huawei-ironic-driver-ci"},"state":"REVIEWER"},{"updated":"2020-06-02 19:16:01.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2020-06-02 21:08:48.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"REVIEWER"},{"updated":"2020-06-04 09:12:51.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-06-04 09:31:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7586293a974350f548099e9b765dbd4dac39f0d6","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-06-02 16:32:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"52c09153d2c8c3c0dcf0c1c8a77bb05e18b2dc3c","author":{"_account_id":30100,"name":"Huawei Ironic CI","email":"zhengxianhui@huawei.com","username":"huawei-ironic-driver-ci"},"date":"2020-06-02 16:32:56.000000000","message":"Patch Set 1:\n\nFAILURE \n\n* huawei-rack-pxe http://116.66.187.11/builds/Ironic-Unsuri-CI/525 : Build failed. \n- To recheck use \u0027run huawei-ironic-ci\u0027. \n- For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems","accounts_in_message":[],"_revision_number":1},{"id":"3090f16bbcd4249b74e0346c684aceb8bf1cf1ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 17:54:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e6ae20d549744594b0634c5fdafa77a6 : SUCCESS in 8m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/820387d5e9d6418ba5981f09f387e5f2 : SUCCESS in 6m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb4106dc2a8e4154961b9a6fec41fd19 : SUCCESS in 13m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b25a7abd9663433d8e856cc6b2629527 : SUCCESS in 7m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f4e036565774d70bc0e6b2bb9d0ab45 : SUCCESS in 12m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1aeefc0ecf242a1bb7bdf54771eedef : SUCCESS in 10m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c74b69e21c034a4181c5082f4effbd6f : SUCCESS in 11m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a4559330d804e9f98c92d9a4d4efbed : SUCCESS in 7m 06s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/d102c444c43d4e118cb61d525b78373a : SUCCESS in 8m 04s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/6b32a2e7a4b04222980a670486f59455 : SUCCESS in 1h 20m 08s\n- ironic-inspector-tempest-managed https://zuul.opendev.org/t/openstack/build/01838d1222354f15baca5f817224c414 : SUCCESS in 41m 44s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fb15d00829d046e19edc8494f3bface2 : SUCCESS in 17m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"11b5aedfba159fa2d142c6e866fb0e441b31edf2","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-06-02 19:16:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fe010945fb38cbdf664d29850f7c889160f4194e","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-06-02 21:08:48.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac/2020-06-02:11:33/732736/1/ : FAILURE in 58m 41s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac/2020-06-02:11:33/732736/1/ : SUCCESS in 53m 23s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac/2020-06-02:11:33/732736/1/ : FAILURE in 1h 46m 53s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish/2020-06-02:11:33/732736/1/ : SUCCESS in 50m 15s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish/2020-06-02:11:33/732736/1/ : SUCCESS in 1h 02m 09s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi/2020-06-02:11:33/732736/1/ : SUCCESS in 49m 47s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi/2020-06-02:11:33/732736/1/ : SUCCESS in 53m 07s\n- dellemc-poweredge-XE-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-XE-UEFI-tempest-dsvm-ironic-redfish/2020-06-02:11:33/732736/1/ : SUCCESS in 49m 16s\n- dellemc-poweredge-R7515-R6515-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dsvm-ironic-idrac/2020-06-02:11:33/732736/1/ : SUCCESS in 57m 07s\n- dellemc-poweredge-R7515-R6515-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-UEFI-tempest-dsvm-ironic-redfish/2020-06-02:11:33/732736/1/ : SUCCESS in 57m 30s","accounts_in_message":[],"_revision_number":1},{"id":"588f8ba4b0832b846f26741471bb2cf6573ca185","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-06-04 09:12:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6cfe043b2dbb576598a39ba600391f1eaba8b06b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 09:13:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4aaae86a02e8429f074e4eae12e8a837ff6340e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 09:31:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/21842ae8de8546639964f6f17e32f6b5 : SUCCESS in 5m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/51347a2295944943b09df89a036699de : SUCCESS in 8m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39368d1ce37b4a2b9c891a4eb4d52ce1 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b8488c8e41741179ecd85ca295b7db9 : SUCCESS in 10m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df7eff84517a4184a4ce2dbefbd3e70a : SUCCESS in 10m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41a7974a4b43414ea350ab5a67bd45e4 : SUCCESS in 17m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c91801697e334016b4e2cf53633563c5 : SUCCESS in 7m 38s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/50c52d1381204b4bbe8ed08fa43d839d : SUCCESS in 8m 47s","accounts_in_message":[],"_revision_number":1},{"id":"752b61a6093a8775accea80fdfd1f049edbc91d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 09:32:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f846b61451ff57734c7b6e342f8e2e566cf6dd48","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 09:33:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/4bc383140b094a6397b6533a7886e3f3 : SUCCESS in 1m 28s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebde6464767d4909ba8c3e387cdb6b20 : SUCCESS in 1m 29s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3948333b9ed2485ab5f6d1792538c758 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"63bd40c89ad1e2ae3d6a7ca261793e7b49dd73cf","revisions":{"63bd40c89ad1e2ae3d6a7ca261793e7b49dd73cf":{"kind":"REWORK","_number":1,"created":"2020-06-02 16:32:40.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/36/732736/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/36/732736/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/36/732736/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/36/732736/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/36/732736/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/36/732736/1"}}},"commit":{"parents":[{"commit":"3daffe07a834ae4814e932a7ab5fb006bff21cdb","subject":"Enforce autospec in test_agent module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/3daffe07a834ae4814e932a7ab5fb006bff21cdb"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-06-02 16:32:02.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-06-02 16:32:02.000000000","tz":120},"subject":"Enforce autospec in test_agent_base module","message":"Enforce autospec in test_agent_base module\n\nAnd remove corresponding H210 filters.\n\nChange-Id: I72990b22a8dfab7c3abe04d976ee0118df015b6e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/63bd40c89ad1e2ae3d6a7ca261793e7b49dd73cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/63bd40c89ad1e2ae3d6a7ca261793e7b49dd73cf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
