)]}'
{"id":"openstack%2Fironic~744539","triplet_id":"openstack%2Fironic~stable%2Ftrain~I13da6275c04ffc6237a7f2edf25c03b4ddee936a","project":"openstack/ironic","branch":"stable/train","hashtags":[],"change_id":"I13da6275c04ffc6237a7f2edf25c03b4ddee936a","subject":"Update number of VM on ironic-base","status":"MERGED","created":"2020-08-03 19:00:27.000000000","updated":"2020-08-07 15:03:44.000000000","submitted":"2020-08-07 14:57:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744539-1596812262921-453d58e8","meta_rev_id":"2a590c5b048361bc90c77919f4701e87ae042441","_number":744539,"virtual_id_number":744539,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-08-06 14:49:35.000000000","_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"date":"2020-08-07 04:48:05.000000000","_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-07 14:57:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":2,"date":"2020-08-07 11:07:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":1,"date":"2020-08-07 11:07:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-03 19:00:27.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2020-08-06 14:49:35.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-08-07 04:48:05.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"REVIEWER"},{"updated":"2020-08-07 11:07:02.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-08-07 14:57:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"981212734e5c56be93f538504ab977280c5e0c14","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-08-03 19:00:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"51363bbf9df3d0fb780a83c6744fe16796b54e9c","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-08-03 19:04:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fe2b1b99d1d83efe3236cd26aaeb7f299652b977","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 20:52:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/250d25b5668f48859fa0add355c19816 : SUCCESS in 13m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af3ad622c6644d38a3dd74a1e8ee803c : FAILURE in 9m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/901d4dbfdb5c4d5db06bc1d7f5040f53 : SUCCESS in 8m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ae6ee08dda4b4b06830dba11722b474a : SUCCESS in 12m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/068d202372a3496786ab85c65e90cdc4 : SUCCESS in 12m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/547c40a3c4f643529bcfb58011cca770 : SUCCESS in 10m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/169e4525a1ab4102a581cde9832c91a0 : SUCCESS in 17m 20s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/2be760b0382842f881d4a1b1e3ba2676 : SUCCESS in 18m 02s\n- ironic-tox-unit-with-driver-libs-python3 https://zuul.opendev.org/t/openstack/build/a214d876d467417b940762387180c6b1 : SUCCESS in 9m 18s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/d24e26149e1c44f28a775dc4ce8ba291 : SUCCESS in 1h 07m 02s\n- ironic-tempest-functional-python2 https://zuul.opendev.org/t/openstack/build/32e9d1c918fc4990bdea64d73ebebc06 : SUCCESS in 23m 12s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/52153e8234a549369d3b61f7fbada59a : SUCCESS in 31m 09s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/971530e4bb1b4a25a15f8ebddca83707 : SUCCESS in 1h 35m 39s (non-voting)\n- ironic-grenade-dsvm-multinode-multitenant https://zuul.opendev.org/t/openstack/build/347f07b21b0d44b69a1b1582a2aa4fa8 : FAILURE in 55m 05s (non-voting)\n- ironic-tempest-ipa-partition-pxe_ipmitool-tinyipa-python3 https://zuul.opendev.org/t/openstack/build/45c6cd3d0d2344af9f39188208416337 : SUCCESS in 53m 34s\n- ironic-tempest-ipa-partition-redfish-tinyipa https://zuul.opendev.org/t/openstack/build/6778692af4a644f0aaaa629aae44538f : SUCCESS in 57m 20s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/b16621d60a864e7f9ba224bbe1b1e000 : SUCCESS in 44m 00s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/663e01f75bf34f23b64463bc61c38332 : POST_FAILURE in 1h 40m 07s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/62986e3ecd6c4f04a4ada6668638c6c4 : SUCCESS in 1h 07m 16s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/bbb57ae98366444a8b295e24d84a639d : SUCCESS in 1h 03m 58s\n- ironic-tempest-ipa-partition-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/3bb408402d8d4bada7f7ca564cb8e58a : SUCCESS in 59m 42s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/278501820f7e4bcca3519d77e58ed5ca : SUCCESS in 34m 18s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/952885a0c3944f54984474827038eb54 : SUCCESS in 51m 54s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/35c959fa961149a4a0e92b38340256e4 : SUCCESS in 5m 44s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-pxe_snmp-tinyipa https://zuul.opendev.org/t/openstack/build/143a04e6c6104c74883d5748216c5171 : SUCCESS in 45m 38s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/3593521d37a6486f85224df954f24570 : SUCCESS in 50m 36s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/7f825864c9564f72929862882396c62d : SUCCESS in 22m 28s (non-voting)\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/5f992e1ef3a742359b5714e432b92602 : RETRY_LIMIT in 37m 34s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/6cc81032a93f46ed9a26ac48780e15bf : SUCCESS in 59m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5c39a0bffd91b669eb3de7b1ddd0a16640373102","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-08-03 23:06:29.000000000","message":"Patch Set 2:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac/2020-08-03:14:04/744539/2/ : FAILURE in 1h 15m 43s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac/2020-08-03:14:04/744539/2/ : FAILURE in 50m 23s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac/2020-08-03:14:04/744539/2/ : FAILURE in 1h 18m 34s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish/2020-08-03:14:04/744539/2/ : SUCCESS in 1h 02m 18s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish/2020-08-03:14:04/744539/2/ : SUCCESS in 1h 20m 08s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi/2020-08-03:14:04/744539/2/ : SUCCESS in 56m 02s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi/2020-08-03:14:04/744539/2/ : FAILURE in 1h 47m 37s\n- dellemc-poweredge-XE-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-XE-UEFI-tempest-dsvm-ironic-redfish/2020-08-03:14:04/744539/2/ : SUCCESS in 1h 01m 36s\n- dellemc-poweredge-R7515-R6515-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dsvm-ironic-idrac/2020-08-03:14:04/744539/2/ : FAILURE in 1h 03m 15s\n- dellemc-poweredge-R7515-R6515-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-UEFI-tempest-dsvm-ironic-redfish/2020-08-03:14:04/744539/2/ : FAILURE in 48m 25s","accounts_in_message":[],"_revision_number":2},{"id":"36a462a280849a7a105859969cf899cc647d8c4e","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-08-06 14:49:35.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8dd706703ba1e0289df8cb1c8e89172994325d36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 17:09:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1cab57fb07524cdf98cb20154656ced3 : SUCCESS in 9m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/60df6fa9c2b14634bfdf140bc6a2c3f0 : SUCCESS in 11m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bad8947b1a4c4704937d30cd2a471787 : SUCCESS in 7m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cafdd71011044ed2b81abcfa7d5de303 : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f298c8595dd45998dd14e4a5d1ceb7b : SUCCESS in 10m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a0302a4886f54a0aa9efbad8e4a59709 : SUCCESS in 12m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0929301a11b84e378aee2c82f0374b18 : SUCCESS in 15m 24s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/21855f8d1a3d474e8af82b2f0024f137 : SUCCESS in 8m 46s\n- ironic-tox-unit-with-driver-libs-python3 https://zuul.opendev.org/t/openstack/build/9e11773f642145c98dadbff4e7612aa6 : SUCCESS in 8m 09s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/ee6382f9f8204ec1979dc2c71e697f4c : SUCCESS in 1h 15m 19s\n- ironic-tempest-functional-python2 https://zuul.opendev.org/t/openstack/build/bef5f72104d0496599aacd77597b452c : SUCCESS in 26m 58s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/e4d4b7e0b0e54cc482b2baa7992b79e0 : SUCCESS in 24m 12s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/5373236ffe2f431bb7db1e88ce6625fa : SUCCESS in 1h 51m 53s (non-voting)\n- ironic-grenade-dsvm-multinode-multitenant https://zuul.opendev.org/t/openstack/build/27858bd22ab2433e82638fbcee01c41e : FAILURE in 54m 31s (non-voting)\n- ironic-tempest-ipa-partition-pxe_ipmitool-tinyipa-python3 https://zuul.opendev.org/t/openstack/build/ec4f505f2d2d4d7ca832b55c84feee55 : SUCCESS in 42m 57s\n- ironic-tempest-ipa-partition-redfish-tinyipa https://zuul.opendev.org/t/openstack/build/a8db42ee56bb4844bed548fb73b04efd : SUCCESS in 40m 26s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f98cd6c810b64176ad4d22fc31d6c304 : SUCCESS in 48m 47s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/4ddb77e2a6924f0ca9c075a4089367f9 : FAILURE in 1h 38m 19s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4c1dd3c74177444cb3bc740e06e40fa2 : SUCCESS in 1h 21m 07s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/5f92d89694c44a6e9507921cc5a2ca13 : SUCCESS in 1h 06m 00s\n- ironic-tempest-ipa-partition-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/112ffd9ab7b04504ad83c208ebe3cad9 : SUCCESS in 1h 08m 45s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/c7229195df5f44f08a32a87363dd2c95 : SUCCESS in 48m 38s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/d47a61d4d92a4ee4b60317079ff34aed : SUCCESS in 49m 15s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/f061415061d0413794a3c681f502b498 : SUCCESS in 5m 00s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-pxe_snmp-tinyipa https://zuul.opendev.org/t/openstack/build/8813c672c4c74058a169866768ff8142 : SUCCESS in 52m 17s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/426e6109647a4618b8c37719c865ffeb : SUCCESS in 52m 49s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/29ba7c2c844941239e5580f182265aab : SUCCESS in 24m 04s (non-voting)\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/c1b3e7d1ef894675900a8de59db1d51e : RETRY_LIMIT in 25m 24s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/a9531769fa6745cea40c8d91a83d6057 : SUCCESS in 1h 02m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c21b6d586626b21a53045674d581a5dd32d3da7b","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-08-07 04:48:05.000000000","message":"Patch Set 2:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac/2020-08-06:09:49/744539/2/ : FAILURE in 1h 11m 56s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac/2020-08-06:09:49/744539/2/ : FAILURE in 51m 11s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac/2020-08-06:09:49/744539/2/ : FAILURE in 1h 42m 55s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish/2020-08-06:09:49/744539/2/ : SUCCESS in 58m 14s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish/2020-08-06:09:49/744539/2/ : SUCCESS in 1h 11m 01s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi/2020-08-06:09:49/744539/2/ : FAILURE in 1h 04m 03s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi/2020-08-06:09:49/744539/2/ : FAILURE in 1h 45m 57s\n- dellemc-poweredge-XE-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-XE-UEFI-tempest-dsvm-ironic-redfish/2020-08-06:09:49/744539/2/ : SUCCESS in 57m 45s\n- dellemc-poweredge-R7515-R6515-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dsvm-ironic-idrac/2020-08-06:09:49/744539/2/ : FAILURE in 1h 01m 35s\n- dellemc-poweredge-R7515-R6515-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-UEFI-tempest-dsvm-ironic-redfish/2020-08-06:09:49/744539/2/ : FAILURE in 51m 31s","accounts_in_message":[],"_revision_number":2},{"id":"3d4205cf7bc891195311c5e0f3d5373f8cc25f3e","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-08-07 11:07:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3ca14cc751bf9c67ffff34938af512872c457d6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 11:07:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d17754adc9eac70d84d7a3e33b74afe630331d3c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 14:57:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f174709c31b4202467aee34e893b3550d0e2b6e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 14:57:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa37ea5efe814613b4618b2a158b527e : SUCCESS in 8m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c744a5102ec4cfb8e9169d8ed0e69d4 : SUCCESS in 6m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1bbc8663ef544af59b98561077cd83e9 : SUCCESS in 9m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a984e6a75c64382a6cba4ddf0a8943c : SUCCESS in 10m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/afd141e6a2974045a557594f12361434 : SUCCESS in 10m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddc675953c7643c58bad2e7c5d16e8ba : SUCCESS in 16m 07s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/b5873ef5c1b245ac81cd7d8ca2592b0b : SUCCESS in 9m 26s\n- ironic-tox-unit-with-driver-libs-python3 https://zuul.opendev.org/t/openstack/build/52642cc2065743ea8247de42af91cf5f : SUCCESS in 11m 27s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/a78402128ac94a839b9f80bd3c51ce12 : SUCCESS in 1h 17m 33s\n- ironic-tempest-functional-python2 https://zuul.opendev.org/t/openstack/build/3a140bf41c91410c8ec0a44eafe46a45 : SUCCESS in 27m 21s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/a008268f250446508a4a4a75052b16f6 : SUCCESS in 29m 30s\n- ironic-tempest-ipa-partition-pxe_ipmitool-tinyipa-python3 https://zuul.opendev.org/t/openstack/build/2e631025722f436da50e0bffe9f7610c : SUCCESS in 1h 00m 42s\n- ironic-tempest-ipa-partition-redfish-tinyipa https://zuul.opendev.org/t/openstack/build/05c1b2ab59274426a3be9ea0703e4f17 : SUCCESS in 1h 07m 30s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f002bbbab97b4284acc15503e974c260 : SUCCESS in 1h 23m 43s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/548e3f4e573d4d94a9b8ecb0634457b0 : SUCCESS in 1h 23m 45s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/04679fcc8069426a8cc260cbacb2f10c : SUCCESS in 1h 19m 37s\n- ironic-tempest-ipa-partition-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/c97361f26f494f04b7c8465e943f9854 : SUCCESS in 1h 08m 14s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/cc6af291e8d145c99a95531f715df7df : SUCCESS in 41m 09s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/b76666fe191c43f4a1eff841cc85b002 : SUCCESS in 53m 58s","accounts_in_message":[],"_revision_number":2},{"id":"2a590c5b048361bc90c77919f4701e87ae042441","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 15:03:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec49020305c344838b5b8b9338dc9daf : SUCCESS in 2m 30s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3df540a35d567ea78b04afb3188c76c64a7e7fae","revisions":{"bd702894caa1b4c169402a8e4e4738b638c6a4f7":{"kind":"REWORK","_number":1,"created":"2020-08-03 19:00:27.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/39/744539/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/744539/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/744539/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/744539/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/744539/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/744539/1"}}},"commit":{"parents":[{"commit":"a54376cef0e8dd32c670fb359a5f7bd525c2c92e","subject":"Remove locks before RPC bus is started","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/a54376cef0e8dd32c670fb359a5f7bd525c2c92e"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2020-07-18 08:55:07.000000000","tz":120},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-08-03 19:00:21.000000000","tz":-420},"subject":"Update number of VM on ironic-base","message":"Update number of VM on ironic-base\n\nSince we merged the change to have partition and wholedisk\ntesting on basic_ops most of the jobs started requiring 2 VMs\nto run the tempest tets.\n\nLet\u0027s increase on the ironic-base so all jobs will be default to 2.\n\nRemoving IRONIC_VM_COUNT\u003d2 from jobs that uses ironic-base as parent.\n\nChange-Id: I13da6275c04ffc6237a7f2edf25c03b4ddee936a\n(cherry picked from commit dc87a189cb0d5bf2e96650ffd699fbaa972b7983)\n(cherry picked from commit 3730c92ce35ca30ff15d5d8310f4dd37092fbb14)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/bd702894caa1b4c169402a8e4e4738b638c6a4f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/bd702894caa1b4c169402a8e4e4738b638c6a4f7"}]},"branch":"refs/heads/stable/train"},"3df540a35d567ea78b04afb3188c76c64a7e7fae":{"kind":"REWORK","_number":2,"created":"2020-08-03 19:04:02.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/39/744539/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/744539/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/744539/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/744539/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/744539/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/744539/2"}}},"commit":{"parents":[{"commit":"a54376cef0e8dd32c670fb359a5f7bd525c2c92e","subject":"Remove locks before RPC bus is started","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/a54376cef0e8dd32c670fb359a5f7bd525c2c92e"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2020-07-18 08:55:07.000000000","tz":120},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-08-03 19:03:57.000000000","tz":-420},"subject":"Update number of VM on ironic-base","message":"Update number of VM on ironic-base\n\nSince we merged the change to have partition and wholedisk\ntesting on basic_ops most of the jobs started requiring 2 VMs\nto run the tempest tets.\n\nLet\u0027s increase on the ironic-base so all jobs will be default to 2.\n\nRemoving IRONIC_VM_COUNT\u003d2 from jobs that uses ironic-base as parent.\n\nChange-Id: I13da6275c04ffc6237a7f2edf25c03b4ddee936a\n(cherry picked from commit dc87a189cb0d5bf2e96650ffd699fbaa972b7983)\n(cherry picked from commit 3730c92ce35ca30ff15d5d8310f4dd37092fbb14)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/3df540a35d567ea78b04afb3188c76c64a7e7fae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/3df540a35d567ea78b04afb3188c76c64a7e7fae"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
