)]}'
{"id":"openstack%2Fironic~745935","triplet_id":"openstack%2Fironic~master~Icb6bb30d1859eb3787bb5f4df4ad97143a098017","project":"openstack/ironic","branch":"master","hashtags":[],"change_id":"Icb6bb30d1859eb3787bb5f4df4ad97143a098017","subject":"Fix time usage in unit tests for agent power interface","status":"MERGED","created":"2020-08-12 17:01:38.000000000","updated":"2020-08-13 15:11:38.000000000","submitted":"2020-08-13 14:58:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745935-1597330713981-c5357c5d","meta_rev_id":"bcb08decebcf0ccad8a173c43fc5c579aac2caf1","_number":745935,"virtual_id_number":745935,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-08-13 13:37:29.000000000","_account_id":30100,"name":"Huawei Ironic CI","email":"zhengxianhui@huawei.com","username":"huawei-ironic-driver-ci"},{"value":0,"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},{"value":0,"date":"2020-08-13 13:24:50.000000000","_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"date":"2020-08-12 22:14:51.000000000","_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-13 14:58:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"recommended":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"all":[{"value":0,"_account_id":30100,"name":"Huawei Ironic CI","email":"zhengxianhui@huawei.com","username":"huawei-ironic-driver-ci"},{"value":1,"date":"2020-08-12 17:07:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":2,"date":"2020-08-13 14:37:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-13 14:37:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":30100,"name":"Huawei Ironic CI","email":"zhengxianhui@huawei.com","username":"huawei-ironic-driver-ci"},{"value":0,"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":1,"date":"2020-08-13 14:38:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30100,"name":"Huawei Ironic CI","email":"zhengxianhui@huawei.com","username":"huawei-ironic-driver-ci"},{"value":0,"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":30100,"name":"Huawei Ironic CI","email":"zhengxianhui@huawei.com","username":"huawei-ironic-driver-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-12 17:07:11.000000000","updated_by":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"reviewer":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"state":"REVIEWER"},{"updated":"2020-08-12 22:14:51.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"REVIEWER"},{"updated":"2020-08-13 13:37:29.000000000","updated_by":{"_account_id":30100,"name":"Huawei Ironic CI","email":"zhengxianhui@huawei.com","username":"huawei-ironic-driver-ci"},"reviewer":{"_account_id":30100,"name":"Huawei Ironic CI","email":"zhengxianhui@huawei.com","username":"huawei-ironic-driver-ci"},"state":"REVIEWER"},{"updated":"2020-08-13 14:37:55.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2020-08-13 14:38:00.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-08-13 14:58:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"06a69c88ea7445e023e905d054d0fc8dac20c36c","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-08-12 17:01:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ece89180c3583118301877ac981c5cd362145905","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2020-08-12 17:07:11.000000000","message":"Patch Set 1: Code-Review+1\n\nLooks good, tests failing with rhel 8 now pass.","accounts_in_message":[],"_revision_number":1},{"id":"ecc443964c1bece3bdce94ea5449f17f2771dd15","author":{"_account_id":30100,"name":"Huawei Ironic CI","email":"zhengxianhui@huawei.com","username":"huawei-ironic-driver-ci"},"date":"2020-08-12 17:15:05.000000000","message":"Patch Set 1:\n\nSUCCESS \n\n* huawei-rack-pxe http://116.66.187.11/builds/Ironic-Unsuri-CI/1310 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"e5fbbf23830784efb884b1b595f6ca6ce7b33df6","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-08-12 17:45:01.000000000","message":"Patch Set 1:\n\nPasses now on Debian, so I think it is good. :)","accounts_in_message":[],"_revision_number":1},{"id":"d38db4edd53cb3389bee1fa3f08ebdaa5206e50a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 18:12:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4eac6c235a8944618459553ad7b45a91 : SUCCESS in 15m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f86704e680064816b7ea63879c33cec1 : SUCCESS in 9m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/285e2de8339a4af484ff0f50c93fb1f3 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3346d0994ef4dcf9941f71b7243d9e0 : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ebd50ecb9ec42a792291535450c1928 : SUCCESS in 8m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60c39928f3ac44c8bb6d7c0c9070d2c0 : SUCCESS in 15m 12s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/6f8e1c87a650434aa28f2f1f62250d67 : FAILURE in 12m 01s\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/ab9bc73471694ca796726c9ac1b1b33f : SUCCESS in 1h 03m 40s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a3875394e1364081a837a77bfb1fca49 : SUCCESS in 22m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b216e2b222f4f4b79476ca24a3d4735d4efc0c6d","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-08-12 22:14:51.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac/2020-08-12:12:01/745935/1/ : SUCCESS in 1h 00m 15s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac/2020-08-12:12:01/745935/1/ : SUCCESS in 1h 01m 21s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac/2020-08-12:12:01/745935/1/ : SUCCESS in 1h 26m 48s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish/2020-08-12:12:01/745935/1/ : SUCCESS in 55m 51s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish/2020-08-12:12:01/745935/1/ : SUCCESS in 1h 45m 37s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi/2020-08-12:12:01/745935/1/ : FAILURE in 57m 27s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi/2020-08-12:12:01/745935/1/ : SUCCESS in 58m 01s\n- dellemc-poweredge-XE-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-XE-UEFI-tempest-dsvm-ironic-redfish/2020-08-12:12:01/745935/1/ : SUCCESS in 51m 54s\n- dellemc-poweredge-R7515-R6515-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dsvm-ironic-idrac/2020-08-12:12:01/745935/1/ : FAILURE in 59m 44s\n- dellemc-poweredge-R7515-R6515-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-UEFI-tempest-dsvm-ironic-redfish/2020-08-12:12:01/745935/1/ : FAILURE in 41m 47s","accounts_in_message":[],"_revision_number":1},{"id":"76ef360902e33184c8e3f51e0f71fe3b967feeaa","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-08-13 13:24:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3af294116bf92cc61e6ab5ea8511acffe4efe85c","author":{"_account_id":30100,"name":"Huawei Ironic CI","email":"zhengxianhui@huawei.com","username":"huawei-ironic-driver-ci"},"date":"2020-08-13 13:37:29.000000000","message":"Patch Set 1:\n\nSUCCESS \n\n* huawei-rack-pxe http://116.66.187.11/builds/Ironic-Unsuri-CI/1323 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"d15ae15b0695879ae7562b2fa30ab4371eedc12d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 14:31:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/939b4fb70ede4005837becde4cede41a : SUCCESS in 8m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/26ee776347df436c94ac7de3aeb41894 : SUCCESS in 14m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5aa2c3386e840d0884e9c83514f37c2 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11eba6d674b4407cb7bd07d4e463df84 : SUCCESS in 10m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f6b4488d6e1404a91e1909e355e73f5 : SUCCESS in 9m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd3cb7ddf34b4b1a827aa29bdef0c922 : SUCCESS in 17m 05s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/39621371eeeb4ba7ae29a30b33ba6d19 : SUCCESS in 8m 27s\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/adb61905f2f748d9bf480d0d342b47c4 : SUCCESS in 57m 41s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/20e8f922748c4d2cbb70554caf449488 : SUCCESS in 23m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"33e3e9334a4df92061daf0b269ea6aa6444b3b37","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-08-13 14:37:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ee64ec9bd9a2901c646d3ccb11b707bd4e3dbe7e","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-08-13 14:37:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b5c486a3c4657cce26fb3c14f1c7b9ec0f3e8215","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-08-13 14:38:00.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6efe0bc26082dfa3fed8703b3f05d0e4603b3d32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 14:38:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5b42d68549a08e0f525cbe837a7823147caa6fbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 14:58:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/edaf1f8dfe6848139950e243b9ff3404 : SUCCESS in 9m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f60766227b447adb9b721851b69c197 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aafa0bf9084b498fb33fe60432d27771 : SUCCESS in 10m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/43b9b1441d8e4c56b07da87b430f8655 : SUCCESS in 9m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e7a64a1cce448f8933f011973d7ab2e : SUCCESS in 15m 36s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/e590e17715b047f1ac89be4484db2420 : SUCCESS in 8m 58s","accounts_in_message":[],"_revision_number":1},{"id":"422b8f824cc04473e538b48b00ebdc0efdc87f94","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 14:58:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"799589633003e7ec66c7999dc5d7c0a7ad09e520","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 15:02:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5302c1f3c2d413cbc832f874a83004d : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":1},{"id":"bcb08decebcf0ccad8a173c43fc5c579aac2caf1","author":{"_account_id":30100,"name":"Huawei Ironic CI","email":"zhengxianhui@huawei.com","username":"huawei-ironic-driver-ci"},"date":"2020-08-13 15:11:38.000000000","message":"Patch Set 1:\n\nSUCCESS \n\n* huawei-rack-pxe http://116.66.187.11/builds/Ironic-Unsuri-CI/1324 : SUCCESS","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0d94ef67cb5d2a82d4a838e99b63a155eb872a7f","revisions":{"0d94ef67cb5d2a82d4a838e99b63a155eb872a7f":{"kind":"REWORK","_number":1,"created":"2020-08-12 17:01:38.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/35/745935/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/35/745935/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/35/745935/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/35/745935/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/35/745935/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/35/745935/1"}}},"commit":{"parents":[{"commit":"3fd9b5588045ef66eb5cffeb4ac9a8614492c75b","subject":"Merge \"Adds support SUM based firmware update as deploy step\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/3fd9b5588045ef66eb5cffeb4ac9a8614492c75b"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-08-12 17:00:14.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-08-12 17:00:14.000000000","tz":-420},"subject":"Fix time usage in unit tests for agent power interface","message":"Fix time usage in unit tests for agent power interface\n\nThe agent power interface unit tests were asking for the now() time\nwhich is system user time relative. For example, for US-Pacific would\ndisplay 10 AM. Where as the rest of ironic was expecting 5 PM for the\ntest as the base code uses utcnow() instead of now().\n\nChange-Id: Icb6bb30d1859eb3787bb5f4df4ad97143a098017\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/0d94ef67cb5d2a82d4a838e99b63a155eb872a7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/0d94ef67cb5d2a82d4a838e99b63a155eb872a7f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
