)]}'
{"id":"openstack%2Fironic~747730","triplet_id":"openstack%2Fironic~bugfix%2F15.1~I72a422553ee992d313b83df091af2c9deb8393b5","project":"openstack/ironic","branch":"bugfix/15.1","hashtags":[],"change_id":"I72a422553ee992d313b83df091af2c9deb8393b5","subject":"Ansible deploy - Ignore invalid devices","status":"MERGED","created":"2020-08-24 13:47:13.000000000","updated":"2020-08-25 08:48:03.000000000","submitted":"2020-08-25 08:45:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747730-1598345107715-1fceb40a","meta_rev_id":"ca4baab49244e9a4a9fd1553b51bae3306cd7997","_number":747730,"virtual_id_number":747730,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-08-24 14:22:53.000000000","_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-25 08:45:07.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":2,"date":"2020-08-25 08:22:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":1,"date":"2020-08-25 08:22:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-24 14:22:53.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"REVIEWER"},{"updated":"2020-08-25 08:22:04.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-08-25 08:45:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ba19246c48f8c93e4ca597cf62dee69db43c7605","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-08-24 13:47:13.000000000","message":"Patch Set 1: Cherry Picked from branch bugfix/15.2.","accounts_in_message":[],"_revision_number":1},{"id":"5475398b51381779a865ae44be106727a78e61a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 14:19:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff6063cc81c34c128c20b083993f9ed5 : SUCCESS in 7m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/abeda1e97df94875bfdcbe30ce96116e : SUCCESS in 7m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2aed0b3f729c4f819d82d7d06e4d3c4a : SUCCESS in 8m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e730ee799ed43b1801705ed9e8ef81e : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4f85aefd0b34f52a072fbcc2d6f8a33 : SUCCESS in 7m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ee458c33d90418cb91daa1467cffae6 : SUCCESS in 16m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14ab52775a294988b594761a3ba947aa : SUCCESS in 6m 52s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/92835c5afe3c4dfbb5090c78e676e474 : SUCCESS in 7m 51s\n- ironic-grenade-dsvm-multinode-multitenant https://zuul.opendev.org/t/openstack/build/2482f265e34848e5a587f38b8236362b : FAILURE in 17m 36s (non-voting)\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/838311b3146e47af80528a832ba73d55 : SUCCESS in 6m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4f1a530f5c7d2f1fe02b779c5a813f8e6f2eca7f","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-08-24 14:22:53.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac/2020-08-24:08:47/747730/1/ : FAILURE in 15m 45s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac/2020-08-24:08:47/747730/1/ : FAILURE in 14m 54s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac/2020-08-24:08:47/747730/1/ : FAILURE in 14m 07s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish/2020-08-24:08:47/747730/1/ : FAILURE in 14m 21s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish/2020-08-24:08:47/747730/1/ : FAILURE in 22m 34s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi/2020-08-24:08:47/747730/1/ : FAILURE in 16m 06s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi/2020-08-24:08:47/747730/1/ : FAILURE in 13m 37s\n- dellemc-poweredge-XE-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-XE-UEFI-tempest-dsvm-ironic-redfish/2020-08-24:08:47/747730/1/ : FAILURE in 13m 34s\n- dellemc-poweredge-R7515-R6515-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dsvm-ironic-idrac/2020-08-24:08:47/747730/1/ : FAILURE in 13m 35s\n- dellemc-poweredge-R7515-R6515-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-UEFI-tempest-dsvm-ironic-redfish/2020-08-24:08:47/747730/1/ : FAILURE in 14m 22s","accounts_in_message":[],"_revision_number":1},{"id":"b72311b838a1a20276ab7d553534cd8c5f5aeb37","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-08-25 08:22:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"523d6e645eb01a076fa93fde8e70dca430b755d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 08:22:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"178ccd5aa0cd61220436fd4c024182c01637b9b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 08:45:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"eaf3f83039e0c827bea96dae2accf6a0cef80739","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 08:45:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c83d4a19458748c79df484df8759c92a : SUCCESS in 9m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25d53f1f872d4b32973e828c7e66000f : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/638039caba7e4949a3b59349f166b22f : SUCCESS in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3457ee1abecb43edbd6464039994b898 : SUCCESS in 7m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02207ad6ef744302b730453f107edba8 : SUCCESS in 16m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9315befcefa245cda710218b171a44d5 : SUCCESS in 6m 43s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/5fe3eed46d9a47e3a798e2de5b446b64 : SUCCESS in 11m 02s","accounts_in_message":[],"_revision_number":1},{"id":"ca4baab49244e9a4a9fd1553b51bae3306cd7997","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 08:48:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebe74327e45040d9b66f0a15a96da20e : SUCCESS in 2m 25s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a48852e44ffe4198b7ae8af04313e4a9 : SUCCESS in 1m 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"03c772c77a5a24968ae7e8ffd15a46248968fb51","revisions":{"03c772c77a5a24968ae7e8ffd15a46248968fb51":{"kind":"REWORK","_number":1,"created":"2020-08-24 13:47:13.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/30/747730/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/30/747730/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/30/747730/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/30/747730/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/30/747730/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/30/747730/1"}}},"commit":{"parents":[{"commit":"df9b79907c96d2c027033cbde2d4a425915b030f","subject":"Use TLS for json_rpc when configured","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/df9b79907c96d2c027033cbde2d4a425915b030f"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-08-21 14:35:55.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-08-24 13:47:13.000000000","tz":0},"subject":"Ansible deploy - Ignore invalid devices","message":"Ansible deploy - Ignore invalid devices\n\nThis change updates the ansible deploy driver to exclude\ndevices with \"sr\", \"loop\", and \"mem\" which can appear in the\ndevices list for consideration as the root device.\n\nThis change effectively causes them to be ignored.\n\nChange-Id: I72a422553ee992d313b83df091af2c9deb8393b5\n(cherry picked from commit 0071d28460f1c0baa449d8f7920f4cf36e876bdc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/03c772c77a5a24968ae7e8ffd15a46248968fb51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/03c772c77a5a24968ae7e8ffd15a46248968fb51"}]},"branch":"refs/heads/bugfix/15.1"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
