)]}'
{"id":"openstack%2Fironic~755713","triplet_id":"openstack%2Fironic~master~Ie75f36d04e7084be921def3a8ee33b2a249434e6","project":"openstack/ironic","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"Ie75f36d04e7084be921def3a8ee33b2a249434e6","subject":"Imported Translations from Zanata","status":"MERGED","created":"2020-10-02 07:13:31.000000000","updated":"2020-10-05 09:20:09.000000000","submitted":"2020-10-05 09:18:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755713-1601889506566-c1ef629e","meta_rev_id":"6cf303f01238fcc14bd91c20e2d05d0898bcdbb8","_number":755713,"virtual_id_number":755713,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-10-03 20:15:08.000000000","_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-05 09:18:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-05 08:56:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-05 08:56:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-03 20:15:08.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"REVIEWER"},{"updated":"2020-10-05 08:56:47.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-10-05 09:18:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5ce28a9b4c62f789ac7228bb5e5e727d39b15348","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-02 07:13:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8714760caf27606c2d030756cb9bcb492a6ca3f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 07:42:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/939c4ccf1400421995747acf23723de2 : SUCCESS in 8m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df9e98470c9641fc96ae30369b8c4746 : SUCCESS in 17m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53bb41790f374df4b0ab26898c655fe1 : SUCCESS in 7m 04s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/fa50c256285e45afbc09e08f33a339e7 : SUCCESS in 9m 07s","accounts_in_message":[],"_revision_number":1},{"id":"5532c52ff6ee0fb433df72865e3a8f6be6ca86f2","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-10-02 09:58:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5887275fbbe774529feb8fe7499565f460ad499f","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-10-02 23:51:17.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac/2020-10-02:02:13/755713/1/ : SUCCESS in 1h 06m 31s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac/2020-10-02:02:13/755713/1/ : SUCCESS in 1h 06m 00s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac/2020-10-02:02:13/755713/1/ : FAILURE in 1h 20m 29s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish/2020-10-02:02:13/755713/1/ : FAILURE in 1h 06m 17s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish/2020-10-02:02:13/755713/1/ : FAILURE in 1h 10m 23s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi/2020-10-02:02:13/755713/1/ : FAILURE in 1h 11m 37s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi/2020-10-02:02:13/755713/1/ : FAILURE in 1h 46m 42s\n- dellemc-poweredge-XE-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-XE-UEFI-tempest-dsvm-ironic-redfish/2020-10-02:02:13/755713/1/ : FAILURE in 58m 11s\n- dellemc-poweredge-R7515-R6515-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dsvm-ironic-idrac/2020-10-02:02:13/755713/1/ : FAILURE in 1h 03m 54s\n- dellemc-poweredge-R7515-R6515-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-UEFI-tempest-dsvm-ironic-redfish/2020-10-02:02:13/755713/1/ : FAILURE in 52m 45s","accounts_in_message":[],"_revision_number":1},{"id":"91248ba126d2f34b040a79654c9abed9fa3af7d4","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-03 06:46:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b12b99a9a13710ae09118e45970ee71a83ce2fa9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-03 07:03:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36de08a02fef408da0b962edd6d48034 : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4c3befbd85e4a1792a92599f126cd34 : SUCCESS in 14m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ffbbaf64ed494ea9b3df62de20cee4a9 : SUCCESS in 7m 23s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/dfa99948ab5e43898915e4301f8090a0 : SUCCESS in 10m 53s","accounts_in_message":[],"_revision_number":2},{"id":"54265ab50f8064ccf00a5b9cc705dc5df6202737","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-10-03 20:15:08.000000000","message":"Patch Set 2:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac/2020-10-03:01:46/755713/2/ : SUCCESS in 1h 12m 45s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac/2020-10-03:01:46/755713/2/ : SUCCESS in 1h 07m 16s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac/2020-10-03:01:46/755713/2/ : FAILURE in 1h 26m 06s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish/2020-10-03:01:46/755713/2/ : FAILURE in 57m 29s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish/2020-10-03:01:46/755713/2/ : FAILURE in 1h 47m 18s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi/2020-10-03:01:46/755713/2/ : FAILURE in 1h 05m 23s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi/2020-10-03:01:46/755713/2/ : SUCCESS in 1h 08m 35s\n- dellemc-poweredge-XE-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-XE-UEFI-tempest-dsvm-ironic-redfish/2020-10-03:01:46/755713/2/ : FAILURE in 1h 11m 16s\n- dellemc-poweredge-R7515-R6515-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dsvm-ironic-idrac/2020-10-03:01:46/755713/2/ : FAILURE in 1h 03m 54s\n- dellemc-poweredge-R7515-R6515-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-UEFI-tempest-dsvm-ironic-redfish/2020-10-03:01:46/755713/2/ : FAILURE in 50m 46s","accounts_in_message":[],"_revision_number":2},{"id":"278c7e055e8f53e7833035a5718aad9ab8125278","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-10-05 08:56:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1535d5a101b96e46083aebb14e46aad5f73e2273","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 08:57:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8b6502e010ed3cfbca8e0c1ca563e0c0a09253c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 09:18:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c09d112ac03d41cbadf1014b1ffa2241 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffb1e47beed842ffb78388fd86707d2d : SUCCESS in 15m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a527a8de7948440fa5db06e6f94e7d8b : SUCCESS in 6m 42s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/53182fe13ef041e595c3bcc39ea4dfb1 : SUCCESS in 12m 16s","accounts_in_message":[],"_revision_number":2},{"id":"827160bccc92f5edc2e70aa4dc6a9973a9e54a61","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 09:18:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"273cf0939a5ad2f989703323885f8110a9af1cbe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 09:20:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42b0cd18beff4695aa8ce79efec8a2bf : SUCCESS in 1m 16s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cbce54ed76c94fc0b04b7a1811d20bf2 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a25da1c580460d552ad208f9344594040ec57318","revisions":{"b30a8291fd94038c1625193409f82e21501bb27a":{"kind":"REWORK","_number":1,"created":"2020-10-02 07:13:31.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/13/755713/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/13/755713/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/755713/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/755713/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/755713/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/13/755713/1"}}},"commit":{"parents":[{"commit":"33a34ce364ce28077a114b7d1a80fd719521580a","subject":"Merge \"Feat: add ibmc hardware info support for devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/33a34ce364ce28077a114b7d1a80fd719521580a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-02 07:13:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-02 07:13:28.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Ie75f36d04e7084be921def3a8ee33b2a249434e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/b30a8291fd94038c1625193409f82e21501bb27a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/b30a8291fd94038c1625193409f82e21501bb27a"}]},"branch":"refs/heads/master"},"a25da1c580460d552ad208f9344594040ec57318":{"kind":"REWORK","_number":2,"created":"2020-10-03 06:46:18.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/13/755713/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/13/755713/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/755713/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/755713/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/755713/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/13/755713/2"}}},"commit":{"parents":[{"commit":"8604f84fd7bda4e30d3f07005c4901f3662303a7","subject":"Merge \"Add Python3 wallaby unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/8604f84fd7bda4e30d3f07005c4901f3662303a7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-03 06:46:16.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-03 06:46:16.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Ie75f36d04e7084be921def3a8ee33b2a249434e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/a25da1c580460d552ad208f9344594040ec57318"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/a25da1c580460d552ad208f9344594040ec57318"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
