)]}'
{"id":"openstack%2Fironic~756574","triplet_id":"openstack%2Fironic~master~I23d1ddaf8ca4acfd1b51ade7d8d7ee476029cb7a","project":"openstack/ironic","branch":"master","topic":"40965","hashtags":[],"change_id":"I23d1ddaf8ca4acfd1b51ade7d8d7ee476029cb7a","subject":"Changes links so we do not have \"here\" as the link","status":"ABANDONED","created":"2020-10-07 19:17:46.000000000","updated":"2021-08-12 18:32:45.000000000","total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"acebc1be0871d61d780de0a7f515e5f285c11dcd","_number":756574,"virtual_id_number":756574,"owner":{"_account_id":32493,"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","username":"kafilat-adeleke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-10-08 14:23:20.000000000","_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"date":"2020-10-08 11:57:51.000000000","_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-10-08 08:40:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-08 08:40:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-08 11:57:51.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"REVIEWER"},{"updated":"2020-10-08 14:23:20.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"}],"messages":[{"id":"11b3d8da5da3a7819170831b0d3bcb1d0c0d373c","author":{"_account_id":32493,"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","username":"kafilat-adeleke"},"date":"2020-10-07 19:17:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"76d47bee30853a18b4d09ea0584011f787cdcc90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 19:36:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0ad491f7e4f468eb62d2e4dd153cd70 : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/220a4158511b420a99d1e3f8e44e7cfd : SUCCESS in 16m 24s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/43db831f5aae490a8eef01bd385be60d : SUCCESS in 9m 47s","accounts_in_message":[],"_revision_number":1},{"id":"9d703ca79bf294645adae8da8bd8b36051599aa4","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-10-07 20:56:52.000000000","message":"Patch Set 1: Code-Review-1\n\n(7 comments)\n\nGreetings,\n\nThank you for this. I think we may have missed communicating the point when we wrote that we didn\u0027t want to use the word \"here\" for links in the documentation. We want to ensure that we\u0027re generally stating what the link is to. We don\u0027t want to have the URL visible in the resulting rendered output. Luckily, it shouldn\u0027t be difficult to change. If you have any questions on how to edit this change, please let us know!","accounts_in_message":[],"_revision_number":1},{"id":"4118756f71c5ede4e61972bcc1935d5b57cd02da","author":{"_account_id":32493,"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","username":"kafilat-adeleke"},"date":"2020-10-08 08:05:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"05d45f709cce46841469546cdee4ec61b8eabff8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 08:40:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15484035c3004cdab3de086f0003c1d4 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dd2b859abd845f599557976e5e2ac14 : TIMED_OUT in 31m 54s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/2d72c32e542e4123bdc4b27026a6c757 : SUCCESS in 7m 31s","accounts_in_message":[],"_revision_number":2},{"id":"a37623c11e41ebf4eebe284d8e69a3ede3a24fd6","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-10-08 11:57:51.000000000","message":"Patch Set 2:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac/2020-10-08:03:05/756574/2/ : SUCCESS in 1h 09m 40s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac/2020-10-08:03:05/756574/2/ : FAILURE in 1h 02m 28s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac/2020-10-08:03:05/756574/2/ : FAILURE in 1h 43m 51s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish/2020-10-08:03:05/756574/2/ : FAILURE in 37m 34s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish/2020-10-08:03:05/756574/2/ : SUCCESS in 52m 01s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi/2020-10-08:03:05/756574/2/ : FAILURE in 1h 07m 02s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi/2020-10-08:03:05/756574/2/ : FAILURE in 1h 25m 54s\n- dellemc-poweredge-XE-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-XE-UEFI-tempest-dsvm-ironic-redfish/2020-10-08:03:05/756574/2/ : FAILURE in 1h 04m 02s\n- dellemc-poweredge-R7515-R6515-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dsvm-ironic-idrac/2020-10-08:03:05/756574/2/ : FAILURE in 1h 09m 33s\n- dellemc-poweredge-R7515-R6515-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-UEFI-tempest-dsvm-ironic-redfish/2020-10-08:03:05/756574/2/ : FAILURE in 50m 18s","accounts_in_message":[],"_revision_number":2},{"id":"6d1a9a23fd1b896d7d64863c1819a58e9b915a0b","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-10-08 14:23:20.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nrecheck Docs job timed out, which it should not have.","accounts_in_message":[],"_revision_number":2},{"id":"acebc1be0871d61d780de0a7f515e5f285c11dcd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32493,"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","username":"kafilat-adeleke"},"date":"2021-08-12 18:32:45.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"df2824b1a85b8e33347af91646109a07bdaae08f","revisions":{"cc568eb2dda193d9a368da1a0d21aee35a16562c":{"kind":"REWORK","_number":1,"created":"2020-10-07 19:17:46.000000000","uploader":{"_account_id":32493,"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","username":"kafilat-adeleke"},"ref":"refs/changes/74/756574/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/74/756574/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/74/756574/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/74/756574/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/74/756574/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/74/756574/1"}}},"commit":{"parents":[{"commit":"4c9cd83461aa28beb170d76084bc1b9e0c965201","subject":"Merge \"Wiping agent tokens on reboot via API - take 2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4c9cd83461aa28beb170d76084bc1b9e0c965201"}]}],"author":{"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","date":"2020-10-07 19:12:15.000000000","tz":-420},"committer":{"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","date":"2020-10-07 19:12:15.000000000","tz":-420},"subject":"Change links so we do not have \"here\" as the link","message":"Change links so we do not have \"here\" as the link\n\nChange links so we do not have \"here\" as the link. Excludes release notes.\nStory: 2008194\nTasks: 40965\n\nChange-Id: I23d1ddaf8ca4acfd1b51ade7d8d7ee476029cb7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/cc568eb2dda193d9a368da1a0d21aee35a16562c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/cc568eb2dda193d9a368da1a0d21aee35a16562c"}]},"branch":"refs/heads/master"},"df2824b1a85b8e33347af91646109a07bdaae08f":{"kind":"REWORK","_number":2,"created":"2020-10-08 08:05:57.000000000","uploader":{"_account_id":32493,"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","username":"kafilat-adeleke"},"ref":"refs/changes/74/756574/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/74/756574/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/74/756574/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/74/756574/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/74/756574/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/74/756574/2"}}},"commit":{"parents":[{"commit":"4c9cd83461aa28beb170d76084bc1b9e0c965201","subject":"Merge \"Wiping agent tokens on reboot via API - take 2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4c9cd83461aa28beb170d76084bc1b9e0c965201"}]}],"author":{"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","date":"2020-10-07 19:12:15.000000000","tz":-420},"committer":{"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","date":"2020-10-08 08:01:13.000000000","tz":-420},"subject":"Changes links so we do not have \"here\" as the link","message":"Changes links so we do not have \"here\" as the link\n\nStory: #2008194\nTask: #40965\n\nChange-Id: I23d1ddaf8ca4acfd1b51ade7d8d7ee476029cb7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/df2824b1a85b8e33347af91646109a07bdaae08f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/df2824b1a85b8e33347af91646109a07bdaae08f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
