)]}'
{"id":"openstack%2Fironic~758367","triplet_id":"openstack%2Fironic~master~Idb2a999517f6450381c364e589c10f58a020cc8c","project":"openstack/ironic","branch":"master","topic":"38975","hashtags":[],"change_id":"Idb2a999517f6450381c364e589c10f58a020cc8c","subject":"Parent:     4c9cd834 (Merge \"Wiping agent tokens on reboot via API - take 2\") Author:     ANURADHAJHA99 \u003canuradha750871@gmail.com\u003e AuthorDate: 2020-10-08 17:23:27 +0530 Commit:     ANURADHAJHA99 \u003canuradha750871@gmail.com\u003e CommitDate: 2020-10-08 20:54:48 +0530","status":"NEW","created":"2020-10-15 08:55:30.000000000","updated":"2020-10-15 20:11:54.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d137e3fe08a944dd61ad7a923c4f84b15390c4f7","_number":758367,"virtual_id_number":758367,"owner":{"_account_id":32518,"name":"Anuradha","email":"anuradha750871@gmail.com","username":"anuradha1904"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"date":"2020-10-15 10:41:44.000000000","_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-10-15 10:47:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"all":[{"value":-1,"date":"2020-10-15 20:11:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-15 10:41:44.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"REVIEWER"},{"updated":"2020-10-15 10:47:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-15 20:11:54.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"}],"messages":[{"id":"9f00e7178c54015ea45b1f0ee8e36df1eb308ebb","author":{"_account_id":32518,"name":"Anuradha","email":"anuradha750871@gmail.com","username":"anuradha1904"},"date":"2020-10-15 08:55:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4a73c566ab93f77c1dbe984608f960186ed51ee4","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-10-15 10:41:44.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-idrac/2020-10-15:03:55/758367/1/ : SUCCESS in 1h 22m 45s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-idrac/2020-10-15:03:55/758367/1/ : SUCCESS in 1h 11m 45s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-idrac/2020-10-15:03:55/758367/1/ : SUCCESS in 1h 09m 11s\n- dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-tempest-dsvm-ironic-redfish/2020-10-15:03:55/758367/1/ : FAILURE in 1h 01m 07s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-redfish/2020-10-15:03:55/758367/1/ : FAILURE in 58m 05s\n- dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R730-R630-tempest-dsvm-ironic-ipmi/2020-10-15:03:55/758367/1/ : FAILURE in 1h 03m 55s\n- dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-R640-UEFI-tempest-dsvm-ironic-ipmi/2020-10-15:03:55/758367/1/ : FAILURE in 1h 45m 16s\n- dellemc-poweredge-XE-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-XE-UEFI-tempest-dsvm-ironic-redfish/2020-10-15:03:55/758367/1/ : FAILURE in 58m 46s\n- dellemc-poweredge-R7515-R6515-tempest-dsvm-ironic-idrac https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dsvm-ironic-idrac/2020-10-15:03:55/758367/1/ : FAILURE in 1h 09m 44s\n- dellemc-poweredge-R7515-R6515-UEFI-tempest-dsvm-ironic-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-UEFI-tempest-dsvm-ironic-redfish/2020-10-15:03:55/758367/1/ : FAILURE in 55m 36s","accounts_in_message":[],"_revision_number":1},{"id":"f5ed8cdbd0df19fcb1a9c13861a2f7b485f86d48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 10:47:03.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7fe5bea934084091a31e44eeecd3c1ca : FAILURE in 7m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2234d61bf4fa4b4691de5b064665dd33 : FAILURE in 11m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25a02fbea0da4dcca21c94c5fba46e7a : FAILURE in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5da2cb38ca064c219ee19f7115aac695 : FAILURE in 7m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9379d0dc19314192b7db99928fe7c8c6 : FAILURE in 7m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5989e7e7b21d4c7bb3856d39e3e85ca2 : SUCCESS in 12m 18s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/8317f983611e4d12bb3c31e63ff9d760 : FAILURE in 10m 17s\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/a7c3de7b1a0c433f8081d5f29936ed79 : SUCCESS in 1h 46m 58s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6a747d4ba28646da9f095ef7cd33d27b : FAILURE in 4m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c049a6dee9a8eda31ecf88ef361f63cd1a3943d5","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-10-15 20:11:54.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"736c7d3f84ca0c505b3723af853586f34b8a3c50","revisions":{"736c7d3f84ca0c505b3723af853586f34b8a3c50":{"kind":"REWORK","_number":1,"created":"2020-10-15 08:55:30.000000000","uploader":{"_account_id":32518,"name":"Anuradha","email":"anuradha750871@gmail.com","username":"anuradha1904"},"ref":"refs/changes/67/758367/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/67/758367/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/67/758367/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/67/758367/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/67/758367/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/67/758367/1"}}},"commit":{"parents":[{"commit":"cde792a8c3b36afd9dd916b38bf60ed6af4f441a","subject":"Merge \"Updates ironic documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/cde792a8c3b36afd9dd916b38bf60ed6af4f441a"}]}],"author":{"name":"ANURADHAJHA99","email":"anuradha750871@gmail.com","date":"2020-10-15 08:49:33.000000000","tz":330},"committer":{"name":"ANURADHAJHA99","email":"anuradha750871@gmail.com","date":"2020-10-15 08:55:25.000000000","tz":330},"subject":"Parent:     4c9cd834 (Merge \"Wiping agent tokens on reboot via API - take 2\") Author:     ANURADHAJHA99 \u003canuradha750871@gmail.com\u003e AuthorDate: 2020-10-08 17:23:27 +0530 Commit:     ANURADHAJHA99 \u003canuradha750871@gmail.com\u003e CommitDate: 2020-10-08 20:54:48 +0530","message":"Parent:     4c9cd834 (Merge \"Wiping agent tokens on reboot via API - take 2\")\nAuthor:     ANURADHAJHA99 \u003canuradha750871@gmail.com\u003e\nAuthorDate: 2020-10-08 17:23:27 +0530\nCommit:     ANURADHAJHA99 \u003canuradha750871@gmail.com\u003e\nCommitDate: 2020-10-08 20:54:48 +0530\n\nChanges to add exception in \"default_interface()\" function\n\nAdding the try exception block in the function default_interface()\nto catch the exception IncompatibleInterface issued by get_interface(),\nand raise NoValidDefaultForInterface instead,\nremoved the variable \"factory\" from the try phrase.\nAdded unit test to check the exception is raised.\n\nStory : #2007390\nTask : #38975\n\nChange-Id: Idb2a999517f6450381c364e589c10f58a020cc8c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/736c7d3f84ca0c505b3723af853586f34b8a3c50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/736c7d3f84ca0c505b3723af853586f34b8a3c50"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
