)]}'
{"id":"openstack%2Fironic~763752","triplet_id":"openstack%2Fironic~master~I40574cad53de8b9f89e1ae0a033b75de39140769","project":"openstack/ironic","branch":"master","topic":"update-jobs-list","hashtags":[],"change_id":"I40574cad53de8b9f89e1ae0a033b75de39140769","subject":"Make standalone jobs voting again","status":"MERGED","created":"2020-11-23 08:45:57.000000000","updated":"2020-11-25 11:47:29.000000000","submitted":"2020-11-25 11:41:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"763752-update-jobs-list","meta_rev_id":"d2ddcd5941bb90045b73ca63190bc90621ec9069","_number":763752,"virtual_id_number":763752,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-25 11:41:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-25 08:03:44.000000000","_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"value":2,"date":"2020-11-23 13:53:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-11-25 08:03:44.000000000","_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-23 11:11:35.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-11-23 11:44:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-25 08:03:44.000000000","updated_by":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"reviewer":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"state":"REVIEWER"}],"messages":[{"id":"bb56a524d836db37851e1e968af04ed592da1a22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-11-23 08:45:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e6d8e96b5353ce469bc75b1e8f35c8b31b4f1c54","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-11-23 11:11:35.000000000","message":"Patch Set 1: Code-Review-1\n\nWe need to make standalone jobs voting ASAP, let\u0027s leave a TODO rather than pretending they were designed to be non-voting.","accounts_in_message":[],"_revision_number":1},{"id":"e6db484af6e402e13406e3f23085378442fa7ccd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 11:44:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ae2cc38dfd842be91a24f8d208ec1b9 : SUCCESS in 17m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32a03955ebd14a01be8c1a0913ff48e4 : SUCCESS in 8m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47a3afc6634a4fac864aeaacf0996dd5 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76c2a181ea5d4835917a3227d4150fc7 : SUCCESS in 7m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4374779dde2e49f8b6ee00713cba6f21 : SUCCESS in 11m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60aa70f16f2d448482b88ee9c09cef6b : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f36b1a85e3949498d05c12371876d99 : SUCCESS in 15m 11s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/6ab60c7edc484f1a8473ec78bd671096 : SUCCESS in 8m 20s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/b7a121d1e2954b2284edc67dd83423ba : SUCCESS in 25m 21s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/17a4d6e95e76455d842e36e5129d0470 : SUCCESS in 1h 10m 29s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/0641850be9f74c5d980b60ff600498b6 : SUCCESS in 48m 58s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/de528289c80c4e42bc43bddefec9efab : SUCCESS in 46m 17s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/e907f6c443b04271bc23edbd6dd1f0f9 : SUCCESS in 48m 38s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/09f1c4e436d14951ae0a2637bf603433 : SUCCESS in 1h 21m 02s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/56c1923466c047da9f8fb7e87eff3bf2 : SUCCESS in 48m 30s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/a2be630d903244f7b79b6302a731d0ca : SUCCESS in 2h 32m 43s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/ac46387297d2456eb34b6e9eb7544cbd : SUCCESS in 1h 07m 26s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/35e947889d804000b8409ab508efb4ae : SUCCESS in 39m 46s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/02e83976895149ac821ef9e0b59dfac3 : SUCCESS in 52m 59s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/8cafa63e01fe490ab905b5da50f4bbd5 : SUCCESS in 1h 03m 35s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/826ac4d670584e13b2bb11862022c9ca : FAILURE in 1h 42m 54s (non-voting)\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/96c89dcd81dd48a49abc5c67bd0f71ec : SUCCESS in 56m 05s (non-voting)\n- ironic-tempest-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/9ae3a6a9b2664a388ff792818928e9b4 : FAILURE in 29m 26s (non-voting)\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/5c2290c89a9f4b89981f049723151b48 : SUCCESS in 4m 39s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/bb99d66d41bc43f7b706577a6e481b6d : SUCCESS in 54m 40s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/4d0660b902a64d8fb0c1cda3745c0ffb : SUCCESS in 56m 52s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/c8c342485f534c9dbc4cb37a6898e409 : SUCCESS in 56m 00s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/662a58a4a83c4508833a69900fc5c729 : SUCCESS in 54m 13s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f44b63e3d47b476696c1e4962e3d0104 : SUCCESS in 25m 17s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/406d8c482ff04d13a4baf905fdd3ed92 : SUCCESS in 53m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ca92f76dd0f89f10ba407c06b6fb1609ce93faf5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-11-23 13:42:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a80cacf05e99c6e9e45d7cb4cc0d10fa9b472f93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-11-23 13:52:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6c9176dc298b1c0be8e2da8427e789ff3575856f","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-11-23 13:53:58.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fe9591c7720cb63133229c3b5fc6119cc4c17820","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 15:41:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/637785579bf6408697e4b786d74202de : SUCCESS in 7m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1e87827317a04482a5078e36ed237f4f : SUCCESS in 7m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d747e3399b80425dacf1904043e72885 : SUCCESS in 9m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/316b81a14dcb42e0b4cfcc7c5bff0c10 : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/695c5e45762148a6b308c5578bf9d907 : SUCCESS in 13m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b84b3519960549fda2d65f6814617d1b : SUCCESS in 15m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cde05208565e45e0a0e2618dd66b0d2c : SUCCESS in 12m 25s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/a1329a9b387244b3a592506d57fff5d4 : SUCCESS in 9m 11s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/45cfd42cc60f4828b9305a48909cc8f3 : SUCCESS in 28m 20s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/20fa814a640f4b88a447179c71056044 : SUCCESS in 1h 09m 50s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/2428d28e2f8a43c1bb91db8873f2e4b1 : SUCCESS in 54m 15s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/d36115c1dd4c4e83b0dd1d34133d0706 : SUCCESS in 58m 09s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/ce6f8e798eaa4ba9af1e4b6276b41ab8 : SUCCESS in 50m 17s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/a79b8b730f634101a3e25b0fb297922c : SUCCESS in 55m 09s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/db3dd674aaa14fa38e8d5ff7b42849c0 : SUCCESS in 50m 47s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/d5a90888493d4d2792bb0643324d9844 : SUCCESS in 1h 15m 59s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/72749eb6e11849f78b82c2f86419ec6f : SUCCESS in 52m 18s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/6527080a972d499e950351cad6cfc5e7 : SUCCESS in 1h 41m 47s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/247dd402eb0d46baa555b25366df21cd : SUCCESS in 1h 09m 07s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/382b03da54d7432abc9701738918c0d1 : SUCCESS in 36m 30s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/10f53a5e60334953bb816353321e9a37 : SUCCESS in 41m 51s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/54fef274b4c2404ea7d8a1d3a91b61b9 : SUCCESS in 1h 01m 23s\n- ironic-tempest-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/0906015631524ba886cdb6f743c5a4f4 : FAILURE in 25m 47s (non-voting)\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/046c71eca07f48cf8b958958e830d87c : SUCCESS in 4m 32s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/6743080a80e24fafb7fe375dafe943eb : SUCCESS in 48m 52s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/bbede2bc29104148820d16bf2ce93c6c : SUCCESS in 1h 00m 40s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/6245583362e149aca563d9b5128044e2 : SUCCESS in 59m 18s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/a67d7c4cc0af408d8f35411b37d52cba : SUCCESS in 53m 52s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5845b2d947934e9fa66b891979a4c1ea : SUCCESS in 24m 51s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/038a4872d37a418d848cba198703688b : SUCCESS in 1h 09m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6dc68c378b985fb9a1c2dd199cb1456821806499","author":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"date":"2020-11-25 08:03:44.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nThanks, Riccardo!","accounts_in_message":[],"_revision_number":3},{"id":"2d2abb1272b989a1a850b26a5583b254ac41f9ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 08:04:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"257fefeff35f7ff03a7e8044b6d45d3f13d50961","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 11:41:34.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5cd30f8b77044db9a362793f71388386 : SUCCESS in 12m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95febc59b6a744be841a96da85766622 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e9baf6fac814e34955ecff19ce84674 : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aecfb27a6f5f49b297c1788cae15e9b3 : SUCCESS in 12m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93b36859b9664a09a286272e3523de96 : SUCCESS in 15m 35s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/11175cb027cf44e58fdb7a168a9ce223 : SUCCESS in 8m 38s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/45fed95d180143a3bd3566adee62e785 : SUCCESS in 38m 36s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/37ef585aed9f426bb60cd509f79d1d48 : SUCCESS in 1h 15m 23s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/ac0f427a08a949e2859dda0792270605 : SUCCESS in 50m 33s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/627b66b35f9f45d2ae899c50be65d8d5 : SUCCESS in 1h 03m 11s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/a309eb12ee0a4ce7aac20407208de455 : SUCCESS in 50m 58s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/fd7709da2233469ca2c90cef74e016e7 : SUCCESS in 48m 42s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/8d718d00e160448e931f5189ec1058ca : SUCCESS in 54m 47s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/c86b8728ddde4a199515646f7bc89e20 : SUCCESS in 1h 20m 51s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/777edba9d3a14b8d81b33495dc5578fd : SUCCESS in 54m 11s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/5c4051ce3cfb44e88ee95bb645b55d23 : SUCCESS in 1h 26m 38s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/dd75c6f1bde445e49daf57d3fabc89dd : SUCCESS in 1h 18m 25s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/d03e4a0e6505457c9c700e607e177be8 : SUCCESS in 47m 47s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/989a73f9da2b4d22802788ac2f4a206b : SUCCESS in 53m 18s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/cb3523e2626a44159f9c4c10446a0233 : SUCCESS in 1h 20m 38s","accounts_in_message":[],"_revision_number":3},{"id":"727626e890edaad8cea010b79ea2a2df4ddc3081","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 11:41:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"d2ddcd5941bb90045b73ca63190bc90621ec9069","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 11:47:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42f0c23250024effa159bcb934151a3c : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"89af9aef8098d529c261fa48d424cbbf462ff0d6","revisions":{"e47b800e170f255eb2792894bacf3d2c944ffa05":{"kind":"REWORK","_number":1,"created":"2020-11-23 08:45:57.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/52/763752/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/52/763752/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/52/763752/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/52/763752/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/52/763752/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/52/763752/1"}}},"commit":{"parents":[{"commit":"585f90212add6275dae9c0c65ebdafbde4c70e63","subject":"Merge \"Expunge the internal version of WSME\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/585f90212add6275dae9c0c65ebdafbde4c70e63"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-11-23 08:45:22.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-11-23 08:45:22.000000000","tz":60},"subject":"Removing 2 non-voting jobs from gate","message":"Removing 2 non-voting jobs from gate\n\nAnd putting non-voting jovs under the non-voting section.\n\nChange-Id: I40574cad53de8b9f89e1ae0a033b75de39140769\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/e47b800e170f255eb2792894bacf3d2c944ffa05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/e47b800e170f255eb2792894bacf3d2c944ffa05"}]},"branch":"refs/heads/master"},"d5b0a1c8e39d903049742c9f705357b1f5a6d2c5":{"kind":"REWORK","_number":2,"created":"2020-11-23 13:42:04.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/52/763752/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/52/763752/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/52/763752/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/52/763752/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/52/763752/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/52/763752/2"}}},"commit":{"parents":[{"commit":"585f90212add6275dae9c0c65ebdafbde4c70e63","subject":"Merge \"Expunge the internal version of WSME\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/585f90212add6275dae9c0c65ebdafbde4c70e63"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-11-23 08:45:22.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-11-23 13:41:54.000000000","tz":60},"subject":"Removing 2 non-voting jobs from gate","message":"Removing 2 non-voting jobs from gate\n\nAnd putting non-voting jovs under the non-voting section.\n\nChange-Id: I40574cad53de8b9f89e1ae0a033b75de39140769\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/d5b0a1c8e39d903049742c9f705357b1f5a6d2c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/d5b0a1c8e39d903049742c9f705357b1f5a6d2c5"}]},"branch":"refs/heads/master"},"89af9aef8098d529c261fa48d424cbbf462ff0d6":{"kind":"REWORK","_number":3,"created":"2020-11-23 13:52:04.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/52/763752/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/52/763752/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/52/763752/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/52/763752/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/52/763752/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/52/763752/3"}}},"commit":{"parents":[{"commit":"585f90212add6275dae9c0c65ebdafbde4c70e63","subject":"Merge \"Expunge the internal version of WSME\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/585f90212add6275dae9c0c65ebdafbde4c70e63"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-11-23 08:45:22.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-11-23 13:51:25.000000000","tz":60},"subject":"Make standalone jobs voting again","message":"Make standalone jobs voting again\n\nAlso remove 2 non-voting jobs from gate.\n\nChange-Id: I40574cad53de8b9f89e1ae0a033b75de39140769\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/89af9aef8098d529c261fa48d424cbbf462ff0d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/89af9aef8098d529c261fa48d424cbbf462ff0d6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
