)]}'
{"id":"openstack%2Fironic~770673","triplet_id":"openstack%2Fironic~master~I9a3eb62bb41e0623df9aa5c34fde6f343053dd76","project":"openstack/ironic","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"I9a3eb62bb41e0623df9aa5c34fde6f343053dd76","subject":"Attempt to slim down protection test base class","status":"MERGED","created":"2021-01-13 22:35:06.000000000","updated":"2021-01-21 11:37:42.000000000","submitted":"2021-01-21 11:33:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"770673-secure-rbac","meta_rev_id":"33f0e75420935aeb6da77046672aff0e8958646e","_number":770673,"virtual_id_number":770673,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-21 11:33:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"all":[{"value":2,"date":"2021-01-21 11:37:42.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-21 11:37:42.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"all":[{"value":1,"date":"2021-01-21 00:44:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-13 22:49:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-13 23:35:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-14 15:08:26.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2021-01-18 04:02:01.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"},{"updated":"2021-01-18 16:10:07.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"57e65b4ba5b949e7fc11c95db60d047b65ce55b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-01-13 22:35:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2562da247a339bbc701f9c04ad10071f61db69b5","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-01-13 22:40:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5567685a48542c91fbaaaa9f8724d37276b70ac2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 22:49:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a33ed1e616bf45b2a95ca6be5e12dcae : SUCCESS in 8m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9fe7bf600c6546038cd22cec21457cc7 : SUCCESS in 9m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b956a251d7bdbeb632d7e123c317aeb86ac785eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 23:35:40.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3da911b462c344c099d52e7529d20270 : SUCCESS in 13m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ded726090a67432f8f8deaf63598abe7 : FAILURE in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8bd46e08d304473a8cc576a9db5c287d : SUCCESS in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b967c86324df48b38fb28bb59cda80fc : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f845803a70964224ba22bd3e19f41347 : SUCCESS in 8m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/257a30c093ec466e9c3827a0987a19b5 : SUCCESS in 16m 49s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/f67e08e0e4694898918d29481744021f : SUCCESS in 8m 47s\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/5bc038917643426ab69e34e05550c21c : SUCCESS in 54m 30s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c15f5fbe344441c91c656aa45d9142d : SUCCESS in 19m 19s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/20cfb1cf1eb84239adf801cb14624536 : SUCCESS in 23m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7e9ef3e9486ff5f0aec4b181e04cd4ccbc45ec28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-01-14 15:08:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7537ff04323b5107659faf53a2ab1d7d0be7927a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 16:16:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b1b4722293d8151a2608c88da536fbce1b3843b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 16:58:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b07d388d2354b25ad842e9efa78424d : SUCCESS in 7m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b73f570e3ed84789a4a2b6cde960911e : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed17b92804304794b7c33705233a83f5 : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed64133ff4a84c4cb0761ec23a1d3343 : SUCCESS in 6m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ed91bea30264ba3aa01f2370b5de5f6 : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/439208a93b634f5e8a5f1b5f46207ea0 : SUCCESS in 13m 41s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/1123739b3426449da1b709dd98f75a5b : SUCCESS in 8m 22s\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/c1e640f6c70e434caa38d615ab96a1fe : SUCCESS in 1h 11m 56s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57c1b3c29cd44f05b1ab964b23a1f7bd : SUCCESS in 27m 38s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/3decb7c3be534cfeaa532bf76979e240 : SUCCESS in 24m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"07ede142fd23615cfc7cd0cc655692586d4146ea","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2021-01-18 04:02:01.000000000","message":"Patch Set 2: Code-Review+2\n\nThat is much nicer, thanks","accounts_in_message":[],"_revision_number":2},{"id":"4fddd1924d28432d6c92e7f440829b9b2d9a8e71","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-01-18 16:10:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fd101d2eb1296305b53b6acaad619367b502a86c","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-01-18 20:00:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1ee7fab3bb8c118fcaf6fadf5e909a3d220ce1d0","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2021-01-18 23:35:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"24985d88b32b3db1eaf1d75069dc919a619d5f65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-01-20 01:07:43.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"8e6ffff2a53f80a5e4e9766efdde7ffdcfdac7da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 01:40:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/45a62026dce24a0080b29bf3a86bd53f : SUCCESS in 16m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c2403e52bacd4a31afa48a63936fecc3 : SUCCESS in 11m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5f76974538c66c06cd1a8788d8eb819ea4f38bd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 02:34:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae7bbb4fae9848c09da18779653fe236 : SUCCESS in 9m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d25f407c78d446018ee573d1a217fbcc : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3fa8d6be365e4462b655a0c822ff58a2 : SUCCESS in 11m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf9fc15a167846b5a69427a45a7dde2f : SUCCESS in 9m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd90c61c05444f25a538f0f924237064 : SUCCESS in 9m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/274e000b95914dd4a4e2fd65bb25c73e : SUCCESS in 15m 30s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/796a110814d5476e9c6a1a192c1050d1 : SUCCESS in 9m 04s\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/0a76d9ab956e4befaac07202ccc47aa1 : SUCCESS in 49m 37s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ae44a39b9824205972a180b29e4c29b : SUCCESS in 22m 59s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/525cd0a6146a4231b7f0db5328fe2b7a : TIMED_OUT in 1h 00m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c9faa6cf7051fc62ef936e13c52e7228311ea24c","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2021-01-20 03:13:56.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"67a2db6a1843ffe488b7d9e94b91fdb5fd92f38f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-01-20 14:47:01.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"380ef67c0fe8c7a8e53978e34d35803763a9be4e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 15:18:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2f977780b60d4f82911b1b2d7ac5fd14 : SUCCESS in 11m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/801e82f25ebb49c69d8a9ccc78260ed1 : SUCCESS in 13m 50s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"cffc488b8ea7b323f81f5d753083aeb07627497d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 16:09:10.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd8bdb52ec5f48ac851f008ccc9c003c : SUCCESS in 8m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40bb650bc7244a6bb278e00a9083e90c : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/acaa5dcba09e48ddbc51ae7dce504280 : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73c1cd3173f34f6e9b052aa238769aaa : SUCCESS in 8m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b0d4058a21824876b79bdfd59298c4bc : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0529f06e7bd84b39b1351dfc2aa0c09d : SUCCESS in 16m 34s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/b946e15e23174400aa67f1764a1756ca : SUCCESS in 10m 18s\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/5b5b191f7953414c9ccd898b77ea2976 : SUCCESS in 53m 13s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10bede764d1247efa7a881622166b9a1 : SUCCESS in 26m 01s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/8c43a8747cc04e3b8f8c8ab4bbf7a7cb : SUCCESS in 26m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c6d89d01eb57ff0d34de4c2b0dd10e79465acc41","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2021-01-21 00:44:26.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"bfedb8a029a89d1e66621be3e0aa07e1aadba630","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 11:08:45.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ce538686bab839be1b0f22a6e6424a455167a6ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 11:33:00.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89e056baa3854c62b95f4df4050a20d4 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d623f41b031a4b54bcb1231cdd5a5e26 : SUCCESS in 7m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b60ce25a6a348fdb132fd029b4dd624 : SUCCESS in 8m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74faae38fc6643f79f9ef66c02ff16f1 : SUCCESS in 16m 49s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/958820078b364909a097d1cfd011b39a : SUCCESS in 10m 25s","accounts_in_message":[],"_revision_number":4},{"id":"803efe1db67eafc2320f6b68c7b79942ac832f64","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 11:33:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"291dc29ae902f997603a0a2675c3257a55643d70","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 11:37:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/588eb7a8b0fb4e09bfb9cb91e4379ee7 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c7f24e8ceb9b49ed0fcdc94353240f5c1cfa65e2","revisions":{"72bc6209a19177d71a794342fa31eec97bba54a8":{"kind":"REWORK","_number":1,"created":"2021-01-13 22:35:06.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/73/770673/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/73/770673/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/73/770673/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/73/770673/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/73/770673/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/73/770673/1"}}},"commit":{"parents":[{"commit":"04a1f17ef8a0d8d2e3462a583689daf05ecb30fb","subject":"Enable testing to dynamicly leverage ACL roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/04a1f17ef8a0d8d2e3462a583689daf05ecb30fb"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-01-13 22:30:27.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-01-13 22:34:31.000000000","tz":0},"subject":"Attempt to slim down protection test base class","message":"Attempt to slim down protection test base class\n\nThis commit builds on an attempt to introduce a base class that we can\nuse for protection testing. I\u0027m removing the majority of the fake tokens\nand context objects, ultimately removing anything that\nkeystonemiddleware_authtoken middleware does.\n\nDoing this allows us to set authorization headers directly in the test,\nwhich makes the test case really clear from an input and outcome\nperspective because we can see who is make the request and the intended\noutcome.\n\nIt also removes any need to mock or understand keystonemiddleware\nimplementation details to implement protection testing in ironic.\n\nChange-Id: I9a3eb62bb41e0623df9aa5c34fde6f343053dd76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/72bc6209a19177d71a794342fa31eec97bba54a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/72bc6209a19177d71a794342fa31eec97bba54a8"}]},"branch":"refs/heads/master"},"b18849c19ffd7d3da6802ff5c2d9e176e9b7b057":{"kind":"REWORK","_number":2,"created":"2021-01-14 15:08:26.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/73/770673/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/73/770673/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/73/770673/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/73/770673/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/73/770673/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/73/770673/2"}}},"commit":{"parents":[{"commit":"023884d38d6a2f5a3144513a000949c7ba51bc11","subject":"Write stub ACL test for every existing API call","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/023884d38d6a2f5a3144513a000949c7ba51bc11"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-01-13 22:30:27.000000000","tz":0},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-01-14 15:08:16.000000000","tz":-480},"subject":"Attempt to slim down protection test base class","message":"Attempt to slim down protection test base class\n\nThis commit builds on an attempt to introduce a base class that we can\nuse for protection testing. I\u0027m removing the majority of the fake tokens\nand context objects, ultimately removing anything that\nkeystonemiddleware_authtoken middleware does.\n\nDoing this allows us to set authorization headers directly in the test,\nwhich makes the test case really clear from an input and outcome\nperspective because we can see who is make the request and the intended\noutcome.\n\nIt also removes any need to mock or understand keystonemiddleware\nimplementation details to implement protection testing in ironic.\n\nChange-Id: I9a3eb62bb41e0623df9aa5c34fde6f343053dd76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/b18849c19ffd7d3da6802ff5c2d9e176e9b7b057"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/b18849c19ffd7d3da6802ff5c2d9e176e9b7b057"}]},"branch":"refs/heads/master"},"20015df2967ae2187f6cc9b60f005d6c1d5cbd77":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-01-20 01:07:43.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/73/770673/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/73/770673/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/73/770673/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/73/770673/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/73/770673/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/73/770673/3"}}},"commit":{"parents":[{"commit":"7b1f2bd6e0f99c1d5b07760f7ac37a59054d26d6","subject":"Write stub ACL test for every existing API call","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/7b1f2bd6e0f99c1d5b07760f7ac37a59054d26d6"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-01-13 22:30:27.000000000","tz":0},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-01-19 19:21:03.000000000","tz":-480},"subject":"Attempt to slim down protection test base class","message":"Attempt to slim down protection test base class\n\nThis commit builds on an attempt to introduce a base class that we can\nuse for protection testing. I\u0027m removing the majority of the fake tokens\nand context objects, ultimately removing anything that\nkeystonemiddleware_authtoken middleware does.\n\nDoing this allows us to set authorization headers directly in the test,\nwhich makes the test case really clear from an input and outcome\nperspective because we can see who is make the request and the intended\noutcome.\n\nIt also removes any need to mock or understand keystonemiddleware\nimplementation details to implement protection testing in ironic.\n\nChange-Id: I9a3eb62bb41e0623df9aa5c34fde6f343053dd76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/20015df2967ae2187f6cc9b60f005d6c1d5cbd77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/20015df2967ae2187f6cc9b60f005d6c1d5cbd77"}]},"branch":"refs/heads/master"},"c7f24e8ceb9b49ed0fcdc94353240f5c1cfa65e2":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-01-20 14:47:01.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/73/770673/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/73/770673/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/73/770673/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/73/770673/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/73/770673/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/73/770673/4"}}},"commit":{"parents":[{"commit":"36d819e2fb8e27f8c900cf50c6c6fde1bde68a41","subject":"Write stub ACL test for every existing API call","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/36d819e2fb8e27f8c900cf50c6c6fde1bde68a41"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-01-13 22:30:27.000000000","tz":0},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-01-20 14:41:47.000000000","tz":-480},"subject":"Attempt to slim down protection test base class","message":"Attempt to slim down protection test base class\n\nThis commit builds on an attempt to introduce a base class that we can\nuse for protection testing. I\u0027m removing the majority of the fake tokens\nand context objects, ultimately removing anything that\nkeystonemiddleware_authtoken middleware does.\n\nDoing this allows us to set authorization headers directly in the test,\nwhich makes the test case really clear from an input and outcome\nperspective because we can see who is make the request and the intended\noutcome.\n\nIt also removes any need to mock or understand keystonemiddleware\nimplementation details to implement protection testing in ironic.\n\nChange-Id: I9a3eb62bb41e0623df9aa5c34fde6f343053dd76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/c7f24e8ceb9b49ed0fcdc94353240f5c1cfa65e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/c7f24e8ceb9b49ed0fcdc94353240f5c1cfa65e2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
