)]}'
{"id":"openstack%2Fironic~772239","triplet_id":"openstack%2Fironic~master~I1b1a6baafd4cc4daa2fbdb82f69ded6253b1fcbf","project":"openstack/ironic","branch":"master","topic":"cache-vendor","hashtags":[],"change_id":"I1b1a6baafd4cc4daa2fbdb82f69ded6253b1fcbf","subject":"For Supermicro BMCs set enable when changing boot device","status":"MERGED","created":"2021-01-25 01:05:53.000000000","updated":"2021-01-31 08:02:27.000000000","submitted":"2021-01-31 08:00:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":25,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"772239-cache-vendor","meta_rev_id":"c51ae623c149cdfae116ac5387f8eedfdb58f9ea","_number":772239,"virtual_id_number":772239,"owner":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-31 08:00:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":2,"date":"2021-01-30 07:14:22.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},{"value":2,"date":"2021-01-29 22:51:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":1,"date":"2021-01-30 07:14:22.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-25 01:07:00.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2021-01-25 01:34:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-25 02:36:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-27 21:59:56.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"},{"updated":"2021-01-28 17:06:58.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2021-01-29 01:21:56.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2021-01-29 07:24:29.000000000","updated_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"reviewer":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"state":"REVIEWER"}],"messages":[{"id":"f19da7aa323d38ed73d78d6ed73d58c4d8b515b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2021-01-25 01:05:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08bf6c4262ec7200b54732e9c03215eeebbe2ab7","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2021-01-25 01:06:25.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"c87e98aefca2e6716439b63cf8e65bd40ac67d5b","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-01-25 01:07:00.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 : NODE_FAILURE in 0s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 : NODE_FAILURE in 0s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 : NODE_FAILURE in 0s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 : NODE_FAILURE in 0s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 : NODE_FAILURE in 0s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"ff57973730ff7798b0bdb2ddf71c5056d285925e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 01:34:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0801792fa6268b192ca9d454d4bb5a1fd2864d6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 02:36:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/343f42136cb94f9cba9b74633fe6c05b : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/480a7276286543a58b47c0b07741d6fc : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af220785c22a4c0fa2b31054ea0fb098 : SUCCESS in 7m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e0811c6689c42919004275ad08a576e : SUCCESS in 7m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8e89096cecef4ab9812eb2bd0e480f0d : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f459539b0e9b43fca82372568b6a0b88 : SUCCESS in 16m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf3028dc2c7a4e1f9abcc3b74eaf87fa : SUCCESS in 5m 49s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/96b0d32aa3b74ea6bf54ad6bec33252c : SUCCESS in 8m 58s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/91c04d389391433da5b116e961b87180 : SUCCESS in 26m 37s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/989c8393b37149fca8c8667c7882a7ec : SUCCESS in 1h 11m 25s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/e8a5213f735e490ba619166f59feade5 : SUCCESS in 1h 13m 39s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/4446aa5e638d4ecbb3f0c6a5e214c2ec : SUCCESS in 59m 17s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/d446cff9fa6443049731533d31d8d4f9 : SUCCESS in 44m 37s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/e83ded83533e41b49b02b67c0eec8797 : SUCCESS in 50m 55s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/f69aa7f33be248d380006e0183cff19b : SUCCESS in 48m 25s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/e2646e500d85417188e71bd2e35c52da : SUCCESS in 1h 01m 48s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/0453f2b36f894f4d855974a155150bab : SUCCESS in 51m 00s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/57b38c3714354911b645df69242a2bb3 : SUCCESS in 1h 25m 17s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/daedfa2f3e3142dab34785847e0f58ef : SUCCESS in 1h 13m 28s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/5b86c1994bf246ad96d8eaf65d289c34 : SUCCESS in 41m 27s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/51b5d68a09374b75a7adb786ff2ddf53 : SUCCESS in 49m 53s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/176cf29396374a4ca4384235cd52ceeb : SUCCESS in 1h 08m 55s\n- ironic-tempest-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/30c3d51eb8364c8b83817661c8b667ac : FAILURE in 33m 53s (non-voting)\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/c6a40bd6bd414404847389b6b8079eab : SUCCESS in 4m 32s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/52cb8f64899f4c7393bae771b9fa5597 : SUCCESS in 54m 42s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/fa1313c0dc524ba4ac5eb28addf6b60e : SUCCESS in 1h 02m 02s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/b4eda9a4982f4f709a4ea52db0f32547 : SUCCESS in 46m 50s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/7f9af4a407e8491d9b4007c2bd8db337 : SUCCESS in 58m 17s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/113c7466c6774a24987214feca3c404c : SUCCESS in 25m 18s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/0bc2452f406d44c4b2cd0ee0ee9fde9a : SUCCESS in 20m 20s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/071b9412c6ae40caa8765f2a6b737748 : SUCCESS in 45m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5c79ae83aa6ae50162d95b2000e4f71bfa293c8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2021-01-26 13:48:45.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"de8a636cf019f8763ece264b2e40b55a1caf9d27","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 14:05:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/89bed74233ad400eacd9e15d9e3ca6b8 : SUCCESS in 9m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/25f968aa3c8a427d9c931544659f07b3 : SUCCESS in 11m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c67acf759e8cfce42606a675ac1613b8a83d4836","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2021-01-26 14:43:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8e90d996f74884563925580b5de48f693849ba1b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 14:57:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f7f308579c29459c861b7e7e67a669cb : SUCCESS in 12m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0639a8079f7946cd9a7611ca5d75ab8c : SUCCESS in 11m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b107022094b54aa66f7bb7368778ba2a93ec8675","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 16:23:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b68b43e00a246f0802bcfe39bfa8fde : SUCCESS in 9m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b05e00be97a42c5b86bf6bc1e6a6b8a : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e678e895179439ebd11c0d1e4dd0fde : SUCCESS in 7m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f56b010e6ad44f9b9cfca152549b608 : SUCCESS in 6m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b4cf935b4de4323a3478c18f74535ce : SUCCESS in 9m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83e125632bc04327b0e7a8f59beab4a4 : SUCCESS in 17m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c44fcfc0804044468e3220e09104b582 : SUCCESS in 6m 26s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/2c935ff318144ee3b7ad8e98d6f57b36 : SUCCESS in 7m 52s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/09b8fbd00ba44dafaa71c798d7b6446d : SUCCESS in 23m 07s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/269b79ab42594857a6ee6daa48212b9b : SUCCESS in 1h 06m 57s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/5fe0579477fc42e0b49571b368da7f2c : SUCCESS in 54m 56s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/1e73d3f4d1c7456d93a3b580a86e5e43 : SUCCESS in 1h 04m 40s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/45d9a43a242f4c31bf37386e3d624498 : SUCCESS in 50m 11s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/2feb293be8ac47e4abdc470c09bedcbe : SUCCESS in 1h 09m 46s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/21a6ee78daf94b77828f4ae8c5b501f7 : SUCCESS in 47m 50s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/94e954a3fed7456982cd1cc941c75b6b : SUCCESS in 1h 20m 50s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/36e13ad6ac6048f8bd2b324eed183731 : SUCCESS in 48m 55s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/2ea9e2857ce74e56bd0b18f68e460369 : SUCCESS in 1h 35m 34s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/1df4b161bb244fcb852aa29915c078d2 : SUCCESS in 1h 31m 57s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/8052a69436bc431a9ba85ba29645201c : SUCCESS in 38m 44s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/716422e7de10496d914b66ff8bd5fe35 : SUCCESS in 56m 04s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/8e945e26b43b4105b179fd7567224189 : SUCCESS in 1h 10m 50s\n- ironic-tempest-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/aa91a18fe41747dbad8167a8f117ca27 : FAILURE in 29m 00s (non-voting)\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/43daca1a6bd945cab293439784c7e283 : SUCCESS in 4m 09s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/64d5763fe6f5455b92535b45c1d5f6ec : SUCCESS in 1h 02m 46s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/7a3f002712e44c399b8551e8bbb47f97 : SUCCESS in 54m 49s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/5cd20a6e79264a10936beceda18ffab0 : SUCCESS in 50m 46s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/7c5cc62086dd4463a392764f70d66db3 : SUCCESS in 55m 20s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/37afd2af37f74ae7be74cdb67272b6a8 : SUCCESS in 26m 50s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/0afd5b3b46fa4cb18efa87495133c7f1 : SUCCESS in 26m 06s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/06db6d67bb464863963bda2087a96028 : SUCCESS in 53m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"553cd8b8e34faee9a4109ae4cd4fcfcff8749542","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-01-26 18:37:04.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/772239/3/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/fcfbd1e/fcfbd1e.html : TIMED_OUT in 3h 01m 19s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/772239/3/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3/9191914/9191914.html : TIMED_OUT in 3h 01m 21s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/772239/3/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3/53cb374/53cb374.html : SUCCESS in 42m 06s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/772239/3/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/9566a38/9566a38.html : TIMED_OUT in 3h 01m 05s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/772239/3/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/132f4c4/132f4c4.html : TIMED_OUT in 3h 01m 03s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/772239/3/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/1db09b3/1db09b3.html : TIMED_OUT in 3h 01m 00s","accounts_in_message":[],"_revision_number":3},{"id":"28677126671b338157555a03b1d82a17751b9101","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2021-01-27 21:59:56.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4c584d0c301779e26369ec03c71f854e051b7b2f","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2021-01-27 22:23:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2d1defaa5637042458526f245b461142029a0b89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2021-01-28 13:15:12.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1159fb40d5459af9eff8becf62fabafc117f515b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 13:29:41.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/26fc593d13cb4affbc3c92f97f9946ac : SUCCESS in 10m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e471418c353e4bcabed9a10630d177ec : SUCCESS in 10m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e1def03be2a9c926f79db7702a1953334602933a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 15:27:50.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65989e49758142e4b2077c57c947229b : SUCCESS in 9m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bafdfaa1ffd64deb98c443793f1e6ba1 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b16819f82964d078c1c65b11a4275ca : SUCCESS in 7m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c676e0d95f414a5fb8ef70b00ec42d38 : SUCCESS in 7m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/88be18a573eb4782961ea6d03f81320c : SUCCESS in 8m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/750c203e4fed4a6fb67bee3ed749e9ca : SUCCESS in 15m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13d51d2016a546c29ef7bf1be4587b34 : SUCCESS in 6m 21s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/19a7caa7229d4198af82cdabf46b5c73 : SUCCESS in 9m 06s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/b48d0ec6ec1c4e17b80816a8010dfa39 : SUCCESS in 30m 24s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/c3aa3bce5440425890cb64e7dbed8c2c : SUCCESS in 1h 06m 20s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/852ea6df23dd406d95cc3e58ff781899 : SUCCESS in 1h 09m 05s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/282e7b2bcc824a79aa1843be64911171 : SUCCESS in 1h 09m 06s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/11b32030ed744dfa9af66b22d31c4d48 : SUCCESS in 50m 59s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/8ca8010b85ce451d895ef96c134d2bee : SUCCESS in 56m 55s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/86463d6222db44d08ae614d0fe8781d6 : SUCCESS in 49m 12s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/29b7ff599bac4acea8a654a4363afc99 : SUCCESS in 1h 00m 45s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/3ab2abdc581547e49880343d78814388 : SUCCESS in 50m 39s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/890c905eb3ec497d8c6bca49cbff0ab7 : SUCCESS in 1h 44m 17s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/224eb408d4914367b3a12ab40a697756 : SUCCESS in 1h 24m 41s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/19ecdff2e576481686cb395e0e1dcad5 : FAILURE in 1h 10m 19s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/0abfb7dcf1a343d8b785d160c1e17832 : SUCCESS in 49m 00s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/18aba065257f4214ba3b2fc103eabe57 : SUCCESS in 1h 13m 27s\n- ironic-tempest-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/96535a32499c4aa69d06c0a13adb9e65 : FAILURE in 25m 26s (non-voting)\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/cb93e70244984cd98e017b4ee501be80 : SUCCESS in 5m 24s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/9e757449a9a14ab6808be074af64dddc : SUCCESS in 54m 03s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/71fd898729b6459f911f9c93c2e252fc : SUCCESS in 52m 53s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/627ea6d9123f4e28bfa98214855399d0 : SUCCESS in 50m 54s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/967c4c3312f941b3a3c337f1bb52571d : SUCCESS in 51m 41s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b381b4561924c37a6f82850d7aa8874 : SUCCESS in 24m 22s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/40ab7cd3684f49d487bb62cd89d09bde : SUCCESS in 24m 30s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/7081cded40874306820cae0ee01699bd : SUCCESS in 51m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"37db4539049c2be35909013b1cb407c55c8b7272","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2021-01-28 17:00:48.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"e36f428c14cb05fc49482555290e229c3804ff45","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-01-28 17:06:58.000000000","message":"Patch Set 5: Code-Review-1\n\n(9 comments)\n\nThanks, a few minor things inline.","accounts_in_message":[],"_revision_number":5},{"id":"18878a58e77371e94935b6278ead24e1899d6b73","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 17:21:35.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/566d74cd53a544b0a999ddea911eb799 : SUCCESS in 11m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9343a85968d2438a8d782ce5c4bf7805 : SUCCESS in 13m 42s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2c2fe35d0a9e65afc9bcc88ee59183a42566eb10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2021-01-28 19:15:51.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"9913b0010f2c3f6cc65dcc566eac4832c70280b9","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2021-01-28 19:19:03.000000000","message":"Patch Set 6:\n\n(9 comments)","accounts_in_message":[],"_revision_number":6},{"id":"39a3e7ae88d4287e6976b8309418377ba1221331","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 19:43:17.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e196d7d2068644678d7cc000e4a2ba75 : SUCCESS in 16m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0ee1ebc8cc0742d9bb3f110e78411f76 : SUCCESS in 22m 18s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"668ce18967796a5f4a5d044bbc9f4e0b085be5ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 21:25:19.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f137d49f9f3a437f8b79d14c4806232e : SUCCESS in 8m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97ac34dd9ed5411f819e93bb101ba116 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f66817a70d04dd09a17d6b2781c4a5e : SUCCESS in 7m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b08ddf2483d4a0a84cf1668f17f7af6 : SUCCESS in 7m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d8e0d980da0423ab4b1e961b36216f7 : SUCCESS in 9m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed440ddd7e1048098eb81e5d3b36ddcb : SUCCESS in 16m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1648559b3d7c47e6a75f307d1005edd1 : SUCCESS in 6m 16s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/cc4c713078e94c99bfe20bce5227e120 : SUCCESS in 8m 32s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/4761216833e748f2af6b08721a60eb6a : SUCCESS in 25m 43s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/41d8ffe4eb50473cbdc4f8cb8139504e : SUCCESS in 1h 06m 39s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/95a7f56513b547828f7758631e619335 : SUCCESS in 59m 17s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/acc3cf713d364fcc9523559f840fdbbc : SUCCESS in 1h 04m 19s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/55d80466af454839bdb4881da2494e2b : SUCCESS in 45m 44s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/85c78fe8291b47f1b3c697a5e40c5158 : SUCCESS in 1h 07m 47s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/4d5d84e669304c3bac9a9cb5589a0b42 : SUCCESS in 42m 23s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/a8dcf2ed91ac48459fa8338c06b35ae2 : SUCCESS in 59m 28s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/cadf17eb0f3c4997829b4c7b608bf8c9 : SUCCESS in 50m 22s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/fb42387d7f89469b8900735b2b0e59d1 : SUCCESS in 2h 00m 56s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/c14a365f6c1a429d9324ee69b0b90d3d : SUCCESS in 1h 22m 37s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/ebc9c210c260421cb0625e7f6d82b45c : SUCCESS in 52m 03s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/c81dab759b0b47df910bf119edc4589a : SUCCESS in 43m 53s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/78ca918aedfe4e779e570252e71809cd : SUCCESS in 1h 02m 40s\n- ironic-tempest-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/7c0b6f191f6140d9ace44ffdc4bb6e21 : FAILURE in 25m 51s (non-voting)\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/c0ed0905e1b244a2b392b79e88266769 : SUCCESS in 4m 26s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/3c586c944105467ba10ae7968f5bdde4 : SUCCESS in 50m 30s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/5c9d977d7e2048a3a2d7eff0127bd78f : SUCCESS in 57m 51s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/775d097930af416f987445a57afef63c : SUCCESS in 48m 03s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/234e47b8e58e4d72a936a73132fbf556 : SUCCESS in 57m 37s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3843af37f149432ea50dc31e1260efdd : SUCCESS in 25m 16s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/b4af84433182405daf361724c1543946 : SUCCESS in 24m 17s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/13cfa6fb46d94e79a2d16d36003daaad : SUCCESS in 49m 30s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"53f3acce962b454000de8ee128b9377095ca52ea","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-01-29 01:21:56.000000000","message":"Patch Set 6: Code-Review+2\n\nLooks good! Thanks!\n\nP.S. Please backport this once it merges.","accounts_in_message":[],"_revision_number":6},{"id":"d9521ce1cc03bfd7b1083276605e39f593124682","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2021-01-29 07:24:29.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)\n\nThank you Bob.","accounts_in_message":[],"_revision_number":6},{"id":"b787f32e62cc6b39a95bb14bd3176db578f37634","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-01-29 07:26:04.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/772239/6/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/d9308f7/d9308f7.html : TIMED_OUT in 3h 01m 27s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/772239/6/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3/83876db/83876db.html : TIMED_OUT in 3h 01m 04s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/772239/6/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3/1ebf7b6/1ebf7b6.html : SUCCESS in 45m 31s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/772239/6/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/96b2812/96b2812.html : TIMED_OUT in 3h 01m 03s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/772239/6/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/50202bd/50202bd.html : TIMED_OUT in 3h 01m 04s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/772239/6/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/dfb7beb/dfb7beb.html : TIMED_OUT in 3h 01m 01s","accounts_in_message":[],"_revision_number":6},{"id":"f5d4d163cbdd36d87152e93482c6f09ae0c79329","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-01-29 14:13:07.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d5b07f15eb9711b273d353787f79bf67b25c31a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2021-01-29 16:46:53.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"44084d90ba482aef45c19f045aa1c05818fd6b21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2021-01-29 16:55:52.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"eb4a6a19fd20e6e451466ddb5185111fea8b263c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 17:11:46.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/764bfb5ffabc4fc08adbd8436cf42332 : SUCCESS in 10m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0a1653fa84de4aafb419172acfa58805 : SUCCESS in 11m 00s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"a7c57e741e6a48be2194b49ede42e3f1aa81ae7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 18:44:08.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af36c635e8624ae09cb763db5ab68ad8 : SUCCESS in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf26dccd763e4d409d5368ee7fc8b1b3 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e1030a2126f433195f08cd43be16a43 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65c627a665524d0ca622971075448796 : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/93527af9265e4202ae65d879af879da9 : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bbdee3de769405492fef8389fb57b7d : SUCCESS in 15m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af5f39496a2943bf986c602b4a296f00 : SUCCESS in 5m 55s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/34bcff0b284348baab829db97569b8c8 : SUCCESS in 8m 32s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/25f797532e4e4570a4a48ba0ab882622 : SUCCESS in 25m 45s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/7f8237d4f2bc41948eee59cbda1b222f : SUCCESS in 1h 11m 58s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/344fe5e221f74cb5b4b462c324a876ea : SUCCESS in 54m 33s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/ccbd1835b855479eacd175b1f39e6c87 : SUCCESS in 1h 02m 57s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/2460e0201223444981bb5ad9ea38941a : SUCCESS in 45m 09s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/c5be1d8dad8340c98e72f01c7b5a8fb0 : SUCCESS in 49m 10s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/e7542f6dc14b489abdb0c1fe2f797f8b : SUCCESS in 45m 49s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/10edbbd0de6844779eaea956bbdb307b : SUCCESS in 58m 32s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/e8d2bcc27cb54565a97d164fe8cb7730 : SUCCESS in 48m 31s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/4656869c970e4f14824262f80f67ad3e : SUCCESS in 1h 42m 24s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/c9d342b9bc4b46c49a25510b55081688 : SUCCESS in 1h 11m 02s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/974ce275b5e941ff988eed7f65588f12 : SUCCESS in 46m 09s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/fa12183600344a5a9731889ba244bca8 : SUCCESS in 1h 02m 49s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/25bc20b2f52b44eab36f52b6c2ba126c : SUCCESS in 1h 12m 35s\n- ironic-tempest-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/84d4710267e5415f823682cc6c105fba : FAILURE in 9m 10s (non-voting)\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/68d9411130624909bdaea759f2685593 : SUCCESS in 3m 52s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/e50b467a9c0a449493c499b220a07f16 : SUCCESS in 48m 41s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/2cbbf1369af341d6a8de031fca3221dc : FAILURE in 40m 31s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/330eb1ffea1c43f6b96b83ca81845210 : SUCCESS in 46m 47s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/18ad414810c54b96a1ac39f7092dd83e : SUCCESS in 48m 01s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8379192b83f4e7fba08ba4178d07f03 : SUCCESS in 27m 15s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/f4feff8fb93b460986107cc748315c4c : SUCCESS in 25m 15s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/c8593a575827438fa4a8e9572d1b910b : SUCCESS in 48m 56s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"2fc6b714f7e91d58b4e954f7ffec952be4163c1e","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-01-29 21:46:10.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/772239/8/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/7c297c6/7c297c6.html : TIMED_OUT in 3h 01m 07s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/772239/8/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3/201885d/201885d.html : TIMED_OUT in 3h 01m 19s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/772239/8/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3/0d688de/0d688de.html : SUCCESS in 41m 03s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/772239/8/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/6ddfad9/6ddfad9.html : TIMED_OUT in 3h 01m 02s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/772239/8/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/c20bddc/c20bddc.html : TIMED_OUT in 3h 01m 03s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/772239/8/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/580a6e5/580a6e5.html : TIMED_OUT in 3h 01m 01s","accounts_in_message":[],"_revision_number":8},{"id":"5c226763e05b46973c550ce375caa940996a1a4b","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2021-01-29 22:51:34.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"ee86af40fa11b6e95b6c1dee02780465440856fc","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-01-29 22:51:36.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"606f332775097a8485f5cc3171aaf14ed3f9078b","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2021-01-30 07:14:22.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(1 comment)\n\nThank you Bob.","accounts_in_message":[],"_revision_number":8},{"id":"5c67c6009afa0c9caab5550aa7033b978a57c131","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-30 07:14:40.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"3816f6d575d337ff4d7f02d351ff43cf79a884a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-30 09:17:39.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a99e17644c0544cdb220445f47d33156 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af1ddce61e7c4eb18b059f3fbc98ecad : SUCCESS in 7m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f0344255caf48708d8be85c19fdd6c1 : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69cdf2b51a504f3da693d1fb0ca0227c : SUCCESS in 15m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd0f7cb30067491e96f5f8cdf2691e70 : SUCCESS in 6m 00s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/51d1f8aad75c4f8daf13e2a04a7ad0a4 : SUCCESS in 9m 03s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/e39657adc77e4df7b8cebccd72da8163 : SUCCESS in 26m 29s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/23fd19dba7174401ab1039950278f3c6 : SUCCESS in 1h 12m 01s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/087bda4748af4295beec9d2b6235edfc : SUCCESS in 1h 08m 04s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/f946b690dad245f4984f9e392e3caa59 : FAILURE in 1h 12m 57s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/036db6a294d9493c8f5496ca66d465cf : SUCCESS in 47m 51s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/957ab0cd08a1460db03ce865f11e131c : SUCCESS in 50m 49s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/aaf0fe1ac51a49fca59b5c810da2d2fa : SUCCESS in 48m 34s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/ddc69969803e48ffa8a2d792a702c8ee : SUCCESS in 59m 11s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/6d7e0f5cbaf2433ba83943024d521692 : SUCCESS in 50m 54s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/194e8f15a60a4275801407b32048440d : SUCCESS in 1h 57m 10s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/7f5e642bdea44e488546b0c57dddc925 : SUCCESS in 1h 11m 30s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/1076a38d42a6427faa9d441c7cd5f94f : SUCCESS in 38m 57s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/5b0bb4c02d394d6eabc5dd380c837e85 : SUCCESS in 50m 58s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/63a942556b464a5fb72229fcdbae403e : SUCCESS in 1h 13m 08s","accounts_in_message":[],"_revision_number":8},{"id":"4ed288e0f04d4ae36e9365ced0750818248c6130","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2021-01-30 19:58:44.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"a0f5ea7fcb3ebfa72073f5296dfb8f6ee92dd4cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-30 20:12:56.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4cb68752d2754501b89d463096970d93 : SUCCESS in 9m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1f2b1e366a244a498be44d954357de5c : SUCCESS in 9m 02s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"1cacd6a7d669c084f2c83e0c727ab3fffab2605a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-30 22:28:26.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/391338fa15944a89a55583347d322b61 : SUCCESS in 9m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4822445a76d44bc09071d83793851ca8 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88032de7e4ee475480d938af5b58425c : SUCCESS in 7m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f75d7a4c10c478489e291066c9f046b : SUCCESS in 6m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dcf4a6cd50244dd5815c56c0044268fb : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/691e4958da9c4dfab0ec053424b88661 : SUCCESS in 17m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d16ea844b1247689ceb36c69e04dcf4 : SUCCESS in 5m 26s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/6f3fb7f2d51643029b39e8e7b040c17b : SUCCESS in 10m 56s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/a2c79e24aa3d4187bbb2532732fff405 : SUCCESS in 25m 56s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/1caa8cf01b344d1595062c05630a269f : SUCCESS in 1h 16m 19s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/1eb19b2527d34f81b4348c3044e80136 : SUCCESS in 58m 12s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/e2212a262b884cd580dd9fe833f7c1cc : SUCCESS in 1h 32m 48s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/4d81a43796f849dfbeec94e51a1375d0 : SUCCESS in 46m 19s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/2586864303a649398130fe567c20e69c : SUCCESS in 1h 06m 59s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/8c3a834a0e92449d96b2f2006582ce5c : SUCCESS in 53m 37s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/fc08f6764f2147d79d2318e4c2589a37 : SUCCESS in 1h 13m 00s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/fbc1a81f4f6e4b908979eb2b2cce04dc : SUCCESS in 48m 01s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/e6cc02fdd47c416087ce72a26c2b4a1b : FAILURE in 2h 27m 32s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/4d167bed4dfd497f832539e457154aa5 : SUCCESS in 1h 14m 29s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/8441dea55b5f4b4198bad9949b596c72 : SUCCESS in 39m 29s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/3a6d11af122240c9a42e25350a9c518a : SUCCESS in 47m 33s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/6ce5420d336e4673a732c6afee599fa3 : SUCCESS in 1h 06m 45s\n- ironic-tempest-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/698895d38eda49eba20e25beba18e1a3 : FAILURE in 30m 18s (non-voting)\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/4ecec10ee3cd42fb90947ad2bab3d09a : SUCCESS in 3m 31s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/247c2688302b4cb19cb202cc09789ac1 : SUCCESS in 48m 33s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/2234e4a5dd1742969ffafe73fc3828a2 : SUCCESS in 51m 07s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/10987614b35a40a5881ed06864f93511 : SUCCESS in 53m 54s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/1199dbef568c4ee796254dfd60c00c5e : SUCCESS in 54m 30s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c0f600aacc38483694b97b183f1e1a08 : SUCCESS in 25m 42s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/5be9221cdcfc4647aed53a5677bb2269 : SUCCESS in 23m 41s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/a5299a1fd6004c24ac46df560821edcb : SUCCESS in 47m 31s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"2bb84e160186a6abd711f7a921b27794e19d24ba","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2021-01-31 05:04:16.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"61cf79eda3771801900ba54e021c8b1d9c15d315","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-31 05:18:30.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6c9b067d231a4ab1856d8523e2db54ec : SUCCESS in 8m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/666aa2560df44e4bb332f0e56d315973 : SUCCESS in 10m 12s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"1f12e71bb78732b2a904952377b0cf1e5668074f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-31 06:36:06.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6497f188edc247cdbbbab6d88ae6af91 : SUCCESS in 7m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f78172e34d246deb0ab2d4686b29f77 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7e40fbb280b4f49a0ff58f1afe8d146 : SUCCESS in 8m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9e6fc36e937421694267bf820f5fdc8 : SUCCESS in 8m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/126f6fdbe834491194488f1bc54e0588 : SUCCESS in 8m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c3aae41ef404a85a1a2001a5be477d0 : SUCCESS in 17m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83a4c87ff67840d8809adcfbd135e0d9 : SUCCESS in 6m 40s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/af2d29be7a324877a72a6349f22a38e0 : SUCCESS in 9m 39s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/997f67d0592b453386ebb5545421617a : SUCCESS in 28m 30s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/2555be4d1bd94a41af373c315d805c67 : SUCCESS in 1h 15m 42s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/8fd40a2ed9a74ed48e1f460dd6c53004 : SUCCESS in 1h 15m 21s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/4e01ede636304d959a646ad02074ca5d : SUCCESS in 1h 03m 06s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/cfa51dbdd05d4809a3f3088ae7fac1eb : SUCCESS in 49m 30s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/3edce68765794896bfeff672b93fc5ca : SUCCESS in 53m 02s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/46624d4446b042b29bbd208d086df13b : SUCCESS in 47m 24s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/a2cda690fbc9479682a8d29df7314bfa : SUCCESS in 1h 02m 49s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/3b657b71a7394516b715d4dcad3a83a7 : SUCCESS in 52m 48s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/a9ab01854a43419ab98567307a8517bf : SUCCESS in 1h 25m 45s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/dcacbe6a331c41cf99fd878246652352 : SUCCESS in 1h 13m 48s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/94270e0d69004012ad8541d4928125e0 : SUCCESS in 40m 06s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/f3e1414cab7b491e8718c328c4ee8fbd : SUCCESS in 1h 11m 31s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/0735aa71eab94db68550a82bbd761d76 : SUCCESS in 1h 13m 55s\n- ironic-tempest-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/1cd19d5c0bd04ec6a08afc63342629c9 : FAILURE in 30m 13s (non-voting)\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/b4ec711f55334af597ab9f9aef8f2837 : SUCCESS in 4m 01s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/3edd22d9f4cd4f0f8b19a3f6b9fdfa4e : SUCCESS in 57m 29s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/cb91a850bd0947e9a07572a8ed91c7da : SUCCESS in 52m 42s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/64d2876430834156aa64a8e14788e18a : SUCCESS in 56m 31s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/6ab98eaad637442cb61b9b5d7b8bafa6 : SUCCESS in 53m 14s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f776e48d90d405e8227487e8d2a4ac6 : SUCCESS in 25m 05s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/7aeb649ce1b844a9bf6760144f3d15ca : SUCCESS in 23m 02s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/9c3941fb414e4bd1b45c02e4df417db5 : SUCCESS in 40m 08s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"9f6029eac0c8f7e64171fef1be8b03667c75d243","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-31 06:36:22.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"e293f38d1c668a4068ec0ad0b8fc595a6679b474","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-31 08:00:49.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83082ec4fbab4f3b8c85eb94f23180f3 : SUCCESS in 6m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5dd3aae9574f43fea55bd99bb1ea4ab0 : SUCCESS in 8m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9890c03c8dc4a2ba5a95d7f0321e63e : SUCCESS in 7m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ed37d200f8641a8b579f262a1cb3340 : SUCCESS in 15m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0043e2dd7c094a5d9db022f49962eaf8 : SUCCESS in 6m 56s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/2a1ea2130e0e4185a3b85518d6364635 : SUCCESS in 8m 27s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/690ae16f6c3d4a40b42f26a82cc7a462 : SUCCESS in 22m 20s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/16799a4a55224638af571a34b601496f : SUCCESS in 56m 48s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/1156047456b640c7982d1903a97951e8 : SUCCESS in 1h 11m 29s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/1b4e8c3d0eb6458e93310cd4b82ae93f : SUCCESS in 52m 05s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/20abccc8f38e4cb4bb71ddad2c188b88 : SUCCESS in 59m 53s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/e315169e22074c69b4f9817d5b911bc7 : SUCCESS in 54m 40s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/4507d110d7cd4aab963358f9cbb1f7ad : SUCCESS in 59m 25s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/b68bdb09d8944198831e318bd61b6740 : SUCCESS in 1h 11m 18s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/8ccb62d810f442bf8b5cfb1c1f226a92 : SUCCESS in 47m 31s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/265560346c374cb4ac145e9707e3541a : SUCCESS in 1h 21m 43s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/f03b72eb5b3345578a93515dd3dadba8 : SUCCESS in 1h 03m 52s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/fcc0af147ffd4c6598c5f43515f54cec : SUCCESS in 36m 44s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/2c03242869574395ba22395fd75a8905 : SUCCESS in 40m 42s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/054a48563049438f8101cd53e3bd1444 : SUCCESS in 57m 34s","accounts_in_message":[],"_revision_number":8},{"id":"7a8e04077613378427a75957924e5f6d4cc9459f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-31 08:00:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"c51ae623c149cdfae116ac5387f8eedfdb58f9ea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-31 08:02:27.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b5078d4d85f4f1b9e73a0fa1f3fb0a3 : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b1efdae5f5348929e89c658824b61c4 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"4a7d50ce56b88c82e95665ac6df44666d3e647ed","revisions":{"cc1bad3c0d37bb7317db43ae002e82e7d6af2038":{"kind":"REWORK","_number":1,"created":"2021-01-25 01:05:53.000000000","uploader":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"ref":"refs/changes/39/772239/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/772239/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/772239/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/772239/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/772239/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/772239/1"}}},"commit":{"parents":[{"commit":"4af9e498591c255336c0444b7c47d2f8cc4532be","subject":"Merge \"Update minversion of tox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4af9e498591c255336c0444b7c47d2f8cc4532be"}]}],"author":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-01-25 00:48:21.000000000","tz":-300},"committer":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-01-25 01:04:27.000000000","tz":-300},"subject":"[WIP] For Supermicro BMCs, set the BootSourceOverrideEnable value when changing boot device","message":"[WIP] For Supermicro BMCs, set the BootSourceOverrideEnable value when changing boot device\n\nWhen setting the boot settings on Supermicro BMCs, the BootSourceOverrideEnabled\nmust be set to the desired value whenever the BootSourceOverrideTarget is set or\nit will revert to the default value (Once).  This is different than what is\ncurrently implemented for other BMCs in which the BootSourceOverrideEnabled is\nnot set if it matches the current setting.\n\nThis change uses the vendor setting to determine if it\u0027s a Supermicro BMC.\n\n(Note the change to detect vendor for Redfish in\nhttps://review.opendev.org/c/openstack/ironic/+/771619 should be added here if\nthat patch doesn\u0027t merge, hence the WIP.)\n\nStory: 2008547\nTask: 41652\n\nChange-Id: I1b1a6baafd4cc4daa2fbdb82f69ded6253b1fcbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/cc1bad3c0d37bb7317db43ae002e82e7d6af2038"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/cc1bad3c0d37bb7317db43ae002e82e7d6af2038"}]},"branch":"refs/heads/master"},"d1f731934cd0fc233aca59d9f1ebd34313872cb2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-01-26 13:48:45.000000000","uploader":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"ref":"refs/changes/39/772239/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/772239/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/772239/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/772239/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/772239/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/772239/2"}}},"commit":{"parents":[{"commit":"4af9e498591c255336c0444b7c47d2f8cc4532be","subject":"Merge \"Update minversion of tox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4af9e498591c255336c0444b7c47d2f8cc4532be"}]}],"author":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-01-25 00:48:21.000000000","tz":-300},"committer":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-01-26 13:47:59.000000000","tz":-300},"subject":"For Supermicro BMCs, set the BootSourceOverrideEnable value when changing boot device","message":"For Supermicro BMCs, set the BootSourceOverrideEnable value when changing boot device\n\nWhen setting the boot settings on Supermicro BMCs, the BootSourceOverrideEnabled\nmust be set to the desired value whenever the BootSourceOverrideTarget is set or\nit will revert to the default value (Once).  This is different than what is\ncurrently implemented for other BMCs in which the BootSourceOverrideEnabled is\nnot set if it matches the current setting.\n\nThis change uses the vendor setting to determine if it\u0027s a Supermicro BMC.\n\nStory: 2008547\nTask: 41652\n\nChange-Id: I1b1a6baafd4cc4daa2fbdb82f69ded6253b1fcbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/d1f731934cd0fc233aca59d9f1ebd34313872cb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/d1f731934cd0fc233aca59d9f1ebd34313872cb2"}]},"branch":"refs/heads/master"},"9edcbfa722ec1e3235c1045f69e721e280441d71":{"kind":"REWORK","_number":3,"created":"2021-01-26 14:43:08.000000000","uploader":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"ref":"refs/changes/39/772239/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/772239/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/772239/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/772239/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/772239/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/772239/3"}}},"commit":{"parents":[{"commit":"4af9e498591c255336c0444b7c47d2f8cc4532be","subject":"Merge \"Update minversion of tox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4af9e498591c255336c0444b7c47d2f8cc4532be"}]}],"author":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-01-25 00:48:21.000000000","tz":-300},"committer":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-01-26 14:42:43.000000000","tz":-300},"subject":"For Supermicro BMCs, set the BootSourceOverrideEnable value when changing boot device","message":"For Supermicro BMCs, set the BootSourceOverrideEnable value when changing boot device\n\nWhen setting the boot settings on Supermicro BMCs, the BootSourceOverrideEnabled\nmust be set to the desired value whenever the BootSourceOverrideTarget is set or\nit will revert to the default value (Once).  This is different than what is\ncurrently implemented for other BMCs in which the BootSourceOverrideEnabled is\nnot set if it matches the current setting.\n\nThis change uses the vendor setting to determine if it\u0027s a Supermicro BMC.\n\nStory: 2008547\nTask: 41652\n\nChange-Id: I1b1a6baafd4cc4daa2fbdb82f69ded6253b1fcbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/9edcbfa722ec1e3235c1045f69e721e280441d71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/9edcbfa722ec1e3235c1045f69e721e280441d71"}]},"branch":"refs/heads/master"},"a9f716cff047a5694e19bd0d96cd7ca7b91b0425":{"kind":"REWORK","_number":4,"created":"2021-01-28 13:15:12.000000000","uploader":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"ref":"refs/changes/39/772239/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/772239/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/772239/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/772239/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/772239/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/772239/4"}}},"commit":{"parents":[{"commit":"4af9e498591c255336c0444b7c47d2f8cc4532be","subject":"Merge \"Update minversion of tox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4af9e498591c255336c0444b7c47d2f8cc4532be"}]}],"author":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-01-25 00:48:21.000000000","tz":-300},"committer":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-01-28 13:14:57.000000000","tz":-300},"subject":"For Supermicro BMCs, set the BootSourceOverrideEnable value when changing boot device","message":"For Supermicro BMCs, set the BootSourceOverrideEnable value when changing boot device\n\nWhen setting the boot settings on Supermicro BMCs, the BootSourceOverrideEnabled\nmust be set to the desired value whenever the BootSourceOverrideTarget is set or\nit will revert to the default value (Once).  This is different than what is\ncurrently implemented for other BMCs in which the BootSourceOverrideEnabled is\nnot set if it matches the current setting.\n\nThis change uses the vendor setting to determine if it\u0027s a Supermicro BMC.\n\nStory: 2008547\nTask: 41652\n\nChange-Id: I1b1a6baafd4cc4daa2fbdb82f69ded6253b1fcbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/a9f716cff047a5694e19bd0d96cd7ca7b91b0425"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/a9f716cff047a5694e19bd0d96cd7ca7b91b0425"}]},"branch":"refs/heads/master"},"55fbe4b3dd426d49a2fbb29f98d0e92c50eb5907":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-01-28 17:00:48.000000000","uploader":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"ref":"refs/changes/39/772239/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/772239/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/772239/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/772239/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/772239/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/772239/5"}}},"commit":{"parents":[{"commit":"4c4c7a869a47d5f9d3bf3244dcb88420189adb1d","subject":"Add a few words about UEFI user images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4c4c7a869a47d5f9d3bf3244dcb88420189adb1d"}]}],"author":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-01-25 00:48:21.000000000","tz":-300},"committer":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-01-28 16:59:53.000000000","tz":-300},"subject":"For Supermicro BMCs, set the BootSourceOverrideEnable value when changing boot device","message":"For Supermicro BMCs, set the BootSourceOverrideEnable value when changing boot device\n\nWhen setting the boot settings on Supermicro BMCs, the BootSourceOverrideEnabled\nmust be set to the desired value whenever the BootSourceOverrideTarget is set or\nit will revert to the default value (Once).  This is different than what is\ncurrently implemented for other BMCs in which the BootSourceOverrideEnabled is\nnot set if it matches the current setting.\n\nThis change uses the vendor setting to determine if it\u0027s a Supermicro BMC.\n\nStory: 2008547\nTask: 41652\n\nChange-Id: I1b1a6baafd4cc4daa2fbdb82f69ded6253b1fcbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/55fbe4b3dd426d49a2fbb29f98d0e92c50eb5907"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/55fbe4b3dd426d49a2fbb29f98d0e92c50eb5907"}]},"branch":"refs/heads/master"},"09203bef925947e7dfc4a643f0835b6753e3fd1a":{"kind":"REWORK","_number":6,"created":"2021-01-28 19:15:51.000000000","uploader":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"ref":"refs/changes/39/772239/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/772239/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/772239/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/772239/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/772239/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/772239/6"}}},"commit":{"parents":[{"commit":"4c4c7a869a47d5f9d3bf3244dcb88420189adb1d","subject":"Add a few words about UEFI user images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4c4c7a869a47d5f9d3bf3244dcb88420189adb1d"}]}],"author":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-01-25 00:48:21.000000000","tz":-300},"committer":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-01-28 19:15:29.000000000","tz":-300},"subject":"For Supermicro BMCs set enable when changing boot device","message":"For Supermicro BMCs set enable when changing boot device\n\nWhen setting the boot settings on Supermicro BMCs, the\nBootSourceOverrideEnabled must be set to the desired\nvalue whenever the BootSourceOverrideTarget is set or\nit will revert to the default value (Once).  This is\ndifferent than what is currently implemented for other\nBMCs in which the BootSourceOverrideEnabled is not set\nif it matches the current setting.\n\nThis change uses the vendor setting to determine if it\u0027s\na Supermicro BMC.\n\nStory: 2008547\nTask: 41652\n\nChange-Id: I1b1a6baafd4cc4daa2fbdb82f69ded6253b1fcbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/09203bef925947e7dfc4a643f0835b6753e3fd1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/09203bef925947e7dfc4a643f0835b6753e3fd1a"}]},"branch":"refs/heads/master"},"8b5f30ad21ce987caecb6609e31cc8e4d670a58d":{"kind":"REWORK","_number":7,"created":"2021-01-29 16:46:53.000000000","uploader":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"ref":"refs/changes/39/772239/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/772239/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/772239/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/772239/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/772239/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/772239/7"}}},"commit":{"parents":[{"commit":"4c4c7a869a47d5f9d3bf3244dcb88420189adb1d","subject":"Add a few words about UEFI user images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4c4c7a869a47d5f9d3bf3244dcb88420189adb1d"}]}],"author":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-01-25 00:48:21.000000000","tz":-300},"committer":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-01-29 16:46:51.000000000","tz":-300},"subject":"For Supermicro BMCs set enable when changing boot device","message":"For Supermicro BMCs set enable when changing boot device\n\nWhen setting the boot settings on Supermicro BMCs, the\nBootSourceOverrideEnabled must be set to the desired\nvalue whenever the BootSourceOverrideTarget is set or\nit will revert to the default value (Once).  This is\ndifferent than what is currently implemented for other\nBMCs in which the BootSourceOverrideEnabled is not set\nif it matches the current setting.\n\nThis change uses the vendor setting to determine if it\u0027s\na Supermicro BMC.\n\nStory: 2008547\nTask: 41652\n\nChange-Id: I1b1a6baafd4cc4daa2fbdb82f69ded6253b1fcbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/8b5f30ad21ce987caecb6609e31cc8e4d670a58d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/8b5f30ad21ce987caecb6609e31cc8e4d670a58d"}]},"branch":"refs/heads/master"},"4a7d50ce56b88c82e95665ac6df44666d3e647ed":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2021-01-29 16:55:52.000000000","uploader":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"ref":"refs/changes/39/772239/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/772239/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/772239/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/772239/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/772239/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/772239/8"}}},"commit":{"parents":[{"commit":"121b3348c813eb075ca38bd264a9315ee3acc2fe","subject":"Refactor vendor detection and add Redfish implementation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/121b3348c813eb075ca38bd264a9315ee3acc2fe"}]}],"author":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-01-25 00:48:21.000000000","tz":-300},"committer":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-01-29 16:55:51.000000000","tz":-300},"subject":"For Supermicro BMCs set enable when changing boot device","message":"For Supermicro BMCs set enable when changing boot device\n\nWhen setting the boot settings on Supermicro BMCs, the\nBootSourceOverrideEnabled must be set to the desired\nvalue whenever the BootSourceOverrideTarget is set or\nit will revert to the default value (Once).  This is\ndifferent than what is currently implemented for other\nBMCs in which the BootSourceOverrideEnabled is not set\nif it matches the current setting.\n\nThis change uses the vendor setting to determine if it\u0027s\na Supermicro BMC.\n\nStory: 2008547\nTask: 41652\n\nChange-Id: I1b1a6baafd4cc4daa2fbdb82f69ded6253b1fcbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4a7d50ce56b88c82e95665ac6df44666d3e647ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4a7d50ce56b88c82e95665ac6df44666d3e647ed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
