)]}'
{"id":"openstack%2Fironic~776439","triplet_id":"openstack%2Fironic~stable%2Fvictoria~I7e48b2e682dc0d5e6109e17b0e73ee9763a29d23","project":"openstack/ironic","branch":"stable/victoria","topic":"gate-stable/victoria","hashtags":[],"change_id":"I7e48b2e682dc0d5e6109e17b0e73ee9763a29d23","subject":"Switch multinode jobs to 512M RAM","status":"MERGED","created":"2021-02-18 13:27:24.000000000","updated":"2021-02-18 18:02:46.000000000","submitted":"2021-02-18 17:55:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"776439-gate-stable/victoria","cherry_pick_of_change":775666,"cherry_pick_of_patch_set":1,"meta_rev_id":"c0e6cad8d2e217bfbe477993a2f7cf6d3517f958","_number":776439,"virtual_id_number":776439,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-18 17:55:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-18 13:29:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-18 13:29:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-18 13:28:49.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2021-02-18 13:29:49.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2021-02-18 15:18:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"78aafa03ec2e9c5a543934faa9a116979e91971d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-02-18 13:27:24.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6b2713402f487326b5870a63e350c0a12896b9d3","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-02-18 13:28:49.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"38f8f918932bc132a85df3a6435ca7bf0c1935cc","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-02-18 13:29:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a077b1a8022601395cad7e185bb03951cce70516","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 15:18:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed78010a3d814a2a904a59ffbf1fc2bb : SUCCESS in 8m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33da7d49b1e8491fad11fd5fd52f5064 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/781f3cf2127b4446868108abfd8ab434 : SUCCESS in 7m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8180041dd8da4ddfa93075a9a53c1c53 : SUCCESS in 6m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ccf557093ef46e2931a82e79169ae89 : SUCCESS in 17m 43s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/395775f1b29341649353280c01525dec : SUCCESS in 10m 02s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/b9607c52c2af42b693d42d78bcf0d1a4 : SUCCESS in 1h 17m 55s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/d345621c992b41d180d0b3da3da6d787 : SUCCESS in 1h 14m 04s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/2676883f7ab84838b99f980c4f57d8e8 : SUCCESS in 26m 57s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/c5773f7397ed40f99ce38d3f17ce44cd : SUCCESS in 1h 17m 22s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/a97905123f674b9399852aaaa0abbb8a : SUCCESS in 49m 39s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/884ca288f48a47e6b6dc0ce086aab28a : SUCCESS in 49m 57s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/932d4bf7404243eba4105a2cfbb00517 : SUCCESS in 53m 47s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/a32be0dd5ef14855aa1b495a6dc688bf : SUCCESS in 1h 04m 31s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/885e924cb965487896141e5773444dff : SUCCESS in 56m 46s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/2315f10dc2df47cda375964708e37df0 : SUCCESS in 1h 46m 06s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/7c9366ea69824883b73e9ca08ebbc4d8 : SUCCESS in 1h 09m 53s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/0d8e52d37f5d456aab2d8c97f5a53e17 : SUCCESS in 40m 44s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/2e5e3ec52be8475fb206ce793520aeed : SUCCESS in 47m 57s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/a37bd46404624fd1b5ab253d4d94e981 : SUCCESS in 1h 06m 00s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/af24b5fea78c4114b6bcf7603bb67bed : SUCCESS in 5m 46s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/d3f21b311e9045ef880427e05de041eb : SUCCESS in 51m 02s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/1fb432d1b5fb49a7bdf241b9c9465069 : SUCCESS in 54m 10s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/21cb4c2fcc0c425aad015d406d3b1e04 : SUCCESS in 48m 28s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/020331999e624dfaa7ee6bbac784affd : SUCCESS in 53m 23s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/21f807f2f8bf4a81be4a279c4724f87e : SUCCESS in 22m 27s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/687bdef53695491b84b04444da4031af : SUCCESS in 54m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"369bfb428f23ca3837f382ffc532487f80cbd28f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 15:18:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7c548b0ae9c58a55738b8f508dd39feb871b37ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 17:55:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/425fd5eb5d8b412fb8711541b3d48fdf : SUCCESS in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/296281ee835d4ea6a6f3e0138e4981a0 : SUCCESS in 10m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f58e597a73ca476381ec9645fa21071e : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3bd0030a34946cdb88e3699a4bfde5e : SUCCESS in 17m 07s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/c66d1d80c36342fcbee0be3239d2ac40 : SUCCESS in 9m 22s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/f56304014de1479da7583f9103bb48db : SUCCESS in 50m 17s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/a391b8d3c77247d185ea7238932977a5 : SUCCESS in 53m 43s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/126b0a61b5424af18fa2a05d49ae5576 : SUCCESS in 28m 14s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/55243d98684f4615aefa09b0db163871 : SUCCESS in 1h 10m 06s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/916772bc7546453fbd103c8ea84a9603 : SUCCESS in 49m 21s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/1677ba9f8e844163a115f46c9e0bc1b4 : SUCCESS in 54m 27s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/e55d22c8723748e8941152a211ecdd57 : SUCCESS in 51m 15s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/0c95eae1ea4c4bcbb31339c6a6295d61 : SUCCESS in 1h 00m 06s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/df8a033e03984cb9a6409a3db3c988f4 : SUCCESS in 50m 32s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/29905bc314a54051b7bcb54e1d0a3f25 : SUCCESS in 2h 31m 45s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/a8e89ce0b77f45928bf4eeba8a561b23 : SUCCESS in 1h 16m 29s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/d023d2c086b8429bb7232f0ff9bb0b71 : SUCCESS in 47m 12s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/28d68e7f8e3845ddb0bfefecc5acaaab : SUCCESS in 49m 31s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/d70edbcf495541fd91b850cc9f5d7174 : SUCCESS in 1h 12m 01s","accounts_in_message":[],"_revision_number":1},{"id":"7378f4b899739f7348868133acd6287ec166aaac","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 17:55:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c0e6cad8d2e217bfbe477993a2f7cf6d3517f958","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 18:02:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1157846ec178477da9e9ef8443eaa4d9 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"73a600afa2a46d24f16f39f8665ea38d7194181a","revisions":{"73a600afa2a46d24f16f39f8665ea38d7194181a":{"kind":"REWORK","_number":1,"created":"2021-02-18 13:27:24.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/39/776439/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/776439/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/776439/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/776439/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/776439/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/776439/1"}}},"commit":{"parents":[{"commit":"78924eca236a7ec81fe53a01ed57719b0458e1f7","subject":"Move the IPv6 job to the experimental pipeline","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/78924eca236a7ec81fe53a01ed57719b0458e1f7"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-02-15 15:20:24.000000000","tz":60},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-02-18 13:27:24.000000000","tz":0},"subject":"Switch multinode jobs to 512M RAM","message":"Switch multinode jobs to 512M RAM\n\n384M no longer works reliably with newer tinyIPA.\n\nChange-Id: I7e48b2e682dc0d5e6109e17b0e73ee9763a29d23\n(cherry picked from commit 414f0ca24e797b65bab6276ba651a62729c72637)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/73a600afa2a46d24f16f39f8665ea38d7194181a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/73a600afa2a46d24f16f39f8665ea38d7194181a"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
