)]}'
{"id":"openstack%2Fironic~777900","triplet_id":"openstack%2Fironic~stable%2Fvictoria~Id5fcd4cc1f73b80e8a9e9d2c50e2e4e1667c01cb","project":"openstack/ironic","branch":"stable/victoria","topic":"gate-stable/victoria","hashtags":[],"change_id":"Id5fcd4cc1f73b80e8a9e9d2c50e2e4e1667c01cb","subject":"devstack: a safeguard for disabled tempurls","status":"MERGED","created":"2021-02-28 16:33:59.000000000","updated":"2021-02-28 20:59:06.000000000","submitted":"2021-02-28 20:57:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"777900-gate-stable/victoria","cherry_pick_of_change":777572,"cherry_pick_of_patch_set":1,"meta_rev_id":"00f7c53a34148eac6d59d24936ede2ccdf2c3d28","_number":777900,"virtual_id_number":777900,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-28 20:57:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-28 19:04:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-28 19:04:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-28 16:35:37.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2021-02-28 17:32:10.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2021-02-28 17:52:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-28 19:04:03.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"}],"messages":[{"id":"27b9408a80aee01cfa3066bc6b17d57330fb23f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-02-28 16:33:59.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"cae5216a9ecec3bbcbd9ce2de94ed705bf512cc5","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-02-28 16:35:37.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"5ad5e6f89e9f243971b7369deb196e18a8ffe130","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-02-28 17:32:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- 14G_uefi_ipmi_victoria https://stash.dellemc-community.org/logs/14G_uefi_ipmi_victoria/7 : SUCCESS in Xs\n- 14G_uefi_idrac_victoria https://stash.dellemc-community.org/logs/14G_uefi_idrac_victoria/8 : SUCCESS in Xs\n- R7515_uefi_redfish_victoria https://stash.dellemc-community.org/logs/R7515_uefi_redfish_victoria/6 : SUCCESS in Xs\n- 14G_uefi_redfish_victoria https://stash.dellemc-community.org/logs/14G_uefi_redfish_victoria/10 : SUCCESS in Xs\n- XE2420_uefi_redfish_victoria https://stash.dellemc-community.org/logs/XE2420_uefi_redfish_victoria/6 : SUCCESS in Xs\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"ee2562bc0a9f67615f6f0340e5a55f9064ae3d28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-28 17:52:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10335154f3c14418a178d7343e1e0aea : SUCCESS in 8m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab66a66e996b48a5ad56e72e9746723d : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7294a1954e284dfaab514d437d038997 : SUCCESS in 8m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06adf626d56e4f1686bc237228fbfed8 : SUCCESS in 9m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87ef4cab34fb428bb03d306c351869d3 : SUCCESS in 14m 43s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/1c2f45235ac048b2839eb871f7150249 : SUCCESS in 7m 28s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/53d4420d6bb147f9ba788301d895b382 : SUCCESS in 45m 14s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/f5567fe6d1ee455e8bc0f676a899a247 : SUCCESS in 51m 53s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/1e230c5b57c243e5bc8954dbc9305cd6 : SUCCESS in 28m 33s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/77dcaddf03344e499a292b1a820c8bd0 : SUCCESS in 51m 01s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/a9645b281f674fbcbacee71692ca8f66 : SUCCESS in 47m 56s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/07236c17fb634b91b6439730605da018 : SUCCESS in 1h 11m 35s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/1feddaa32ab74b93899e2e1d313c84f1 : SUCCESS in 48m 20s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/fc68e3d1bbd1460ba4d9d44f57009258 : SUCCESS in 49m 41s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/a03d7a69eb124ff19927f391165ae174 : SUCCESS in 52m 23s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/1c0415c4ce144412b454148f06d1f77e : SUCCESS in 1h 06m 28s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/760f1cd198b34d9a92577b5a51e280f3 : SUCCESS in 56m 47s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/68b101429f9f45778566d2c8ea5958d1 : SUCCESS in 34m 43s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/094e7e04fa3e45dbbdc7d7794fa758ab : SUCCESS in 42m 01s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/873f0ccea4c34d049ec617aff856d362 : SUCCESS in 48m 51s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/6eb7b496d6f643e19ee7a1ebf632d77b : SUCCESS in 5m 33s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/5617b52c6c2245cd81936916d2572d23 : SUCCESS in 1h 14m 08s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/b615fe49a7884b2e81d33da2bfe78095 : SUCCESS in 1h 13m 30s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/b36e88594703469d86afd7bb14d86f25 : SUCCESS in 42m 25s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/abcd123428bc4e1ab0bda121ad3e5a65 : SUCCESS in 38m 54s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a46566c9d2fd4f67b70e7e5461e9bb3a : SUCCESS in 18m 03s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/7fa63f58ac824fd49fa07695f48518d8 : SUCCESS in 51m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9999f8423c906ab52f3b99d0f5570108851930c5","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-02-28 19:04:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bfca967882efca3a3d8ed5d10ef153b97fa8af2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-28 19:04:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0c10edf1c4a296292dda8aa58a72b6304e4cfdbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-28 20:57:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/886c1388c3c74d9f847fd9e53a9daec5 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f68c82b95264dd1a1e3b7806336b044 : SUCCESS in 7m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db9cecc3812e4227bf75f929fed2204a : SUCCESS in 7m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5489434ea9e047d7a764353878d42285 : SUCCESS in 16m 36s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/b5254794210742a7b23844da4e4a26cb : SUCCESS in 8m 43s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/56564d3ce3ce4c1483845514748d08c4 : SUCCESS in 1h 02m 21s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/005d769c1df944649614056266af0242 : SUCCESS in 1h 10m 11s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/8b3f7265fd374b3da285f4d4c9b68d27 : SUCCESS in 29m 59s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/0c2f952e00744f5e95b213fe65867077 : SUCCESS in 1h 15m 51s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/a75e2c3e0d0a4beca7f284659b7ea007 : SUCCESS in 49m 07s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/614c858b468b489f8783cfbabda76b18 : SUCCESS in 51m 40s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/3dc0eb17b72e4cacb7d9bac51da3bd6e : SUCCESS in 49m 26s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/edfcfaed122f41ad9ff72cd8a40e0692 : SUCCESS in 1h 00m 22s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/430a9cdf190149a5a99d2d51e60fb352 : SUCCESS in 53m 23s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/c67e33a7c8424f9294426a4bbfbbab74 : SUCCESS in 1h 47m 46s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/eae769b338674f35a4df27babae4d924 : SUCCESS in 1h 12m 40s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/254a63458082434586f3403464314fd5 : SUCCESS in 43m 02s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/1c3a03fc50034595a589e0c221a0e6a2 : SUCCESS in 51m 44s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/ffa5c1a48b0c4addace90106b4c729f7 : SUCCESS in 1h 16m 25s","accounts_in_message":[],"_revision_number":1},{"id":"5c20c4f38786f3abb72b9a6ab53cea5e42137e9e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-28 20:57:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"00f7c53a34148eac6d59d24936ede2ccdf2c3d28","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-28 20:59:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b5c9c01f9744fc1b2963a0ae31446b4 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"90da180a17852eab6af4b038fbd92989feccb5f8","revisions":{"90da180a17852eab6af4b038fbd92989feccb5f8":{"kind":"REWORK","_number":1,"created":"2021-02-28 16:33:59.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/00/777900/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/00/777900/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/00/777900/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/00/777900/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/00/777900/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/00/777900/1"}}},"commit":{"parents":[{"commit":"a1f596590235d7b76be1d6184a2f6d44857f5eb0","subject":"Enable swift temporary URLs in grenade and provide a good error message","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/a1f596590235d7b76be1d6184a2f6d44857f5eb0"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-02-25 11:09:30.000000000","tz":60},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-02-28 16:33:59.000000000","tz":0},"subject":"devstack: a safeguard for disabled tempurls","message":"devstack: a safeguard for disabled tempurls\n\nChange-Id: Id5fcd4cc1f73b80e8a9e9d2c50e2e4e1667c01cb\n(cherry picked from commit 7abac806a7f889fc87120c9e2c3a61b3a35f6458)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/90da180a17852eab6af4b038fbd92989feccb5f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/90da180a17852eab6af4b038fbd92989feccb5f8"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
