)]}'
{"id":"openstack%2Fironic~779552","triplet_id":"openstack%2Fironic~master~Iac0c6fdb116c14a402c0ea86f98f6a72064adbbc","project":"openstack/ironic","branch":"master","topic":"task/42024","hashtags":[],"change_id":"Iac0c6fdb116c14a402c0ea86f98f6a72064adbbc","subject":"Update Redfish RAID release note","status":"MERGED","created":"2021-03-09 18:22:35.000000000","updated":"2021-03-15 15:44:08.000000000","submitted":"2021-03-15 15:40:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"779552-task/42024","meta_rev_id":"87db7d7e741cdf6a797af7c26cd08aad19b13330","_number":779552,"virtual_id_number":779552,"owner":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-15 15:40:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"recommended":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"all":[{"value":1,"date":"2021-03-10 15:13:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":2,"date":"2021-03-15 15:12:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":1,"date":"2021-03-15 15:12:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"}],"CC":[{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-09 18:24:46.000000000","updated_by":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"reviewer":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"state":"REVIEWER"},{"updated":"2021-03-09 18:24:46.000000000","updated_by":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2021-03-09 18:24:46.000000000","updated_by":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2021-03-09 18:48:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-09 19:29:42.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2021-03-10 13:59:51.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"ac9c41ca1db9a34971e548847b3ba888734cd95d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"date":"2021-03-09 18:22:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f8c3937518267217cdd919598631edb069531a56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 18:48:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68b8d365607e43c9bbca9a836907f139 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/857a6fa9978640c49d21a6205de82303 : SUCCESS in 16m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27aed5dafb114fceb401630d4b0999d8 : SUCCESS in 7m 20s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/0d13971682774806beee2b58c9762ee1 : SUCCESS in 13m 37s","accounts_in_message":[],"_revision_number":1},{"id":"a4219dd0f06c8defd7bec93d37efe2525daf6af4","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-03-09 19:29:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).\n\n- 14G_uefi_ipmi https://stash.dellemc-community.org/logs/14G_uefi_ipmi/311 : SUCCESS in Xs\n- R6515_bios_idrac https://stash.dellemc-community.org/logs/R6515_bios_idrac/69 : SUCCESS in Xs\n- 14G_bios_idrac https://stash.dellemc-community.org/logs/14G_bios_idrac/69 : FAILURE in Xs\n- 14G_uefi_idrac https://stash.dellemc-community.org/logs/14G_uefi_idrac/318 : FAILURE in Xs\n- 13G_bios_ipmi https://stash.dellemc-community.org/logs/13G_bios_ipmi/69 : SUCCESS in Xs\n- R7515_uefi_redfish https://stash.dellemc-community.org/logs/R7515_uefi_redfish/351 : SUCCESS in Xs\n- XE2420_uefi_redfish https://stash.dellemc-community.org/logs/XE2420_uefi_redfish/353 : SUCCESS in Xs\n- 14G_bios_redfish https://stash.dellemc-community.org/logs/14G_bios_redfish/69 : SUCCESS in Xs\n- 14G_uefi_redfish https://stash.dellemc-community.org/logs/14G_uefi_redfish/354 : SUCCESS in Xs\n- 13G_bios_idrac https://stash.dellemc-community.org/logs/13G_bios_idrac/69 : FAILURE in Xs\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"5ecba1a9c2141fc15f89d22f97a936656446d772","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2021-03-10 08:21:21.000000000","message":"Patch Set 1: Code-Review+1\n\nthank you","accounts_in_message":[],"_revision_number":1},{"id":"be69a1bbcbc866eae4b0ecc579977addf531c11f","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-03-10 13:59:51.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cd9ee7e3b2eeb7d7b91eba26c2d64516d4c4b80f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"date":"2021-03-10 14:40:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"511465e326157a775243325d856fb434c3037ce5","author":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"date":"2021-03-10 14:44:09.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nThanks Dmitry and Aija for the reviews. I collapsed the note into one item.","accounts_in_message":[],"_revision_number":2},{"id":"b5e254dcc58fefef382d0859e295317f2a668c1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 15:01:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/730115515afc42048333708f413a386b : SUCCESS in 7m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1674301fbb15493bb4e54768f7b9ea3b : SUCCESS in 14m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/68ea8a95ebeb4f0b93cac4bb69fa4283 : SUCCESS in 7m 31s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/464c1760523241c4929a9bbb8b63e866 : SUCCESS in 16m 55s","accounts_in_message":[],"_revision_number":2},{"id":"3209b2f4b2e2a89c573d4f08ee7c8fd6e9854d3d","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2021-03-10 15:13:42.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"22a4db645fda4c45e67868bd0b49dfd250cb45ec","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-03-10 16:32:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- 13G_bios_ipmi https://stash.dellemc-community.org/logs/13G_bios_ipmi/78 : SUCCESS in Xs\n- 14G_bios_redfish https://stash.dellemc-community.org/logs/14G_bios_redfish/78 : SUCCESS in Xs\n- R6515_bios_idrac https://stash.dellemc-community.org/logs/R6515_bios_idrac/78 : SUCCESS in Xs\n- 14G_bios_idrac https://stash.dellemc-community.org/logs/14G_bios_idrac/78 : SUCCESS in Xs\n- R7515_uefi_redfish https://stash.dellemc-community.org/logs/R7515_uefi_redfish/360 : SUCCESS in Xs\n- 14G_uefi_ipmi https://stash.dellemc-community.org/logs/14G_uefi_ipmi/320 : SUCCESS in Xs\n- 14G_uefi_idrac https://stash.dellemc-community.org/logs/14G_uefi_idrac/327 : SUCCESS in Xs\n- XE2420_uefi_redfish https://stash.dellemc-community.org/logs/XE2420_uefi_redfish/362 : SUCCESS in Xs\n- 14G_uefi_redfish https://stash.dellemc-community.org/logs/14G_uefi_redfish/363 : SUCCESS in Xs\n- 13G_bios_idrac https://stash.dellemc-community.org/logs/13G_bios_idrac/78 : SUCCESS in Xs\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":2},{"id":"56e09af855fdab8a7252c9db545b995c79a876a5","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-03-15 15:12:20.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7e05ec31458d1725be7cff76e5f95be8c93ae52c","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-03-15 15:12:26.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"dbbb10432374eb5c8461ef14aa58fcf0f64b41e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 15:12:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"13f764edd008d5700f82141297664c374101d335","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 15:40:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3c80a88e4ab4a269da94e0072256eda : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a8ce9b901474ad3b140d4b20395c16f : SUCCESS in 17m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5afacd276e4403da0fb65448a863674 : SUCCESS in 7m 33s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/c7a50f15c46a46d5be199794f92e890d : SUCCESS in 16m 58s","accounts_in_message":[],"_revision_number":2},{"id":"cd2aac505d701891d88a48a8b0f2312928cdc2db","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 15:40:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"87db7d7e741cdf6a797af7c26cd08aad19b13330","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 15:44:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4f847f2165c40cb84148cc9c7a4c0ad : SUCCESS in 1m 25s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3aebe09e5c24c4e8495c92a4535e775 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5f85a896643a823361d564612cef677f3438dfac","revisions":{"1d025150f241dc49fef2ad4dfc7c5c3aa5113428":{"kind":"REWORK","_number":1,"created":"2021-03-09 18:22:35.000000000","uploader":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"ref":"refs/changes/52/779552/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/52/779552/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/52/779552/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/52/779552/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/52/779552/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/52/779552/1"}}},"commit":{"parents":[{"commit":"4a3bbf6d28eeede0bab36cf7ed412192c2300a1f","subject":"Merge \"Adds config parameter kernel_append_param for iLO\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4a3bbf6d28eeede0bab36cf7ed412192c2300a1f"}]}],"author":{"name":"Bill Dodd","email":"billdodd@gmail.com","date":"2021-03-09 18:20:43.000000000","tz":-360},"committer":{"name":"Bill Dodd","email":"billdodd@gmail.com","date":"2021-03-09 18:20:43.000000000","tz":-360},"subject":"Update Redfish RAID release note","message":"Update Redfish RAID release note\n\nThe Redfish RAID release note did not reference the related\nstory. This patch updates the release note to include a link to\nthe story.\n\nThe formatting of the release note is also updated to respect the\nmaximum 80 characters per line convention.\n\nChange-Id: Iac0c6fdb116c14a402c0ea86f98f6a72064adbbc\nStory: 2003514\nTask: 42024\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1d025150f241dc49fef2ad4dfc7c5c3aa5113428"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1d025150f241dc49fef2ad4dfc7c5c3aa5113428"}]},"branch":"refs/heads/master"},"5f85a896643a823361d564612cef677f3438dfac":{"kind":"REWORK","_number":2,"created":"2021-03-10 14:40:29.000000000","uploader":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"ref":"refs/changes/52/779552/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/52/779552/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/52/779552/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/52/779552/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/52/779552/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/52/779552/2"}}},"commit":{"parents":[{"commit":"4a3bbf6d28eeede0bab36cf7ed412192c2300a1f","subject":"Merge \"Adds config parameter kernel_append_param for iLO\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4a3bbf6d28eeede0bab36cf7ed412192c2300a1f"}]}],"author":{"name":"Bill Dodd","email":"billdodd@gmail.com","date":"2021-03-09 18:20:43.000000000","tz":-360},"committer":{"name":"Bill Dodd","email":"billdodd@gmail.com","date":"2021-03-10 14:39:18.000000000","tz":-360},"subject":"Update Redfish RAID release note","message":"Update Redfish RAID release note\n\nThe Redfish RAID release note did not reference the related\nstory. This patch updates the release note to include a link to\nthe story.\n\nThe formatting of the release note is also updated to respect the\nmaximum 80 characters per line convention.\n\nChange-Id: Iac0c6fdb116c14a402c0ea86f98f6a72064adbbc\nStory: 2003514\nTask: 42024\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/5f85a896643a823361d564612cef677f3438dfac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/5f85a896643a823361d564612cef677f3438dfac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
