)]}'
{"id":"openstack%2Fironic~780973","triplet_id":"openstack%2Fironic~master~I2799fb1634747cde451a787c567711dcdad2d688","project":"openstack/ironic","branch":"master","hashtags":["ironic-week-prio"],"change_id":"I2799fb1634747cde451a787c567711dcdad2d688","subject":"Revise release notes for 17.0 release","status":"MERGED","created":"2021-03-16 20:40:19.000000000","updated":"2021-03-23 11:43:56.000000000","submitted":"2021-03-23 11:36:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"780973","meta_rev_id":"4ebd21d95cc521ef38effc277b2a65b83e619d0d","_number":780973,"virtual_id_number":780973,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-23 11:36:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2021-03-22 18:11:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-23 09:57:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-23 09:57:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}],"CC":[{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-16 21:04:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-16 22:50:06.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2021-03-17 11:08:50.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"CC"},{"updated":"2021-03-22 15:35:49.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2021-03-23 09:57:51.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"d8df7b221cc6ef0cbf292b7a7ba2521b4c39e626","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-03-16 20:40:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a6ecbd876912131a39301555361cb75c6ffa8c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 21:04:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe9c458317fc4f238fc4d0fe81ffd9a5 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02f1df24da1446b99c960c55cef92506 : SUCCESS in 16m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9734022cfce249c2b3fd4b36aafaa752 : SUCCESS in 7m 07s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/bfe811a3008c44eab80dabaeb5391cf8 : SUCCESS in 11m 03s","accounts_in_message":[],"_revision_number":1},{"id":"52e429bc9d669ae98f2bb0315437f39ab2031cb6","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-03-16 22:50:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).\n\n- 14G_uefi_redfish https://stash.dellemc-community.org/logs/14G_uefi_redfish/427 : SUCCESS in 2h 8m 11s\n- XE2420_uefi_redfish https://stash.dellemc-community.org/logs/XE2420_uefi_redfish/426 : SUCCESS in 1h 41m 15s\n- R6515_bios_idrac https://stash.dellemc-community.org/logs/R6515_bios_idrac/141 : SUCCESS in 1h 36m 29s\n- 13G_bios_idrac https://stash.dellemc-community.org/logs/13G_bios_idrac/142 : SUCCESS in 1h 19m 15s\n- 13G_bios_ipmi https://stash.dellemc-community.org/logs/13G_bios_ipmi/142 : SUCCESS in 1h 2m 48s\n- 14G_uefi_ipmi https://stash.dellemc-community.org/logs/14G_uefi_ipmi/384 : SUCCESS in 53m 27s\n- 14G_bios_idrac https://stash.dellemc-community.org/logs/14G_bios_idrac/142 : SUCCESS in 51m 50s\n- 14G_uefi_idrac https://stash.dellemc-community.org/logs/14G_uefi_idrac/391 : SUCCESS in 27m 1s\n- 14G_bios_redfish https://stash.dellemc-community.org/logs/14G_bios_redfish/142 : SUCCESS in 20m 46s\n- R7515_uefi_redfish https://stash.dellemc-community.org/logs/R7515_uefi_redfish/424 : FAILURE in 10m 54s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"c290680437a5970e3579e99d39f523bb3a89d976","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-03-17 11:08:50.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"4156b47c61b0ad7fde434710a8f27e4d5f36eea0","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-03-22 15:33:51.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":1},{"id":"85dfbb915382774071dffe48d74a686ce668cc32","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-03-22 15:35:49.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"b98ddc5858aedf95061b750ea60a06fa3cbf2947","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-03-22 17:19:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a6143579ab3ac66bdb12718c156c67194a3752fb","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-03-22 17:19:55.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"1c9563bec9593fd1e5b4dbc0ecc1345657834672","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-03-22 17:39:06.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"8365d0c84fe8ee6753e3624f60bf9d5439268abf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 17:57:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e24a2897cfc54027a8948d0475b8a3be : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7eb35dc5673450db842524b8ec7019f : SUCCESS in 16m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec2582daad304ec3b5d275a82b17dd07 : SUCCESS in 7m 55s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/60753a11037042c5bb4f2b623a0d1900 : SUCCESS in 13m 24s","accounts_in_message":[],"_revision_number":3},{"id":"e781b9fb40a85c2be20b861d03d38289788bf417","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-03-22 18:11:19.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"6484e25bc90eb70524556e9d081c0f1d6dd6b1d0","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-03-22 19:40:28.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).\n\n- 13G_bios_idrac https://stash.dellemc-community.org/logs/13G_bios_idrac/205 : SUCCESS in 2h 20m 12s\n- XE2420_uefi_redfish https://stash.dellemc-community.org/logs/XE2420_uefi_redfish/489 : SUCCESS in 2h 7m 2s\n- 14G_bios_idrac https://stash.dellemc-community.org/logs/14G_bios_idrac/205 : SUCCESS in 1h 55m 48s\n- 14G_uefi_idrac https://stash.dellemc-community.org/logs/14G_uefi_idrac/454 : SUCCESS in 1h 41m 9s\n- 14G_bios_redfish https://stash.dellemc-community.org/logs/14G_bios_redfish/205 : SUCCESS in 1h 19m 19s\n- 14G_uefi_ipmi https://stash.dellemc-community.org/logs/14G_uefi_ipmi/447 : SUCCESS in 1h 3m 52s\n- R7515_uefi_redfish https://stash.dellemc-community.org/logs/R7515_uefi_redfish/487 : SUCCESS in 1h 3m 12s\n- 13G_bios_ipmi https://stash.dellemc-community.org/logs/13G_bios_ipmi/205 : SUCCESS in 48m 37s\n- 14G_uefi_redfish https://stash.dellemc-community.org/logs/14G_uefi_redfish/490 : SUCCESS in 31m 6s\n- R6515_bios_idrac https://stash.dellemc-community.org/logs/R6515_bios_idrac/205 : FAILURE in 31m 6s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":2},{"id":"c9e5436bce1e60390926fc2646267d1c214f5b26","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-03-23 01:16:38.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).\n\n- XE2420_uefi_redfish https://stash.dellemc-community.org/logs/XE2420_uefi_redfish/490 : SUCCESS in 7h 36m 57s\n- 14G_uefi_ipmi https://stash.dellemc-community.org/logs/14G_uefi_ipmi/448 : SUCCESS in 6h 37m 51s\n- 14G_bios_redfish https://stash.dellemc-community.org/logs/14G_bios_redfish/206 : SUCCESS in 4h 18m 26s\n- 14G_bios_idrac https://stash.dellemc-community.org/logs/14G_bios_idrac/206 : SUCCESS in 3h 48m 39s\n- 13G_bios_ipmi https://stash.dellemc-community.org/logs/13G_bios_ipmi/206 : SUCCESS in 3h 46m 22s\n- 14G_uefi_redfish https://stash.dellemc-community.org/logs/14G_uefi_redfish/491 : SUCCESS in 3h 45m 22s\n- 13G_bios_idrac https://stash.dellemc-community.org/logs/13G_bios_idrac/206 : SUCCESS in 3h 13m 15s\n- R6515_bios_idrac https://stash.dellemc-community.org/logs/R6515_bios_idrac/206 : SUCCESS in 2h 34m 10s\n- R7515_uefi_redfish https://stash.dellemc-community.org/logs/R7515_uefi_redfish/488 : FAILURE in 1h 48m 29s\n- 14G_uefi_idrac https://stash.dellemc-community.org/logs/14G_uefi_idrac/455 : FAILURE in 1h 37m 35s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":3},{"id":"11ae5fb4c093ebea9efcd704eadecff76cc440d2","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-03-23 09:57:51.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f7a19585ae322017cf36570f954548b3cb9ac081","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 09:58:14.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c49b59892a39305b6135d5667515b8cb7b7e6b2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 11:36:17.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00cc25feaea0417c831a7d6517fcb0fe : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22ab114275214725be99f7244f4bdd58 : SUCCESS in 20m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f3727eefb774201b59bf3592892cb9a : SUCCESS in 7m 33s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/4757a427fd2c49ad8b4acdab17150fe5 : SUCCESS in 14m 24s","accounts_in_message":[],"_revision_number":3},{"id":"acb445e701c3e21ce486cddbe6386edf3863b1f2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 11:36:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"4ebd21d95cc521ef38effc277b2a65b83e619d0d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 11:43:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1946e76015874fd9a2be7b56114805c5 : SUCCESS in 1m 26s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da64e38e2fe04421963672c16f4f65f0 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"67394c3cd4502f4534c6c3e3d603d5a8d84dff2d","revisions":{"84e1a074b3fd0c16624dceb34c0f9d228654ab2d":{"kind":"REWORK","_number":1,"created":"2021-03-16 20:40:19.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/73/780973/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/73/780973/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/73/780973/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/73/780973/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/73/780973/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/73/780973/1"}}},"commit":{"parents":[{"commit":"5584cc4cbbc5dd7eb25280532bc8cad8c497bd0f","subject":"Merge \"Update dev quickstart docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/5584cc4cbbc5dd7eb25280532bc8cad8c497bd0f"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-03-16 17:23:31.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-03-16 17:23:31.000000000","tz":-420},"subject":"Revise release notes for 17.0 release","message":"Revise release notes for 17.0 release\n\nChange-Id: I2799fb1634747cde451a787c567711dcdad2d688\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/84e1a074b3fd0c16624dceb34c0f9d228654ab2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/84e1a074b3fd0c16624dceb34c0f9d228654ab2d"}]},"branch":"refs/heads/master"},"caeb9edf3c50e4fe86e813ce476d325f55eca8c3":{"kind":"REWORK","_number":2,"created":"2021-03-22 17:19:40.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/73/780973/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/73/780973/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/73/780973/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/73/780973/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/73/780973/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/73/780973/2"}}},"commit":{"parents":[{"commit":"5584cc4cbbc5dd7eb25280532bc8cad8c497bd0f","subject":"Merge \"Update dev quickstart docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/5584cc4cbbc5dd7eb25280532bc8cad8c497bd0f"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-03-16 17:23:31.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-03-22 17:19:30.000000000","tz":-420},"subject":"Revise release notes for 17.0 release","message":"Revise release notes for 17.0 release\n\nChange-Id: I2799fb1634747cde451a787c567711dcdad2d688\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/caeb9edf3c50e4fe86e813ce476d325f55eca8c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/caeb9edf3c50e4fe86e813ce476d325f55eca8c3"}]},"branch":"refs/heads/master"},"67394c3cd4502f4534c6c3e3d603d5a8d84dff2d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-03-22 17:39:06.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/73/780973/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/73/780973/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/73/780973/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/73/780973/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/73/780973/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/73/780973/3"}}},"commit":{"parents":[{"commit":"acd4b451dc866f4aee48b46f958cc46fd2220d09","subject":"Merge \"Use OOB inspection to fetch MACs for IB inspection\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/acd4b451dc866f4aee48b46f958cc46fd2220d09"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-03-16 17:23:31.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-03-22 17:24:52.000000000","tz":-420},"subject":"Revise release notes for 17.0 release","message":"Revise release notes for 17.0 release\n\nChange-Id: I2799fb1634747cde451a787c567711dcdad2d688\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/67394c3cd4502f4534c6c3e3d603d5a8d84dff2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/67394c3cd4502f4534c6c3e3d603d5a8d84dff2d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
