)]}'
{"id":"openstack%2Fironic~786335","triplet_id":"openstack%2Fironic~stable%2Ftrain~I1b1a6baafd4cc4daa2fbdb82f69ded6253b1fcbf","project":"openstack/ironic","branch":"stable/train","topic":"train/cache-vendor","hashtags":[],"change_id":"I1b1a6baafd4cc4daa2fbdb82f69ded6253b1fcbf","subject":"For Supermicro BMCs set enable when changing boot device","status":"MERGED","created":"2021-04-14 22:08:07.000000000","updated":"2021-04-20 15:19:00.000000000","submitted":"2021-04-20 15:19:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"786335-train/cache-vendor","meta_rev_id":"4130aa5e000361c4a47dff33f882e93e6ef55a08","_number":786335,"virtual_id_number":786335,"owner":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-20 15:18:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-15 11:09:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":2,"date":"2021-04-20 10:59:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":1,"date":"2021-04-20 10:59:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"}],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-14 22:08:07.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"state":"REVIEWER"},{"updated":"2021-04-14 22:08:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-14 22:08:57.000000000","updated_by":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-14 22:13:20.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2021-04-15 05:54:49.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2021-04-15 11:09:17.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2021-04-20 10:59:47.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"18a44f1484a1b53447daa2125124df39f8442a0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2021-04-14 22:08:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"030c073d19895a72c9fa949556a28aca67c05fc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 22:08:26.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"f1f054270bdd255207e5f2f2f105a2cdc3c704e6","author":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"date":"2021-04-14 22:08:57.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nAbout this CI test, see https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_iRMC_CI","accounts_in_message":[],"_revision_number":1},{"id":"cfd4964526a9419e2f1d809dbec188fcaa9f84ae","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-04-14 22:13:20.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"c0bdc90da92807d728da05b1f6cdc2210ff3cb1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2021-04-14 23:02:14.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"318f7bfe8df8047fdc8874d6f45bb08cdd3b8692","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-04-15 00:27:48.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/35/786335/2/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/359d038/359d038.html : FAILURE in 6m 58s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/35/786335/2/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3/34c8e28/34c8e28.html : FAILURE in 8m 27s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/35/786335/2/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3/e171084/e171084.html : FAILURE in 8m 24s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/35/786335/2/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/4cffd38/4cffd38.html : FAILURE in 8m 16s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/35/786335/2/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/371e793/371e793.html : FAILURE in 8m 06s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/35/786335/2/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/0cfc020/0cfc020.html : FAILURE in 7m 50s","accounts_in_message":[],"_revision_number":2},{"id":"2a3e3120a531d112770caa7f058995cb7ec8fe8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 00:40:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/831a891a07894e16bc20e06826a0fd27 : SUCCESS in 9m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/523ec4fd68d6439bbaf6d3d4d184f77f : FAILURE in 6m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/89a12690d4c648f68d04b0cc4bacef62 : SUCCESS in 8m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89cff0e96bee47c3900885a420b63c3c : SUCCESS in 7m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/24bcd51186424be1a130f678ff226d1d : SUCCESS in 9m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d2b401f2f7a4032b5b0538364afae40 : SUCCESS in 19m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e77cc282efeb45caa32baba2de6143ec : SUCCESS in 6m 31s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/030867033a644c3a97b38c8c9a7baf96 : SUCCESS in 11m 19s\n- ironic-tox-unit-with-driver-libs-python3 https://zuul.opendev.org/t/openstack/build/6bd124baf134409ca98c4087be6992eb : SUCCESS in 10m 09s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/33d38de90c65445f836e152a8efd6066 : SUCCESS in 1h 15m 12s\n- ironic-tempest-functional-python2 https://zuul.opendev.org/t/openstack/build/69c20819bcf64fb0979251453018ef04 : SUCCESS in 29m 25s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/060ce3fb64a2455b8b95a195be9259b3 : SUCCESS in 33m 28s\n- ironic-tempest-ipa-partition-pxe_ipmitool-tinyipa-python3 https://zuul.opendev.org/t/openstack/build/2a0938a0ec9c45ce9e9ea2071aa2fbd5 : SUCCESS in 46m 35s\n- ironic-tempest-ipa-partition-redfish-tinyipa https://zuul.opendev.org/t/openstack/build/11623101d234435096cec7a437cc72e2 : SUCCESS in 46m 45s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/34f9e4aacd664758b9a88ca09d2a959d : SUCCESS in 50m 10s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/276519862ee84354a1aa8c3d666d5d4e : POST_FAILURE in 1h 34m 53s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/0ea92b99d540460d9f302bca32d1a3e5 : SUCCESS in 1h 03m 17s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/dcca4cd2fde24eb2ae24a9193e96dda7 : SUCCESS in 57m 20s\n- ironic-tempest-ipa-partition-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/b89f05ead1b24434be46b7b61b7bcec1 : SUCCESS in 50m 24s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/693b97a6481547979523287ba86d138d : SUCCESS in 31m 28s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/c8f517701312444aa36724da3fbdaf20 : SUCCESS in 40m 41s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/ba91b6119e1a48b095cd02f3a68d63ce : SUCCESS in 4m 57s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-pxe_snmp-tinyipa https://zuul.opendev.org/t/openstack/build/f8585898ae97411ea8aaa199f91875fa : SUCCESS in 37m 31s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/b8e4d406dedd46d48faa427c3e16bb7e : SUCCESS in 41m 06s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/673489ef0c2949a9b86acaba021117d5 : FAILURE in 3m 26s (non-voting)\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/e7401f187249485d8c3f62cb587cda93 : RETRY_LIMIT in 24m 13s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/c800079b4ef84b36aee5aa8a5782317c : SUCCESS in 45m 09s (non-voting)\n\nWarning:\n  Comments left for invalid file ironic/drivers/modules/ipmitool.py","accounts_in_message":[],"_revision_number":2},{"id":"59355fadb5ad240950bb66b7f57fbb639b2e434a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2021-04-15 01:14:53.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"a86af0f6c6673b5b4cb1826a305e75481ea66f36","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-04-15 02:24:14.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/35/786335/3/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/7ec5084/7ec5084.html : FAILURE in 7m 26s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 : MERGER_FAILURE\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/35/786335/3/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3/662147b/662147b.html : FAILURE in 7m 46s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/35/786335/3/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/fc38d6e/fc38d6e.html : FAILURE in 8m 04s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/35/786335/3/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/7157575/7157575.html : FAILURE in 8m 22s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/35/786335/3/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/4a474a6/4a474a6.html : FAILURE in 8m 03s","accounts_in_message":[],"_revision_number":3},{"id":"4e00738e98db520ed5ea494674fd546ec0f988c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 03:00:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7fa6cc498364f42a55fa38613ef21ba : SUCCESS in 9m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31c12e5973934d0a80fb0e5aada2a0ba : SUCCESS in 7m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7e2dc8d166e3456692d33703de6d6078 : SUCCESS in 8m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54de95a53851424db83c3ef9accb2035 : SUCCESS in 8m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/813ffb27f60b4d95a458d04afbb641d5 : SUCCESS in 9m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2b115baba9945b697261a577a2f1b90 : SUCCESS in 20m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b5a9e8a74cc44f38ffc913f174e51a9 : SUCCESS in 6m 17s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/d6a5d3c65eb340c5aac221ecdfa1e282 : SUCCESS in 10m 03s\n- ironic-tox-unit-with-driver-libs-python3 https://zuul.opendev.org/t/openstack/build/87228e8f4ba849cda7cd61e26b837b01 : SUCCESS in 10m 31s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/560db49bd16e411fb2f8f3a8129fe94c : SUCCESS in 1h 30m 18s\n- ironic-tempest-functional-python2 https://zuul.opendev.org/t/openstack/build/a59b3a5fe86044e2b6e118787d2ab71d : SUCCESS in 27m 46s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/2f89eaad25604bd09f88c911c89fc76e : SUCCESS in 27m 46s\n- ironic-tempest-ipa-partition-pxe_ipmitool-tinyipa-python3 https://zuul.opendev.org/t/openstack/build/3ac679b3cc3840b5afa376277b09467e : SUCCESS in 1h 02m 01s\n- ironic-tempest-ipa-partition-redfish-tinyipa https://zuul.opendev.org/t/openstack/build/73c2ec86ef334b49a4d639dc89a07694 : SUCCESS in 51m 14s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/24b4683914ae42a28f9be48390e2c58d : SUCCESS in 49m 11s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/7e64f3776c894c44947a53f1c88c0f6c : POST_FAILURE in 1h 26m 34s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/bc1423d3ee584c94b5f14d51eb8c4428 : SUCCESS in 1h 21m 24s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/9931a17380ca41d4b589d740e3810383 : SUCCESS in 1h 03m 19s\n- ironic-tempest-ipa-partition-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/fb6ddb8dc8d940d593e95b71d8c4912d : SUCCESS in 1h 02m 43s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/676dd087b69f4843ad6186aea413d31f : SUCCESS in 38m 39s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/79ab682223904e009f5a6f07cc6b10bc : SUCCESS in 49m 01s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/06004174f6b24a80958cf498e6ed217b : SUCCESS in 5m 28s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-pxe_snmp-tinyipa https://zuul.opendev.org/t/openstack/build/fcac637f373846e09db36afbf0cd792a : SUCCESS in 49m 47s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/71f85c3eeeb545aa89705f62036db5c4 : SUCCESS in 46m 52s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/a300b7dd8d514a9299b41217d023075d : FAILURE in 3m 57s (non-voting)\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/7e3c43b99a07491f9abea2036cd05bc8 : RETRY_LIMIT in 44m 18s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/826cac3641e24464a55e9b7a67f6be21 : SUCCESS in 1h 05m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"aba5e9e122929c7e94d18e4c18cc7c673c811a4e","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-04-15 05:54:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).\n\n- XE2420_uefi_redfish_train https://stash.dellemc-community.org/logs/XE2420_uefi_redfish_train/17 : SUCCESS in 7h 46m 18s\n- 14G_uefi_ipmi_train https://stash.dellemc-community.org/logs/14G_uefi_ipmi_train/17 : SUCCESS in 7h 19m 43s\n- R7515_uefi_redfish_train https://stash.dellemc-community.org/logs/R7515_uefi_redfish_train/18 : SUCCESS in 6h 16m 7s\n- R6515_bios_idrac_train https://stash.dellemc-community.org/logs/R6515_bios_idrac_train/15 : SUCCESS in 5h 43m 55s\n- 13G_bios_idrac_train https://stash.dellemc-community.org/logs/13G_bios_idrac_train/15 : SUCCESS in 4h 26m 47s\n- 14G_bios_redfish_train https://stash.dellemc-community.org/logs/14G_bios_redfish_train/15 : FAILURE in 2h 51m 29s\n- 14G_uefi_redfish_train https://stash.dellemc-community.org/logs/14G_uefi_redfish_train/26 : SUCCESS in 2h 27m 0s\n- 13G_bios_ipmi_train https://stash.dellemc-community.org/logs/13G_bios_ipmi_train/15 : SUCCESS in 2h 1m 59s\n- 14G_uefi_idrac_train https://stash.dellemc-community.org/logs/14G_uefi_idrac_train/18 : SUCCESS in 1h 19m 59s\n- 14G_bios_idrac_train https://stash.dellemc-community.org/logs/14G_bios_idrac_train/15 : SUCCESS in 43m 43s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"2d699df6c61314160d8a80153a31328682582938","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-04-15 11:09:17.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"08c6ebfd70ef5e48c75d5dd3bee1d2953d0bafd1","author":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"date":"2021-04-15 13:23:36.000000000","message":"Patch Set 3:\n\nTempest run failed in Fujitsu iRMC CI environment\n\n- fujitsu-irmc-ci-job http://openstackci.jp.fujitsu.com/irmcci/35/786335/3/Fujitsu-iRMC-CI/fujitsu-irmc-ci-job/c8f2d34/ : FAILURE in 8m 00s\n\nAbout this CI test, see https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_iRMC_CI","accounts_in_message":[],"_revision_number":3},{"id":"48efa2ab8eeac11784c1e99fce1acbd8cd114b2a","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-04-16 08:53:03.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).\n\n- 13G_bios_idrac_train https://stash.dellemc-community.org/logs/13G_bios_idrac_train/19 : SUCCESS in 4h 59m 23s\n- XE2420_uefi_redfish_train https://stash.dellemc-community.org/logs/XE2420_uefi_redfish_train/21 : SUCCESS in 4h 28m 1s\n- 13G_bios_ipmi_train https://stash.dellemc-community.org/logs/13G_bios_ipmi_train/19 : SUCCESS in 4h 2m 36s\n- 14G_uefi_ipmi_train https://stash.dellemc-community.org/logs/14G_uefi_ipmi_train/21 : SUCCESS in 3h 37m 1s\n- 14G_bios_idrac_train https://stash.dellemc-community.org/logs/14G_bios_idrac_train/19 : SUCCESS in 3h 10m 29s\n- 14G_bios_redfish_train https://stash.dellemc-community.org/logs/14G_bios_redfish_train/19 : SUCCESS in 2h 40m 13s\n- 14G_uefi_redfish_train https://stash.dellemc-community.org/logs/14G_uefi_redfish_train/30 : SUCCESS in 2h 15m 43s\n- R6515_bios_idrac_train https://stash.dellemc-community.org/logs/R6515_bios_idrac_train/19 : FAILURE in 1h 50m 19s\n- R7515_uefi_redfish_train https://stash.dellemc-community.org/logs/R7515_uefi_redfish_train/22 : SUCCESS in 1h 8m 35s\n- 14G_uefi_idrac_train https://stash.dellemc-community.org/logs/14G_uefi_idrac_train/22 : SUCCESS in 32m 8s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":2},{"id":"a342636b5698138b5b04ed8261e87705cde3d1b3","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-04-17 06:34:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- 14G_bios_redfish_train https://stash.dellemc-community.org/logs/14G_bios_redfish_train/22 : SUCCESS in 5h 58m 18s\n- XE2420_uefi_redfish_train https://stash.dellemc-community.org/logs/XE2420_uefi_redfish_train/24 : SUCCESS in 5h 26m 53s\n- 14G_uefi_idrac_train https://stash.dellemc-community.org/logs/14G_uefi_idrac_train/25 : SUCCESS in 4h 28m 52s\n- 14G_uefi_redfish_train https://stash.dellemc-community.org/logs/14G_uefi_redfish_train/33 : SUCCESS in 3h 51m 41s\n- 13G_bios_idrac_train https://stash.dellemc-community.org/logs/13G_bios_idrac_train/22 : SUCCESS in 3h 25m 22s\n- 13G_bios_ipmi_train https://stash.dellemc-community.org/logs/13G_bios_ipmi_train/22 : SUCCESS in 2h 54m 26s\n- 14G_bios_idrac_train https://stash.dellemc-community.org/logs/14G_bios_idrac_train/22 : SUCCESS in 2h 17m 49s\n- R7515_uefi_redfish_train https://stash.dellemc-community.org/logs/R7515_uefi_redfish_train/25 : SUCCESS in 1h 35m 12s\n- 14G_uefi_ipmi_train https://stash.dellemc-community.org/logs/14G_uefi_ipmi_train/24 : SUCCESS in 1h 3m 39s\n- R6515_bios_idrac_train https://stash.dellemc-community.org/logs/R6515_bios_idrac_train/22 : SUCCESS in 35m 27s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":3},{"id":"e966552cd3c2c45a14d2c6c654b96abb0f1dfe31","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-04-20 10:59:47.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"db838fc348bc8b30c8dbf52f459613c75bc174d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 11:00:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"47ce130380022a8122fc63319d912a0975b213ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 15:18:58.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2020ba5ffe747aaa0449fe47d406d66 : SUCCESS in 6m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6f0e4864b2aa4f7e9e9f12044bb4a88c : SUCCESS in 9m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c11a49a5f5f42a0beaf6760730ee040 : SUCCESS in 8m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4694dfef038344de886138cb2b2e0b16 : SUCCESS in 13m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d0753e82de1498fba74f1647cc78623 : SUCCESS in 21m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d8d5dfebec34cab8d943482b5f9c1f4 : SUCCESS in 6m 33s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/97fb02ccc5f54d66a028dc4f9ce8627b : SUCCESS in 12m 18s\n- ironic-tox-unit-with-driver-libs-python3 https://zuul.opendev.org/t/openstack/build/8d3b054cdf1f47079a30a1eb4bbbbc74 : SUCCESS in 11m 57s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/2f60408ed0ae4fbe96ab8b98acbc94a5 : SUCCESS in 1h 02m 05s\n- ironic-tempest-functional-python2 https://zuul.opendev.org/t/openstack/build/79e86589fd6547d8b80e80b6e56da7ad : SUCCESS in 20m 43s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/97828b0eccd240e29a706ea796c717f2 : SUCCESS in 21m 03s\n- ironic-tempest-ipa-partition-pxe_ipmitool-tinyipa-python3 https://zuul.opendev.org/t/openstack/build/9e7df5b455cb41f787b595a350cddc3c : SUCCESS in 51m 53s\n- ironic-tempest-ipa-partition-redfish-tinyipa https://zuul.opendev.org/t/openstack/build/df406aa887fe442d88fd946fad209fda : SUCCESS in 38m 22s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4ce215a81f1c4f79aa1ca5c49859ab3a : SUCCESS in 40m 15s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/38c88b22c63349a28f17afeb4e310063 : SUCCESS in 1h 05m 56s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/a3e8689ad7164703acf0b141b0b8ade6 : SUCCESS in 52m 10s\n- ironic-tempest-ipa-partition-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/243a65294b9043f5b7abc001ef3b3f4c : SUCCESS in 55m 00s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/bccd3415d2b8414b8d199f9151907eb9 : SUCCESS in 36m 24s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/15557bec94fa409cbc6b47769a3cd64f : SUCCESS in 39m 40s","accounts_in_message":[],"_revision_number":3},{"id":"4130aa5e000361c4a47dff33f882e93e6ef55a08","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 15:19:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f51097b69640cbcdde2430c7a390587f3d3922c6","revisions":{"5387665e47783c450e19b88fea7ea5590e0b0c77":{"kind":"REWORK","_number":1,"created":"2021-04-14 22:08:07.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/35/786335/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/35/786335/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/35/786335/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/35/786335/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/35/786335/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/35/786335/1"}}},"commit":{"parents":[{"commit":"08d616cbebdda6089541db6bfa470589a177eee1","subject":"Refactor vendor detection and add Redfish implementation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/08d616cbebdda6089541db6bfa470589a177eee1"}]}],"author":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-01-25 00:48:21.000000000","tz":-300},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2021-04-14 22:07:39.000000000","tz":720},"subject":"For Supermicro BMCs set enable when changing boot device","message":"For Supermicro BMCs set enable when changing boot device\n\nWhen setting the boot settings on Supermicro BMCs, the\nBootSourceOverrideEnabled must be set to the desired\nvalue whenever the BootSourceOverrideTarget is set or\nit will revert to the default value (Once).  This is\ndifferent than what is currently implemented for other\nBMCs in which the BootSourceOverrideEnabled is not set\nif it matches the current setting.\n\nThis change uses the vendor setting to determine if it\u0027s\na Supermicro BMC.\n\nStory: 2008547\nTask: 41652\n\nChange-Id: I1b1a6baafd4cc4daa2fbdb82f69ded6253b1fcbf\n(cherry picked from commit 4a7d50ce56b88c82e95665ac6df44666d3e647ed)\n(cherry picked from commit 7d74ea0eee8f935f5b6c8406e6997d4143fce01e)\n(cherry picked from commit 2e4e8755680080fca2a31adf7a1dfec7b1c10d0a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/5387665e47783c450e19b88fea7ea5590e0b0c77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/5387665e47783c450e19b88fea7ea5590e0b0c77"}]},"branch":"refs/heads/stable/train"},"3ad3d4257d688e4343393d65036d05a7af1106e2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-04-14 23:02:14.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/35/786335/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/35/786335/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/35/786335/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/35/786335/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/35/786335/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/35/786335/2"}}},"commit":{"parents":[{"commit":"85111bebecb316bd64ba7323b132ba5ac83ab5f2","subject":"Refactor vendor detection and add Redfish implementation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/85111bebecb316bd64ba7323b132ba5ac83ab5f2"}]}],"author":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-01-25 00:48:21.000000000","tz":-300},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2021-04-14 22:33:20.000000000","tz":720},"subject":"For Supermicro BMCs set enable when changing boot device","message":"For Supermicro BMCs set enable when changing boot device\n\nWhen setting the boot settings on Supermicro BMCs, the\nBootSourceOverrideEnabled must be set to the desired\nvalue whenever the BootSourceOverrideTarget is set or\nit will revert to the default value (Once).  This is\ndifferent than what is currently implemented for other\nBMCs in which the BootSourceOverrideEnabled is not set\nif it matches the current setting.\n\nThis change uses the vendor setting to determine if it\u0027s\na Supermicro BMC.\n\nStory: 2008547\nTask: 41652\n\nChange-Id: I1b1a6baafd4cc4daa2fbdb82f69ded6253b1fcbf\n(cherry picked from commit 4a7d50ce56b88c82e95665ac6df44666d3e647ed)\n(cherry picked from commit 7d74ea0eee8f935f5b6c8406e6997d4143fce01e)\n(cherry picked from commit 2e4e8755680080fca2a31adf7a1dfec7b1c10d0a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/3ad3d4257d688e4343393d65036d05a7af1106e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/3ad3d4257d688e4343393d65036d05a7af1106e2"}]},"branch":"refs/heads/stable/train"},"f51097b69640cbcdde2430c7a390587f3d3922c6":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-04-15 01:14:53.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/35/786335/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/35/786335/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/35/786335/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/35/786335/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/35/786335/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/35/786335/3"}}},"commit":{"parents":[{"commit":"418d2355ee8b5c5bb131a40a1f811d264075387b","subject":"Refactor vendor detection and add Redfish implementation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/418d2355ee8b5c5bb131a40a1f811d264075387b"}]}],"author":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-01-25 00:48:21.000000000","tz":-300},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2021-04-15 01:12:34.000000000","tz":720},"subject":"For Supermicro BMCs set enable when changing boot device","message":"For Supermicro BMCs set enable when changing boot device\n\nWhen setting the boot settings on Supermicro BMCs, the\nBootSourceOverrideEnabled must be set to the desired\nvalue whenever the BootSourceOverrideTarget is set or\nit will revert to the default value (Once).  This is\ndifferent than what is currently implemented for other\nBMCs in which the BootSourceOverrideEnabled is not set\nif it matches the current setting.\n\nThis change uses the vendor setting to determine if it\u0027s\na Supermicro BMC.\n\nStory: 2008547\nTask: 41652\n\nChange-Id: I1b1a6baafd4cc4daa2fbdb82f69ded6253b1fcbf\n(cherry picked from commit 4a7d50ce56b88c82e95665ac6df44666d3e647ed)\n(cherry picked from commit 7d74ea0eee8f935f5b6c8406e6997d4143fce01e)\n(cherry picked from commit 2e4e8755680080fca2a31adf7a1dfec7b1c10d0a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/f51097b69640cbcdde2430c7a390587f3d3922c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/f51097b69640cbcdde2430c7a390587f3d3922c6"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
