)]}'
{"id":"openstack%2Fironic~788955","triplet_id":"openstack%2Fironic~master~If876d5bbb7e18f293673d56912ecab3170fe5dfb","project":"openstack/ironic","branch":"master","topic":"iscsi-gate","hashtags":[],"change_id":"If876d5bbb7e18f293673d56912ecab3170fe5dfb","subject":"Stop testing the iscsi deploy interface","status":"MERGED","created":"2021-04-30 13:54:41.000000000","updated":"2021-05-03 07:48:34.000000000","submitted":"2021-05-03 07:46:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"788955-iscsi-gate","meta_rev_id":"986d2473cfd4d6d5736e19cadee546baf26d4eaf","_number":788955,"virtual_id_number":788955,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-03 07:45:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":2,"date":"2021-05-03 06:19:10.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":2,"date":"2021-04-30 21:24:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":1,"date":"2021-05-03 06:19:10.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-30 14:11:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-30 15:32:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-30 16:41:58.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2021-04-30 16:51:36.000000000","updated_by":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-30 21:24:31.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2021-05-03 06:19:10.000000000","updated_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"reviewer":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"state":"REVIEWER"}],"messages":[{"id":"872722d28ffe57da7ba9cd070b88936edef558ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-04-30 13:54:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e57694a715c89d35dc785a3b4d5cbca980ef7edf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 14:11:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ff54d482b389410599412ab42310f36d : SUCCESS in 14m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/447058fc536b4ac0ac1cfdb352007584 : SUCCESS in 14m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e73ec92c53cacaf574adab691fc4c2733c4f27b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 15:32:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ffbd3ca9f80c47cbaf9210cee2cfafc6 : SUCCESS in 12m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c264da5d1557420f99f2539b0f1fe939 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e01a3da5e7842d19d3869b16760e970 : SUCCESS in 11m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8627732df7d4ed79e745957e8e6ed82 : SUCCESS in 15m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/198c49beccaa466d9a49b71be8eaf7ae : SUCCESS in 15m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf96039931784aa096e7af9097d461d2 : SUCCESS in 16m 21s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/29a759b336234081925a5d579313eed5 : SUCCESS in 13m 02s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/dba693b9ef324c7b94a5b3140fa5e282 : SUCCESS in 25m 34s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/0115e09bdac245fea936cafc926e298c : SUCCESS in 1h 15m 50s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/dcdd31fac66c4716a4ed26c5ad3856e9 : SUCCESS in 49m 18s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/78c343cf559742cea828bd1a72a2e226 : SUCCESS in 45m 35s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/424d3b9882dc496e89672aadcae533b6 : SUCCESS in 48m 53s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/357190747e4649619007fb5a8f52ace5 : SUCCESS in 1h 01m 28s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/e3cac016f9264f06a05a8fc56734a3ec : SUCCESS in 50m 47s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/356e4a3b4ffe419cb486e24ba65d86e4 : SUCCESS in 1h 19m 04s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/0c4cd51a0878406d926ee9ce349eff43 : SUCCESS in 49m 34s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/1146dafd7dea49c29b4c74c306f6f534 : SUCCESS in 1h 35m 40s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/8a782f0a6710415581f57153fef584e2 : SUCCESS in 1h 09m 15s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/3a9763a4e1944ae3bd41765dd4e4a55c : SUCCESS in 38m 52s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/667bec1a97c44deeabea2e702bed150f : SUCCESS in 51m 24s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/a1babed2065c4cdd9d0e9e784fe35449 : SUCCESS in 1h 31m 51s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/3570747b17cf45f8ae6a4826d6a371d4 : SUCCESS in 6m 24s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/b98c5fc8ca5b4098a06b2e78bacf813d : SUCCESS in 52m 38s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/f5ba34cadf5c49dbad9c23be202745e1 : SUCCESS in 52m 51s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/61c26139b776410ca456522c36883fde : SUCCESS in 47m 47s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/82500ac3d63e4541b03fdb44e2554098 : SUCCESS in 48m 49s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96c49d8f1c3c4e258b6de9f93f691fd6 : SUCCESS in 24m 09s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/dde801bf27cf4ffa8afd15b4eb29e911 : SUCCESS in 25m 21s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/5c55e70e405a4c5e92f0e8af499a928e : SUCCESS in 52m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3ecb2181c9f0cfcad50ab25387b9d2460590beb3","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-04-30 16:41:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).\n\n- 13G_bios_ipmi https://stash.dellemc-community.org/logs/13G_bios_ipmi/529 : FAILURE in 1h 58m 27s\n- 14G_bios_redfish https://stash.dellemc-community.org/logs/14G_bios_redfish/529 : SUCCESS in 1h 40m 51s\n- 13G_bios_idrac https://stash.dellemc-community.org/logs/13G_bios_idrac/529 : SUCCESS in 1h 29m 27s\n- 14G_uefi_ipmi https://stash.dellemc-community.org/logs/14G_uefi_ipmi/771 : SUCCESS in 1h 28m 32s\n- XE2420_uefi_redfish https://stash.dellemc-community.org/logs/XE2420_uefi_redfish/813 : SUCCESS in 1h 27m 37s\n- 14G_bios_idrac https://stash.dellemc-community.org/logs/14G_bios_idrac/529 : FAILURE in 1h 4m 2s\n- R6515_bios_idrac https://stash.dellemc-community.org/logs/R6515_bios_idrac/529 : SUCCESS in 49m 18s\n- R7515_uefi_redfish https://stash.dellemc-community.org/logs/R7515_uefi_redfish/811 : SUCCESS in 49m 17s\n- 14G_uefi_redfish https://stash.dellemc-community.org/logs/14G_uefi_redfish/814 : SUCCESS in 49m 11s\n- 14G_uefi_idrac https://stash.dellemc-community.org/logs/14G_uefi_idrac/778 : SUCCESS in 46m 31s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"d6373774e26f4d9073ddd98ac47d9c45ebd70477","author":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"date":"2021-04-30 16:51:36.000000000","message":"Patch Set 1:\n\nTempest run failed in Fujitsu iRMC CI environment\n\n- fujitsu-irmc-ci-job http://openstackci.jp.fujitsu.com/irmcci/55/788955/1/Fujitsu-iRMC-CI/fujitsu-irmc-ci-job/06d74c6/ : FAILURE in 2h 23m 08s\n\nAbout this CI test, see https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_iRMC_CI","accounts_in_message":[],"_revision_number":1},{"id":"bc5fc8d9ef1725d033243026093759c83c09aabd","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2021-04-30 21:24:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"16610cbaa5eafc2057f81c1dbee8164b40b910ec","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2021-05-03 06:19:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThank you Dmitry.","accounts_in_message":[],"_revision_number":1},{"id":"a5f7465e540c23907c585433c05ce0e9e4cf09da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 06:19:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7a8f0a4d76060c74019444db747f8af180bda4b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 07:45:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57fb4ea3a8f04729998d19e522f2a981 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9177427cf46144d0b208f9666b150183 : SUCCESS in 11m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b579789c394040f182d93b9e6d051380 : SUCCESS in 12m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd1621d495604a89855f01a8bc45bf9d : SUCCESS in 19m 26s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/a01a29de023a425482f0047a089e507a : SUCCESS in 14m 22s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/10e6f5db23ab400bb639a4c44ff5b553 : SUCCESS in 32m 48s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/6db81f518db647c88f2341ba22fd60cc : SUCCESS in 1h 19m 11s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/71c3359690d14c34bc3ba584130afcba : SUCCESS in 46m 51s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/d2cbbbf6023a45a7b443d3f56896d5c0 : SUCCESS in 44m 21s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/16c36e96a11e4e4ebc8192a23652f760 : SUCCESS in 48m 31s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/337e0c069fdb428389d56433fdd38a25 : SUCCESS in 53m 02s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/36cf457df5ad4700a3648a1f1119d1dd : SUCCESS in 44m 39s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/db9ab79218a04cfebca84d9e54b68c5e : SUCCESS in 1h 24m 20s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/e75a10225e34439bb591b7ebd7889750 : SUCCESS in 1h 01m 15s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/d984ac33d0fe4d4483b088220ebf924c : SUCCESS in 59m 22s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/f444adef9e394fb1a2ccb2c61b8bf144 : SUCCESS in 40m 24s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/c2b40abc9e704ec4af06ac66520329d3 : SUCCESS in 1h 02m 54s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/6d07aeb1f21941e8b05e3da04c0f4b5a : SUCCESS in 1h 11m 32s","accounts_in_message":[],"_revision_number":1},{"id":"411c63858623f295e7b59c3fee198bea201db6f3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 07:46:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"986d2473cfd4d6d5736e19cadee546baf26d4eaf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 07:48:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae860cd5b1a04abca655cdff2efd7d38 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a7f4b4a21c0d9a8cd2b15b845f2ef7b3bdd9aea0","revisions":{"a7f4b4a21c0d9a8cd2b15b845f2ef7b3bdd9aea0":{"kind":"REWORK","_number":1,"created":"2021-04-30 13:54:41.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/55/788955/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/55/788955/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/55/788955/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/55/788955/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/55/788955/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/55/788955/1"}}},"commit":{"parents":[{"commit":"193d2e65f100917b8cd71dd3e4da1e341f940592","subject":"Merge \"Process in-band deploy steps on fast-track\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/193d2e65f100917b8cd71dd3e4da1e341f940592"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-04-30 13:54:23.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-04-30 13:54:23.000000000","tz":120},"subject":"Stop testing the iscsi deploy interface","message":"Stop testing the iscsi deploy interface\n\nChange-Id: If876d5bbb7e18f293673d56912ecab3170fe5dfb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/a7f4b4a21c0d9a8cd2b15b845f2ef7b3bdd9aea0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/a7f4b4a21c0d9a8cd2b15b845f2ef7b3bdd9aea0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
