)]}'
{"id":"openstack%2Fironic~790282","triplet_id":"openstack%2Fironic~stable%2Fwallaby~I2ea21c9ec440fa7ddf8578cf7b34d6d0ebbb5dc8","project":"openstack/ironic","branch":"stable/wallaby","hashtags":[],"change_id":"I2ea21c9ec440fa7ddf8578cf7b34d6d0ebbb5dc8","subject":"Fix deployment when executing a command fails after the command starts","status":"MERGED","created":"2021-05-07 17:02:22.000000000","updated":"2021-05-10 12:22:14.000000000","submitted":"2021-05-10 12:20:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"790282","meta_rev_id":"81116af18786c7ff52f51eae496591e59639908e","_number":790282,"virtual_id_number":790282,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-10 12:20:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2021-05-10 10:37:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-10 10:01:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2021-05-10 10:37:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"date":"2021-05-10 10:35:12.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}],"CC":[{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-07 17:20:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-07 18:47:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-07 19:52:09.000000000","updated_by":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-08 00:14:47.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2021-05-10 10:01:04.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2021-05-10 10:37:12.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"b7e165ea051af53e0132b10b9072495582d836b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-05-07 17:02:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79f2045597fe0e2c92787530c42275228c979ed8","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-05-07 17:12:36.000000000","message":"Patch Set 1: Workflow-1\n\nThe master change hasn\u0027t merged yet","accounts_in_message":[],"_revision_number":1},{"id":"39e5e8a4a4362711ee94cb455029f88f29658ebe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 17:20:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/08d39b3757dd4fd2a56cab17e095425e : SUCCESS in 16m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/af8b8208541e4328af1cb1f529083386 : SUCCESS in 13m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0d514313f3de95f17892e27834f35417b5f56e2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 18:47:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29fc179d76b34b02892c021d8354c4bb : SUCCESS in 12m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c490912114bd449aa054bac6610d821d : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/822ea05ccef147c196abe92b349a29a6 : SUCCESS in 10m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a1444f3728e4f5daba4277f8f904ba2 : SUCCESS in 10m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ecd73e697c541d0b7f2e9520777c0f8 : SUCCESS in 13m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ae79774ea4445b2815d8dc3b8a05598 : SUCCESS in 14m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7041b0347a1944949dccd66d70a32676 : SUCCESS in 6m 57s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/b00bce4540834c09add55207a53b833d : SUCCESS in 13m 00s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/41a867af38c84c779944d9c6f2d521c1 : SUCCESS in 28m 42s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/337e0c0b85b64d83ba581806160cfbd7 : SUCCESS in 1h 13m 36s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/67479f63d73f45e580f74a3cf6569f69 : SUCCESS in 1h 03m 05s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/ffb2f0570d8a43d3ba5587b6802478b5 : SUCCESS in 1h 07m 22s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/7094f1377dfc470d9634233d2047ccfb : SUCCESS in 50m 56s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/35714aca5eda48eca8beaeb9bd2d96da : SUCCESS in 51m 29s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/3462025d98194b3a9f72b7ede24f5547 : SUCCESS in 1h 01m 31s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/99c7767f28a94a2b8d198df7f23cf8ee : SUCCESS in 1h 24m 52s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/ad413ba7292c49d8b0253363fb64563a : SUCCESS in 50m 45s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/4644ff58f557451995a4d85934368e9b : SUCCESS in 1h 44m 03s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/4dc2d1e09fa14c61b8c5b84d63881b08 : SUCCESS in 1h 22m 22s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/e244d444ccb746e69ae43a057febcc9d : SUCCESS in 33m 32s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/041e6e5936374dd7b7b780a739592108 : SUCCESS in 1h 10m 12s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/4038b82b6a9a423ca595729a29308baf : SUCCESS in 1h 25m 08s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/6fd920d47c9a432586429c23dc430917 : SUCCESS in 4m 43s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/fbac4d8709e24099817b273ee26d1f97 : SUCCESS in 54m 49s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/10067924d07843bdbf721bbbd326b394 : SUCCESS in 54m 02s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/346642050b9f4b23a05c0fc1b01ec545 : SUCCESS in 54m 52s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/e862264104724c2a94f63be3e4c97829 : SUCCESS in 47m 36s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9b2ca59c47749988061aba5dcdff844 : SUCCESS in 32m 39s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/48a086cd9d734720bab247ea6e4d77ac : SUCCESS in 23m 53s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/f992510617a24de0b15eb6143d4a84d0 : SUCCESS in 55m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b8430ac52bb8f7bd3e60ee98e16d1d44229818bf","author":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"date":"2021-05-07 19:52:09.000000000","message":"Patch Set 1:\n\nTempest run failed in Fujitsu iRMC CI environment\n\n- fujitsu-irmc-ci-job http://openstackci.jp.fujitsu.com/irmcci/82/790282/1/Fujitsu-iRMC-CI/fujitsu-irmc-ci-job/e1c70ac/ : FAILURE in 1h 27m 18s\n\nAbout this CI test, see https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_iRMC_CI","accounts_in_message":[],"_revision_number":1},{"id":"6a31b7140e38184252c8c6287ed0741d38ae45ee","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-05-08 00:14:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- 14G_bios_idrac_wallaby https://stash.dellemc-community.org/logs/14G_bios_idrac_wallaby/11 : SUCCESS in 7h 11m 53s\n- 14G_bios_redfish_wallaby https://stash.dellemc-community.org/logs/14G_bios_redfish_wallaby/11 : SUCCESS in 6h 28m 48s\n- 13G_bios_ipmi_wallaby https://stash.dellemc-community.org/logs/13G_bios_ipmi_wallaby/11 : SUCCESS in 5h 35m 15s\n- 14G_uefi_redfish_wallaby https://stash.dellemc-community.org/logs/14G_uefi_redfish_wallaby/11 : SUCCESS in 4h 55m 48s\n- 14G_uefi_ipmi_wallaby https://stash.dellemc-community.org/logs/14G_uefi_ipmi_wallaby/11 : SUCCESS in 4h 18m 56s\n- 14G_uefi_idrac_wallaby https://stash.dellemc-community.org/logs/14G_uefi_idrac_wallaby/11 : SUCCESS in 3h 39m 55s\n- XE2420_uefi_redfish_wallaby https://stash.dellemc-community.org/logs/XE2420_uefi_redfish_wallaby/11 : SUCCESS in 2h 49m 58s\n- 13G_bios_idrac_wallaby https://stash.dellemc-community.org/logs/13G_bios_idrac_wallaby/11 : SUCCESS in 2h 11m 31s\n- R6515_bios_idrac_wallaby https://stash.dellemc-community.org/logs/R6515_bios_idrac_wallaby/11 : SUCCESS in 1h 26m 48s\n- R7515_uefi_redfish_wallaby https://stash.dellemc-community.org/logs/R7515_uefi_redfish_wallaby/11 : SUCCESS in 39m 38s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"21cd1b5b6ca6433bd1f36e07e478bc06976c498b","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-05-10 10:01:04.000000000","message":"Patch Set 1: Code-Review+2\n\n+2 since Master\u0027s change is already on gate","accounts_in_message":[],"_revision_number":1},{"id":"506a6cde3aaadbef921cc34549a636c5a351bd48","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-05-10 10:35:12.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"1a4eb94bdcaf0151c9938cab40185bcc775616f8","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-05-10 10:37:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6ea6df4402ea92c804e0f8ebe48ad29aec252495","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 10:37:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eb92dab362e67221160c3c470ae45ca79cf63aa2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 12:20:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cbe14aeb9334f3280cd721249584dee : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/442c6e1d78534839b527f17bbc4018e8 : SUCCESS in 12m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47d77ac79ee44228a31f449948cb4ef1 : SUCCESS in 10m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6249dc6d482244c6bb6820dcd610ebea : SUCCESS in 16m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8ff0fea648148b0bca9227b3e4fa01d : SUCCESS in 6m 56s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/32891b5a5ee14ee289f18e66d6f122cc : SUCCESS in 16m 19s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/ce111cb5ff8d4fc0af909106ea7fd946 : SUCCESS in 29m 58s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/0933237cdaff47248672ac158abd0dc9 : SUCCESS in 1h 22m 44s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/1795fdaa875045a991a179f53af96985 : SUCCESS in 1h 01m 17s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/4c7e35781b9140e08aa1872bced96dc8 : SUCCESS in 1h 15m 17s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/cf5a4b94785e4b0e924bec3ef7632ce5 : SUCCESS in 52m 49s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/039a2bb618244a87a2f82801cf4f2ecc : SUCCESS in 56m 37s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/cafbc414e41d4c35a778c815d6906a0e : SUCCESS in 1h 00m 38s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/34553a844d2d42cba72299ea8812e729 : SUCCESS in 1h 09m 21s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/1a8a0c94021946a6a482aee502a68457 : SUCCESS in 59m 37s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/b922643cb0b0402aaf6f6aedd032d048 : SUCCESS in 1h 29m 17s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/d65c79b188ac468dbc8b171a514011dd : SUCCESS in 41m 21s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/e1cda7659234465e8920fcfbab337aa6 : SUCCESS in 59m 42s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/6cec8ade0fbe4d04b59deb2f82dbbc11 : SUCCESS in 1h 39m 00s","accounts_in_message":[],"_revision_number":1},{"id":"70aea0b7a5f2559d79b285ea8b4dcad9f760c17b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 12:20:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"81116af18786c7ff52f51eae496591e59639908e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 12:22:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/392678c717344172be0721f686e3fc78 : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3b1f4636de644d18e71c412005ac76c : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fce4cc11bc0a270283898d9357a73b0ca85994c9","revisions":{"fce4cc11bc0a270283898d9357a73b0ca85994c9":{"kind":"REWORK","_number":1,"created":"2021-05-07 17:02:22.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/82/790282/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/82/790282/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/82/790282/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/82/790282/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/82/790282/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/82/790282/1"}}},"commit":{"parents":[{"commit":"659eef72ee59fdeb639577c8499406f59644a433","subject":"Provide an option to not cache bootable iso ramdisks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/659eef72ee59fdeb639577c8499406f59644a433"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-05-07 08:38:16.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-05-07 16:03:47.000000000","tz":120},"subject":"Fix deployment when executing a command fails after the command starts","message":"Fix deployment when executing a command fails after the command starts\n\nIf the agent accepts a command, but is unable to reply to Ironic (which\nsporadically happens before of the eventlet\u0027s TLS implementation), we\ncurrently retry the request and fail because the command is already\nexecuting. Ironic now detects this situation by checking the list of\nexecuting commands after receiving a connection error. If the requested\ncommand is last one, we assume that the command request succeeded.\n\nIdeally, we should pass a request ID to IPA and then compare it. Such\na change would affect the API contract between the agent and Ironic\nand thus would not be backportable.\n\nChange-Id: I2ea21c9ec440fa7ddf8578cf7b34d6d0ebbb5dc8\n(cherry picked from commit abfe383cc32d63dd3df2b90f0feace08d73b8364)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/fce4cc11bc0a270283898d9357a73b0ca85994c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/fce4cc11bc0a270283898d9357a73b0ca85994c9"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
