)]}'
{"id":"openstack%2Fironic~790439","triplet_id":"openstack%2Fironic~stable%2Fwallaby~I60787243fdc6ed2741522355ec79970bdb912f41","project":"openstack/ironic","branch":"stable/wallaby","hashtags":[],"change_id":"I60787243fdc6ed2741522355ec79970bdb912f41","subject":"Delete unavailable py2 package","status":"MERGED","created":"2021-05-10 15:15:44.000000000","updated":"2021-05-18 11:47:38.000000000","submitted":"2021-05-18 11:46:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"790439","cherry_pick_of_change":790384,"cherry_pick_of_patch_set":2,"meta_rev_id":"db7b2c9106ac4b3269245bde408c8207845baf0b","_number":790439,"virtual_id_number":790439,"owner":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-18 11:46:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":2,"date":"2021-05-18 10:54:58.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":2,"date":"2021-05-11 14:05:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":1,"date":"2021-05-18 10:54:58.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}],"CC":[{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-10 15:49:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-11 00:51:28.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2021-05-11 14:05:15.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2021-05-18 10:54:58.000000000","updated_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"reviewer":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"state":"REVIEWER"}],"messages":[{"id":"b155810813173352182911326a3b4564c40ae18b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"date":"2021-05-10 15:15:44.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c5610fccd7291e4e3c82d8c71cfbceb6dc69345f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 15:49:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac1ceb1633a24a2cac16cff1fee999d8 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e88c482fb51d45638922383cac60ce81 : SUCCESS in 17m 59s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/fb0d4bf252914f11bdaea2a3a29fa4e3 : SUCCESS in 25m 42s","accounts_in_message":[],"_revision_number":1},{"id":"e300ba8e769554ae18c99c513f13edc3f66460d4","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-05-11 00:51:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).\n\n- 14G_bios_redfish_wallaby https://stash.dellemc-community.org/logs/14G_bios_redfish_wallaby/12 : SUCCESS in 9h 34m 39s\n- 14G_uefi_ipmi_wallaby https://stash.dellemc-community.org/logs/14G_uefi_ipmi_wallaby/12 : SUCCESS in 8h 46m 52s\n- R7515_uefi_redfish_wallaby https://stash.dellemc-community.org/logs/R7515_uefi_redfish_wallaby/12 : SUCCESS in 7h 17m 23s\n- 14G_uefi_idrac_wallaby https://stash.dellemc-community.org/logs/14G_uefi_idrac_wallaby/12 : SUCCESS in 6h 38m 16s\n- 14G_uefi_redfish_wallaby https://stash.dellemc-community.org/logs/14G_uefi_redfish_wallaby/12 : SUCCESS in 5h 52m 4s\n- 13G_bios_idrac_wallaby https://stash.dellemc-community.org/logs/13G_bios_idrac_wallaby/12 : FAILURE in 5h 14m 43s\n- XE2420_uefi_redfish_wallaby https://stash.dellemc-community.org/logs/XE2420_uefi_redfish_wallaby/12 : SUCCESS in 3h 48m 51s\n- R6515_bios_idrac_wallaby https://stash.dellemc-community.org/logs/R6515_bios_idrac_wallaby/12 : SUCCESS in 3h 9m 52s\n- 13G_bios_ipmi_wallaby https://stash.dellemc-community.org/logs/13G_bios_ipmi_wallaby/12 : FAILURE in 2h 22m 22s\n- 14G_bios_idrac_wallaby https://stash.dellemc-community.org/logs/14G_bios_idrac_wallaby/12 : SUCCESS in 43m 44s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"54d9469e48ec226fe88ff42bd26895316c10d632","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-05-11 14:05:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"147c6bfce8202e3ba5069f7a70b0f697dc0c2b94","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2021-05-18 10:54:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThank you Lin","accounts_in_message":[],"_revision_number":1},{"id":"281bddb58c44f299bbe5d9c7cffefba58e73f91d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 10:55:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7c9a3a41e093c5e7bf08cfe29c3ffaade6baeee0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 11:46:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d05c03201e1b43fd9ed05b95b89bbc96 : SUCCESS in 10m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f57a44f37b7f418d975dce05a61061d6 : SUCCESS in 17m 01s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/0e63f0cb4dd14999a4fcbd9bead3b0a8 : SUCCESS in 10m 04s","accounts_in_message":[],"_revision_number":1},{"id":"8b11b6b7af8443a9d756ec5176733408321e0371","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 11:46:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"db7b2c9106ac4b3269245bde408c8207845baf0b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 11:47:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc7dec3fe36e4780acaaa71bd8aefcc9 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"77be4c6c690cfa8f52a10ada2b9c1ecabccd423f","revisions":{"77be4c6c690cfa8f52a10ada2b9c1ecabccd423f":{"kind":"REWORK","_number":1,"created":"2021-05-10 15:15:44.000000000","uploader":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"ref":"refs/changes/39/790439/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/790439/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/790439/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/790439/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/790439/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/790439/1"}}},"commit":{"parents":[{"commit":"fce4cc11bc0a270283898d9357a73b0ca85994c9","subject":"Fix deployment when executing a command fails after the command starts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/fce4cc11bc0a270283898d9357a73b0ca85994c9"}]}],"author":{"name":"LinPeiWen","email":"591171850@qq.com","date":"2021-05-10 08:15:54.000000000","tz":0},"committer":{"name":"Lin PeiWen","email":"591171850@qq.com","date":"2021-05-10 15:15:44.000000000","tz":0},"subject":"Delete unavailable py2 package","message":"Delete unavailable py2 package\n\nThe openstack Ussuri and Victoria versions no longer support the\nCentos7 and pyrhon2 environment packages. Correct the missing\nproblems in the latest document\n\nChange-Id: I60787243fdc6ed2741522355ec79970bdb912f41\n(cherry picked from commit 35dea078974261e83f594e798855893401a53511)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/77be4c6c690cfa8f52a10ada2b9c1ecabccd423f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/77be4c6c690cfa8f52a10ada2b9c1ecabccd423f"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
