)]}'
{"id":"openstack%2Fironic~792007","triplet_id":"openstack%2Fironic~master~I29d97e5e734d38100f2d5a49b8dbad8af0f11b21","project":"openstack/ironic","branch":"master","topic":"idrac-message","hashtags":[],"change_id":"I29d97e5e734d38100f2d5a49b8dbad8af0f11b21","subject":"Trivial: fix unfinished message in iDRAC boot","status":"ABANDONED","created":"2021-05-18 16:14:37.000000000","updated":"2021-05-19 10:37:58.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"adf2b8372fcdaff1becd7c16b363afb68932dce9","_number":792007,"virtual_id_number":792007,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-05-18 18:11:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"all":[{"value":-1,"date":"2021-05-18 17:31:25.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-18 16:33:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-18 16:37:15.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2021-05-18 17:31:25.000000000","updated_by":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"reviewer":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"state":"REVIEWER"},{"updated":"2021-05-18 18:11:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-18 21:10:23.000000000","updated_by":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"b77b6655f52b9b3de331c0d908da90d54fb5bd63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-05-18 16:14:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8340bfe97fa519fef5afda626bf9e00b00ee24be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 16:33:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/140f01ee66be49a9958787d8c44bd50c : SUCCESS in 15m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1bc87a80918c475fa3d737ebbdf76ec4 : SUCCESS in 16m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0720582adc353b01a1fb6dd6cf1bfe41a5a1ffa6","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-05-18 16:37:15.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/07/792007/1/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/27a111b/27a111b.html : FAILURE in 13m 40s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/07/792007/1/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3/e91b77d/e91b77d.html : FAILURE in 13m 47s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/07/792007/1/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3/b8a20b6/b8a20b6.html : FAILURE in 14m 04s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/07/792007/1/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/2c144cb/2c144cb.html : FAILURE in 13m 21s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/07/792007/1/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/f93e2d9/f93e2d9.html : FAILURE in 13m 48s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/07/792007/1/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/b55d2e4/b55d2e4.html : FAILURE in 9m 00s","accounts_in_message":[],"_revision_number":1},{"id":"e7d5076d17b0c7865a56afebf5188fec419a0f59","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2021-05-18 17:31:25.000000000","message":"Patch Set 1: Code-Review-1\n\nThanks for the patch.\n\nThis OEM manager pattern is repeating itself in other places as noted by many before. I\u0027m working on eliminating duplicated code and improving the pattern itself.\nOne of the changes I will propose will include the errors in the final message.\n\nI suggest wait for this patch - I need to finish unit tests and will post it if not tomorrow, then this week.","accounts_in_message":[],"_revision_number":1},{"id":"c9e1160ed5944056a125c2ce530d411e95cc01ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 18:11:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0912bf027528412cbe4d3675288228d9 : SUCCESS in 11m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fea25b0965594765b4ad47dda9378520 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac6aa5826dc740bd9e572d08d520922c : SUCCESS in 9m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9538343cedde44b0bab469cdefa661d5 : SUCCESS in 10m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed8c73f6fe3b4ea39d786279455d999b : SUCCESS in 15m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/465d91799cf44e7e9882e0b237dfb034 : SUCCESS in 14m 56s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/8bc82821c6894a14b8f74fa2eb13d66b : SUCCESS in 12m 51s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/8146c30c7a63411fbc2a327568ea12b8 : SUCCESS in 27m 37s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/7f05edb905644ffcae6222fa24c2739b : SUCCESS in 1h 19m 09s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/263c7c7b23354c0e9b75617122f72f24 : SUCCESS in 55m 18s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/066632f078244d9d8ea7746d8e54bd1d : SUCCESS in 57m 36s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/c519855487624027b24881ad9d032b51 : SUCCESS in 48m 31s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/22ffc0b693334171af91768a0a751d01 : SUCCESS in 50m 23s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/c8a458a435e94bf8b9a64580f97db358 : SUCCESS in 57m 30s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/21cd77c8c7404900a5e11aadff0b4551 : SUCCESS in 1h 14m 08s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/e0ae26d2307c4d5e8625d5d07dc4f0c4 : SUCCESS in 39m 55s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/99c96330a47745849faabf760e090e60 : SUCCESS in 1h 51m 18s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/771f3b9a991c4ecabbfccdf277dd3eb8 : SUCCESS in 1h 01m 06s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/b6d1fab4ddb84b81a789930b9c408f8f : SUCCESS in 44m 31s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/a8606ae464794843b9aafe5afc1171a7 : SUCCESS in 38m 49s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/88b9f85c1bca4551b127973176609f40 : SUCCESS in 1h 36m 12s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/104bef7e9c15411089cf12dcb3aad693 : SUCCESS in 4m 44s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/38ffaf27bc5f4afa837ab68425aea72c : SUCCESS in 38m 09s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/aff3cfcbde5640f7973f2ce55cdf0f2f : SUCCESS in 53m 33s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/e86ccb0099d842698c6b818a118189fc : SUCCESS in 1h 17m 01s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/834515c3530641ee9344a7e6b8a9e9d8 : SUCCESS in 50m 56s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf90145ca5cf4bb0820054ed240fe41c : SUCCESS in 23m 08s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/58e371b6e0454d098122a2b7bcda5d10 : SUCCESS in 26m 02s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/a80feee33b864851bf9e573f097e5eb0 : SUCCESS in 39m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"281ab2a509081b7045580c8c34d8c2017230c0cb","author":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"date":"2021-05-18 21:10:23.000000000","message":"Patch Set 1:\n\nTempest run failed in Fujitsu iRMC CI environment\n\n- fujitsu-irmc-ci-job http://openstackci.jp.fujitsu.com/irmcci/07/792007/1/Fujitsu-iRMC-CI/fujitsu-irmc-ci-job/7bdaf61/ : FAILURE in 1h 08m 12s\n\nAbout this CI test, see https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_iRMC_CI","accounts_in_message":[],"_revision_number":1},{"id":"43a89549ffed399c9d1266bb7137c7bd0089b21b","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2021-05-19 08:47:38.000000000","message":"Patch Set 1:\n\n\u003e I suggest wait for this patch - I need to finish unit tests and will post it if not tomorrow, then this week.\n\nthe patch covering all the other cases and logged errors having this pattern - https://review.opendev.org/c/openstack/ironic/+/792088","accounts_in_message":[],"_revision_number":1},{"id":"adf2b8372fcdaff1becd7c16b363afb68932dce9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-05-19 10:37:58.000000000","message":"Abandoned\n\nJust please make sure your change is backportable.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cdadc5125465395dd7a8bd34928893ecfbfec6eb","revisions":{"cdadc5125465395dd7a8bd34928893ecfbfec6eb":{"kind":"REWORK","_number":1,"created":"2021-05-18 16:14:37.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/07/792007/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/07/792007/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/07/792007/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/07/792007/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/07/792007/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/07/792007/1"}}},"commit":{"parents":[{"commit":"a362bbc9e496a195079e561ff57bc0f0e8fed0af","subject":"Merge \"Rework the user guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/a362bbc9e496a195079e561ff57bc0f0e8fed0af"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-05-18 16:13:51.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-05-18 16:13:51.000000000","tz":120},"subject":"Trivial: fix unfinished message in iDRAC boot","message":"Trivial: fix unfinished message in iDRAC boot\n\nChange-Id: I29d97e5e734d38100f2d5a49b8dbad8af0f11b21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/cdadc5125465395dd7a8bd34928893ecfbfec6eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/cdadc5125465395dd7a8bd34928893ecfbfec6eb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
