)]}'
{"id":"openstack%2Fironic~793573","triplet_id":"openstack%2Fironic~master~I17337ed083e6133bfa1c30a1811c6ceb2d43510b","project":"openstack/ironic","branch":"master","topic":"bug/1526266","hashtags":[],"change_id":"I17337ed083e6133bfa1c30a1811c6ceb2d43510b","subject":"API: Node Tags API","status":"NEW","created":"2021-05-28 08:36:51.000000000","updated":"2021-06-06 05:53:42.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"2346ace7d19b20ee73a98d12b0bf7660a40b4a1e","_number":793573,"virtual_id_number":793573,"owner":{"_account_id":26580,"name":"liuzhuangzhuang","email":"vpbvmw651078@gmail.com","username":"LiuZhuangzhuang"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-06-01 10:37:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-28 08:41:38.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2021-05-28 09:10:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-28 10:05:29.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2021-05-28 10:15:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-28 15:06:14.000000000","updated_by":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-28 20:24:19.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2021-06-06 05:53:42.000000000","updated_by":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"reviewer":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"state":"CC"}],"messages":[{"id":"40888c38d3b2466d4a7b59ca792c4ef4dc4546e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26580,"name":"liuzhuangzhuang","email":"vpbvmw651078@gmail.com","username":"LiuZhuangzhuang"},"date":"2021-05-28 08:36:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3885456500dec0a8a014f2d5d3028061cc376f01","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-05-28 08:41:38.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"63582e3e8ad36346e8bfe621653582c950b6167f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 09:10:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/458fa08ec20e4c58864a662be50c15b0 : FAILURE in 17m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f155c55cedcf4dcbb46893dcd4c5fd6e : FAILURE in 24m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"53cfd52f6cdd4138432b1c5b88680c411d153ae7","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-05-28 10:05:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).\n\n- 13G_bios_idrac https://stash.dellemc-community.org/logs/13G_bios_idrac/666 : FAILURE in 1h 11m 4s\n- 14G_uefi_redfish https://stash.dellemc-community.org/logs/14G_uefi_redfish/951 : FAILURE in 1h 1m 45s\n- 14G_uefi_idrac https://stash.dellemc-community.org/logs/14G_uefi_idrac/915 : FAILURE in 1h 1m 40s\n- 14G_bios_redfish https://stash.dellemc-community.org/logs/14G_bios_redfish/666 : FAILURE in 1h 1m 24s\n- 13G_bios_ipmi https://stash.dellemc-community.org/logs/13G_bios_ipmi/666 : FAILURE in 49m 47s\n- R6515_bios_idrac https://stash.dellemc-community.org/logs/R6515_bios_idrac/666 : FAILURE in 31m 9s\n- R7515_uefi_redfish https://stash.dellemc-community.org/logs/R7515_uefi_redfish/949 : FAILURE in 30m 52s\n- 14G_uefi_ipmi https://stash.dellemc-community.org/logs/14G_uefi_ipmi/908 : FAILURE in 27m 38s\n- 14G_bios_idrac https://stash.dellemc-community.org/logs/14G_bios_idrac/668 : FAILURE in 25m 11s\n- XE2420_uefi_redfish https://stash.dellemc-community.org/logs/XE2420_uefi_redfish/951 : FAILURE in 23m 37s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"5cb7b987e378243cf7d8aae971f60b6a75387650","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 10:15:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b4c9312ab904548947faa452011624e : FAILURE in 12m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/723d24679917416a98d0a6e62ef942c8 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/426abfeb91e74abe94f232f6af6e4dbd : FAILURE in 11m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7dc183752cde4850a51e152484ca5373 : FAILURE in 13m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5645c8e6f1be43ebb3b3894d64b72087 : FAILURE in 12m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15caaf7ab0a544e591a0f4528a4fb1b9 : SUCCESS in 14m 48s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/2eeb8de2c81c4e86a463e56c366a4dfa : FAILURE in 12m 59s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/0a6223a2146345309c331df961a110c4 : FAILURE in 34m 20s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/4f5935ba05fe4797a11ad66c998ede4d : SUCCESS in 1h 14m 49s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/da263ce072444aa9b46d48e42653fdd6 : SUCCESS in 1h 08m 23s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/96690525288145b0905ad98906d37189 : SUCCESS in 48m 30s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/de2b27a3a748442889c77f287c33fecd : SUCCESS in 49m 31s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/913b4ae31e0f4cd1a630a4dd6c3cbfd4 : SUCCESS in 49m 59s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/11c0be8f556d4b758acfe919038c10f9 : SUCCESS in 1h 02m 00s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/ee1859af4f294bfd944d15fc74075f21 : SUCCESS in 1h 16m 58s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/35062e7fc6244f3ebae8d3062848c53a : SUCCESS in 51m 39s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/25515e3fe4ff449eaadae50914c5736a : SUCCESS in 1h 37m 18s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/f058a22f3a784822b39f8a3b0ef88a34 : POST_FAILURE in 21m 46s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/183284336c7046908ff21b6178eda05d : SUCCESS in 36m 30s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/2d715dca3fd74d3a936a0b1911345256 : POST_FAILURE in 21m 42s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/2173f4ce68554ab89f077e9e8ee1a7f6 : SUCCESS in 1h 20m 40s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/ca63a278debe4febb1676353b6b566bf : SUCCESS in 5m 47s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/208a6483a30d444abe862a4708320fe9 : SUCCESS in 50m 01s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/27510d76bdc1465f853c154c8289027d : SUCCESS in 1h 07m 31s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/8044daa21dfd4a6aa339e421b066c4ad : SUCCESS in 48m 31s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/4ec2bf57c6364cc6ad76607e2e106b73 : SUCCESS in 36m 01s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91eb1b61cbdd47c291417efaf7c9a75d : SUCCESS in 30m 38s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/54110137f58148c486b6033fbc0cd96b : SUCCESS in 24m 46s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/3de4d226a33c4dc284325f53842ee1fc : SUCCESS in 50m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"351f7b5a5b82129a770001a7790b245818bad34d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26580,"name":"liuzhuangzhuang","email":"vpbvmw651078@gmail.com","username":"LiuZhuangzhuang"},"date":"2021-05-28 11:15:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"62c9f2079bc614572c1f1c5ba6fba89745fb9014","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-05-28 11:16:58.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 : NODE_FAILURE in 0s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 : NODE_FAILURE in 0s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 : NODE_FAILURE in 0s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 : NODE_FAILURE in 0s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 : NODE_FAILURE in 0s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"339abc58d2999358c4ec37d380eceacccef2af1e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 11:33:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c818c3b3a4874a9b8c3faee6c8129473 : SUCCESS in 14m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8624d23edcc744cd9eaf6efdfacf342e : SUCCESS in 15m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3727a98fd9f42da43d4575b8e4c88f816d41f9c3","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-05-28 12:25:40.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).\n\n- 13G_bios_ipmi https://stash.dellemc-community.org/logs/13G_bios_ipmi/670 : FAILURE in 1h 9m 24s\n- 14G_bios_idrac https://stash.dellemc-community.org/logs/14G_bios_idrac/672 : FAILURE in 59m 32s\n- 14G_uefi_ipmi https://stash.dellemc-community.org/logs/14G_uefi_ipmi/912 : FAILURE in 59m 27s\n- XE2420_uefi_redfish https://stash.dellemc-community.org/logs/XE2420_uefi_redfish/955 : FAILURE in 59m 22s\n- 13G_bios_idrac https://stash.dellemc-community.org/logs/13G_bios_idrac/670 : FAILURE in 48m 0s\n- R6515_bios_idrac https://stash.dellemc-community.org/logs/R6515_bios_idrac/670 : FAILURE in 28m 56s\n- 14G_uefi_redfish https://stash.dellemc-community.org/logs/14G_uefi_redfish/955 : FAILURE in 28m 56s\n- 14G_bios_redfish https://stash.dellemc-community.org/logs/14G_bios_redfish/670 : FAILURE in 28m 41s\n- 14G_uefi_idrac https://stash.dellemc-community.org/logs/14G_uefi_idrac/919 : FAILURE in 25m 52s\n- R7515_uefi_redfish https://stash.dellemc-community.org/logs/R7515_uefi_redfish/953 : FAILURE in 23m 17s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":2},{"id":"977dc9ec3b551d8f156a2d217b946aea56af5e16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 12:50:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc9d20ecc56e467c86412b2e56b86c0a : SUCCESS in 10m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4ca26e63b56452caba2359096c069e0 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b6e93cc50d340b787b2b26f4348cc6b : SUCCESS in 9m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/17ee610ed4c545b0ad61c115c611fd7c : SUCCESS in 10m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41e35ef7b2064414b22fde4fd6779b2a : SUCCESS in 10m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c765624b6e4f419aa5f8243d2267a2a9 : SUCCESS in 14m 35s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/be31de9ed2b44f178f6dfd6c2d43589b : SUCCESS in 12m 09s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/1dc1ecc7fbde4c61b8228cb622f23489 : SUCCESS in 24m 17s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/15fc3332b56f4cb386661e75575d000c : SUCCESS in 1h 04m 22s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/8611f5bb01224dd7a047017332a6de6a : SUCCESS in 44m 07s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/9882ef613c5b4c56bb9c0005334d53ba : SUCCESS in 38m 28s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/52881bfe63ea46c9bdb97a7377068e9b : SUCCESS in 46m 52s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/2588dbd0937140e790ccca25ed531058 : SUCCESS in 51m 02s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/d86cc1017fa44d9cbc7ddbbdeded8fda : SUCCESS in 42m 42s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/8503b7bbd5bf4eeab8ff309515fde564 : SUCCESS in 1h 21m 32s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/ec9baec7836f4282a0b373eddbadd0ce : SUCCESS in 50m 47s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/d24cf3e7232d418c9c99e1b44b48c1ef : SUCCESS in 1h 31m 46s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/f1109a29e1c84c528e9559d9aaa46c6a : SUCCESS in 1h 10m 59s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/56f7a9977049474085eec2b7d42fbc4b : SUCCESS in 30m 18s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/6365a56f01814c059d5d352bbdd6a5d4 : SUCCESS in 48m 46s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/f6c62d0a267e487ca9901ff538d53107 : SUCCESS in 1h 17m 07s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/99d5a7b7f17449268090d1168d7fe252 : SUCCESS in 5m 45s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/302feb094f18497299e02df4e2844f93 : SUCCESS in 59m 59s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/54497f7a65e3444cbea7fe571ede02ea : SUCCESS in 49m 59s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/9720d20b77b0447a8e1484ea5657a381 : SUCCESS in 42m 07s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/725bee4800534a1ab77c46fbf6adfc99 : SUCCESS in 45m 57s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21b2519e1c044f08abbcb45d73c83fa7 : SUCCESS in 18m 14s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/187efc094c62414a85f408f818800a00 : SUCCESS in 19m 50s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/943ea4bcfd1f425899e3f18f238a727a : SUCCESS in 53m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2906ee5df67f2195b555a6eca8bae689cb6c6ba9","author":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"date":"2021-05-28 15:06:14.000000000","message":"Patch Set 2:\n\nTempest run failed in Fujitsu iRMC CI environment\n\n- fujitsu-irmc-ci-job http://openstackci.jp.fujitsu.com/irmcci/73/793573/2/Fujitsu-iRMC-CI/fujitsu-irmc-ci-job/d44aa4a/ : FAILURE in 1h 37m 40s\n\nAbout this CI test, see https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_iRMC_CI","accounts_in_message":[],"_revision_number":2},{"id":"c4608a8309d17455a3cf69f6def10a1040195421","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-05-28 20:24:19.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nGreetings!  Looks like your going to need to add more API  and ACL tests given the controller and changes your making.","accounts_in_message":[],"_revision_number":2},{"id":"4101f8aa044c983b4f4b5c08d0e5c44d9d661d05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26580,"name":"liuzhuangzhuang","email":"vpbvmw651078@gmail.com","username":"LiuZhuangzhuang"},"date":"2021-06-01 08:49:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"25995b806b952c63e607f1358ceefa9ea6523f0b","author":{"_account_id":26580,"name":"liuzhuangzhuang","email":"vpbvmw651078@gmail.com","username":"LiuZhuangzhuang"},"date":"2021-06-01 08:51:16.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 2: Code-Review-1\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e Greetings!  Looks like your going to need to add more API  and ACL tests given the controller and changes your making.\n\nThanks for your review. I have added some rest API and acl tests.","accounts_in_message":[],"_revision_number":3},{"id":"21ec7659d9cbaac1cd7451a1f68d36de9a81848e","author":{"_account_id":26580,"name":"liuzhuangzhuang","email":"vpbvmw651078@gmail.com","username":"LiuZhuangzhuang"},"date":"2021-06-01 08:55:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6891af8f001ba22cefbc50472a3e9fe01bfe0f80","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 09:11:40.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eef5d592d8b24a9ca440edbe610c1424 : SUCCESS in 16m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3c869d643ab04f01bac6f354948843c8 : SUCCESS in 17m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"252c09d2fe04806bf719682089bceaa548bc5fac","author":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"date":"2021-06-01 10:17:39.000000000","message":"Patch Set 3:\n\nTempest run failed in Fujitsu iRMC CI environment\n\n- fujitsu-irmc-ci-job http://openstackci.jp.fujitsu.com/irmcci/73/793573/3/Fujitsu-iRMC-CI/fujitsu-irmc-ci-job/2286dd8/ : FAILURE in 1h 22m 35s\n\nAbout this CI test, see https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_iRMC_CI","accounts_in_message":[],"_revision_number":3},{"id":"16e245b27ab085ce14dd8f4b4fee17c00431f42b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 10:37:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09bf0b44947e49b78041f96ad8fcd47b : SUCCESS in 12m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ee6f8d96cb1455da43a74efc512ff1b : SUCCESS in 8m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d87fbdb19a1402196d3bad8d69667d8 : SUCCESS in 13m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/831e8239c763495db058495429395afe : SUCCESS in 18m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ab2e5715eb54fd8adb6ebed4c25a726 : SUCCESS in 10m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34d32a52327a47ebbeed8f48acf02814 : SUCCESS in 22m 32s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/96c904c43de04fcfbf7a38c06450cb56 : SUCCESS in 14m 20s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/bfeeeb586bf9454d841f95c8861b2e5c : SUCCESS in 26m 49s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/b3d5fbfc25e2418bb1fdfba87ce63185 : SUCCESS in 56m 40s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/e437a69e50da478ab40d2838efc03fc8 : SUCCESS in 43m 38s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/4af86082342e428a84cd3d99a38bf335 : SUCCESS in 39m 16s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/1155b400834f458b8f69866fd98a3b2c : SUCCESS in 47m 56s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/95f008a0535d4fb88bf402396f5a925b : SUCCESS in 38m 35s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/b190a5b42aee44679367587fc904447b : SUCCESS in 49m 49s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/8fa8cdfc05a94bff980d3ad4d2c80f64 : SUCCESS in 45m 59s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/d5b92f454e34429681f33a8d3913ec3f : SUCCESS in 49m 38s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/3832b564573740218b2f0ec2f87f9bf0 : SUCCESS in 1h 46m 54s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/0a1ba6fbc62a4dd0b4cba43d0fd6982d : SUCCESS in 1h 40m 10s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/aa09bbce7e9249098ba8c1e8b09a58f8 : SUCCESS in 40m 21s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/55d995e41b584ce196132b01989ee4b6 : SUCCESS in 44m 57s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/91142292dacd40cb813b53d6f3e20b01 : SUCCESS in 1h 45m 08s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/a1cc1cdaebe74561a2ab56825f6e37df : SUCCESS in 4m 52s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/aaf09553fdb34c3cb59ce3a0632d66b4 : SUCCESS in 1h 27m 46s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/e3eed8565eea45208b3c96fc1ed19c4e : SUCCESS in 45m 26s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/79fb505450c5499abe63753b8711af36 : SUCCESS in 48m 47s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/ff0b2f299f1c46fc831c173328ccadb0 : SUCCESS in 58m 59s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e73708a669748048d774cf8f279896e : SUCCESS in 25m 24s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/21c54ba97fe14f4dac30746e590c5edb : SUCCESS in 17m 55s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/95125fb5581a4fa0a36dd68c03a5ab87 : SUCCESS in 51m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"86ca006ec2f249159f34ce358a91f270166a595b","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-06-01 21:17:29.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).\n\n- 14G_bios_idrac https://stash.dellemc-community.org/logs/14G_bios_idrac/678 : FAILURE in 32m 13s\n- 14G_uefi_idrac https://stash.dellemc-community.org/logs/14G_uefi_idrac/925 : FAILURE in 32m 12s\n- 14G_uefi_redfish https://stash.dellemc-community.org/logs/14G_uefi_redfish/961 : FAILURE in 32m 10s\n- R7515_uefi_redfish https://stash.dellemc-community.org/logs/R7515_uefi_redfish/959 : FAILURE in 32m 9s\n- XE2420_uefi_redfish https://stash.dellemc-community.org/logs/XE2420_uefi_redfish/961 : FAILURE in 32m 8s\n- R6515_bios_idrac https://stash.dellemc-community.org/logs/R6515_bios_idrac/676 : FAILURE in 32m 6s\n- 13G_bios_ipmi https://stash.dellemc-community.org/logs/13G_bios_ipmi/676 : FAILURE in 32m 5s\n- 14G_bios_redfish https://stash.dellemc-community.org/logs/14G_bios_redfish/676 : FAILURE in 32m 4s\n- 13G_bios_idrac https://stash.dellemc-community.org/logs/13G_bios_idrac/677 : FAILURE in 32m 3s\n- 14G_uefi_ipmi https://stash.dellemc-community.org/logs/14G_uefi_ipmi/920 : FAILURE in 32m 1s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":2},{"id":"2346ace7d19b20ee73a98d12b0bf7660a40b4a1e","author":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"date":"2021-06-06 05:53:42.000000000","message":"Patch Set 3:\n\nThis is a quite old feature discontinued, I suspect we need to re-investigate after the long evolution. It seems the needs of tags is already covered by traits in some extend.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"36b0f30d5d59042c58e19cdf4f99648ba100b488","revisions":{"0177db4283567278c276093cada92d6f502c2ddf":{"kind":"REWORK","_number":1,"created":"2021-05-28 08:36:51.000000000","uploader":{"_account_id":26580,"name":"liuzhuangzhuang","email":"vpbvmw651078@gmail.com","username":"LiuZhuangzhuang"},"ref":"refs/changes/73/793573/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/73/793573/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/73/793573/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/73/793573/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/73/793573/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/73/793573/1"}}},"commit":{"parents":[{"commit":"caa4c8fd2964cd0c75025705497b7040c973dbe0","subject":"Merge \"Delay rendering configdrive\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/caa4c8fd2964cd0c75025705497b7040c973dbe0"}]}],"author":{"name":"Armstrong Liu","email":"vpbvmw651078@gmail.com","date":"2021-05-28 08:32:27.000000000","tz":480},"committer":{"name":"Armstrong Liu","email":"vpbvmw651078@gmail.com","date":"2021-05-28 08:34:50.000000000","tz":480},"subject":"WIP:API: Node Tags API","message":"WIP:API: Node Tags API\n\nBumps the API microversion to 1.74.\n\nAdds basic CRUD on tags, including set and delete of individual tags.\n\nThis adss six APIs to manage node\u0027s tags:\n* GET /nodes/{node_uuid}/tags. Lists all tags for the node.\n* PUT /nodes/{node_uuid}/tags. Replaces all tags on specified node\n  with the new set of tags.\n* DELETE /nodes/{node_uuid}/tags. Deletes all tags from the specified\n  node.\n* PUT /nodes/{node_uuid}/tags/{tag}. Adds a single tag to the node.\n* DELETE /nodes/{node_uuid}/tags/{tag}. Deletes a single tag from\n\nPartial-Bug:#1526266\nChange-Id: I17337ed083e6133bfa1c30a1811c6ceb2d43510b\nSigned-off-by: Armstrong Liu \u003cvpbvmw651078@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/0177db4283567278c276093cada92d6f502c2ddf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/0177db4283567278c276093cada92d6f502c2ddf"}]},"branch":"refs/heads/master"},"3db06bc04c376672fc750141b30f9efe3ce3a349":{"kind":"REWORK","_number":2,"created":"2021-05-28 11:15:45.000000000","uploader":{"_account_id":26580,"name":"liuzhuangzhuang","email":"vpbvmw651078@gmail.com","username":"LiuZhuangzhuang"},"ref":"refs/changes/73/793573/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/73/793573/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/73/793573/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/73/793573/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/73/793573/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/73/793573/2"}}},"commit":{"parents":[{"commit":"caa4c8fd2964cd0c75025705497b7040c973dbe0","subject":"Merge \"Delay rendering configdrive\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/caa4c8fd2964cd0c75025705497b7040c973dbe0"}]}],"author":{"name":"Armstrong Liu","email":"vpbvmw651078@gmail.com","date":"2021-05-28 08:32:27.000000000","tz":480},"committer":{"name":"Armstrong Liu","email":"vpbvmw651078@gmail.com","date":"2021-05-28 11:14:26.000000000","tz":480},"subject":"API: Node Tags API","message":"API: Node Tags API\n\nBumps the API microversion to 1.74.\n\nAdds basic CRUD on tags, including set and delete of individual tags.\n\nThis adss six APIs to manage node\u0027s tags:\n* GET /nodes/{node_uuid}/tags. Lists all tags for the node.\n* PUT /nodes/{node_uuid}/tags. Replaces all tags on specified node\n  with the new set of tags.\n* DELETE /nodes/{node_uuid}/tags. Deletes all tags from the specified\n  node.\n* PUT /nodes/{node_uuid}/tags/{tag}. Adds a single tag to the node.\n* DELETE /nodes/{node_uuid}/tags/{tag}. Deletes a single tag from\n\nPartial-Bug:#1526266\nChange-Id: I17337ed083e6133bfa1c30a1811c6ceb2d43510b\nSigned-off-by: Armstrong Liu \u003cvpbvmw651078@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/3db06bc04c376672fc750141b30f9efe3ce3a349"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/3db06bc04c376672fc750141b30f9efe3ce3a349"}]},"branch":"refs/heads/master"},"36b0f30d5d59042c58e19cdf4f99648ba100b488":{"kind":"REWORK","_number":3,"created":"2021-06-01 08:49:16.000000000","uploader":{"_account_id":26580,"name":"liuzhuangzhuang","email":"vpbvmw651078@gmail.com","username":"LiuZhuangzhuang"},"ref":"refs/changes/73/793573/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/73/793573/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/73/793573/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/73/793573/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/73/793573/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/73/793573/3"}}},"commit":{"parents":[{"commit":"e15440370cca1f1a998d3607910697c3129d040a","subject":"Include bios registry fields in bios API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/e15440370cca1f1a998d3607910697c3129d040a"}]}],"author":{"name":"Armstrong Liu","email":"vpbvmw651078@gmail.com","date":"2021-05-28 08:32:27.000000000","tz":480},"committer":{"name":"Armstrong Liu","email":"vpbvmw651078@gmail.com","date":"2021-06-01 08:48:43.000000000","tz":480},"subject":"API: Node Tags API","message":"API: Node Tags API\n\nBumps the API microversion to 1.74.\n\nAdds basic CRUD on tags, including set and delete of individual tags.\n\nThis adss six APIs to manage node\u0027s tags:\n* GET /nodes/{node_uuid}/tags. Lists all tags for the node.\n* PUT /nodes/{node_uuid}/tags. Replaces all tags on specified node\n  with the new set of tags.\n* DELETE /nodes/{node_uuid}/tags. Deletes all tags from the specified\n  node.\n* PUT /nodes/{node_uuid}/tags/{tag}. Adds a single tag to the node.\n* DELETE /nodes/{node_uuid}/tags/{tag}. Deletes a single tag from\n\nPartial-Bug:#1526266\nChange-Id: I17337ed083e6133bfa1c30a1811c6ceb2d43510b\nSigned-off-by: Armstrong Liu \u003cvpbvmw651078@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/36b0f30d5d59042c58e19cdf4f99648ba100b488"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/36b0f30d5d59042c58e19cdf4f99648ba100b488"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
