)]}'
{"id":"openstack%2Fironic~794880","triplet_id":"openstack%2Fironic~master~I342e7900cac388cb4749480684418a5a15ac60eb","project":"openstack/ironic","branch":"master","topic":"db-perf-1","hashtags":["ironic-week-prio"],"change_id":"I342e7900cac388cb4749480684418a5a15ac60eb","subject":"Allow node_sanitize function to be provided overrides","status":"MERGED","created":"2021-06-04 18:57:52.000000000","updated":"2021-07-07 18:35:03.000000000","submitted":"2021-07-07 18:24:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"794880-db-perf-1","meta_rev_id":"a27f17bdbf3b06c6a95e175c4eb2d3b609399a4a","_number":794880,"virtual_id_number":794880,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-07 18:24:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":2,"date":"2021-07-06 20:32:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-07 16:52:46.000000000","_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-07 16:56:35.000000000","_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}],"REVIEWER":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-04 19:21:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-04 21:28:52.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2021-06-04 22:17:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-05 04:03:08.000000000","updated_by":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-05 15:11:50.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2021-07-02 22:52:24.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2021-07-07 16:52:46.000000000","updated_by":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"reviewer":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"state":"REVIEWER"}],"messages":[{"id":"61e729cf7ebb5ee7b62c36fe20b75e3a9014fedc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-06-04 18:57:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"365bf1e33c836b85f24fbc41a5976b34e31d0c10","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 19:21:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/74d8bb2b508d439b84f44f97c114882c : SUCCESS in 13m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6c8bfd163871441c80642e86562dbef2 : SUCCESS in 13m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5d637d38381c93cfabc7c878708544d15e4f6ffc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-06-04 20:26:30.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"57ccf8b00f7481509dc06670632c384263379131","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 20:45:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8c6434a9e91a4af2b0abb91807cbd5da : SUCCESS in 13m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f6d2d68cce564aab94dac3b4358a1ca4 : SUCCESS in 14m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8e752e26bc98887d89d5642be70bf7d914d36867","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-06-04 20:52:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3447afedd994b8faf5f1cd7e132c0d14b1ff582f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 21:25:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f30f617aeb504ec1aa19632001a738e3 : FAILURE in 25m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6357960caf544d1993ee784744291079 : FAILURE in 27m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5a28e2a3c7c24eb175ecf2323d1fd7c3cb55b262","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-06-04 21:28:52.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/80/794880/3/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/e2454b5/e2454b5.html : FAILURE in 6m 45s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/80/794880/3/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3/b2e4c58/b2e4c58.html : FAILURE in 8m 09s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 finger://HPE-CI/598adea208d640d6b7dcb883700cc424 : POST_FAILURE in 1m 08s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 : MERGER_FAILURE\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/80/794880/3/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/116f656/116f656.html : FAILURE in 7m 02s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/80/794880/3/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/5ae78de/5ae78de.html : FAILURE in 6m 27s","accounts_in_message":[],"_revision_number":3},{"id":"a8bca23521e76ad75c1b5d69d4b31ad3fa99734e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 22:17:13.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd47fbfca45f4093b53f4c756a537038 : SUCCESS in 10m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d46e77af6d542d4b5651e9f47207cac : FAILURE in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55c966b0ae2f4803b5662f0de9e956b0 : SUCCESS in 11m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8a539b3dbf54b2f98421e8d57af132d : SUCCESS in 10m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/79683144f5094e459944a7dc70607ae0 : SUCCESS in 10m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59f8f885722f4c33aa6e5a9c4488490e : SUCCESS in 14m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de0f7f0e4f314ada9afa0bd2998979b8 : SUCCESS in 6m 31s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/15b6aa92d26d4755903146fee390e372 : SUCCESS in 13m 05s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/90e16744d21d4b679f82051f04af3249 : SUCCESS in 1h 19m 37s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/91ff770c7bae421db02574fd12fd6519 : FAILURE in 4m 52s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8013847dc8254899b8564c4933a5defa : SUCCESS in 21m 46s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/4e5c437152a64d6598b02e4397077f07 : SUCCESS in 22m 42s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/5f8c420ddd104286959527457e64b574 : SUCCESS in 26m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"25c66fd9d9332ca9c81b23d00019897a008b6d38","author":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"date":"2021-06-05 04:03:08.000000000","message":"Patch Set 3:\n\nTempest run failed in Fujitsu iRMC CI environment\n\n- fujitsu-irmc-ci-job http://openstackci.jp.fujitsu.com/irmcci/80/794880/3/Fujitsu-iRMC-CI/fujitsu-irmc-ci-job/707dac6/ : FAILURE in 2h 12m 30s\n\nAbout this CI test, see https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_iRMC_CI","accounts_in_message":[],"_revision_number":3},{"id":"404fe1dfea13417901cc3894a18ae72dfdc265b4","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-06-05 15:11:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).\n\n- XE2420_uefi_redfish https://stash.dellemc-community.org/logs/XE2420_uefi_redfish/980 : FAILURE in 3h 15m 27s\n- 14G_uefi_ipmi https://stash.dellemc-community.org/logs/14G_uefi_ipmi/939 : SUCCESS in 3h 12m 5s\n- R7515_uefi_redfish https://stash.dellemc-community.org/logs/R7515_uefi_redfish/978 : SUCCESS in 3h 11m 15s\n- 14G_uefi_redfish https://stash.dellemc-community.org/logs/14G_uefi_redfish/980 : SUCCESS in 3h 1m 46s\n- 13G_bios_ipmi https://stash.dellemc-community.org/logs/13G_bios_ipmi/695 : FAILURE in 2h 38m 11s\n- 14G_bios_idrac https://stash.dellemc-community.org/logs/14G_bios_idrac/697 : SUCCESS in 1h 39m 56s\n- 14G_uefi_idrac https://stash.dellemc-community.org/logs/14G_uefi_idrac/944 : SUCCESS in 1h 35m 1s\n- 13G_bios_idrac https://stash.dellemc-community.org/logs/13G_bios_idrac/696 : FAILURE in 1h 34m 21s\n- R6515_bios_idrac https://stash.dellemc-community.org/logs/R6515_bios_idrac/694 : SUCCESS in 1h 32m 0s\n- 14G_bios_redfish https://stash.dellemc-community.org/logs/14G_bios_redfish/696 : SUCCESS in 1h 18m 14s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"014fb86b470f26627672f52632e80cbc413712ab","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-06-05 18:17:22.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).\n\n- 14G_uefi_redfish https://stash.dellemc-community.org/logs/14G_uefi_redfish/981 : SUCCESS in 3h 5m 18s\n- 13G_bios_ipmi https://stash.dellemc-community.org/logs/13G_bios_ipmi/696 : FAILURE in 2h 45m 49s\n- R7515_uefi_redfish https://stash.dellemc-community.org/logs/R7515_uefi_redfish/979 : SUCCESS in 2h 41m 3s\n- XE2420_uefi_redfish https://stash.dellemc-community.org/logs/XE2420_uefi_redfish/981 : FAILURE in 2h 18m 20s\n- 14G_bios_redfish https://stash.dellemc-community.org/logs/14G_bios_redfish/697 : SUCCESS in 1h 41m 57s\n- R6515_bios_idrac https://stash.dellemc-community.org/logs/R6515_bios_idrac/695 : SUCCESS in 1h 37m 44s\n- 14G_uefi_ipmi https://stash.dellemc-community.org/logs/14G_uefi_ipmi/940 : SUCCESS in 1h 34m 43s\n- 14G_uefi_idrac https://stash.dellemc-community.org/logs/14G_uefi_idrac/945 : FAILURE in 1h 18m 31s\n- 13G_bios_idrac https://stash.dellemc-community.org/logs/13G_bios_idrac/697 : FAILURE in 1h 10m 20s\n- 14G_bios_idrac https://stash.dellemc-community.org/logs/14G_bios_idrac/698 : FAILURE in 34m 49s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":3},{"id":"2043308680332e816eea140d12e0008886315737","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-06-07 15:24:54.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":3},{"id":"984d082db0258c683497358757a7dcb3f00017cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-06-17 20:27:14.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7f8519ca4c752e86526b6c7193ed244e056a7779","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 20:45:17.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9d3e5d6c633a48bf9db46cf085c37fa2 : SUCCESS in 14m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a8827e612ea642528231329d0bfadf1a : SUCCESS in 15m 52s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6e2016002ab9839670646eddf76f6a76b3cae02e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 22:00:18.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e07272e83cd14a2a8e67c61d80d1a160 : SUCCESS in 12m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7658d27046364ae89c72a5fd5ae13ee8 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0957d54dc6fc419eab64986abb153e9f : SUCCESS in 10m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75e723d80bef453c85be90b56b8b6d5c : SUCCESS in 11m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f255876a379495b94eaaff4d26d4614 : SUCCESS in 11m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bbc0c76bb6440a4a1b657251319f889 : SUCCESS in 15m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28bd85a8792d4545acf60c8ecdf9cfbe : SUCCESS in 6m 29s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/2c75e592999a42a3bde7927e2509822e : SUCCESS in 13m 36s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/c6a6cc4f5eea4b22a472a388213a165b : FAILURE in 29m 40s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/d7edf324279e46ee8e73fd6529bed16a : SUCCESS in 1h 14m 49s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/1c7d1da06dca4ab298bd3db6ccaaf39f : SUCCESS in 48m 21s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/d667439b676544d28833b16076bef7df : SUCCESS in 54m 59s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/ab6548053b654281bb9582cb1e8b8bc2 : SUCCESS in 52m 05s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/a518712355e54f41affcfa76c2167954 : SUCCESS in 49m 16s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/3eb17f0085284580bf42903037315852 : SUCCESS in 47m 10s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/fc63ccab2b9248289b868b87b15d0074 : SUCCESS in 1h 07m 59s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/3760a7ae55dc428799a0ef59c0b2178e : SUCCESS in 54m 27s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/a206228adc8848f498f36eebc7d0fb89 : SUCCESS in 1h 18m 37s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/64dae51c397a44daa9dac82a8078e64e : SUCCESS in 1h 14m 23s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/35dcb67c1a24463993296523102a3df9 : SUCCESS in 43m 04s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/bce6db72762a44a2823c4448b7b475e5 : SUCCESS in 44m 51s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/efaab6846a4c43d5a85b99a5c8fbf905 : SUCCESS in 1h 26m 30s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/70f86e23c7b5423e8f2d02ee6a08af58 : FAILURE in 5m 44s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/17233a3004ea43c9bc6105f3e4f9d287 : SUCCESS in 47m 22s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/0e75105fa41142eaa5746578bdc526e9 : SUCCESS in 1h 05m 15s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/202b10cc4446434bbb5393c48e1c4162 : SUCCESS in 50m 07s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/092bae456ab940e0b796846d702978d5 : SUCCESS in 49m 26s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9530b36ced654d10a7dea48bdffa5800 : SUCCESS in 22m 35s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/97892c830ac24c669cc4f4afbfe23d73 : SUCCESS in 23m 01s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/03e127b0301a4eed909cf07da3fe7254 : FAILURE in 29m 21s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/7adb29e9ff4d47b89dcaffa628042b03 : SUCCESS in 25m 13s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c0b16f0a2c7ffd66cb1b59a52287bb2999223ef4","author":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"date":"2021-06-18 07:30:47.000000000","message":"Patch Set 4:\n\nTempest run failed in Fujitsu iRMC CI environment\n\n- fujitsu-irmc-ci-job http://openstackci.jp.fujitsu.com/irmcci/80/794880/4/Fujitsu-iRMC-CI/fujitsu-irmc-ci-job/fff867b/ : FAILURE in 1h 39m 25s\n\nAbout this CI test, see https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_iRMC_CI","accounts_in_message":[],"_revision_number":4},{"id":"359282846d4bac98c91fdd398dd88b575a3e1b4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-06-18 18:42:06.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"af9d33c068b54d9663b8639d26011bd40ed57732","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-06-18 18:45:36.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 : NODE_FAILURE in 0s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 : NODE_FAILURE in 0s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 : NODE_FAILURE in 0s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 : NODE_FAILURE in 0s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 : NODE_FAILURE in 0s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"153bb3008b5c1ecd4b81dd0c2118c8fefc0c1732","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 19:06:42.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/677a2543059a40d5ac0709edfad7539e : SUCCESS in 13m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/68b34b2422da46dc9a312255ab2b1efb : SUCCESS in 14m 18s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"20d2e7144b4f1e93b1f58bf32205062676e3d823","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 20:35:14.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/560907f65be2401ba1672e26ce7ec43b : SUCCESS in 10m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed7925af4d0243b395cd349ee72b933f : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa1aedc9292f4e54917aa5eb4c42a556 : SUCCESS in 15m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab926809c36f412981f00d774e469f8f : SUCCESS in 9m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53e4de8abe5f4654b1545ab3fb1005f7 : SUCCESS in 14m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29fac35d929a4b49a70162e17ae23366 : SUCCESS in 15m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5cb3d58970b74812b95fde69868f9065 : SUCCESS in 6m 36s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/8c6f4e70b24349fb90919f782b13ee2f : SUCCESS in 13m 39s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/0466cba57a7e4eb889f5547cc79d776b : FAILURE in 26m 26s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/de44baa4df574d18b898f5d593bcea5c : SUCCESS in 1h 20m 51s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/8f0fe348b808404dab176d7f7756bebd : SUCCESS in 57m 18s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/d4fb67d2da284bf08af2a857a8b4241d : SUCCESS in 57m 55s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/c38f74a403db4211a3226f9ab2f266c6 : SUCCESS in 47m 44s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/5785a7af97f24bcda44dcdc1ee2477ec : SUCCESS in 49m 36s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/f2e0005b89cf41b88ba06b8ab8c89ea4 : SUCCESS in 49m 38s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/1368fe781ba04a4bae97b7e5aaef2dc4 : SUCCESS in 1h 13m 53s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/5905c45d7aaa4cd1848fb133df813e8e : SUCCESS in 51m 07s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/c2af6a535b14400a95bcaac3b1c02d35 : SUCCESS in 1h 24m 42s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/f502e533ffac4e84b5e0f8a45f81513d : SUCCESS in 1h 08m 27s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/1b129d9ccbbf41f4b30e565f11fa18f9 : SUCCESS in 37m 18s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/4dae8c7515df4419be5901f8b140f129 : SUCCESS in 46m 21s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/b53412f156124f5ca04a604785f2424f : SUCCESS in 1h 49m 21s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/a70d6dd7497e4c9c96212f8a6db5397f : SUCCESS in 5m 48s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/85df889a060f4e538459ce3fed9feaaf : SUCCESS in 59m 10s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/c8f6f4532a504188af2dd7227e476d6c : SUCCESS in 48m 13s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/7ccc9f76a8914d1ca94435de19880fe6 : SUCCESS in 49m 40s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/712f2b4e1f5e4eee865cb9c8eedf127b : SUCCESS in 50m 08s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f23c86e84c1461e9c858bf3c0b6618f : SUCCESS in 22m 25s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/181ef10c6afa436692d06bebd3f0dc17 : SUCCESS in 24m 25s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/5db6491d88ac4ce5af3a1ca7a9d505c9 : SUCCESS in 49m 15s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/3a030318a1c645ab96479feb01241178 : SUCCESS in 26m 56s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7643ebc0c39a76963abf2ccb8d7d07300d3f749a","author":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"date":"2021-06-19 05:33:19.000000000","message":"Patch Set 5:\n\nTempest run failed in Fujitsu iRMC CI environment\n\n- fujitsu-irmc-ci-job http://openstackci.jp.fujitsu.com/irmcci/80/794880/5/Fujitsu-iRMC-CI/fujitsu-irmc-ci-job/4271094/ : FAILURE in 2h 11m 54s\n\nAbout this CI test, see https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_iRMC_CI","accounts_in_message":[],"_revision_number":5},{"id":"0d192bb26bca63bbe0ccea3da5a807aa37a0a114","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-06-21 18:23:44.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"7d48995a8e3d25fa3ba3a47178b0917ea26a267a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 18:50:26.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a7c1f9b70146470180e10c2493f14bb3 : SUCCESS in 20m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a6107c6e270f42d48762a82772714302 : SUCCESS in 20m 59s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"66b674c9dd2447a42f909a4bfe0732d82b865ce2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 19:54:06.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd77c218506a464792a2a9a181cb7dfc : SUCCESS in 20m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb281d2a040c473db383598b8680cb30 : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/783f6315ca504cd1a8d6dd47c3c241c9 : SUCCESS in 17m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/15a9dcf80eab44b9bd859afc9fba5673 : SUCCESS in 16m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e539ab8e5164cbfb7ae0226d6f6c413 : SUCCESS in 18m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0eb51f9da344227941591d00d39be36 : SUCCESS in 15m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd01a3e069334758a0ec6584b4f809b8 : SUCCESS in 7m 40s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/ec7a356c53a94ae99e81c22a793eeabb : SUCCESS in 19m 53s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/8e4eeb7e116545c686c06eb50a5dd0d6 : FAILURE in 27m 41s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/a5e08d712ac544f394cc98386afd3d77 : SUCCESS in 1h 22m 41s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/84afe37719eb41aa839aa5b0cb4a31d0 : SUCCESS in 52m 43s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/9151198ab80f463097b99de8a6e5ff6a : SUCCESS in 46m 08s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/12d77d7b1e944009aadac5ebdf7da2c6 : SUCCESS in 47m 29s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/8ea6051b33e74d2e94c310379d3c2297 : SUCCESS in 59m 27s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/16a132e940ee4c7f957ff5a1918e16a4 : SUCCESS in 53m 27s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/b463febd5342400eb1922865c941a902 : SUCCESS in 1h 15m 23s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/9d9f3b97352a4f209f4bf7cc3aefbb2e : SUCCESS in 49m 05s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/5aa15fd567b84fcbacff157628272ae3 : SUCCESS in 1h 27m 18s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/8a8db8b4de554f47acabdcf64c83f2dc : SUCCESS in 1h 14m 30s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/3f5704fa44a34bf7acf89da84ac00675 : SUCCESS in 41m 31s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/72247f1d78cf41d6b7b9ddb107500df9 : SUCCESS in 52m 25s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/12b9cad9a87d4dd49d84bef3c3adc8cf : SUCCESS in 1h 10m 55s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/f6a8b511fde54d39b7c5d75c7918074c : SUCCESS in 6m 43s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/0f8ca2fdca4b4072b7aa298226f8607e : SUCCESS in 51m 14s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/551f93c7af974f64807478d33e69df16 : SUCCESS in 55m 27s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/8bbf46303d35416883fe29a7303d39a3 : SUCCESS in 44m 37s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/4615e6c61554474ea94c9deb6d0ec0b2 : SUCCESS in 42m 29s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62236c04001747e780ad859624bc618c : SUCCESS in 25m 19s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/a43a1f4d433e49518699629807362d7d : SUCCESS in 27m 56s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/ba9923313bad49b9992ad31e959b5eca : SUCCESS in 50m 01s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/1403a561965a4756b36be7a2e4f1db6b : SUCCESS in 25m 53s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"31707d22b6042ff1766014376111c7548a61d5f8","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-06-21 19:58:45.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/80/794880/6/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/f1e6604/f1e6604.html : FAILURE in 8m 35s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/80/794880/6/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3/39e1ed1/39e1ed1.html : FAILURE in 9m 49s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/80/794880/6/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3/1bc6a52/1bc6a52.html : FAILURE in 10m 06s (non-voting)\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/80/794880/6/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/1d6623b/1d6623b.html : FAILURE in 9m 53s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/80/794880/6/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/69e1862/69e1862.html : FAILURE in 8m 28s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/80/794880/6/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/345039f/345039f.html : FAILURE in 12m 17s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"4a26c45b975e6d953f27058231bbccad2189ac6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-06-21 22:31:42.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"3fb490ca1a27c443541722687d78562f9519a85d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 23:11:16.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/04405f8e009c4dd181282d09563f1877 : SUCCESS in 30m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ce602945c8cb48ae9f5d1609837e1ac3 : SUCCESS in 17m 15s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"c66572926783735b376ea4d4eeb558ead3c8b585","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 00:04:08.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8bfc6de417ac4eb09f15737ad3ef0343 : SUCCESS in 12m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a27f76c24e434149bc21c3745fdb687d : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb9ac2af49094bc9bb05f70bbe98526f : SUCCESS in 20m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/301fec8863ef48668c83e751ba88de25 : SUCCESS in 10m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/040450c8770949cb96e25593174f6c2a : SUCCESS in 16m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f1f21b7d5f04b31b8329cfd541df457 : SUCCESS in 14m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/555b39a1da7f471da80960ea2772da84 : SUCCESS in 7m 42s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/f0fd76dd14cf4e7f9ed2cfa1e64fbe13 : SUCCESS in 18m 07s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/7b472914b5694e428f2786f50a707ce5 : SUCCESS in 30m 31s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/699d8f96a3244915a180bceabcabec60 : SUCCESS in 1h 06m 08s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/3d1228a6d3354d0d9f5c3af59c6b016b : SUCCESS in 58m 42s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/ac0aa6c4d4994b7c8ef2fadc62c25a3d : SUCCESS in 50m 14s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/5b1a4b8af8c149e68031cb03cd3b7b8c : SUCCESS in 48m 25s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/2ed9474fd23a4a7983dd48872a25c07c : SUCCESS in 51m 38s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/93f33359136b47bfb261d64f1fbdf161 : SUCCESS in 56m 36s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/964f638fa76645c4b861bcd78a4c2b59 : SUCCESS in 1h 13m 18s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/7eab35d6872b41f891c4d7d56288538f : SUCCESS in 1h 03m 27s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/fb558f8bfeec40e8bff07ddbc6440860 : FAILURE in 54m 44s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/730d6565e1e841ecbd161266789294e9 : SUCCESS in 1h 07m 31s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/71c1563f1ca14983b0bb2a89e26d0812 : SUCCESS in 40m 10s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/02a5dd0f74ed414ba5d857d804b1aa46 : SUCCESS in 44m 07s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/f7c20bfbef2448cba06f89b1b7848877 : SUCCESS in 1h 27m 28s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/22155fcc43bc4a1bba6bbdd06edc8df7 : SUCCESS in 6m 05s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/abef9de41f0d476097f21057721582da : SUCCESS in 1h 04m 17s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/bebd2755a18b482f92a7de29bf4d4fb1 : SUCCESS in 46m 57s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/cd43a20cc9f04f919224637a5e77e8ee : SUCCESS in 54m 52s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/852469f7112c4f61ac5eec32cc051cc8 : SUCCESS in 44m 25s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f337b0d65164c439c8726db74b2c4b0 : SUCCESS in 27m 30s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/1c3211cd29484490a39bdffb92fdda5a : SUCCESS in 28m 00s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/0f6d86fa69bd47bd8aada130896bca11 : SUCCESS in 47m 58s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/9c30cd94546049ac8f1378a79037a71c : SUCCESS in 28m 29s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"1e90716facf125c51e7aba8e076190277f7d3aff","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-06-22 00:16:29.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/80/794880/7/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/08e3247/08e3247.html : FAILURE in 8m 29s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/80/794880/7/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3/16f84fd/16f84fd.html : FAILURE in 10m 07s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/80/794880/7/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3/00d2aee/00d2aee.html : FAILURE in 10m 51s (non-voting)\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/80/794880/7/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/4924b10/4924b10.html : FAILURE in 11m 36s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/80/794880/7/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/0b30079/0b30079.html : FAILURE in 10m 45s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/80/794880/7/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/18022f8/18022f8.html : FAILURE in 9m 53s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e86e4b717f1acf136b1a4475fa444ec0bc26f35b","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-06-22 08:54:35.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).\n\n- 14G_bios_idrac https://stash.dellemc-community.org/logs/14G_bios_idrac/755 : SUCCESS in 2h 54m 52s\n- 14G_uefi_ipmi https://stash.dellemc-community.org/logs/14G_uefi_ipmi/998 : SUCCESS in 2h 51m 49s\n- 13G_bios_ipmi https://stash.dellemc-community.org/logs/13G_bios_ipmi/754 : SUCCESS in 2h 51m 27s\n- 14G_uefi_redfish https://stash.dellemc-community.org/logs/14G_uefi_redfish/1039 : FAILURE in 2h 32m 20s\n- R6515_bios_idrac https://stash.dellemc-community.org/logs/R6515_bios_idrac/752 : FAILURE in 2h 26m 34s\n- 13G_bios_idrac https://stash.dellemc-community.org/logs/13G_bios_idrac/755 : SUCCESS in 1h 33m 46s\n- 14G_uefi_idrac https://stash.dellemc-community.org/logs/14G_uefi_idrac/1003 : SUCCESS in 1h 33m 20s\n- 14G_bios_redfish https://stash.dellemc-community.org/logs/14G_bios_redfish/754 : SUCCESS in 1h 24m 48s\n- XE2420_uefi_redfish https://stash.dellemc-community.org/logs/XE2420_uefi_redfish/1039 : SUCCESS in 1h 21m 17s\n- R7515_uefi_redfish https://stash.dellemc-community.org/logs/R7515_uefi_redfish/1037 : SUCCESS in 1h 19m 52s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":6},{"id":"b872eb28a9ea725f2a664547c5cb518a591846fb","author":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"date":"2021-06-22 09:08:20.000000000","message":"Patch Set 7:\n\nTempest run failed in Fujitsu iRMC CI environment\n\n- fujitsu-irmc-ci-job http://openstackci.jp.fujitsu.com/irmcci/80/794880/7/Fujitsu-iRMC-CI/fujitsu-irmc-ci-job/7a87f4f/ : FAILURE in 1h 41m 43s\n\nAbout this CI test, see https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_iRMC_CI","accounts_in_message":[],"_revision_number":7},{"id":"fc58e416611fd079300fcbd58a535c37e827c5da","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-06-23 03:49:57.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- 14G_bios_idrac https://stash.dellemc-community.org/logs/14G_bios_idrac/760 : SUCCESS in 3h 17m 34s\n- 13G_bios_ipmi https://stash.dellemc-community.org/logs/13G_bios_ipmi/759 : SUCCESS in 2h 56m 56s\n- 14G_uefi_idrac https://stash.dellemc-community.org/logs/14G_uefi_idrac/1008 : SUCCESS in 2h 56m 17s\n- 14G_bios_redfish https://stash.dellemc-community.org/logs/14G_bios_redfish/759 : SUCCESS in 2h 55m 26s\n- R6515_bios_idrac https://stash.dellemc-community.org/logs/R6515_bios_idrac/757 : SUCCESS in 2h 55m 2s\n- 13G_bios_idrac https://stash.dellemc-community.org/logs/13G_bios_idrac/760 : SUCCESS in 1h 31m 18s\n- XE2420_uefi_redfish https://stash.dellemc-community.org/logs/XE2420_uefi_redfish/1044 : SUCCESS in 1h 27m 9s\n- 14G_uefi_ipmi https://stash.dellemc-community.org/logs/14G_uefi_ipmi/1003 : SUCCESS in 1h 25m 49s\n- 14G_uefi_redfish https://stash.dellemc-community.org/logs/14G_uefi_redfish/1044 : SUCCESS in 1h 21m 55s\n- R7515_uefi_redfish https://stash.dellemc-community.org/logs/R7515_uefi_redfish/1042 : SUCCESS in 1h 19m 1s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":7},{"id":"ace1a870c76eaad57a0385312a995c50f67e75c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-06-25 23:04:51.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"af1afce9222809ebadd9005fb1756d26e72ac4d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-25 23:27:44.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e4ab6ff29f2d4de49dde23a77d9f7af6 : SUCCESS in 17m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/288963ef0d7a451aa54bb243475f74d9 : SUCCESS in 16m 21s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"aaf66fff0723469265ecbc18ada26aeb85a507b2","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-06-26 00:29:34.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/80/794880/8/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/9bf678b/9bf678b.html : FAILURE in 9m 09s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/80/794880/8/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3/686808c/686808c.html : FAILURE in 9m 37s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/80/794880/8/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3/72c9ed7/72c9ed7.html : FAILURE in 9m 14s (non-voting)\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/80/794880/8/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/3a59b8c/3a59b8c.html : FAILURE in 9m 18s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/80/794880/8/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/5f48367/5f48367.html : FAILURE in 12m 40s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/80/794880/8/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/c2daa85/c2daa85.html : FAILURE in 13m 34s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"f2d728c17b08a0bc14095aa067859a0e6a4dec93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-26 00:36:12.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a006b4e3a16f4d739314564f1a3675f9 : SUCCESS in 12m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fb1e9f4461d41b38a8fd35d3cc0b76c : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/616b1b5d629549afaeece898bc66e830 : SUCCESS in 10m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5763f92212324f94a2fe859f2c48603c : SUCCESS in 12m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1eb3b96943d540cc9788ecd441a9b2b4 : SUCCESS in 11m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2d72626fcc040c0a4d6fcea457065c7 : SUCCESS in 17m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ab1e61fb0f94ef68d5617a51d558f45 : SUCCESS in 10m 37s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/e25ce87df21e44efb81db5f5b38ab673 : SUCCESS in 14m 32s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/f013e8aad399420aa5b68a80f2e6485b : SUCCESS in 29m 44s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/ffcd1bcdd4d7487b9a3826f4ecaa95df : SUCCESS in 1h 25m 07s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/b37640f330a948858ed3a0d3fdd348a2 : SUCCESS in 55m 12s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/9b20141b91194a438d8b9890543e40b0 : SUCCESS in 59m 10s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/c8f516fb62aa4918afded1b5ac6c8406 : SUCCESS in 49m 28s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/7522c245456f45f0a2f54500759c6a41 : SUCCESS in 56m 41s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/753f0c57600f4da18df1ac9cdae65073 : SUCCESS in 43m 53s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/aee941988ab24db994cb230ac674994d : SUCCESS in 1h 16m 11s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/92b5ca08683f4e048e22911e204434dc : SUCCESS in 45m 25s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/43e4298a07814356bd0dcd9d1f14d022 : SUCCESS in 1h 26m 27s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/16f3ec899ec446efa52fcfde362e5cd1 : SUCCESS in 1h 19m 09s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/c1ceb669139846d990924a1af46bbd51 : SUCCESS in 38m 01s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/9fc73c273ec846769ad323e9efb7efe1 : SUCCESS in 43m 38s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/76328946d2144d4fa38645307642a18d : SUCCESS in 1h 23m 21s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/c5b032642e314e7c9b11ee4aa426427a : SUCCESS in 6m 35s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/1b04113cc32b4b9592a632b7ff158c09 : SUCCESS in 54m 24s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/509874f364c84005a7f55cd2c954fca4 : SUCCESS in 57m 21s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/1a91381e72ae434a876e6924fdd962e6 : SUCCESS in 51m 11s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/14a5edff21c743b28987c320664a8daf : SUCCESS in 54m 51s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93dd258c0d784435877543d7e882e5dc : SUCCESS in 28m 18s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/bf28b7b178f343c4920b6aebba714abb : SUCCESS in 22m 00s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/90e500ddf6ec400eac0a30f6ece88ada : SUCCESS in 50m 51s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/ca4f2ddd27314d9b9432ac953e3bc95d : SUCCESS in 31m 18s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"0b2f528a68d399a2fb223f6d576a6f9b474d94fe","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2021-07-02 22:52:24.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)\n\nLooks OK, please fix the spelling issue since it\u0027s in a method name.","accounts_in_message":[],"_revision_number":8},{"id":"65b31a31c8660ce19ad67260cd4189b3264aaa48","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-07-06 17:59:23.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"2699f90c884adfadbde92f0b5d971e1817653797","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-07-06 18:07:51.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"78bde04d53c96023c60311ce4d0b4746ba057ecc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-07-06 20:06:33.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"963e6746be4f5e4c0d4af1b49f2d95a8dfb72f68","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 20:25:13.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f7dc54b04d3e48109bee04d42bab7bec : SUCCESS in 16m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ec65689857ed4333b729733694e994d9 : SUCCESS in 16m 40s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"8981af1effc1462ad63361930eec940049a19802","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2021-07-06 20:32:04.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"c2ac1608ecf66ccf1838b75eb9efe53cde8d6765","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-07-06 21:20:25.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/80/794880/9/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/c032567/c032567.html : SUCCESS in 54m 50s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/80/794880/9/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3/a431865/a431865.html : SUCCESS in 54m 21s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/80/794880/9/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3/1f4ef1b/1f4ef1b.html : FAILURE in 45m 31s (non-voting)\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/80/794880/9/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/fe35587/fe35587.html : SUCCESS in 53m 39s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/80/794880/9/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/600db3a/600db3a.html : SUCCESS in 52m 43s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/80/794880/9/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/0579306/0579306.html : FAILURE in 1h 05m 06s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"69434cacc5f0026a20c1b545a1eb8661f695053a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 22:14:58.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb37ef1600bd4f3897a0f2d0f54bfcb1 : SUCCESS in 16m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b36e5f3d06742d2bcb31d7726eef464 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09794946409740a59f387c7e126f4fd3 : SUCCESS in 9m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/122e19ac66be4882ab36050573772f86 : SUCCESS in 8m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d98392596b2544caae5b4ae0f14a8462 : SUCCESS in 9m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7aa6ee4072a14f84ad23418fd42bb11c : SUCCESS in 14m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70dbbe163a4940c19221f9e8ff9ed3b7 : SUCCESS in 6m 59s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/18e8aebcf88347f3927759169b235bec : SUCCESS in 10m 46s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/ebb5237b0cac407680c998e1ddacc475 : SUCCESS in 24m 57s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/e6add8df3137417a8f40fdd340126d4f : SUCCESS in 1h 12m 09s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/a9ce35d88a544702b9e1e403ad141077 : SUCCESS in 42m 50s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/866e010caa44498a933ef97618ae05a9 : SUCCESS in 44m 27s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/e6757d09c50e4aaaac2ef1a71bc42ecc : SUCCESS in 48m 10s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/cbf9ad07ac38446b80f42d1e9ed388fc : SUCCESS in 58m 28s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/d1631ea4cccd475ebbe9135abe91b70b : SUCCESS in 53m 35s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/c13c92457d294efe90506efba99fcdac : SUCCESS in 1h 16m 52s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/6e697c0a7d3746cb992cd591f73197a4 : SUCCESS in 53m 42s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/4f101384ecc54adc9cf89c6598797ef4 : SUCCESS in 1h 23m 52s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/8209f4208ac84f21afa714ab8207bbe2 : SUCCESS in 1h 08m 25s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/1888dfa357324ee99c849c8554273412 : SUCCESS in 37m 56s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/3801eca1cd4d4ecf9ae56f75ec71d540 : SUCCESS in 45m 06s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/b9c22fad216347479b0f01f44eb1597d : SUCCESS in 1h 48m 42s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/b0a07ecc03ef4c3e81747434697eb5a3 : SUCCESS in 5m 26s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/19524026c2bd4700bffdd096c235db85 : SUCCESS in 47m 14s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/d25f8fe88bb04b64899dddbdd5fed7e5 : SUCCESS in 49m 09s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/25aa8c805f9844e0bfd94153638823a6 : SUCCESS in 48m 56s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/4da03b41e82848d28834929ac27b17af : SUCCESS in 48m 06s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd625c8a700b42448dc609678e306508 : SUCCESS in 22m 52s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/e36ad3a00dea44a59b072717ad810a34 : SUCCESS in 23m 33s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/957d2280a1c14708b6e364a55609a131 : SUCCESS in 53m 15s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/50daabd70f1d4ee18e0a32c1ec07db48 : SUCCESS in 23m 38s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"908ad849896f192c2f4a85611ceefe446db51604","author":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"date":"2021-07-07 16:52:46.000000000","message":"Patch Set 9: Code-Review+2\n\n(4 comments)\n\nThanks, Julia. Some typos inline, otherwise LGTM.","accounts_in_message":[],"_revision_number":9},{"id":"458d18e2c47a167048f271fb11404daa6312c536","author":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"date":"2021-07-07 16:56:35.000000000","message":"Patch Set 9: Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"4ba9e29159b90dce131d650445017e845209a991","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 16:56:57.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"c97dd132cff7fa74dd67984b3bbd3df1bb661723","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-07-07 17:12:33.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"aef8c26220afac23123eb96018e097131e8e7c16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 18:24:40.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b6873dbcffb45d6920ae4cf1451703d : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b11f3a91c464d6b8b1e37bab4a8c5d6 : SUCCESS in 8m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b9a62ef3f0f487eae2795f40adb1e93 : SUCCESS in 9m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54bafce984b7416988a476575903e975 : SUCCESS in 13m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35ba12b59e4c46bca271ddb5ded326f6 : SUCCESS in 7m 08s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/536598e2509d438592132829ae06ef53 : SUCCESS in 10m 40s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/b71349c3ee544675adf7b79747a7d1fa : SUCCESS in 29m 29s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/16dc087e2cf24fca9c80e961295f8b95 : SUCCESS in 1h 26m 20s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/bf60ffb2d5fd4942b16de282c2ea824f : SUCCESS in 47m 30s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/060bbb0801964c52a18ce6560736abef : SUCCESS in 51m 36s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/94550e4a4bf842f0b12b715239219b53 : SUCCESS in 40m 47s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/77e08722f2db48259423afeca5bd79f6 : SUCCESS in 52m 22s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/baa8228e2ad344d29edf41fbe93b0a81 : SUCCESS in 50m 10s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/46ae32f2cb5844b1b0389e947b565fb2 : SUCCESS in 1h 15m 03s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/51328c1fb93640aab9705d5c2638bbb3 : SUCCESS in 50m 46s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/a3f8d64f11734c49b25ab893157bbfeb : SUCCESS in 1h 10m 33s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/2c5912e7714349d98f26630d211c24af : SUCCESS in 32m 45s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/86b97b43ba354da89909a771cd399bb3 : SUCCESS in 51m 16s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/ac0885c356ce4e10865f80e911f7d139 : SUCCESS in 1h 19m 11s","accounts_in_message":[],"_revision_number":9},{"id":"64223d76e1cc524b38f1529a633f5edc572e0582","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 18:24:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"07f5c9c013a4ec7140e9d5c5571394b4c3941b2b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 18:27:05.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2632d069838a4114b846f1534152a00f : SUCCESS in 1m 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/730f2da2493b41fa836a50f160034dc6 : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":9},{"id":"a27f17bdbf3b06c6a95e175c4eb2d3b609399a4a","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-07-07 18:35:03.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- 14G_bios_idrac https://stash.dellemc-community.org/logs/14G_bios_idrac/825 : SUCCESS in 3h 13m 20s\n- R6515_bios_idrac https://stash.dellemc-community.org/logs/R6515_bios_idrac/818 : SUCCESS in 3h 1m 50s\n- 13G_bios_ipmi https://stash.dellemc-community.org/logs/13G_bios_ipmi/826 : SUCCESS in 2h 54m 14s\n- 14G_uefi_redfish https://stash.dellemc-community.org/logs/14G_uefi_redfish/1116 : SUCCESS in 2h 53m 32s\n- 14G_bios_redfish https://stash.dellemc-community.org/logs/14G_bios_redfish/823 : SUCCESS in 2h 51m 47s\n- 14G_uefi_idrac https://stash.dellemc-community.org/logs/14G_uefi_idrac/1078 : SUCCESS in 1h 30m 50s\n- 13G_bios_idrac https://stash.dellemc-community.org/logs/13G_bios_idrac/831 : SUCCESS in 1h 25m 51s\n- XE2420_uefi_redfish https://stash.dellemc-community.org/logs/XE2420_uefi_redfish/1113 : SUCCESS in 1h 25m 11s\n- R7515_uefi_redfish https://stash.dellemc-community.org/logs/R7515_uefi_redfish/1110 : SUCCESS in 1h 24m 46s\n- 14G_uefi_ipmi https://stash.dellemc-community.org/logs/14G_uefi_ipmi/1073 : SUCCESS in 1h 22m 2s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"9851b68ee9d19d53313853fe93126dcfec7ba1e1","revisions":{"05dfdd40742298dde072246f14948dab7c972973":{"kind":"REWORK","_number":1,"created":"2021-06-04 18:57:52.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/80/794880/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/80/794880/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/80/794880/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/80/794880/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/80/794880/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/80/794880/1"}}},"commit":{"parents":[{"commit":"d0fb50c809d16ac4e1c8351686b2aeba7900aa3f","subject":"WIP: API to pass fields to node object list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/d0fb50c809d16ac4e1c8351686b2aeba7900aa3f"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-06-04 18:51:48.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-06-04 18:57:35.000000000","tz":-420},"subject":"WIP: Allow node_sanitize function to be provided overrides","message":"WIP: Allow node_sanitize function to be provided overrides\n\nChange-Id: I342e7900cac388cb4749480684418a5a15ac60eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/05dfdd40742298dde072246f14948dab7c972973"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/05dfdd40742298dde072246f14948dab7c972973"}]},"branch":"refs/heads/master"},"153a1165dd529059e525d5b7edda38f2f5d12036":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-06-04 20:26:30.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/80/794880/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/80/794880/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/80/794880/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/80/794880/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/80/794880/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/80/794880/2"}}},"commit":{"parents":[{"commit":"d0fb50c809d16ac4e1c8351686b2aeba7900aa3f","subject":"WIP: API to pass fields to node object list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/d0fb50c809d16ac4e1c8351686b2aeba7900aa3f"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-06-04 18:51:48.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-06-04 20:25:30.000000000","tz":-420},"subject":"Allow node_sanitize function to be provided overrides","message":"Allow node_sanitize function to be provided overrides\n\nThe biggest amount of API overhead is the node sanitization\nprocess, at least at this point in time.\n\nWe have streamlined the database interaction to ensure specific\nfield selection lists are as orderly as possible, but the\nnode sanitization code re-executes some methods over and over\nwhich do not require variable data from the underlying node.\n\nThese are blanket settings \"is the user allowed to see x, or y\".\n\nWhich means we can call node_sanitize pre-seeding these\narguments and execute the calls once, instead of a thousand times\nto have the same exact result.\n\nStory: 2008885\nTask: 42433\n\nChange-Id: I342e7900cac388cb4749480684418a5a15ac60eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/153a1165dd529059e525d5b7edda38f2f5d12036"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/153a1165dd529059e525d5b7edda38f2f5d12036"}]},"branch":"refs/heads/master"},"97473fe4229d2b618cd4fbaf65ce12362a585f13":{"kind":"REWORK","_number":3,"created":"2021-06-04 20:52:09.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/80/794880/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/80/794880/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/80/794880/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/80/794880/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/80/794880/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/80/794880/3"}}},"commit":{"parents":[{"commit":"c10cd998686babee36371a65b2d944b6aac2ca0e","subject":"API to pass fields to node object list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/c10cd998686babee36371a65b2d944b6aac2ca0e"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-06-04 18:51:48.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-06-04 20:51:49.000000000","tz":-420},"subject":"Allow node_sanitize function to be provided overrides","message":"Allow node_sanitize function to be provided overrides\n\nThe biggest amount of API overhead is the node sanitization\nprocess, at least at this point in time.\n\nWe have streamlined the database interaction to ensure specific\nfield selection lists are as orderly as possible, but the\nnode sanitization code re-executes some methods over and over\nwhich do not require variable data from the underlying node.\n\nThese are blanket settings \"is the user allowed to see x, or y\".\n\nWhich means we can call node_sanitize pre-seeding these\narguments and execute the calls once, instead of a thousand times\nto have the same exact result.\n\nStory: 2008885\nTask: 42433\n\nChange-Id: I342e7900cac388cb4749480684418a5a15ac60eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/97473fe4229d2b618cd4fbaf65ce12362a585f13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/97473fe4229d2b618cd4fbaf65ce12362a585f13"}]},"branch":"refs/heads/master"},"d00fdff6da45f9053e3561a89c845f8a47dbf490":{"kind":"REWORK","_number":4,"created":"2021-06-17 20:27:14.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/80/794880/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/80/794880/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/80/794880/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/80/794880/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/80/794880/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/80/794880/4"}}},"commit":{"parents":[{"commit":"b4a63102febe8e1ff175ae481ee648c43857e1a3","subject":"API to pass fields to node object list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/b4a63102febe8e1ff175ae481ee648c43857e1a3"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-06-04 18:51:48.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-06-17 20:26:38.000000000","tz":-420},"subject":"Allow node_sanitize function to be provided overrides","message":"Allow node_sanitize function to be provided overrides\n\nThe biggest amount of API overhead is the node sanitization\nprocess, at least at this point in time.\n\nWe have streamlined the database interaction to ensure specific\nfield selection lists are as orderly as possible, but the\nnode sanitization code re-executes some methods over and over\nwhich do not require variable data from the underlying node.\n\nThese are blanket settings \"is the user allowed to see x, or y\".\n\nWhich means we can call node_sanitize pre-seeding these\narguments and execute the calls once, instead of a thousand times\nto have the same exact result.\n\nStory: 2008885\nTask: 42433\n\nChange-Id: I342e7900cac388cb4749480684418a5a15ac60eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/d00fdff6da45f9053e3561a89c845f8a47dbf490"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/d00fdff6da45f9053e3561a89c845f8a47dbf490"}]},"branch":"refs/heads/master"},"8483378c27949f43bb3e0377a619c76efb33ed57":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-06-18 18:42:06.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/80/794880/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/80/794880/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/80/794880/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/80/794880/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/80/794880/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/80/794880/5"}}},"commit":{"parents":[{"commit":"69d39077c4a0621a5bcce9cf5fb9e4348c1fd53f","subject":"API to pass fields to node object list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/69d39077c4a0621a5bcce9cf5fb9e4348c1fd53f"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-06-04 18:51:48.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-06-18 18:41:55.000000000","tz":-420},"subject":"Allow node_sanitize function to be provided overrides","message":"Allow node_sanitize function to be provided overrides\n\nThe biggest amount of API overhead is the node sanitization\nprocess, at least at this point in time.\n\nWe have streamlined the database interaction to ensure specific\nfield selection lists are as orderly as possible, but the\nnode sanitization code re-executes some methods over and over\nwhich do not require variable data from the underlying node.\n\nThese are blanket settings \"is the user allowed to see x, or y\".\n\nWhich means we can call node_sanitize pre-seeding these\narguments and execute the calls once, instead of a thousand times\nto have the same exact result.\n\nStory: 2008885\nTask: 42433\n\nChange-Id: I342e7900cac388cb4749480684418a5a15ac60eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/8483378c27949f43bb3e0377a619c76efb33ed57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/8483378c27949f43bb3e0377a619c76efb33ed57"}]},"branch":"refs/heads/master"},"0ef6005c61898b45cb1b9f21a87fe9f70681178c":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-06-21 18:23:44.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/80/794880/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/80/794880/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/80/794880/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/80/794880/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/80/794880/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/80/794880/6"}}},"commit":{"parents":[{"commit":"74f2c33c5a53bec0b996e0862b7b381edf2ee386","subject":"API to pass fields to node object list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/74f2c33c5a53bec0b996e0862b7b381edf2ee386"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-06-04 18:51:48.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-06-21 17:41:00.000000000","tz":-420},"subject":"Allow node_sanitize function to be provided overrides","message":"Allow node_sanitize function to be provided overrides\n\nThe biggest amount of API overhead is the node sanitization\nprocess, at least at this point in time.\n\nWe have streamlined the database interaction to ensure specific\nfield selection lists are as orderly as possible, but the\nnode sanitization code re-executes some methods over and over\nwhich do not require variable data from the underlying node.\n\nThese are blanket settings \"is the user allowed to see x, or y\".\n\nWhich means we can call node_sanitize pre-seeding these\narguments and execute the calls once, instead of a thousand times\nto have the same exact result.\n\nStory: 2008885\nTask: 42433\n\nChange-Id: I342e7900cac388cb4749480684418a5a15ac60eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/0ef6005c61898b45cb1b9f21a87fe9f70681178c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/0ef6005c61898b45cb1b9f21a87fe9f70681178c"}]},"branch":"refs/heads/master"},"6e1fb208b128ae6708644e94588bff5236a982ea":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-06-21 22:31:42.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/80/794880/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/80/794880/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/80/794880/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/80/794880/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/80/794880/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/80/794880/7"}}},"commit":{"parents":[{"commit":"60de93213a462230178b0eb1116299d268e991dc","subject":"API to pass fields to node object list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/60de93213a462230178b0eb1116299d268e991dc"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-06-04 18:51:48.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-06-21 22:31:42.000000000","tz":0},"subject":"Allow node_sanitize function to be provided overrides","message":"Allow node_sanitize function to be provided overrides\n\nThe biggest amount of API overhead is the node sanitization\nprocess, at least at this point in time.\n\nWe have streamlined the database interaction to ensure specific\nfield selection lists are as orderly as possible, but the\nnode sanitization code re-executes some methods over and over\nwhich do not require variable data from the underlying node.\n\nThese are blanket settings \"is the user allowed to see x, or y\".\n\nWhich means we can call node_sanitize pre-seeding these\narguments and execute the calls once, instead of a thousand times\nto have the same exact result.\n\nStory: 2008885\nTask: 42433\n\nChange-Id: I342e7900cac388cb4749480684418a5a15ac60eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/6e1fb208b128ae6708644e94588bff5236a982ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/6e1fb208b128ae6708644e94588bff5236a982ea"}]},"branch":"refs/heads/master","description":"Rebase"},"a5ce2d9d63d1b8b2357486ec93892b0b621bb3c3":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2021-06-25 23:04:51.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/80/794880/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/80/794880/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/80/794880/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/80/794880/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/80/794880/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/80/794880/8"}}},"commit":{"parents":[{"commit":"87e42afb9e98204a683baff22c67171dca2100fe","subject":"API to pass fields to node object list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/87e42afb9e98204a683baff22c67171dca2100fe"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-06-04 18:51:48.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-06-25 23:04:51.000000000","tz":0},"subject":"Allow node_sanitize function to be provided overrides","message":"Allow node_sanitize function to be provided overrides\n\nThe biggest amount of API overhead is the node sanitization\nprocess, at least at this point in time.\n\nWe have streamlined the database interaction to ensure specific\nfield selection lists are as orderly as possible, but the\nnode sanitization code re-executes some methods over and over\nwhich do not require variable data from the underlying node.\n\nThese are blanket settings \"is the user allowed to see x, or y\".\n\nWhich means we can call node_sanitize pre-seeding these\narguments and execute the calls once, instead of a thousand times\nto have the same exact result.\n\nStory: 2008885\nTask: 42433\n\nChange-Id: I342e7900cac388cb4749480684418a5a15ac60eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/a5ce2d9d63d1b8b2357486ec93892b0b621bb3c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/a5ce2d9d63d1b8b2357486ec93892b0b621bb3c3"}]},"branch":"refs/heads/master","description":"Rebase"},"9851b68ee9d19d53313853fe93126dcfec7ba1e1":{"kind":"REWORK","_number":9,"created":"2021-07-06 20:06:33.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/80/794880/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/80/794880/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/80/794880/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/80/794880/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/80/794880/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/80/794880/9"}}},"commit":{"parents":[{"commit":"87e42afb9e98204a683baff22c67171dca2100fe","subject":"API to pass fields to node object list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/87e42afb9e98204a683baff22c67171dca2100fe"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-06-04 18:51:48.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-07-06 18:07:34.000000000","tz":-420},"subject":"Allow node_sanitize function to be provided overrides","message":"Allow node_sanitize function to be provided overrides\n\nThe biggest amount of API overhead is the node sanitization\nprocess, at least at this point in time.\n\nWe have streamlined the database interaction to ensure specific\nfield selection lists are as orderly as possible, but the\nnode sanitization code re-executes some methods over and over\nwhich do not require variable data from the underlying node.\n\nThese are blanket settings \"is the user allowed to see x, or y\".\n\nWhich means we can call node_sanitize pre-seeding these\narguments and execute the calls once, instead of a thousand times\nto have the same exact result.\n\nStory: 2008885\nTask: 42433\n\nChange-Id: I342e7900cac388cb4749480684418a5a15ac60eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/9851b68ee9d19d53313853fe93126dcfec7ba1e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/9851b68ee9d19d53313853fe93126dcfec7ba1e1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
