)]}'
{"id":"openstack%2Fironic~796569","triplet_id":"openstack%2Fironic~stable%2Fwallaby~I72eb4d7f0d26dd5deee9167c5e29cf6af1b683ae","project":"openstack/ironic","branch":"stable/wallaby","topic":"dhcp-less-configdrive-stable/wallaby","hashtags":[],"change_id":"I72eb4d7f0d26dd5deee9167c5e29cf6af1b683ae","subject":"dhcp-less: mention how to provide network_data to instance","status":"MERGED","created":"2021-06-16 12:06:09.000000000","updated":"2021-06-16 17:39:34.000000000","submitted":"2021-06-16 13:39:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"796569-dhcp-less-configdrive-stable/wallaby","cherry_pick_of_change":796478,"cherry_pick_of_patch_set":1,"meta_rev_id":"e1faef16c7b2ec9802f5e8833e21662e6accc331","_number":796569,"virtual_id_number":796569,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-16 13:39:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2021-06-16 13:19:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2021-06-16 13:19:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}],"CC":[{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-16 12:23:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-16 13:19:11.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2021-06-16 17:39:34.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"}],"messages":[{"id":"ea9e79e63f2ece92b8baf88385d0a51428231d63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-06-16 12:06:09.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a05f5b062e7fbe9e516a02dd225115484699a9bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 12:23:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb96d4c29d9741faba8113902eca6044 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ef336b41c944fe9824eb949e41dd216 : SUCCESS in 15m 17s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/2d8ba975486f419c8d2ae161a1f5152e : SUCCESS in 14m 22s","accounts_in_message":[],"_revision_number":1},{"id":"dc796046bb015a9a8bb1425cad7ad8309f196b10","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-06-16 13:19:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c9adb522937e49525546e6c03091a2903d64b97a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 13:19:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2ca437ec2317f208fd7171b825647704f8292693","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 13:39:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6c168057ddd44fab78a39e0804fed38 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0250f58b41804a6fac24f5a4319b8313 : SUCCESS in 15m 44s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/017bcff25f8348c8b667d2d5cc95f641 : SUCCESS in 13m 49s","accounts_in_message":[],"_revision_number":1},{"id":"efa087aac26d2c4f95970bef7cff6e8b1cf01cbd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 13:39:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"664708dfb8fc8ab4c1a781fdd2a39473a68c3515","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 13:41:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00f01e4911ef4840b1b1672a483ddb85 : SUCCESS in 1m 42s","accounts_in_message":[],"_revision_number":1},{"id":"e1faef16c7b2ec9802f5e8833e21662e6accc331","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-06-16 17:39:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- 13G_bios_idrac_wallaby https://stash.dellemc-community.org/logs/13G_bios_idrac_wallaby/31 : SUCCESS in 5h 32m 46s\n- 13G_bios_ipmi_wallaby https://stash.dellemc-community.org/logs/13G_bios_ipmi_wallaby/31 : SUCCESS in 4h 8m 9s\n- 14G_uefi_ipmi_wallaby https://stash.dellemc-community.org/logs/14G_uefi_ipmi_wallaby/31 : SUCCESS in 4h 7m 21s\n- XE2420_uefi_redfish_wallaby https://stash.dellemc-community.org/logs/XE2420_uefi_redfish_wallaby/31 : SUCCESS in 3h 46m 58s\n- 14G_uefi_redfish_wallaby https://stash.dellemc-community.org/logs/14G_uefi_redfish_wallaby/31 : SUCCESS in 3h 23m 54s\n- R6515_bios_idrac_wallaby https://stash.dellemc-community.org/logs/R6515_bios_idrac_wallaby/31 : SUCCESS in 2h 38m 51s\n- 14G_bios_redfish_wallaby https://stash.dellemc-community.org/logs/14G_bios_redfish_wallaby/31 : SUCCESS in 2h 38m 33s\n- 14G_uefi_idrac_wallaby https://stash.dellemc-community.org/logs/14G_uefi_idrac_wallaby/31 : SUCCESS in 2h 36m 30s\n- 14G_bios_idrac_wallaby https://stash.dellemc-community.org/logs/14G_bios_idrac_wallaby/31 : SUCCESS in 2h 21m 57s\n- R7515_uefi_redfish_wallaby https://stash.dellemc-community.org/logs/R7515_uefi_redfish_wallaby/31 : SUCCESS in 2h 4m 55s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"05c838073545b93141b0f6717350a9af9542e61d","revisions":{"05c838073545b93141b0f6717350a9af9542e61d":{"kind":"REWORK","_number":1,"created":"2021-06-16 12:06:09.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/69/796569/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/69/796569/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/69/796569/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/69/796569/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/69/796569/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/69/796569/1"}}},"commit":{"parents":[{"commit":"f99d6870712f51cffb4107b9ef6a117a3d86a8c1","subject":"Refactor iDRAC OEM extension manager calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/f99d6870712f51cffb4107b9ef6a117a3d86a8c1"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-06-15 13:51:13.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-06-16 12:06:09.000000000","tz":0},"subject":"dhcp-less: mention how to provide network_data to instance","message":"dhcp-less: mention how to provide network_data to instance\n\nChange-Id: I72eb4d7f0d26dd5deee9167c5e29cf6af1b683ae\n(cherry picked from commit fd7cde9b9ed7e331a4a4932b903534591e42d562)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/05c838073545b93141b0f6717350a9af9542e61d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/05c838073545b93141b0f6717350a9af9542e61d"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
