)]}'
{"id":"openstack%2Fironic~796898","triplet_id":"openstack%2Fironic~master~I1654e1333f95d4c3ffd25daa2b80ae4cf1033d91","project":"openstack/ironic","branch":"master","topic":"gate","hashtags":[],"change_id":"I1654e1333f95d4c3ffd25daa2b80ae4cf1033d91","subject":"Don\u0027t run the inspector job on changes to inspector tests","status":"MERGED","created":"2021-06-17 16:56:38.000000000","updated":"2021-06-19 05:06:33.000000000","submitted":"2021-06-18 22:15:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"796898-gate","meta_rev_id":"1afcaf21258d12e1b2884d01ddc9065b86295c24","_number":796898,"virtual_id_number":796898,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-18 22:15:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":2,"date":"2021-06-18 20:58:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":2,"date":"2021-06-18 08:22:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":1,"date":"2021-06-18 20:58:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}],"CC":[{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-17 17:19:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-17 18:52:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-18 00:39:37.000000000","updated_by":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-18 08:22:55.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2021-06-18 20:58:17.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2021-06-19 02:01:41.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"}],"messages":[{"id":"6db79171d513357db234e48e39fdb6e14c61e09d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-06-17 16:56:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46a5b7bda75f699ec44902c0038cc87a56a9c0e0","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-06-17 16:56:49.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"32543dac47222dae25a50d17e19958a2865b3bf6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-06-17 17:01:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4057cf5b918f5865910fb85f62f8ce2a2fa07d32","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 17:19:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/db3bd35e180247019904b9378abf739d : SUCCESS in 14m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b78f98f3c9664a3caf4e2da733e70b95 : SUCCESS in 17m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a21b3251eceeff78945c1d8419adc384929396ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 18:52:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c38a6de2a1f487e9e95a6031ed42422 : SUCCESS in 13m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3dd0b5020d747f88f2d57dbd1adf131 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd06d62e989841d993bf1683b22e0510 : SUCCESS in 13m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd3bc8492a3e4f39994319bd6d44bf92 : SUCCESS in 12m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/856c94cfe624429b8666bfc52ba55228 : SUCCESS in 11m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0514fa803c84314890ec082db400654 : SUCCESS in 16m 08s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/efdedc53c6e741f682cae11539fe2f3b : SUCCESS in 15m 36s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/ba7808a69ecb4dca97b4a066419c327c : SUCCESS in 28m 28s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/5a27b66c5e7148d0b2810369e16dc732 : SUCCESS in 1h 26m 29s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/d8b819b65e844ba6a67bf15571be8270 : SUCCESS in 51m 32s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/8f64a249825c4bd0a8e05bb365a6aa6c : SUCCESS in 50m 24s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/7608844b228a4fd09846841457f0eb68 : SUCCESS in 56m 33s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/42809b87f7504ce6bad01d641b6a906f : SUCCESS in 53m 51s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/cde53e3c6fac4d3bbc37300dee7a6842 : SUCCESS in 52m 08s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/51cf9240d7af4e73836434e3cda86050 : SUCCESS in 1h 14m 46s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/865f9f6d814746cbab343deeb9d4a055 : SUCCESS in 52m 03s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/a819a3c6361148158f29768090e13c99 : SUCCESS in 1h 46m 32s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/8b9ffa7e8aff4ac9958d1bba62e4fbe7 : SUCCESS in 1h 21m 37s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/0de50fdfd0454e1d973b66765e60a518 : SUCCESS in 40m 51s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/d8574d66338f4ef2b3ede209a4d17014 : SUCCESS in 47m 31s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/5f09fb1c75bd4370ac5f651cf3c3fc41 : SUCCESS in 1h 22m 47s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/14227ced730f4272a68f222feeb29eac : SUCCESS in 6m 18s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/2f12bd1ff796469e897b457b876c828c : SUCCESS in 56m 49s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/ec12a8d55a1d476187734491263169fa : SUCCESS in 57m 56s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/204f71ee1ab64646ad3e388f38281aae : SUCCESS in 48m 49s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/4d5b5ef3cd884c53b1edf2b5e4aa03b4 : SUCCESS in 50m 27s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77e1b6eefeac43fc9f3acda9d66796a0 : SUCCESS in 28m 38s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/47114899a259407787d2134d998fbac8 : SUCCESS in 25m 36s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/2c52689dd2e14c50a01d0ca9a47bf02b : SUCCESS in 55m 04s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/da5e6bfc4c434bfbb7da77b233a2f670 : SUCCESS in 29m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"985b4d93dc10f23efc4fd3cb83dc0e2cdd550272","author":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"date":"2021-06-18 00:39:37.000000000","message":"Patch Set 2:\n\nTempest run failed in Fujitsu iRMC CI environment\n\n- fujitsu-irmc-ci-job http://openstackci.jp.fujitsu.com/irmcci/98/796898/2/Fujitsu-iRMC-CI/fujitsu-irmc-ci-job/cc89b5d/ : FAILURE in 1h 49m 22s\n\nAbout this CI test, see https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_iRMC_CI","accounts_in_message":[],"_revision_number":2},{"id":"85ce3f0649e64c8c98c78534434caaae88229593","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-06-18 08:22:55.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"52f8a436955ffe6097c9edd80b88f7a7ae401a0a","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2021-06-18 20:58:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"20fe7489d061a794269d905cedf870a55bf90f4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 20:58:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ac183bc89d964abedca0fb3130b7c746e6145987","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 22:15:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04ed2771ba684e3c9a451c66575a8da2 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f5a231219bf429cb7c37301df9a03be : SUCCESS in 10m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b0c64cb62294696bb1d10e0c7e02202 : SUCCESS in 10m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7df5b508ad14f8b834d8c958631888a : SUCCESS in 14m 21s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/a4c21a2a95c942a99002c9f6c0c43743 : SUCCESS in 18m 09s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/6524720defc94538ba1251ffc8ca867a : SUCCESS in 25m 02s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/0d9b52e48bb94893a7a7b51611005f2d : SUCCESS in 1h 13m 14s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/10e0e4b45a6248d3907edf522783e5fe : SUCCESS in 42m 04s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/62123de28f1c4b6783c305baf6f71c50 : SUCCESS in 52m 51s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/d76e5df163f84132a8ef639e69c63c50 : SUCCESS in 45m 25s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/1f10566dc04b45e488e1c2d94d13a10b : SUCCESS in 44m 14s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/050ecaf118b144fa8fe8f2314ac7e96a : SUCCESS in 47m 53s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/18846a47a8f04b75b87552a67269e880 : SUCCESS in 1h 09m 52s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/d3d4a0dfefe24c61be082e2a29ada3dd : SUCCESS in 1h 02m 23s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/9f7b40435bfa412d965cbf5987f682a9 : SUCCESS in 1h 06m 46s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/784d1a2959704ff78cad6c9a43a627a9 : SUCCESS in 35m 17s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/c6e0f6cfa9bc40fdbc5f5fbb09edcccc : SUCCESS in 41m 37s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/3255a0575ea346a4ac44c2edfdc2fe89 : SUCCESS in 1h 10m 26s","accounts_in_message":[],"_revision_number":2},{"id":"0c9a16460a5914ce50f9a7ae67413f5e0455c236","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 22:15:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"eac136c4161cce6180c7331ba67b759dd3d85c61","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 22:17:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10c7cb577aee431d933cbd3979381923 : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":2},{"id":"b01e48479aaa5a3652cf270bcec876aa2279d684","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-06-19 02:01:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).\n\n- 13G_bios_idrac https://stash.dellemc-community.org/logs/13G_bios_idrac/744 : SUCCESS in 3h 9m 14s\n- 14G_uefi_redfish https://stash.dellemc-community.org/logs/14G_uefi_redfish/1028 : SUCCESS in 2h 54m 54s\n- R7515_uefi_redfish https://stash.dellemc-community.org/logs/R7515_uefi_redfish/1026 : SUCCESS in 2h 45m 18s\n- 14G_uefi_ipmi https://stash.dellemc-community.org/logs/14G_uefi_ipmi/987 : SUCCESS in 2h 8m 4s\n- 14G_uefi_idrac https://stash.dellemc-community.org/logs/14G_uefi_idrac/992 : SUCCESS in 1h 37m 48s\n- 14G_bios_idrac https://stash.dellemc-community.org/logs/14G_bios_idrac/745 : SUCCESS in 1h 37m 43s\n- 14G_bios_redfish https://stash.dellemc-community.org/logs/14G_bios_redfish/744 : FAILURE in 1h 33m 46s\n- 13G_bios_ipmi https://stash.dellemc-community.org/logs/13G_bios_ipmi/743 : SUCCESS in 1h 26m 8s\n- XE2420_uefi_redfish https://stash.dellemc-community.org/logs/XE2420_uefi_redfish/1028 : SUCCESS in 1h 19m 39s\n- R6515_bios_idrac https://stash.dellemc-community.org/logs/R6515_bios_idrac/742 : FAILURE in 48m 25s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"1afcaf21258d12e1b2884d01ddc9065b86295c24","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-06-19 05:06:33.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).\n\n- 14G_bios_redfish https://stash.dellemc-community.org/logs/14G_bios_redfish/745 : SUCCESS in 3h 4m 6s\n- 14G_bios_idrac https://stash.dellemc-community.org/logs/14G_bios_idrac/746 : SUCCESS in 2h 56m 56s\n- 13G_bios_ipmi https://stash.dellemc-community.org/logs/13G_bios_ipmi/744 : SUCCESS in 2h 51m 2s\n- R7515_uefi_redfish https://stash.dellemc-community.org/logs/R7515_uefi_redfish/1027 : SUCCESS in 2h 50m 6s\n- R6515_bios_idrac https://stash.dellemc-community.org/logs/R6515_bios_idrac/743 : FAILURE in 2h 19m 32s\n- 14G_uefi_idrac https://stash.dellemc-community.org/logs/14G_uefi_idrac/993 : SUCCESS in 1h 34m 41s\n- 14G_uefi_redfish https://stash.dellemc-community.org/logs/14G_uefi_redfish/1029 : SUCCESS in 1h 25m 3s\n- XE2420_uefi_redfish https://stash.dellemc-community.org/logs/XE2420_uefi_redfish/1029 : SUCCESS in 1h 24m 23s\n- 13G_bios_idrac https://stash.dellemc-community.org/logs/13G_bios_idrac/745 : SUCCESS in 1h 23m 23s\n- 14G_uefi_ipmi https://stash.dellemc-community.org/logs/14G_uefi_ipmi/988 : SUCCESS in 1h 20m 45s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5a8a77ec5be12fed105623bc2de6fd79a5fb5158","revisions":{"4986b1cfea083755cf180820d0cae48744bb5cef":{"kind":"REWORK","_number":1,"created":"2021-06-17 16:56:38.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/98/796898/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/98/796898/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/98/796898/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/98/796898/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/98/796898/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/98/796898/1"}}},"commit":{"parents":[{"commit":"ccde474b26b61907287572f8047fe45cd5146a39","subject":"Merge \"Add documentation for anaconda deploy interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/ccde474b26b61907287572f8047fe45cd5146a39"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-06-17 16:55:52.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-06-17 16:55:52.000000000","tz":120},"subject":"Don\u0027t run the inspector job on changes to inspector tests","message":"Don\u0027t run the inspector job on changes to inspector tests\n\nThe current irrelevant-files contains the ironic tests directory, but\nnot the inspector one.\n\nChange-Id: I1654e1333f95d4c3ffd25daa2b80ae4cf1033d91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4986b1cfea083755cf180820d0cae48744bb5cef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4986b1cfea083755cf180820d0cae48744bb5cef"}]},"branch":"refs/heads/master"},"5a8a77ec5be12fed105623bc2de6fd79a5fb5158":{"kind":"REWORK","_number":2,"created":"2021-06-17 17:01:32.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/98/796898/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/98/796898/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/98/796898/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/98/796898/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/98/796898/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/98/796898/2"}}},"commit":{"parents":[{"commit":"ccde474b26b61907287572f8047fe45cd5146a39","subject":"Merge \"Add documentation for anaconda deploy interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/ccde474b26b61907287572f8047fe45cd5146a39"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-06-17 16:55:52.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-06-17 17:01:24.000000000","tz":120},"subject":"Don\u0027t run the inspector job on changes to inspector tests","message":"Don\u0027t run the inspector job on changes to inspector tests\n\nThe current irrelevant-files contains the ironic tests directory, but\nnot the inspector one.\n\nChange-Id: I1654e1333f95d4c3ffd25daa2b80ae4cf1033d91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/5a8a77ec5be12fed105623bc2de6fd79a5fb5158"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/5a8a77ec5be12fed105623bc2de6fd79a5fb5158"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
