)]}'
{"id":"openstack%2Fironic~797679","triplet_id":"openstack%2Fironic~stable%2Fvictoria~I0c458c8d9ae673181beb6d85c2ee68235ccef239","project":"openstack/ironic","branch":"stable/victoria","hashtags":[],"change_id":"I0c458c8d9ae673181beb6d85c2ee68235ccef239","subject":"Cache AgentClient on Task, not globally","status":"MERGED","created":"2021-06-23 15:27:26.000000000","updated":"2021-07-16 12:14:14.000000000","submitted":"2021-07-16 12:12:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"797679","meta_rev_id":"a4296b78e0fa5151b2d90111c2528c2c7ef84d18","_number":797679,"virtual_id_number":797679,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-16 12:12:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":2,"date":"2021-07-16 09:57:36.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":2,"date":"2021-07-06 07:40:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":1,"date":"2021-07-16 09:57:36.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}],"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-23 17:33:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-23 18:50:59.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2021-06-24 01:48:02.000000000","updated_by":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-05 17:20:02.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2021-07-06 07:40:45.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2021-07-16 09:57:36.000000000","updated_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"reviewer":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"state":"REVIEWER"}],"messages":[{"id":"73e7aff238c6876652c5e4a0c2214de2e7848f50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-06-23 15:27:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac91a3b4f86db6e08dc42cf59d81012c39f37ee3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 17:33:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94bbe24455c345bfbe8ba896c0ca8cf3 : FAILURE in 8m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79aa6fce857a4360b238f49a15deb11d : SUCCESS in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eae77857b78d46598054ef6b44ee0b13 : FAILURE in 8m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/514f0732060b41368dc2437bc89c964c : FAILURE in 7m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aec818343ed04647a82fdccf74228932 : SUCCESS in 17m 56s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/f0ec254093884741bc8bed2f95d56122 : FAILURE in 11m 00s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/9ee25e3349d4438686064dec94f13848 : SUCCESS in 1h 25m 47s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/7a7793ae14574e2182e26dff9a3db9da : SUCCESS in 1h 59m 53s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/b3ab4ef2a3a842b8a00d4aff899995c4 : SUCCESS in 49m 29s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/89d9dc656a0a42a98907709f0549e4d4 : SUCCESS in 1h 04m 07s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/62e32600a5c045a2a897f935c7573dd8 : SUCCESS in 54m 04s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/a918899437f34ac2831316e32e2c8fe7 : SUCCESS in 58m 13s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/c13c11dd95754167927b2f495fd20758 : SUCCESS in 54m 46s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/bdaa8ba2b1054eb89fa5c2f33ab01fa5 : SUCCESS in 1h 13m 19s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/94f4535eb2904fd98f5b220b46086fd4 : SUCCESS in 56m 20s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/1a521925bcaa45179260ffd35a8fedeb : SUCCESS in 2h 03m 32s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/21845c2422d5450ab81f4b9dd059536c : SUCCESS in 1h 14m 00s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/c3906fd9c8a44af9a9b7abe0eeb018b9 : SUCCESS in 45m 26s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/658541a8fb394daa8654bd11d5f60ed6 : SUCCESS in 1h 03m 22s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/700b552949de4147a85466b7b04bfefc : SUCCESS in 1h 29m 01s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/708b0ed9ee9345d5a132f8984f954d1e : SUCCESS in 5m 46s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/81aefdd6a7d944508526e822bb20977c : SUCCESS in 1h 02m 52s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/0e661a02131044178149c055b30b2629 : SUCCESS in 59m 15s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/ee1b46b1a62d433e96003d9554711b5f : FAILURE in 40m 33s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/10753c2f70e94a18adb278efda3e52f0 : SUCCESS in 59m 30s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/988039d52c06459a8d183e14c655e3c5 : SUCCESS in 27m 09s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/3cedbb44764c466ab8018833d4e519f6 : SUCCESS in 57m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0f9e2621160cb75757002fb488e1490acf36154b","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-06-23 18:50:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).\n\n- 14G_uefi_redfish_victoria https://stash.dellemc-community.org/logs/14G_uefi_redfish_victoria/61 : SUCCESS in 3h 23m 1s\n- 14G_uefi_idrac_victoria https://stash.dellemc-community.org/logs/14G_uefi_idrac_victoria/59 : FAILURE in 3h 21m 51s\n- 14G_uefi_ipmi_victoria https://stash.dellemc-community.org/logs/14G_uefi_ipmi_victoria/58 : SUCCESS in 3h 13m 12s\n- 14G_bios_redfish_victoria https://stash.dellemc-community.org/logs/14G_bios_redfish_victoria/48 : SUCCESS in 2h 51m 34s\n- 13G_bios_ipmi_victoria https://stash.dellemc-community.org/logs/13G_bios_ipmi_victoria/48 : SUCCESS in 2h 44m 35s\n- R7515_uefi_redfish_victoria https://stash.dellemc-community.org/logs/R7515_uefi_redfish_victoria/57 : FAILURE in 1h 56m 52s\n- XE2420_uefi_redfish_victoria https://stash.dellemc-community.org/logs/XE2420_uefi_redfish_victoria/57 : FAILURE in 1h 56m 44s\n- R6515_bios_idrac_victoria https://stash.dellemc-community.org/logs/R6515_bios_idrac_victoria/48 : SUCCESS in 1h 42m 48s\n- 13G_bios_idrac_victoria https://stash.dellemc-community.org/logs/13G_bios_idrac_victoria/49 : SUCCESS in 1h 41m 44s\n- 14G_bios_idrac_victoria https://stash.dellemc-community.org/logs/14G_bios_idrac_victoria/49 : SUCCESS in 1h 39m 35s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"cf3f2a5c3ee414287e1a8aa856c3b960a6b56427","author":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"date":"2021-06-24 01:48:02.000000000","message":"Patch Set 1:\n\nTempest run failed in Fujitsu iRMC CI environment\n\n- fujitsu-irmc-ci-job http://openstackci.jp.fujitsu.com/irmcci/79/797679/1/Fujitsu-iRMC-CI/fujitsu-irmc-ci-job/09566ae/ : FAILURE in 2h 39m 33s\n\nAbout this CI test, see https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_iRMC_CI","accounts_in_message":[],"_revision_number":1},{"id":"96c90c8c5196d0acccf7160ae23f8f21e3b55ccd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-07-05 16:48:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b88f10663351dbd24e58509285caf730083f9b00","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-07-05 17:20:02.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/79/797679/2/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/461b30d/461b30d.html : FAILURE in 7m 07s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/79/797679/2/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3/54b3e6e/54b3e6e.html : FAILURE in 5m 37s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/79/797679/2/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3/c879824/c879824.html : FAILURE in 6m 07s (non-voting)\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/79/797679/2/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/361ce6f/361ce6f.html : FAILURE in 6m 12s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/79/797679/2/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/b56d38c/b56d38c.html : FAILURE in 4m 51s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/79/797679/2/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/334b6d0/334b6d0.html : FAILURE in 4m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"df1da9174c95bee154c25a3de9d9282d9283f463","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 19:04:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/102e1cc688e0422bb26c712be03d7067 : SUCCESS in 9m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae409fd9dcd54ecaa6fb83a36cb92a6c : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/339cae4b60fc4f868ac3bf359958b86c : SUCCESS in 8m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e438bfa5c5894b74a3f790f4b130bd4f : SUCCESS in 8m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6d7cc0dd134425aaa8a03700c31f191 : SUCCESS in 17m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f0ec673a61c4d4ea91224b09b178660 : SUCCESS in 6m 57s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/42dd70acb3b84cd1a4c3b261350d747b : SUCCESS in 9m 23s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/dcbb0ea743cb4cba8da529407293ff30 : SUCCESS in 1h 00m 11s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/77142775cc5549868a1166e39c355ece : SUCCESS in 1h 19m 56s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/b299e16433424898ba07009f37d1d2f8 : SUCCESS in 31m 38s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/90f984babd1541fa8970b7c6fafd8f3d : SUCCESS in 57m 16s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/08297e374418408cbbf99ba0889a707e : SUCCESS in 54m 05s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/8f38a3a9fcbc40bc90476dce9bf4ff41 : SUCCESS in 57m 19s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/0c926bce5dc94bf9aa5cbe0a8d9f6106 : SUCCESS in 52m 45s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/2d7b49ba252f440b929fd8e1274bee11 : SUCCESS in 1h 08m 22s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/921f6a8a40ad40bcbec06ab0c5381e5a : SUCCESS in 57m 00s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/7bd3eb03b3bf40bb98455b78e37e4677 : SUCCESS in 2h 00m 52s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/86c7484198ef405893c480b0209083da : SUCCESS in 1h 26m 19s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/6dd240160e7247c7841d62288351fe1e : SUCCESS in 43m 41s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/82f4ac5f6c1d4a489c51c2c56d721c9e : SUCCESS in 57m 58s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/f9652011eb6045fabc0650fd7ef297ca : SUCCESS in 1h 13m 19s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/20473bc3a5ca401f89ac5aa493d72d77 : SUCCESS in 6m 00s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/a617022bab73452abc57526e097d4597 : SUCCESS in 54m 24s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/91aeec0b4eca488dba0084128f135bb3 : SUCCESS in 52m 19s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/3850d11e9c9b408ea167fbe77e67bbc8 : SUCCESS in 1h 01m 06s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/d073fd3c262e41bebc22808282bc8d1c : SUCCESS in 45m 38s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0d69c7150ffe4fe9a946c26b69315aaa : SUCCESS in 21m 20s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/79bb0a5cb96f499a8486caf50b2c495c : SUCCESS in 51m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"72b8068a5670b144dc9c45369e29a23473459ef5","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-07-06 07:40:45.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3430ca8fd843e07c59a56d5ab2bb05728562ec0d","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2021-07-16 09:57:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThank you Dmitry.","accounts_in_message":[],"_revision_number":2},{"id":"b8e92a26f2f899b69624f8a401b89838beee52c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 09:58:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fc2970eff5e0b9a382dd419b43f0f2c7f5602e2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 12:12:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/715542eeb6994b44a06e35bf14288b02 : SUCCESS in 9m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a64d338cf1fd407e89205dabd53707ab : SUCCESS in 7m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cab45ce3bb12426eb5e429e388fd713b : SUCCESS in 7m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/060bb966b8014079a3b2696588309391 : SUCCESS in 16m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa0383c0afd14d2e9e5b44b8688fa25c : SUCCESS in 7m 13s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/687868708e334c31a710b639d4c9774b : SUCCESS in 10m 46s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/dfda4091a2294c5385d66452d92e542b : SUCCESS in 1h 10m 13s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/055d06579f0443b4a0577ba62d1a521a : SUCCESS in 1h 44m 04s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/600da72810e246d09eaa33696fd599fc : SUCCESS in 30m 26s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/0fd8bda83bbc48afbe0028d2b7e80bc3 : SUCCESS in 1h 16m 06s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/8b57fe03d6474763a05f559e2b6fd7e6 : SUCCESS in 54m 37s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/e6166f0b464e43e28962dece7c3afd15 : SUCCESS in 58m 17s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/f774b106acac4fc2b6e2973468f9ab77 : SUCCESS in 56m 18s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/5db67d3118fc4db1ac172287502ed2e1 : SUCCESS in 1h 12m 02s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/81e17437ee874fa0abfa427b6de7e128 : SUCCESS in 54m 39s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/7fa48d4c4e9b4149a3998210332999d7 : SUCCESS in 2h 08m 51s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/09cc53dfa68a44469a0fdff03967d0fe : SUCCESS in 1h 24m 54s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/5e1102980b27430d9805ea5cd57d88eb : SUCCESS in 46m 42s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/789c67d917054fc4a3b283e561cf738c : SUCCESS in 56m 19s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/0509b9592eb44211825040c3a70d2a39 : SUCCESS in 1h 30m 06s","accounts_in_message":[],"_revision_number":2},{"id":"843723e6ff4de5e08ea0f533c7b6333c8993613b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 12:12:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a4296b78e0fa5151b2d90111c2528c2c7ef84d18","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 12:14:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6c2b4aeedee43adbc58783e0d84e8cd : SUCCESS in 1m 24s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/023444edb64245b9b96f583723ea6a31 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0cb15a223a24927a203d84dded451de461008945","revisions":{"5e7de11342bdb4bc103f475d59c978e46a3b9cf3":{"kind":"REWORK","_number":1,"created":"2021-06-23 15:27:26.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/79/797679/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/79/797679/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/79/797679/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/79/797679/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/79/797679/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/79/797679/1"}}},"commit":{"parents":[{"commit":"755c75e2e82ee6a7addfac3b7a30fafefb36ab56","subject":"Fix node detail instance_uuid request handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/755c75e2e82ee6a7addfac3b7a30fafefb36ab56"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-06-23 15:11:16.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-06-23 15:26:49.000000000","tz":120},"subject":"[WIP] Cache AgentClient on Task, not globally","message":"[WIP] Cache AgentClient on Task, not globally\n\nChange-Id: I0c458c8d9ae673181beb6d85c2ee68235ccef239\n(cherry picked from commit d2117576dfa671a0aeee396d86c05941e11db2f4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/5e7de11342bdb4bc103f475d59c978e46a3b9cf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/5e7de11342bdb4bc103f475d59c978e46a3b9cf3"}]},"branch":"refs/heads/stable/victoria"},"0cb15a223a24927a203d84dded451de461008945":{"kind":"REWORK","_number":2,"created":"2021-07-05 16:48:20.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/79/797679/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/79/797679/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/79/797679/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/79/797679/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/79/797679/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/79/797679/2"}}},"commit":{"parents":[{"commit":"4ac6ad7317e20e1f2ebd99b1ce53d3ad2b1f09b0","subject":"Update the clear job id\u0027s constant","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4ac6ad7317e20e1f2ebd99b1ce53d3ad2b1f09b0"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-06-23 15:11:16.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-07-05 16:42:20.000000000","tz":120},"subject":"Cache AgentClient on Task, not globally","message":"Cache AgentClient on Task, not globally\n\nIn order to avoid potential cache coherency issues\nwhen using a globally cached AgentClient, e.g. with\nTSL certificates from the IPA, cache the AgentClient\non a per task basis.\n\nCo-Authored-By: Arne Wiebalck \u003carne.wiebalck@cern.ch\u003e\n\nConflicts:\n\tironic/drivers/modules/agent.py\n\tironic/drivers/modules/agent_base.py\n\tironic/drivers/modules/ansible/deploy.py\n\tironic/drivers/modules/iscsi_deploy.py\n\tironic/tests/unit/drivers/modules/test_agent.py\n\nStory: #2009004\nTask: #42678\n\nChange-Id: I0c458c8d9ae673181beb6d85c2ee68235ccef239\n(cherry picked from commit fcb6a1096f48336f19a5e8aa80445f57d4480c08)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/0cb15a223a24927a203d84dded451de461008945"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/0cb15a223a24927a203d84dded451de461008945"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
