)]}'
{"id":"openstack%2Fironic~797739","triplet_id":"openstack%2Fironic~master~If026b9afd0356a4fd1ac0a2157a06b8d5b0b176c","project":"openstack/ironic","branch":"master","topic":"fips-compatibility","attention_set":{"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2021-11-18 16:51:49.000000000","reason":"Someone else replied on the change"},"9914":{"account":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"last_update":"2021-11-18 16:51:49.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"If026b9afd0356a4fd1ac0a2157a06b8d5b0b176c","subject":"Add FIPS jobs","status":"NEW","created":"2021-06-23 18:55:07.000000000","updated":"2022-01-17 20:15:12.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"e4eb5def5a4bebe54ea931e2f5290fd2c2cfa73b","_number":797739,"virtual_id_number":797739,"owner":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-11-24 00:28:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}],"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-23 18:56:34.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2021-06-23 19:25:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-24 03:39:13.000000000","updated_by":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-26 07:32:24.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2021-07-29 21:26:32.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"},{"updated":"2021-07-29 22:57:52.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"}],"messages":[{"id":"94447e1693c6f8a9cc3573142e0d917ad9685cac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-06-23 18:55:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"23abf504673e7c0ecc87b2c0aab77f9c013b7e71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 19:25:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- ironic-tempest-functional-fips https://zuul.opendev.org/t/openstack/build/6b0d543046144921bbf7e239d50c520f : POST_FAILURE in 18m 13s\n- ironic-inspector-tempest-fips https://zuul.opendev.org/t/openstack/build/bb44a6bc67c14524ad175b15af3345ba : POST_FAILURE in 21m 04s","accounts_in_message":[],"_revision_number":1},{"id":"4e127d1e76cddf88a70cf4f2bfbc03f285f0ecf5","author":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"date":"2021-06-24 03:39:13.000000000","message":"Patch Set 1:\n\nTempest run failed in Fujitsu iRMC CI environment\n\n- fujitsu-irmc-ci-job http://openstackci.jp.fujitsu.com/irmcci/39/797739/1/Fujitsu-iRMC-CI/fujitsu-irmc-ci-job/d61a056/ : FAILURE in 1h 44m 31s\n\nAbout this CI test, see https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_iRMC_CI","accounts_in_message":[],"_revision_number":1},{"id":"4e17e99f3117dfbfcd4ae94b94be0515480adf88","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-06-26 07:32:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).\n\n- 14G_uefi_idrac https://stash.dellemc-community.org/logs/14G_uefi_idrac/1031 : SUCCESS in 3h 1m 55s\n- 13G_bios_ipmi https://stash.dellemc-community.org/logs/13G_bios_ipmi/782 : SUCCESS in 2h 56m 9s\n- 14G_uefi_ipmi https://stash.dellemc-community.org/logs/14G_uefi_ipmi/1026 : SUCCESS in 2h 53m 3s\n- 13G_bios_idrac https://stash.dellemc-community.org/logs/13G_bios_idrac/783 : SUCCESS in 2h 52m 13s\n- R6515_bios_idrac https://stash.dellemc-community.org/logs/R6515_bios_idrac/780 : SUCCESS in 1h 35m 29s\n- 14G_bios_redfish https://stash.dellemc-community.org/logs/14G_bios_redfish/782 : FAILURE in 1h 33m 50s\n- 14G_bios_idrac https://stash.dellemc-community.org/logs/14G_bios_idrac/783 : SUCCESS in 1h 27m 38s\n- XE2420_uefi_redfish https://stash.dellemc-community.org/logs/XE2420_uefi_redfish/1067 : SUCCESS in 1h 24m 25s\n- 14G_uefi_redfish https://stash.dellemc-community.org/logs/14G_uefi_redfish/1067 : SUCCESS in 1h 21m 40s\n- R7515_uefi_redfish https://stash.dellemc-community.org/logs/R7515_uefi_redfish/1065 : SUCCESS in 1h 17m 31s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"b949244501ca21b6dcb94332c4a81ca46eff5a8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-07-29 21:31:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3c72f9e5261b5a1f0023cee3584774c8d86095c3","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-07-29 22:57:52.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 finger://HPE-CI/004d7a4a72dc47639679e92c52042c56 : RETRY_LIMIT in 1m 02s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/2/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3/6bb4594/6bb4594.html : FAILURE in 35m 52s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 finger://HPE-CI/9f99a7e594484340b345dbac5b25033f : RETRY_LIMIT in 1m 04s (non-voting)\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/2/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/f62e907/f62e907.html : SUCCESS in 53m 19s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/2/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/c2dcac8/c2dcac8.html : FAILURE in 36m 18s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 finger://HPE-CI/cd5b38667e824b2b84fd7a799fee0dc5 : RETRY_LIMIT in 1m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"19bba2933757c56fe00b4c7df993f91b263018fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 00:37:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- ironic-tempest-functional-fips https://zuul.opendev.org/t/openstack/build/50c7aa8ab5fa4b66825d2e7e3d70e6d9 : RETRY_LIMIT in 1h 01m 31s\n- ironic-inspector-tempest-fips https://zuul.opendev.org/t/openstack/build/328cc9bf598f42e1a7a78c7a87ce6179 : RETRY_LIMIT in 1h 02m 21s","accounts_in_message":[],"_revision_number":2},{"id":"18fec39abf3da0ee81dbe2c735622e02d61c9d7f","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-07-30 00:39:09.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).\n\n- 14G_uefi_ipmi https://stash.dellemc-community.org/logs/14G_uefi_ipmi/1180 : SUCCESS in 3h 7m 3s\n- 14G_bios_idrac https://stash.dellemc-community.org/logs/14G_bios_idrac/932 : SUCCESS in 2h 24m 46s\n- 14G_uefi_redfish https://stash.dellemc-community.org/logs/14G_uefi_redfish/1223 : SUCCESS in 2h 23m 27s\n- R6515_bios_idrac https://stash.dellemc-community.org/logs/R6515_bios_idrac/925 : SUCCESS in 2h 18m 59s\n- 13G_bios_idrac https://stash.dellemc-community.org/logs/13G_bios_idrac/938 : SUCCESS in 2h 17m 28s\n- 14G_uefi_idrac https://stash.dellemc-community.org/logs/14G_uefi_idrac/1185 : SUCCESS in 2h 5m 23s\n- 14G_bios_redfish https://stash.dellemc-community.org/logs/14G_bios_redfish/930 : FAILURE in 1h 21m 56s\n- 13G_bios_ipmi https://stash.dellemc-community.org/logs/13G_bios_ipmi/933 : SUCCESS in 1h 13m 31s\n- XE2420_uefi_redfish https://stash.dellemc-community.org/logs/XE2420_uefi_redfish/1220 : SUCCESS in 1h 12m 15s\n- R7515_uefi_redfish https://stash.dellemc-community.org/logs/R7515_uefi_redfish/1218 : SUCCESS in 1h 1m 50s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":2},{"id":"71e97f6862f5c9e38f3f0194ca0d88fbda7b52e2","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2021-08-01 23:40:04.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"8d0e6bc6a6b80890e1d2d4c685fc21a604336fe7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2021-08-01 23:47:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5987e48741785bfd45bae9d7cef7dd092b202170","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 00:03:12.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- ironic-tempest-functional-fips https://zuul.opendev.org/t/openstack/build/9deb2b38fdf745aba14fd4a901cf7c31 : FAILURE in 10m 36s\n- ironic-inspector-tempest-fips https://zuul.opendev.org/t/openstack/build/9ecd830ca38f474ebfe40f33eee91308 : FAILURE in 11m 32s","accounts_in_message":[],"_revision_number":3},{"id":"99ba950d2e9f4a50eab1fb38435a7325f4390b7a","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-08-02 01:13:45.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/3/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/99ff81a/99ff81a.html : SUCCESS in 53m 17s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 finger://HPE-CI/518492f9a5254dffa309efe2a8250cd0 : RETRY_LIMIT in 1m 02s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 finger://HPE-CI/14762c5661b140649a6b23c80d75e321 : RETRY_LIMIT in 1m 02s (non-voting)\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 finger://HPE-CI/b967b5fa9be248c781aba8e300802bc0 : RETRY_LIMIT in 1m 02s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/3/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/44230b5/44230b5.html : FAILURE in 34m 01s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 finger://HPE-CI/ea2e186cdfa844da852fed6a56e16152 : RETRY_LIMIT in 1m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ef6b589c6acd737f0965f6bdbee987a644d1a43c","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-08-02 04:56:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- 14G_uefi_redfish https://stash.dellemc-community.org/logs/14G_uefi_redfish/1227 : SUCCESS in 5h 7m 37s\n- 13G_bios_ipmi https://stash.dellemc-community.org/logs/13G_bios_ipmi/937 : SUCCESS in 3h 46m 24s\n- 14G_uefi_ipmi https://stash.dellemc-community.org/logs/14G_uefi_ipmi/1184 : SUCCESS in 3h 12m 6s\n- 14G_uefi_idrac https://stash.dellemc-community.org/logs/14G_uefi_idrac/1189 : SUCCESS in 2h 28m 1s\n- XE2420_uefi_redfish https://stash.dellemc-community.org/logs/XE2420_uefi_redfish/1224 : SUCCESS in 2h 18m 57s\n- R7515_uefi_redfish https://stash.dellemc-community.org/logs/R7515_uefi_redfish/1222 : SUCCESS in 2h 9m 55s\n- 14G_bios_redfish https://stash.dellemc-community.org/logs/14G_bios_redfish/934 : SUCCESS in 1h 30m 3s\n- R6515_bios_idrac https://stash.dellemc-community.org/logs/R6515_bios_idrac/929 : SUCCESS in 1h 24m 7s\n- 13G_bios_idrac https://stash.dellemc-community.org/logs/13G_bios_idrac/942 : SUCCESS in 1h 23m 32s\n- 14G_bios_idrac https://stash.dellemc-community.org/logs/14G_bios_idrac/936 : SUCCESS in 1h 16m 3s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":3},{"id":"69d0e5cd7cc2072198a3ef426ebedfe6d950f597","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2021-08-03 01:28:32.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"70c85fb1cdb8e63448d8fba2e1c455ec3247270f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 01:54:00.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- ironic-tempest-functional-fips https://zuul.opendev.org/t/openstack/build/b45e8cde88134cf29c3aefe44ac7efc3 : FAILURE in 19m 53s\n- ironic-inspector-tempest-fips https://zuul.opendev.org/t/openstack/build/23a79c8552d24d44965b5d7b7a88ac70 : FAILURE in 22m 25s","accounts_in_message":[],"_revision_number":4},{"id":"2dcae26324dfdd85c961fc5ac733d893efa5638a","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-08-03 03:07:31.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 finger://HPE-CI/cf06360028db48aebfed6536ae05ae49 : RETRY_LIMIT in 1m 04s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/4/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3/0ab198e/0ab198e.html : FAILURE in 37m 55s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 finger://HPE-CI/eee17ebe095c4427968a14c88a695939 : RETRY_LIMIT in 1m 10s (non-voting)\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/4/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/2c0adcd/2c0adcd.html : SUCCESS in 50m 40s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/4/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/df46845/df46845.html : FAILURE in 37m 59s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/4/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/9f04c02/9f04c02.html : FAILURE in 1h 06m 36s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"74f3a73090c02e74549f8e5da7686fbb52b83f9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2021-08-03 04:20:25.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"efe7819d601189000731e891a99148372a0972b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 04:35:48.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- ironic-tempest-functional-fips https://zuul.opendev.org/t/openstack/build/0ecedb879d144ebf87811ecda4f469cd : FAILURE in 11m 35s\n- ironic-inspector-tempest-fips https://zuul.opendev.org/t/openstack/build/1fa6ecff585f4c15885132d36a7cc8fa : FAILURE in 12m 53s","accounts_in_message":[],"_revision_number":5},{"id":"7521a42e5f89577d8af9f7c90ad3951bd7f29f4f","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-08-03 06:00:57.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/5/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/7e570f5/7e570f5.html : SUCCESS in 53m 43s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 finger://HPE-CI/ff03ef84d20549078578a49a9712868b : RETRY_LIMIT in 1m 04s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 finger://HPE-CI/15a45bc2def6477fadc15654d6e4f180 : RETRY_LIMIT in 1m 02s (non-voting)\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 finger://HPE-CI/c85a9d65e1744e278eeb9fbb801a20da : RETRY_LIMIT in 1m 03s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/5/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/2c7df0d/2c7df0d.html : FAILURE in 35m 50s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/5/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/967b480/967b480.html : FAILURE in 1h 04m 49s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"984b3455358edbdb8baa62dadb68c989993ca8f5","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-08-03 06:07:06.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).\n\n- 14G_uefi_idrac https://stash.dellemc-community.org/logs/14G_uefi_idrac/1196 : FAILURE in 1h 45m 22s\n- 14G_bios_idrac https://stash.dellemc-community.org/logs/14G_bios_idrac/943 : FAILURE in 1h 44m 45s\n- 13G_bios_ipmi https://stash.dellemc-community.org/logs/13G_bios_ipmi/944 : FAILURE in 1h 40m 55s\n- 14G_uefi_ipmi https://stash.dellemc-community.org/logs/14G_uefi_ipmi/1191 : FAILURE in 1h 40m 17s\n- 14G_bios_redfish https://stash.dellemc-community.org/logs/14G_bios_redfish/941 : FAILURE in 1h 35m 19s\n- R6515_bios_idrac https://stash.dellemc-community.org/logs/R6515_bios_idrac/936 : FAILURE in 57m 41s\n- R7515_uefi_redfish https://stash.dellemc-community.org/logs/R7515_uefi_redfish/1229 : FAILURE in 56m 36s\n- XE2420_uefi_redfish https://stash.dellemc-community.org/logs/XE2420_uefi_redfish/1231 : FAILURE in 55m 26s\n- 13G_bios_idrac https://stash.dellemc-community.org/logs/13G_bios_idrac/949 : FAILURE in 54m 52s\n- 14G_uefi_redfish https://stash.dellemc-community.org/logs/14G_uefi_redfish/1234 : FAILURE in 49m 47s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":4},{"id":"ed2ba08df13b26beefeb15a064d68084a5112dff","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-08-03 08:41:32.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).\n\n- 14G_bios_redfish https://stash.dellemc-community.org/logs/14G_bios_redfish/943 : FAILURE in 2h 34m 21s\n- 14G_uefi_ipmi https://stash.dellemc-community.org/logs/14G_uefi_ipmi/1193 : FAILURE in 2h 34m 13s\n- 13G_bios_idrac https://stash.dellemc-community.org/logs/13G_bios_idrac/951 : FAILURE in 2h 34m 2s\n- 14G_uefi_idrac https://stash.dellemc-community.org/logs/14G_uefi_idrac/1198 : FAILURE in 2h 30m 41s\n- 14G_bios_idrac https://stash.dellemc-community.org/logs/14G_bios_idrac/945 : FAILURE in 2h 30m 22s\n- R6515_bios_idrac https://stash.dellemc-community.org/logs/R6515_bios_idrac/938 : FAILURE in 1h 57m 14s\n- R7515_uefi_redfish https://stash.dellemc-community.org/logs/R7515_uefi_redfish/1231 : FAILURE in 1h 53m 22s\n- 13G_bios_ipmi https://stash.dellemc-community.org/logs/13G_bios_ipmi/946 : FAILURE in 1h 52m 47s\n- 14G_uefi_redfish https://stash.dellemc-community.org/logs/14G_uefi_redfish/1236 : FAILURE in 1h 52m 25s\n- XE2420_uefi_redfish https://stash.dellemc-community.org/logs/XE2420_uefi_redfish/1233 : FAILURE in 1h 51m 55s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":5},{"id":"0c8a8afc9596234fefb880019299eb5d15263e45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2021-08-03 22:55:36.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"7c401a1ca859e31a3dbdee33a74a800e22c65ee1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 23:21:21.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- ironic-tempest-functional-fips https://zuul.opendev.org/t/openstack/build/3d27a695cd84478587937bd25a4f357b : FAILURE in 19m 37s\n- ironic-inspector-tempest-fips https://zuul.opendev.org/t/openstack/build/72ea359d4eb8455d87b7f479a740f6ed : FAILURE in 20m 35s","accounts_in_message":[],"_revision_number":6},{"id":"8ce765fca159b3f0f6a893f9797d78336de147c1","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-08-04 00:27:42.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/6/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/ea318b5/ea318b5.html : SUCCESS in 54m 26s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 finger://HPE-CI/d1c45f8800d24259a54c68943228f06e : RETRY_LIMIT in 1m 06s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 finger://HPE-CI/e8ccf6a9f3ee41229569223335a60649 : RETRY_LIMIT in 1m 03s (non-voting)\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 finger://HPE-CI/8da3250ecbd74e1c8de2e27dd189b247 : RETRY_LIMIT in 1m 02s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/6/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/ba2227f/ba2227f.html : FAILURE in 35m 44s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 finger://HPE-CI/4ec09f17c1094b6d9e6a6ba80ee226ec : RETRY_LIMIT in 1m 04s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9283553e45c5d79b814577c72c2f9727dd10a278","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2021-08-04 01:18:29.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"40b662c2a16fb7670096e63713fd4a977a7fb3ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 01:44:19.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- ironic-tempest-functional-fips https://zuul.opendev.org/t/openstack/build/41c2633f3739413ebe785be0eef390b5 : FAILURE in 20m 03s\n- ironic-inspector-tempest-fips https://zuul.opendev.org/t/openstack/build/a9cd3a8915b54c44ada72c66a31588a2 : FAILURE in 21m 05s","accounts_in_message":[],"_revision_number":7},{"id":"c7498df6c05b5ef530f475646762b3a0f720f1c0","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-08-04 02:25:55.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).\n\n- 14G_uefi_redfish https://stash.dellemc-community.org/logs/14G_uefi_redfish/1241 : FAILURE in 2h 44m 24s\n- 14G_uefi_idrac https://stash.dellemc-community.org/logs/14G_uefi_idrac/1203 : FAILURE in 2h 35m 48s\n- 14G_bios_idrac https://stash.dellemc-community.org/logs/14G_bios_idrac/950 : FAILURE in 2h 25m 44s\n- 14G_uefi_ipmi https://stash.dellemc-community.org/logs/14G_uefi_ipmi/1198 : FAILURE in 2h 8m 43s\n- R6515_bios_idrac https://stash.dellemc-community.org/logs/R6515_bios_idrac/943 : FAILURE in 2h 5m 48s\n- R7515_uefi_redfish https://stash.dellemc-community.org/logs/R7515_uefi_redfish/1236 : FAILURE in 2h 5m 41s\n- 13G_bios_ipmi https://stash.dellemc-community.org/logs/13G_bios_ipmi/951 : FAILURE in 1h 56m 40s\n- 14G_bios_redfish https://stash.dellemc-community.org/logs/14G_bios_redfish/948 : FAILURE in 1h 45m 52s\n- 13G_bios_idrac https://stash.dellemc-community.org/logs/13G_bios_idrac/956 : FAILURE in 1h 28m 46s\n- XE2420_uefi_redfish https://stash.dellemc-community.org/logs/XE2420_uefi_redfish/1238 : FAILURE in 1h 27m 38s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":6},{"id":"f8d891370dc0e8ba44ce5823b5f0c86b0be857e8","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-08-04 04:04:23.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/7/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/a0d47a6/a0d47a6.html : SUCCESS in 1h 00m 14s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/7/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3/a4ca26a/a4ca26a.html : FAILURE in 36m 03s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/7/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3/361df30/361df30.html : FAILURE in 48m 20s (non-voting)\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/7/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/04add44/04add44.html : SUCCESS in 55m 33s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/7/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/a3f9399/a3f9399.html : FAILURE in 38m 37s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/7/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/43140aa/43140aa.html : FAILURE in 1h 06m 32s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"17737270b12337d7054a622d2a918c388a3b0d09","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-08-04 04:55:01.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).\n\n- 14G_uefi_ipmi https://stash.dellemc-community.org/logs/14G_uefi_ipmi/1200 : FAILURE in 2h 29m 0s\n- 14G_uefi_redfish https://stash.dellemc-community.org/logs/14G_uefi_redfish/1243 : FAILURE in 2h 28m 52s\n- 14G_bios_idrac https://stash.dellemc-community.org/logs/14G_bios_idrac/952 : FAILURE in 2h 27m 32s\n- 14G_uefi_idrac https://stash.dellemc-community.org/logs/14G_uefi_idrac/1205 : FAILURE in 2h 4m 23s\n- R6515_bios_idrac https://stash.dellemc-community.org/logs/R6515_bios_idrac/945 : FAILURE in 1h 54m 44s\n- R7515_uefi_redfish https://stash.dellemc-community.org/logs/R7515_uefi_redfish/1238 : FAILURE in 1h 50m 55s\n- 13G_bios_idrac https://stash.dellemc-community.org/logs/13G_bios_idrac/958 : FAILURE in 1h 50m 10s\n- 14G_bios_redfish https://stash.dellemc-community.org/logs/14G_bios_redfish/950 : FAILURE in 1h 46m 57s\n- XE2420_uefi_redfish https://stash.dellemc-community.org/logs/XE2420_uefi_redfish/1240 : FAILURE in 1h 23m 35s\n- 13G_bios_ipmi https://stash.dellemc-community.org/logs/13G_bios_ipmi/953 : FAILURE in 1h 14m 47s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":7},{"id":"ecc4da170f9f3c45f1b3040392beee1040f96b3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2021-08-04 20:52:11.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"fc829c8a53c4c5755cf1c6e7063ae32979737c94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 21:17:59.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- ironic-tempest-functional-fips https://zuul.opendev.org/t/openstack/build/3a12bbd4777b4f658e850590b187672d : FAILURE in 15m 37s\n- ironic-inspector-tempest-fips https://zuul.opendev.org/t/openstack/build/85135924b91846da9259f6407d147dd9 : FAILURE in 21m 38s","accounts_in_message":[],"_revision_number":8},{"id":"a17a384fbca9aaf9b3fd77e139437c2c1100c462","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2021-08-04 21:28:45.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"789a7c2e3c88f0d70d481843842b14074956191d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 21:45:51.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- ironic-tempest-functional-fips https://zuul.opendev.org/t/openstack/build/9b7fdb1432854eaeaa5d55fe6d24ad91 : FAILURE in 13m 14s\n- ironic-inspector-tempest-fips https://zuul.opendev.org/t/openstack/build/295d52726baf4a1cacab4e725038b1df : FAILURE in 14m 34s","accounts_in_message":[],"_revision_number":9},{"id":"9356ee8a3f2b29c0b58b6e766b738a794bea1456","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-08-04 22:58:26.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/9/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/be80b1f/be80b1f.html : SUCCESS in 56m 37s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/9/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3/07149b6/07149b6.html : FAILURE in 36m 21s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/9/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3/20dbc81/20dbc81.html : FAILURE in 44m 43s (non-voting)\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/9/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/46756f4/46756f4.html : SUCCESS in 52m 55s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/9/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/95c8202/95c8202.html : FAILURE in 37m 07s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/9/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/41920f6/41920f6.html : SUCCESS in 53m 32s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"edb4b358060c41e0e06ee9e46867b3c849778063","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2021-08-04 23:47:53.000000000","message":"Uploaded patch set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"7318ea8919f07f1b61d19307a152948ea3e593ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 00:51:31.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- ironic-tempest-functional-fips https://zuul.opendev.org/t/openstack/build/b47f54da3073456e8864a0ad2ca0e24b : FAILURE in 18m 00s\n- ironic-inspector-tempest-fips https://zuul.opendev.org/t/openstack/build/89fe7d77833b4567883494ea8d0829c5 : FAILURE in 26m 16s","accounts_in_message":[],"_revision_number":10},{"id":"1873b9b3d25c0882bc59c2a390c77748bc789a61","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-08-05 01:18:31.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- 14G_uefi_redfish https://stash.dellemc-community.org/logs/14G_uefi_redfish/1253 : SUCCESS in 4h 25m 47s\n- R7515_uefi_redfish https://stash.dellemc-community.org/logs/R7515_uefi_redfish/1248 : SUCCESS in 4h 17m 38s\n- 14G_bios_idrac https://stash.dellemc-community.org/logs/14G_bios_idrac/962 : SUCCESS in 3h 20m 16s\n- R6515_bios_idrac https://stash.dellemc-community.org/logs/R6515_bios_idrac/955 : SUCCESS in 3h 17m 19s\n- 14G_bios_redfish https://stash.dellemc-community.org/logs/14G_bios_redfish/960 : SUCCESS in 2h 30m 7s\n- 13G_bios_ipmi https://stash.dellemc-community.org/logs/13G_bios_ipmi/963 : SUCCESS in 2h 27m 44s\n- 14G_uefi_ipmi https://stash.dellemc-community.org/logs/14G_uefi_ipmi/1210 : SUCCESS in 2h 14m 48s\n- XE2420_uefi_redfish https://stash.dellemc-community.org/logs/XE2420_uefi_redfish/1250 : SUCCESS in 1h 16m 59s\n- 14G_uefi_idrac https://stash.dellemc-community.org/logs/14G_uefi_idrac/1215 : SUCCESS in 1h 16m 34s\n- 13G_bios_idrac https://stash.dellemc-community.org/logs/13G_bios_idrac/968 : SUCCESS in 1h 13m 16s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":8},{"id":"533c04fd59f81098e08438e66d375344ab1eba90","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2021-08-05 01:19:56.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"b983c7438957e934a561184dba63655d8fbef576","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-08-05 01:20:27.000000000","message":"Patch Set 10:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":10},{"id":"0fcb90a23c7b29c0421cb0b05e965f0e7cec2f51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 01:53:43.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- ironic-tempest-functional-fips https://zuul.opendev.org/t/openstack/build/63a5a7a846e64d76b396b0dcf001cd6b : FAILURE in 16m 16s\n- ironic-inspector-tempest-fips https://zuul.opendev.org/t/openstack/build/80015ddd929144caaa5f36c2a36c461f : FAILURE in 26m 16s","accounts_in_message":[],"_revision_number":10},{"id":"d468d8a8eeff888d2aef8aa4ad44bf8273258ef5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2021-08-05 02:26:31.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"097e7cf2d9ea7a801d0035d0550ff97b4ab628e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 02:51:10.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- ironic-tempest-functional-fips https://zuul.opendev.org/t/openstack/build/2eca52e0aaee4bddbb43676a47100ee8 : FAILURE in 15m 22s\n- ironic-inspector-tempest-fips https://zuul.opendev.org/t/openstack/build/ef4826301d8146f58863c18525abb0a4 : FAILURE in 20m 01s","accounts_in_message":[],"_revision_number":11},{"id":"c0168e3ef1674aad255cd505f0ae740b95ab88da","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-08-05 04:19:07.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/11/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/9f2a00e/9f2a00e.html : SUCCESS in 1h 01m 34s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 finger://HPE-CI/45b96535dc5947a2976974f41db02cd8 : RETRY_LIMIT in 1m 07s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/11/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3/f7f999d/f7f999d.html : FAILURE in 59m 28s (non-voting)\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/11/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/2d2b800/2d2b800.html : SUCCESS in 1h 02m 44s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/11/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/390fa62/390fa62.html : FAILURE in 40m 55s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/11/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/2a088ec/2a088ec.html : FAILURE in 1h 10m 56s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"5d0683588d91e7a6abb3c5846ce7ada5e7ae3756","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-08-05 05:09:48.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- 14G_bios_redfish https://stash.dellemc-community.org/logs/14G_bios_redfish/963 : SUCCESS in 3h 31m 8s\n- 13G_bios_ipmi https://stash.dellemc-community.org/logs/13G_bios_ipmi/966 : SUCCESS in 3h 18m 44s\n- 14G_bios_idrac https://stash.dellemc-community.org/logs/14G_bios_idrac/965 : SUCCESS in 3h 6m 38s\n- 14G_uefi_ipmi https://stash.dellemc-community.org/logs/14G_uefi_ipmi/1213 : SUCCESS in 3h 6m 32s\n- R6515_bios_idrac https://stash.dellemc-community.org/logs/R6515_bios_idrac/958 : SUCCESS in 2h 48m 39s\n- 14G_uefi_idrac https://stash.dellemc-community.org/logs/14G_uefi_idrac/1218 : SUCCESS in 2h 19m 31s\n- 14G_uefi_redfish https://stash.dellemc-community.org/logs/14G_uefi_redfish/1256 : SUCCESS in 2h 15m 3s\n- XE2420_uefi_redfish https://stash.dellemc-community.org/logs/XE2420_uefi_redfish/1253 : SUCCESS in 2h 3m 42s\n- 13G_bios_idrac https://stash.dellemc-community.org/logs/13G_bios_idrac/971 : SUCCESS in 2h 1m 54s\n- R7515_uefi_redfish https://stash.dellemc-community.org/logs/R7515_uefi_redfish/1251 : SUCCESS in 1h 39m 38s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":9},{"id":"ecc48dd16b264a8e7cf21a4246ddc8a283394384","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-08-05 11:49:18.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).\n\n- 14G_bios_redfish https://stash.dellemc-community.org/logs/14G_bios_redfish/967 : SUCCESS in 3h 23m 35s\n- 14G_uefi_redfish https://stash.dellemc-community.org/logs/14G_uefi_redfish/1260 : SUCCESS in 3h 9m 27s\n- 13G_bios_ipmi https://stash.dellemc-community.org/logs/13G_bios_ipmi/970 : SUCCESS in 3h 8m 39s\n- 14G_bios_idrac https://stash.dellemc-community.org/logs/14G_bios_idrac/969 : SUCCESS in 3h 6m 47s\n- 13G_bios_idrac https://stash.dellemc-community.org/logs/13G_bios_idrac/975 : SUCCESS in 2h 21m 40s\n- 14G_uefi_ipmi https://stash.dellemc-community.org/logs/14G_uefi_ipmi/1217 : SUCCESS in 2h 16m 14s\n- R6515_bios_idrac https://stash.dellemc-community.org/logs/R6515_bios_idrac/962 : FAILURE in 2h 8m 33s\n- R7515_uefi_redfish https://stash.dellemc-community.org/logs/R7515_uefi_redfish/1255 : SUCCESS in 2h 4m 19s\n- 14G_uefi_idrac https://stash.dellemc-community.org/logs/14G_uefi_idrac/1222 : SUCCESS in 2h 2m 57s\n- XE2420_uefi_redfish https://stash.dellemc-community.org/logs/XE2420_uefi_redfish/1257 : SUCCESS in 2h 1m 41s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":10},{"id":"0f3430a5734516dac3a1355f3dfe57b185b2a8ff","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-08-05 13:59:51.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).\n\n- 14G_bios_idrac https://stash.dellemc-community.org/logs/14G_bios_idrac/970 : SUCCESS in 4h 16m 57s\n- 14G_uefi_idrac https://stash.dellemc-community.org/logs/14G_uefi_idrac/1223 : FAILURE in 4h 15m 33s\n- 13G_bios_idrac https://stash.dellemc-community.org/logs/13G_bios_idrac/976 : SUCCESS in 3h 58m 19s\n- 14G_bios_redfish https://stash.dellemc-community.org/logs/14G_bios_redfish/968 : SUCCESS in 3h 53m 54s\n- 13G_bios_ipmi https://stash.dellemc-community.org/logs/13G_bios_ipmi/971 : SUCCESS in 3h 9m 24s\n- 14G_uefi_ipmi https://stash.dellemc-community.org/logs/14G_uefi_ipmi/1218 : SUCCESS in 3h 8m 59s\n- R6515_bios_idrac https://stash.dellemc-community.org/logs/R6515_bios_idrac/963 : SUCCESS in 2h 57m 51s\n- R7515_uefi_redfish https://stash.dellemc-community.org/logs/R7515_uefi_redfish/1256 : SUCCESS in 2h 53m 28s\n- 14G_uefi_redfish https://stash.dellemc-community.org/logs/14G_uefi_redfish/1261 : SUCCESS in 2h 52m 27s\n- XE2420_uefi_redfish https://stash.dellemc-community.org/logs/XE2420_uefi_redfish/1258 : SUCCESS in 2h 7m 15s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":11},{"id":"36eca1015caebb400a8b6dce409c21d5d91455a2","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-09-15 04:03:55.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"3076803b02774d484371304fd90a52fd2c08fbc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 04:26:29.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- ironic-tempest-functional-fips https://zuul.opendev.org/t/openstack/build/8a64ba92a8284398b1f7e7fd27bf878c : FAILURE in 18m 03s\n- ironic-inspector-tempest-fips https://zuul.opendev.org/t/openstack/build/71bf2a3086f345cbafa8e7bcdcc9fa56 : FAILURE in 20m 49s","accounts_in_message":[],"_revision_number":11},{"id":"3fcebf1fdfa04884d27c4f51c5582202c646c148","author":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"date":"2021-09-15 13:04:43.000000000","message":"Patch Set 11:\n\nTempest test failed in Fujitsu iRMC CI environment.\n\nWe run following Tempest tests\n  - ironic_tempest_plugin.tests.scenario.test_baremetal_basic_ops.BaremetalBasicOps.test_baremetal_server_ops_partition_image\n  - ironic_tempest_plugin.tests.scenario.test_baremetal_basic_ops.BaremetalBasicOps.test_baremetal_server_ops_wholedisk_image\n\nIt\u0027ll take few minutes till log appears in following URL.\n\n\n- fujitsu-irmc-ci-job http://openstackci.jp.fujitsu.com/irmcci/39/797739/11/Fujitsu-iRMC-CI/fujitsu-irmc-ci-job/386dfb5/ : FAILURE in 1h 04m 00s\n\nAbout this CI test, see https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_iRMC_CI","accounts_in_message":[],"_revision_number":11},{"id":"b4d959031771d6e3fe8305aba4bea2d04ae6efe6","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-09-16 11:07:53.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).\n\n- 14G_bios_idrac https://stash.dellemc-community.org/logs/14G_bios_idrac/1249 : SUCCESS in 3h 30m 49s\n- R6515_bios_idrac https://stash.dellemc-community.org/logs/R6515_bios_idrac/1241 : SUCCESS in 3h 19m 18s\n- 14G_uefi_redfish https://stash.dellemc-community.org/logs/14G_uefi_redfish/1544 : SUCCESS in 3h 19m 12s\n- 13G_bios_ipmi https://stash.dellemc-community.org/logs/13G_bios_ipmi/1249 : SUCCESS in 2h 55m 15s\n- R7515_uefi_redfish https://stash.dellemc-community.org/logs/R7515_uefi_redfish/1534 : SUCCESS in 2h 31m 49s\n- 14G_bios_redfish https://stash.dellemc-community.org/logs/14G_bios_redfish/1246 : SUCCESS in 2h 20m 13s\n- 13G_bios_idrac https://stash.dellemc-community.org/logs/13G_bios_idrac/1255 : SUCCESS in 2h 15m 10s\n- XE2420_uefi_redfish https://stash.dellemc-community.org/logs/XE2420_uefi_redfish/1536 : SUCCESS in 2h 9m 45s\n- 14G_uefi_idrac https://stash.dellemc-community.org/logs/14G_uefi_idrac/1501 : FAILURE in 1h 36m 39s\n- 14G_uefi_ipmi https://stash.dellemc-community.org/logs/14G_uefi_ipmi/1496 : SUCCESS in 1h 31m 32s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":11},{"id":"45437b1bc6f4db033386fc0b8d27d02e103a8dd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-09-21 14:07:35.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"a42107d63da1aa29ed11de7dc5256ac644e76122","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-21 14:36:37.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- ironic-tempest-functional-fips https://zuul.opendev.org/t/openstack/build/916181b1c8394f5e860809518c1e1eec : FAILURE in 22m 46s\n- ironic-inspector-tempest-fips https://zuul.opendev.org/t/openstack/build/99952c00a0fe4acd8f4ec87b2917c4a8 : FAILURE in 26m 25s","accounts_in_message":[],"_revision_number":12},{"id":"3c121efff1996e630b0ea8c924eec0605a33750c","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-09-21 15:19:57.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/12/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/ab330d8/ab330d8.html : SUCCESS in 54m 04s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/12/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3/6c6bd3d/6c6bd3d.html : FAILURE in 33m 33s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/12/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3/360dd6a/360dd6a.html : FAILURE in 42m 12s (non-voting)\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/12/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/d9df332/d9df332.html : SUCCESS in 50m 10s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/12/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/59dffed/59dffed.html : FAILURE in 34m 02s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/12/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/e14ed0d/e14ed0d.html : FAILURE in 35m 20s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"68add0d10a9b08d5179a446ce35dd74bc7f0f257","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-09-21 18:50:00.000000000","message":"Patch Set 12:\n\n*sigh*\n\n_SYSTEMD_UNIT\u003dhttpd.service\n_STREAM_ID\u003d3fa9f5beacbf46aaa9cad497736fedb3\nMESSAGE\u003dAH00526: Syntax error on line 4 of /etc/httpd/conf.d/http-services-tls-proxy.conf:\n_PID\u003d51735\n\n_PID\u003d51735\n_SYSTEMD_INVOCATION_ID\u003daa3e13ca83694a26ae0eca1713c4576d\nMESSAGE\u003dInvalid command \u0027SSLEngine\u0027, perhaps misspelled or defined by a module not included in the server configuration\n\nle-sigh.","accounts_in_message":[],"_revision_number":12},{"id":"7e5e40a9cc00e973f23e0864e9f1b727c81c8992","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-09-21 18:56:35.000000000","message":"Patch Set 12:\n\n\u003e Patch Set 12:\n\u003e \n\u003e *sigh*\n\u003e \n\u003e _SYSTEMD_UNIT\u003dhttpd.service\n\u003e _STREAM_ID\u003d3fa9f5beacbf46aaa9cad497736fedb3\n\u003e MESSAGE\u003dAH00526: Syntax error on line 4 of /etc/httpd/conf.d/http-services-tls-proxy.conf:\n\u003e _PID\u003d51735\n\u003e \n\u003e _PID\u003d51735\n\u003e _SYSTEMD_INVOCATION_ID\u003daa3e13ca83694a26ae0eca1713c4576d\n\u003e MESSAGE\u003dInvalid command \u0027SSLEngine\u0027, perhaps misspelled or defined by a module not included in the server configuration\n\u003e \n\u003e le-sigh.\n\nNice find -- so are we just missing mod_ssl? (as described here, for instance - https://www.namecheap.com/support/knowledgebase/article.aspx/9793/2238/invalid-command-sslengine-error-on-apache/ )","accounts_in_message":[],"_revision_number":12},{"id":"c1440488eda122ea1224c185ddc34361281bc681","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-09-21 18:59:16.000000000","message":"Patch Set 12:\n\n\u003e Patch Set 12:\n\u003e \n\u003e \u003e Patch Set 12:\n\u003e \u003e \n\u003e \u003e *sigh*\n\u003e \u003e \n\u003e \u003e _SYSTEMD_UNIT\u003dhttpd.service\n\u003e \u003e _STREAM_ID\u003d3fa9f5beacbf46aaa9cad497736fedb3\n\u003e \u003e MESSAGE\u003dAH00526: Syntax error on line 4 of /etc/httpd/conf.d/http-services-tls-proxy.conf:\n\u003e \u003e _PID\u003d51735\n\u003e \u003e \n\u003e \u003e _PID\u003d51735\n\u003e \u003e _SYSTEMD_INVOCATION_ID\u003daa3e13ca83694a26ae0eca1713c4576d\n\u003e \u003e MESSAGE\u003dInvalid command \u0027SSLEngine\u0027, perhaps misspelled or defined by a module not included in the server configuration\n\u003e \u003e \n\u003e \u003e le-sigh.\n\u003e \n\u003e Nice find -- so are we just missing mod_ssl? (as described here, for instance - https://www.namecheap.com/support/knowledgebase/article.aspx/9793/2238/invalid-command-sslengine-error-on-apache/ )\n\nI\u0027m not sure yet.... YET! digging through logs now.","accounts_in_message":[],"_revision_number":12},{"id":"cb2637d98d822386673be700f53a37b5647635a0","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-09-21 19:19:28.000000000","message":"Patch Set 12:\n\nSo we don\u0027t install it.\n\nLooks like we silently fail package installs because where invoked, we also can\u0027t find packages related to it.","accounts_in_message":[],"_revision_number":12},{"id":"38cc106d54107a4edb54979a57c17de4a28b5360","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-09-21 19:26:17.000000000","message":"Patch Set 12:\n\n2021-09-21 14:22:51.019 | Package which-2.21-16.el8.x86_64 is already installed.\n2021-09-21 14:22:51.032 | No match for argument: docker-io\n2021-09-21 14:22:51.040 | Package iptables-1.8.4-20.el8.x86_64 is already installed.\n2021-09-21 14:22:51.046 | Package gnupg2-2.2.20-2.el8.x86_64 is already installed.\n2021-09-21 14:22:51.058 | No match for argument: libvirt-python\n2021-09-21 14:22:51.073 | No match for argument: qemu-system-x86\n2021-09-21 14:22:51.080 | Package openssh-clients-8.0p1-9.el8.x86_64 is already installed.\n2021-09-21 14:22:51.097 | No match for argument: shellinabox\n2021-09-21 14:22:51.135 | Error: Unable to find a match: docker-io libvirt-python qemu-system-x86 shellinabox","accounts_in_message":[],"_revision_number":12},{"id":"4d6d8744909c34ee71f50bbdc3b0c54adf265682","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-09-21 19:39:47.000000000","message":"Patch Set 13: Commit message was updated.","accounts_in_message":[],"_revision_number":13},{"id":"7a4caa02edf7b811e22fdba8afef61696a193edc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-21 20:10:08.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- ironic-tempest-functional-fips https://zuul.opendev.org/t/openstack/build/c2c3aef8e7d3412ca8400749df000273 : FAILURE in 19m 48s\n- ironic-inspector-tempest-fips https://zuul.opendev.org/t/openstack/build/637898930ece41cdbf1b43bc71c70a6d : FAILURE in 28m 28s","accounts_in_message":[],"_revision_number":13},{"id":"0c3c2d936997ebce49b4e9b0b650c50747f29efb","author":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"date":"2021-09-21 20:36:56.000000000","message":"Patch Set 13:\n\nThis change depends on a change that failed to merge.\n\nAbout this CI test, see https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_iRMC_CI","accounts_in_message":[],"_revision_number":13},{"id":"179b908dffb528eefba4bbb4f65347404e9653e4","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-09-21 20:37:12.000000000","message":"Patch Set 13:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":13},{"id":"311acd51130857d8a7fc8cf1f4ec2977a488d731","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-09-21 20:40:44.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"8f0f062c47646f4ac787f66b87c7e3ebbbbc69a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-21 21:19:14.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- ironic-tempest-functional-fips https://zuul.opendev.org/t/openstack/build/088f7bcd1d564e0aa3a455bdf3974a42 : FAILURE in 28m 18s\n- ironic-inspector-tempest-fips https://zuul.opendev.org/t/openstack/build/d278977d5189453eb41245bdce521212 : FAILURE in 34m 04s","accounts_in_message":[],"_revision_number":13},{"id":"626a87810517d5de6f3ea9f69b7af1861e381dce","author":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"date":"2021-09-21 21:33:20.000000000","message":"Patch Set 13:\n\nThis change depends on a change that failed to merge.\n\nAbout this CI test, see https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_iRMC_CI","accounts_in_message":[],"_revision_number":13},{"id":"7c093dfb11d68716538fe863ad7ec26a8f208361","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-09-21 21:33:28.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"35aa306a87ed7d541a939b77f376cb580ef3a2b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-21 23:41:18.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- ironic-tempest-functional-fips https://zuul.opendev.org/t/openstack/build/2bfe002f857845d6b1222695ce6ba060 : SUCCESS in 30m 00s\n- ironic-inspector-tempest-fips https://zuul.opendev.org/t/openstack/build/1ae796c679e741009f8eda59ab3bbdea : FAILURE in 42m 42s","accounts_in_message":[],"_revision_number":13},{"id":"d2768b34f40e9935d1b229c81de8a6eb97715e65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-09-22 00:05:18.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"587b2295946f0b3cfe74dfafff3a27045a1ed17a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 00:41:31.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- ironic-tempest-functional-fips https://zuul.opendev.org/t/openstack/build/5e9fef63e33d4bb9b403bccb465910d3 : SUCCESS in 30m 46s\n- ironic-inspector-tempest-fips https://zuul.opendev.org/t/openstack/build/68c2c5963bc64f81bc63f9e4f7f4b14d : FAILURE in 34m 39s","accounts_in_message":[],"_revision_number":14},{"id":"12cb14582add7b07185b1667286e98483f9e4ecd","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-09-22 03:13:19.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/14/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/0ee8457/0ee8457.html : TIMED_OUT in 3h 01m 33s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/14/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3/66f08b5/66f08b5.html : FAILURE in 33m 50s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/14/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3/2c9c07e/2c9c07e.html : FAILURE in 44m 09s (non-voting)\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/14/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/00f7f2c/00f7f2c.html : SUCCESS in 50m 53s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/14/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/7309192/7309192.html : FAILURE in 34m 42s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/14/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/e0d0474/e0d0474.html : SUCCESS in 53m 36s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"c88e0b8fa862860b95007e105c2ef2afcdb14993","author":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"date":"2021-09-22 08:23:16.000000000","message":"Patch Set 14:\n\nTempest test failed in Fujitsu iRMC CI environment.\n\nWe run following Tempest tests\n  - ironic_tempest_plugin.tests.scenario.test_baremetal_basic_ops.BaremetalBasicOps.test_baremetal_server_ops_partition_image\n  - ironic_tempest_plugin.tests.scenario.test_baremetal_basic_ops.BaremetalBasicOps.test_baremetal_server_ops_wholedisk_image\n\nIt\u0027ll take few minutes till log appears in following URL.\n\n\n- fujitsu-irmc-ci-job http://openstackci.jp.fujitsu.com/irmcci/39/797739/14/Fujitsu-iRMC-CI/fujitsu-irmc-ci-job/c8744a0/ : FAILURE in 17m 02s\n\nAbout this CI test, see https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_iRMC_CI","accounts_in_message":[],"_revision_number":14},{"id":"8cb6a42a07699d8e6459dd000d15c1fa4eba3d5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-09-22 13:52:34.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"f4b9f8acd0d249726684912a74000e4d3c06b3b2","author":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"date":"2021-09-22 13:53:53.000000000","message":"Patch Set 15:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nAbout this CI test, see https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_iRMC_CI","accounts_in_message":[],"_revision_number":15},{"id":"6554c2b917584561e30a90955a9474faea045553","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-09-22 13:54:46.000000000","message":"Patch Set 15:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":15},{"id":"e7f5f589e8ec02ac01ada936e5638421177578da","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-09-22 14:19:23.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"8c6245aea0f15031e66ae31c94aa0b8e20cbc400","author":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"date":"2021-09-22 14:20:24.000000000","message":"Patch Set 15:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nAbout this CI test, see https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_iRMC_CI","accounts_in_message":[],"_revision_number":15},{"id":"431cf7db838dbd7537330bedb70d87c5fd1cbfc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 14:26:33.000000000","message":"Patch Set 15: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":15},{"id":"7186adb52a1af6fb666c8276aa8875f001e52302","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 14:26:38.000000000","message":"Patch Set 15:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":15},{"id":"2940f272418ef443c5c40575f49f236ee0869613","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 15:20:04.000000000","message":"Patch Set 15:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":15},{"id":"665f199ae6c3c20a10659b630fdcb69997b3329f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 15:20:10.000000000","message":"Patch Set 15:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":15},{"id":"e2e0d384787f05d70e800343bcf9c6140ff40ddf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-09-22 15:52:13.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"759375fa13877208fc87201ee77f14fb303d9351","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-09-22 18:03:34.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/16/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/069e247/069e247.html : FAILURE in 35m 50s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/16/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3/8630c51/8630c51.html : FAILURE in 34m 59s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/16/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3/a623ab4/a623ab4.html : FAILURE in 43m 16s (non-voting)\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/16/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/d93f44f/d93f44f.html : SUCCESS in 51m 41s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/16/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/861c523/861c523.html : FAILURE in 34m 46s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/16/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/32b8f1e/32b8f1e.html : FAILURE in 1h 05m 18s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"15f4d5115ee62c0aad3a1aca9b055d8baec3ae2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 19:02:27.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- ironic-tempest-functional-fips https://zuul.opendev.org/t/openstack/build/35da2c96845b4cbca98e178fe18cd1e0 : SUCCESS in 37m 23s\n- ironic-inspector-tempest-fips https://zuul.opendev.org/t/openstack/build/c71dd1471bfd49c9922399421ce7897e : FAILURE in 34m 14s","accounts_in_message":[],"_revision_number":16},{"id":"234c4e8358f08af97c2fde5ccc36c5eeb9ec7921","author":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"date":"2021-09-23 22:50:50.000000000","message":"Patch Set 16:\n\nTempest test passed in Fujitsu iRMC CI environment.\n\nWe run following Tempest tests\n  - ironic_tempest_plugin.tests.scenario.test_baremetal_basic_ops.BaremetalBasicOps.test_baremetal_server_ops_partition_image\n  - ironic_tempest_plugin.tests.scenario.test_baremetal_basic_ops.BaremetalBasicOps.test_baremetal_server_ops_wholedisk_image\n\nIt\u0027ll take few minutes till log appears in following URL.\n\n\n- fujitsu-irmc-ci-job http://openstackci.jp.fujitsu.com/irmcci/39/797739/16/Fujitsu-iRMC-CI/fujitsu-irmc-ci-job/fa4330f/ : SUCCESS in 1h 14m 15s\n\nAbout this CI test, see https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_iRMC_CI","accounts_in_message":[],"_revision_number":16},{"id":"bbeb7fea16788d4210b5ef8ac8f29dcbbd3e307d","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-09-24 15:26:19.000000000","message":"Patch Set 16:\n\nBuild failed (check pipeline).\n\n- 13G_bios_idrac https://stash.dellemc-community.org/logs/13G_bios_idrac/1329 : SUCCESS in 4h 31m 10s\n- 14G_bios_idrac https://stash.dellemc-community.org/logs/14G_bios_idrac/1323 : SUCCESS in 4h 11m 54s\n- 14G_bios_redfish https://stash.dellemc-community.org/logs/14G_bios_redfish/1320 : SUCCESS in 3h 48m 26s\n- XE2420_uefi_redfish https://stash.dellemc-community.org/logs/XE2420_uefi_redfish/1610 : SUCCESS in 3h 35m 37s\n- 13G_bios_ipmi https://stash.dellemc-community.org/logs/13G_bios_ipmi/1323 : SUCCESS in 3h 33m 24s\n- R7515_uefi_redfish https://stash.dellemc-community.org/logs/R7515_uefi_redfish/1608 : SUCCESS in 3h 16m 28s\n- R6515_bios_idrac https://stash.dellemc-community.org/logs/R6515_bios_idrac/1315 : SUCCESS in 2h 54m 20s\n- 14G_uefi_redfish https://stash.dellemc-community.org/logs/14G_uefi_redfish/1618 : FAILURE in 2h 45m 31s\n- 14G_uefi_ipmi https://stash.dellemc-community.org/logs/14G_uefi_ipmi/1570 : FAILURE in 2h 28m 3s\n- 14G_uefi_idrac https://stash.dellemc-community.org/logs/14G_uefi_idrac/1575 : SUCCESS in 2h 21m 36s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":16},{"id":"79ba770a40d3abca9beca9f4d3f99908156a0e79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-09-24 16:45:12.000000000","message":"Patch Set 17: Patch Set 16 was rebased","accounts_in_message":[],"_revision_number":17},{"id":"961426cd9784e1b435b3b3f365254958683fa77d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 17:30:59.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- ironic-tempest-functional-fips https://zuul.opendev.org/t/openstack/build/66889e03932846f195262df89480be92 : RETRY_LIMIT in 12m 11s\n- ironic-inspector-tempest-fips https://zuul.opendev.org/t/openstack/build/f4e7900cc9694e4ab4f056fc54c6b305 : RETRY_LIMIT in 13m 59s","accounts_in_message":[],"_revision_number":17},{"id":"be7904952904dd66c8ad317fc4990bb5b77d57ea","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-09-24 20:04:29.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/17/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/30f2e8b/30f2e8b.html : TIMED_OUT in 3h 01m 24s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/17/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3/6b2ae7a/6b2ae7a.html : FAILURE in 34m 11s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/17/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3/78a3720/78a3720.html : FAILURE in 42m 30s (non-voting)\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/17/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/f56e115/f56e115.html : SUCCESS in 48m 59s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/17/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/37ee64b/37ee64b.html : FAILURE in 33m 21s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/17/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/d172d69/d172d69.html : SUCCESS in 50m 45s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"033796fe0bade8d988512a34d6d4e19e5123c917","author":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"date":"2021-09-25 10:31:14.000000000","message":"Patch Set 17:\n\nTempest test failed in Fujitsu iRMC CI environment.\n\nWe run following Tempest tests\n  - ironic_tempest_plugin.tests.scenario.test_baremetal_basic_ops.BaremetalBasicOps.test_baremetal_server_ops_partition_image\n  - ironic_tempest_plugin.tests.scenario.test_baremetal_basic_ops.BaremetalBasicOps.test_baremetal_server_ops_wholedisk_image\n\nIt\u0027ll take few minutes till log appears in following URL.\n\n\n- fujitsu-irmc-ci-job http://openstackci.jp.fujitsu.com/irmcci/39/797739/17/Fujitsu-iRMC-CI/fujitsu-irmc-ci-job/c954923/ : POST_FAILURE in 50m 41s\n\nAbout this CI test, see https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_iRMC_CI","accounts_in_message":[],"_revision_number":17},{"id":"50acb1715f24f75e53fcb1e1a12315c7f6a7beba","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-09-26 02:42:56.000000000","message":"Patch Set 17:\n\nBuild succeeded (check pipeline).\n\n- 13G_bios_ipmi https://stash.dellemc-community.org/logs/13G_bios_ipmi/1336 : SUCCESS in 4h 16m 41s\n- 14G_bios_redfish https://stash.dellemc-community.org/logs/14G_bios_redfish/1333 : SUCCESS in 3h 38m 2s\n- R6515_bios_idrac https://stash.dellemc-community.org/logs/R6515_bios_idrac/1328 : SUCCESS in 3h 30m 10s\n- XE2420_uefi_redfish https://stash.dellemc-community.org/logs/XE2420_uefi_redfish/1623 : SUCCESS in 3h 21m 20s\n- 14G_uefi_ipmi https://stash.dellemc-community.org/logs/14G_uefi_ipmi/1583 : SUCCESS in 3h 18m 37s\n- 13G_bios_idrac https://stash.dellemc-community.org/logs/13G_bios_idrac/1342 : SUCCESS in 3h 13m 28s\n- 14G_bios_idrac https://stash.dellemc-community.org/logs/14G_bios_idrac/1336 : SUCCESS in 2h 33m 49s\n- R7515_uefi_redfish https://stash.dellemc-community.org/logs/R7515_uefi_redfish/1621 : SUCCESS in 2h 14m 54s\n- 14G_uefi_redfish https://stash.dellemc-community.org/logs/14G_uefi_redfish/1631 : SUCCESS in 2h 7m 49s\n- 14G_uefi_idrac https://stash.dellemc-community.org/logs/14G_uefi_idrac/1588 : SUCCESS in 2h 7m 40s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":17},{"id":"34210c9eeb31d7b5a07bed7e151c9048e2758b19","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-09-29 20:48:24.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"04dd0b84ebaabd37601bdf7276000e5248784037","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 21:36:53.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- ironic-tempest-functional-fips https://zuul.opendev.org/t/openstack/build/40ba3732ebf74f71985b9d813fb59fd3 : SUCCESS in 43m 29s\n- ironic-inspector-tempest-fips https://zuul.opendev.org/t/openstack/build/f2eae945e9d348e7b5243897c039ef7e : FAILURE in 42m 49s","accounts_in_message":[],"_revision_number":17},{"id":"81406c4f49453446f2c15653b7bc940ee19ab95d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-09-29 21:43:17.000000000","message":"Patch Set 18: Published edit on patch set 17.","accounts_in_message":[],"_revision_number":18},{"id":"41791cbf64c9568a7ec353c371c76e26dd856468","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 22:50:57.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- ironic-tempest-functional-fips https://zuul.opendev.org/t/openstack/build/aa054ceb1b31463d8141c68437381fc1 : SUCCESS in 30m 40s\n- ironic-inspector-tempest-fips https://zuul.opendev.org/t/openstack/build/e86c0eecca634e77a2ffc7144b6ff809 : SUCCESS in 1h 05m 43s","accounts_in_message":[],"_revision_number":18},{"id":"d21dc00acb3c040738aa414704dcf3b84ead0a63","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-09-30 01:17:36.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/18/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/12f43b3/12f43b3.html : TIMED_OUT in 3h 01m 26s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/18/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3/d480b5a/d480b5a.html : FAILURE in 31m 53s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/18/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3/097eaac/097eaac.html : FAILURE in 39m 59s (non-voting)\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/18/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/77f1728/77f1728.html : TIMED_OUT in 3h 01m 21s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/18/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/94e60f8/94e60f8.html : FAILURE in 32m 09s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/18/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/6d91384/6d91384.html : TIMED_OUT in 3h 01m 21s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"9bd9d3b1faa2ab0e466bdd1ee15684dab3c95c60","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-09-30 02:53:43.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).\n\n- XE2420_uefi_redfish https://stash.dellemc-community.org/logs/XE2420_uefi_redfish/1641 : SUCCESS in 4h 3m 38s\n- 14G_uefi_ipmi https://stash.dellemc-community.org/logs/14G_uefi_ipmi/1601 : SUCCESS in 3h 58m 51s\n- R6515_bios_idrac https://stash.dellemc-community.org/logs/R6515_bios_idrac/1346 : SUCCESS in 3h 54m 43s\n- 14G_bios_redfish https://stash.dellemc-community.org/logs/14G_bios_redfish/1351 : SUCCESS in 3h 48m 36s\n- 13G_bios_ipmi https://stash.dellemc-community.org/logs/13G_bios_ipmi/1354 : SUCCESS in 3h 15m 58s\n- 13G_bios_idrac https://stash.dellemc-community.org/logs/13G_bios_idrac/1360 : SUCCESS in 3h 4m 34s\n- R7515_uefi_redfish https://stash.dellemc-community.org/logs/R7515_uefi_redfish/1639 : SUCCESS in 3h 0m 13s\n- 14G_uefi_redfish https://stash.dellemc-community.org/logs/14G_uefi_redfish/1649 : SUCCESS in 2h 53m 18s\n- 14G_bios_idrac https://stash.dellemc-community.org/logs/14G_bios_idrac/1354 : FAILURE in 2h 51m 24s\n- 14G_uefi_idrac https://stash.dellemc-community.org/logs/14G_uefi_idrac/1606 : SUCCESS in 2h 45m 56s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":17},{"id":"56dfb9e330bf66995faf0c81bdfb8a1ed1c75617","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-09-30 08:23:51.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).\n\n- XE2420_uefi_redfish https://stash.dellemc-community.org/logs/XE2420_uefi_redfish/1644 : SUCCESS in 4h 25m 23s\n- 14G_uefi_ipmi https://stash.dellemc-community.org/logs/14G_uefi_ipmi/1604 : SUCCESS in 4h 9m 58s\n- 14G_bios_idrac https://stash.dellemc-community.org/logs/14G_bios_idrac/1357 : SUCCESS in 4h 9m 58s\n- R6515_bios_idrac https://stash.dellemc-community.org/logs/R6515_bios_idrac/1349 : SUCCESS in 3h 40m 18s\n- 13G_bios_ipmi https://stash.dellemc-community.org/logs/13G_bios_ipmi/1357 : SUCCESS in 3h 33m 45s\n- 14G_uefi_idrac https://stash.dellemc-community.org/logs/14G_uefi_idrac/1609 : SUCCESS in 3h 28m 3s\n- 14G_bios_redfish https://stash.dellemc-community.org/logs/14G_bios_redfish/1354 : SUCCESS in 3h 11m 43s\n- 13G_bios_idrac https://stash.dellemc-community.org/logs/13G_bios_idrac/1363 : FAILURE in 3h 10m 39s\n- R7515_uefi_redfish https://stash.dellemc-community.org/logs/R7515_uefi_redfish/1642 : SUCCESS in 3h 2m 11s\n- 14G_uefi_redfish https://stash.dellemc-community.org/logs/14G_uefi_redfish/1652 : SUCCESS in 2h 19m 58s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":18},{"id":"040209a3def430732df3fc57a0571a84d384c2fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-10-06 18:34:49.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"5254da2e4ad0b3c2989128c5c390d42b0f78c016","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 19:42:59.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/92be6c6f27ad4e7d941d48d92775ec56 : SUCCESS in 15m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9bc17def99b94dd481b0eb4ac1990b00 : SUCCESS in 18m 51s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"18109869b3cb6a6525d8dfb4b23028bd82ac7e3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 20:27:59.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/023778df8b85494581603f6bb71f4bce : SUCCESS in 12m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27860e7a57694ba2aaabd3f95cd6f5df : SUCCESS in 11m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b5dd418bb5e4cd8b3cba9ab963ae3fd : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce39e1348da6430a87209309c5403376 : SUCCESS in 18m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7541b8ecd97149189981d37ce058a129 : SUCCESS in 9m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c36b4ab486446099026dfce424f9fc0 : SUCCESS in 14m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18a214f144c7412b9a5c7e21c1928801 : SUCCESS in 17m 05s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/d863f625689a4b4a87884fd2efed70f5 : SUCCESS in 15m 29s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/8476c457fa1542238152153540ac657a : SUCCESS in 27m 47s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/6718874291784c44b5a3a2826c9e5757 : SUCCESS in 27m 48s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/de2e9840b18641359407dfc9510931d5 : SUCCESS in 1h 15m 05s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/cae152c011164f259cadd218ea3b3c9c : SUCCESS in 1h 12m 10s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/2f3ffc354750475488487a99db4b635c : FAILURE in 37m 28s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/516c74985421426f8b7f206e2414a553 : SUCCESS in 48m 38s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/b0f3ce1e9bc7441489dbaea9a8d28f32 : SUCCESS in 51m 48s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/2b45334f02944e50966d9c69544fe565 : SUCCESS in 50m 49s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/687afbdd415c47c392a1a5782052224d : SUCCESS in 1h 08m 24s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/6d498038e8664e8cb17e0d155d233e0b : FAILURE in 36m 36s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/33fdb57e18fb4d42b6e8e8ea0af2bf1e : SUCCESS in 1h 45m 52s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/ce784e396b3349308078d921ae3ad036 : SUCCESS in 1h 20m 10s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/d2414eed8d7846b98bc6387a09c1391d : SUCCESS in 41m 48s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/80d68ea0f6944dd8941fcabe1e15daea : SUCCESS in 51m 13s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/41c70fbcc4264532891f0dedc831333d : SUCCESS in 1h 37m 16s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/1c530aa872d3467db17a09a238d1286a : SUCCESS in 5m 19s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/0586d5c988e64b5683d6719bc7c6dc6a : SUCCESS in 53m 50s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/8fc5bd2861ca4080a96ce10f11a5d05c : SUCCESS in 1h 06m 55s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/627b08485efb4e238d7dcfe4dda5a6e0 : SUCCESS in 50m 55s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/bc3c9256cd9d424c92b267daf58f8112 : SUCCESS in 51m 46s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/afa106e06d824bdfbb3b7c79ad4c26a3 : SUCCESS in 53m 05s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21def2f741484e03adc93c0e32b2ebbb : SUCCESS in 33m 28s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/40d40a8f24164064a0df81ca9faf5fa2 : SUCCESS in 29m 52s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/b27e8eff7439484e892a7395168d935b : SUCCESS in 56m 11s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/adafa9a51c0846348cadc0e0d9780e8e : SUCCESS in 31m 31s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"6c5d0ff0b9dabcd83c1cf32eb86337210015828d","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-10-06 22:19:11.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 finger://HPE-CI/0c416591a1314fd28c80fb570956ca2e : RETRY_LIMIT in 1m 02s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 finger://HPE-CI/1aa40595cb9b4c8199e90cd08cf3a034 : RETRY_LIMIT in 1m 07s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/19/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3/71718dc/71718dc.html : FAILURE in 44m 15s (non-voting)\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 finger://HPE-CI/8f0fb811ce024a30a38c7cbcb0ccb7e0 : RETRY_LIMIT in 1m 03s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/19/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/dd7bd8c/dd7bd8c.html : FAILURE in 34m 56s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/19/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/faac258/faac258.html : TIMED_OUT in 3h 03m 14s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"d5c1b59011d72c9f56533c17f05b54cb8d7c4076","author":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"date":"2021-10-06 23:44:00.000000000","message":"Patch Set 19:\n\nTempest test passed in Fujitsu iRMC CI environment.\n\nWe run following Tempest tests\n  - ironic_tempest_plugin.tests.scenario.test_baremetal_basic_ops.BaremetalBasicOps.test_baremetal_server_ops_partition_image\n  - ironic_tempest_plugin.tests.scenario.test_baremetal_basic_ops.BaremetalBasicOps.test_baremetal_server_ops_wholedisk_image\n\nIt\u0027ll take few minutes till log appears in following URL.\n\n\n- fujitsu-irmc-ci-job http://openstackci.jp.fujitsu.com/irmcci/39/797739/19/Fujitsu-iRMC-CI/fujitsu-irmc-ci-job/61c83c7/ : SUCCESS in 1h 10m 19s\n\nAbout this CI test, see https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_iRMC_CI","accounts_in_message":[],"_revision_number":19},{"id":"eeba4547c435bee36910bc902603fba67b6b3465","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-10-07 00:18:46.000000000","message":"Patch Set 19:\n\nBuild succeeded (check pipeline).\n\n- 13G_bios_ipmi https://stash.dellemc-community.org/logs/13G_bios_ipmi/1401 : SUCCESS in 3h 10m 22s\n- 14G_bios_redfish https://stash.dellemc-community.org/logs/14G_bios_redfish/1400 : SUCCESS in 3h 5m 2s\n- R6515_bios_idrac https://stash.dellemc-community.org/logs/R6515_bios_idrac/1393 : SUCCESS in 3h 2m 33s\n- 14G_uefi_idrac https://stash.dellemc-community.org/logs/14G_uefi_idrac/1653 : SUCCESS in 2h 52m 55s\n- XE2420_uefi_redfish https://stash.dellemc-community.org/logs/XE2420_uefi_redfish/1688 : SUCCESS in 2h 4m 33s\n- 14G_uefi_ipmi https://stash.dellemc-community.org/logs/14G_uefi_ipmi/1648 : SUCCESS in 2h 1m 50s\n- 14G_bios_idrac https://stash.dellemc-community.org/logs/14G_bios_idrac/1401 : SUCCESS in 1h 57m 48s\n- R7515_uefi_redfish https://stash.dellemc-community.org/logs/R7515_uefi_redfish/1686 : SUCCESS in 1h 49m 47s\n- 13G_bios_idrac https://stash.dellemc-community.org/logs/13G_bios_idrac/1407 : SUCCESS in 1h 3m 45s\n- 14G_uefi_redfish https://stash.dellemc-community.org/logs/14G_uefi_redfish/1696 : SUCCESS in 59m 9s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":19},{"id":"f24a619af7ad05aa375eae61952d12f0809d192d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-11-18 16:51:26.000000000","message":"Patch Set 20: Commit message was updated.","accounts_in_message":[],"_revision_number":20},{"id":"4a56cd8d9527fa54adb8f06c309f6e6c9cb94a39","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-11-18 16:51:49.000000000","message":"Patch Set 20:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/ironic on branch master.  The error was:\n\n  Unknown projects: opendev.org/openstack/ironic-inspector\n\nThe error appears in the following job stanza:\n\n  job:\n      name: ironic-inspector-tempest-partition-bios-redfish-vmedia\n      description: \"Inspect and deploy ironic node over Redfish virtual media using legacy BIOS boot mode\"\n      parent: ironic-tempest-partition-uefi-redfish-vmedia\n      required-projects:\n        - opendev.org/openstack/ironic-inspector\n      vars:\n        # NOTE(dtantsur): the inspector job includes booting an instance too.\n        # Excluding the abort tests since it hits \"node locked\" too often.\n        tempest_test_regex: test_baremetal_introspection\n        devstack_localrc:\n          IRONIC_BOOT_MODE: bios\n          IRONIC_INSPECTOR_MANAGED_BOOT: True\n          IRONIC_INSPECTOR_NODE_NOT_FOUND_HOOK: \u0027\u0027\n          IRONIC_AUTOMATED_CLEAN_ENABLED: False\n          IRONIC_VM_COUNT: 1\n        devstack_plugins:\n          ironic-inspector: https://opendev.org/openstack/ironic-inspector\n        devstack_services:\n          ironic-inspector: True\n          ironic-inspector-dhcp: True\n\n  in \"openstack/ironic/zuul.d/ironic-jobs.yaml@master\", line 266, column 3\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":20},{"id":"d539ba09fd1afa7ae539cfb10611a3ad4878a4b8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 17:12:21.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/51f5401b59be4deab252bd8bd666ec3f : SUCCESS in 14m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bb23a470f52c48479663b1bc2b51cd07 : SUCCESS in 18m 15s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"d6fd8483e7398114f0282effb50dd8c723240128","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-11-18 18:02:13.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/20/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/bfc5f05/bfc5f05.html : SUCCESS in 54m 42s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/20/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3/e58ae68/e58ae68.html : FAILURE in 29m 01s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/20/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3/650792e/650792e.html : FAILURE in 34m 16s (non-voting)\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/20/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/133aa18/133aa18.html : FAILURE in 31m 49s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/20/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/dee1eea/dee1eea.html : FAILURE in 32m 12s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/20/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/2450fa5/2450fa5.html : FAILURE in 37m 46s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"59503e8fc93aa6d7ecad5710454284c6d9d825ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 18:39:00.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/528f728f133d4be8a35c901193beb719 : SUCCESS in 12m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70f29f65062e43e3b81fecfc776602cf : SUCCESS in 13m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b931603185624ddbb3edd4b86f490dd1 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb47c2c3fe5d41f08ff16a3d010aee21 : SUCCESS in 9m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4c0f3fae13d410aa078dc8293d64924 : SUCCESS in 7m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2077725e02cb4a8fb3db17bd33f18bed : SUCCESS in 9m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb41eacb3bc94311903373143855f631 : SUCCESS in 15m 40s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/24ec1d76fec04f8bbee3ba6cced0f890 : SUCCESS in 12m 23s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/9a8784cc8be74500a68d577de182353b : SUCCESS in 24m 56s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/b71820301cdb4346ad3f480717ca1fa6 : FAILURE in 21m 26s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/18c196b288d94248a54d3f16cc206db4 : SUCCESS in 56m 48s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/bfd1f1d3a25b47fdb81eb3f48ef997c7 : SUCCESS in 50m 03s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/22a44ff1aebd49a3983a863ff72b1a2a : FAILURE in 30m 47s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/1f22821b6c054d7ba407ae82dc941749 : SUCCESS in 51m 50s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/32c2fc49c3364dd19ceece4fc7b6ef60 : SUCCESS in 1h 02m 47s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/65dcf5091ecb4d26899d77c1a9505b12 : FAILURE in 30m 39s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/4451473897804f92b32c0967c3013c5e : SUCCESS in 1h 37m 21s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/a028d601233f468f8bc7e7d7dc3fbbb6 : FAILURE in 37m 02s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/3a06a6ddb3b54520907f3c87e7c45fdd : SUCCESS in 1h 33m 33s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/537b757ec52c45f8860ff74b294f10c2 : SUCCESS in 1h 15m 50s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/161067cc77f84a449b5cd7dd2bd6f37d : FAILURE in 39m 28s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/f4568c37f31740b09a066feca424badc : SUCCESS in 49m 54s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/965c899f004e4b57b1e64cf8c228f410 : SUCCESS in 1h 28m 45s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/1efed42a922d4a54a916ef963ebbb2bb : SUCCESS in 6m 34s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/885dad0b9d5c4ffd8c563f4f9f1f2235 : FAILURE in 51m 17s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/4c44f21156584b76a6bbbe6055c748b9 : SUCCESS in 55m 06s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/544c17675edb4082a1ab7d4560d1b70c : FAILURE in 35m 56s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/e2a9a963e15b486e80eab1071dcaab97 : SUCCESS in 50m 49s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/5a9cb3b6679644fe9f85a796cea898d9 : FAILURE in 34m 32s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a7d28e57f69f4cdcab708c1313a58227 : SUCCESS in 44m 43s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/8849203d2c674fb99705199ff2980ee5 : SUCCESS in 48m 08s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/853b71516e934fb998e28cafb712f78c : SUCCESS in 56m 13s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/07a7324b2e4d4ca58584e021a50bdb5d : FAILURE in 44m 39s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"ee900c669cb14c5acc0ca747ba16725075b4e7a8","author":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"date":"2021-11-18 23:17:13.000000000","message":"Patch Set 20:\n\nTempest test passed in Fujitsu iRMC CI environment.\n\nWe run following Tempest tests\n  - ironic_tempest_plugin.tests.scenario.test_baremetal_basic_ops.BaremetalBasicOps.test_baremetal_server_ops_partition_image\n  - ironic_tempest_plugin.tests.scenario.test_baremetal_basic_ops.BaremetalBasicOps.test_baremetal_server_ops_wholedisk_image\n\nIt\u0027ll take few minutes till log appears in following URL.\n\n\n- fujitsu-irmc-ci-job http://openstackci.jp.fujitsu.com/irmcci/39/797739/20/Fujitsu-iRMC-CI/fujitsu-irmc-ci-job/44574ab/ : SUCCESS in 1h 02m 21s\n\nAbout this CI test, see https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_iRMC_CI","accounts_in_message":[],"_revision_number":20},{"id":"0b46945e8b7fe3edef1b054580a8d88f8f7904f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-11-22 23:03:54.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"9da9e919fec0d5c25dda60754ad6aa5404a312b9","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-11-22 23:04:12.000000000","message":"Patch Set 21:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/ironic on branch master.  The error was:\n\n  Unknown projects: opendev.org/openstack/ironic-inspector\n\nThe error appears in the following job stanza:\n\n  job:\n      name: ironic-inspector-tempest-partition-bios-redfish-vmedia\n      description: \"Inspect and deploy ironic node over Redfish virtual media using legacy BIOS boot mode\"\n      parent: ironic-tempest-partition-uefi-redfish-vmedia\n      required-projects:\n        - opendev.org/openstack/ironic-inspector\n      vars:\n        # NOTE(dtantsur): the inspector job includes booting an instance too.\n        # Excluding the abort tests since it hits \"node locked\" too often.\n        tempest_test_regex: test_baremetal_introspection\n        devstack_localrc:\n          IRONIC_BOOT_MODE: bios\n          IRONIC_INSPECTOR_MANAGED_BOOT: True\n          IRONIC_INSPECTOR_NODE_NOT_FOUND_HOOK: \u0027\u0027\n          IRONIC_AUTOMATED_CLEAN_ENABLED: False\n          IRONIC_VM_COUNT: 1\n        devstack_plugins:\n          ironic-inspector: https://opendev.org/openstack/ironic-inspector\n        devstack_services:\n          ironic-inspector: True\n          ironic-inspector-dhcp: True\n\n  in \"openstack/ironic/zuul.d/ironic-jobs.yaml@master\", line 273, column 3\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":21},{"id":"f7e29ae1f7ae4de64658f0cf1c3d93f758ef840a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 23:20:43.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c17d43ae3fef4e5eac6774d41621003e : SUCCESS in 10m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/293cca9affa54494a90bd2564800e7ee : SUCCESS in 11m 18s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"3f3600b1be2dac37bdb19ee965f1fc79d4ecbe16","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-11-23 00:44:28.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/21/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/b1d04fd/b1d04fd.html : SUCCESS in 49m 21s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/21/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3/dd8be03/dd8be03.html : SUCCESS in 40m 46s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/21/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3/b3dc354/b3dc354.html : FAILURE in 31m 53s (non-voting)\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/21/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/ec71ef6/ec71ef6.html : SUCCESS in 39m 28s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/21/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/88cbb6f/88cbb6f.html : SUCCESS in 37m 26s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/21/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/3156a41/3156a41.html : SUCCESS in 39m 34s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"e13612c4888d7bb1accf6b6394139d9b274a54eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 01:07:19.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3aced7860ef42d8a19139635a7499bb : SUCCESS in 9m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/26bf63754d6246c4be20af8822a66425 : SUCCESS in 12m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/808082acb9834a0fac5ec83ae8cfa1a0 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15e3ab9d83ec47c18188b25a22bd0c67 : SUCCESS in 8m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df60ded47baa43869455f02b50a40424 : SUCCESS in 8m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd571f4e8e5f4f1585de61f629df217b : SUCCESS in 9m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fddfa7f32229480ba422659cc979d0c4 : SUCCESS in 15m 11s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/e3623824800c48a089c7df649c97c8c0 : SUCCESS in 11m 17s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/d88042859d0340f999db2fcd9cb6bb60 : SUCCESS in 46m 34s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/2910dc9b301245a680ddaec588908694 : SUCCESS in 20m 16s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/b3db9e5f6e084ff8bf6d9e54021bbfc0 : SUCCESS in 1h 03m 01s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/75662bdf28104338afac8e3ef069537b : SUCCESS in 37m 02s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/a258451b0eb0498f8be6cee86be4165f : FAILURE in 31m 10s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/21846882a80c40fdb10417f95616363b : SUCCESS in 55m 39s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/7763703498a441afa455f8fe53fdef5f : SUCCESS in 45m 50s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/ab3451ccfccd456a925a5b60fab2e061 : SUCCESS in 40m 27s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/e57fad89463142e9a2e5bd0df91574fe : SUCCESS in 1h 08m 11s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/9584813c05334423a2e82846dd1108f3 : FAILURE in 34m 16s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/c3f03010efb740849c95e38d4cb9914e : FAILURE in 1h 57m 49s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/4814ed2216914820900b68e8d34fd5ac : SUCCESS in 1h 17m 25s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/ce019a9692ee437bb9fd56ac53e8f348 : SUCCESS in 30m 16s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/ea17de1d246d4d6bb4b0d6f60b21896e : SUCCESS in 1h 07m 39s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/dd64221305b94f51a331d05f7b671ada : SUCCESS in 1h 36m 21s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/f4ee2f26000e4ec1808135cb444a5bc6 : SUCCESS in 3m 47s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/2d84965adddc4e088983935752512bf1 : SUCCESS in 46m 56s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/fea3732794fc45fd9575528795e555b8 : SUCCESS in 46m 48s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/d71aaaa1acce4d6f973b82fa18225a9d : FAILURE in 24m 45s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/b1959822dfdd4590b9d627e21c2b5f8a : SUCCESS in 40m 21s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/804baaa7f3b446c9b3caf5d4b66ebb86 : SUCCESS in 59m 41s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/438e9db79cad432e9fab31793af7111e : SUCCESS in 24m 52s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/d40d8e6c28e0460f92c4afa65cab10f5 : SUCCESS in 39m 37s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/d49359b9057c40d2a3a08b82e924bd7b : SUCCESS in 59m 12s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/971f979f0f764b5ebb354052b4b392a4 : SUCCESS in 27m 52s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"cb515a1f2e50911ee379b7e68c58bd02c3d3b136","author":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"date":"2021-11-23 07:43:20.000000000","message":"Patch Set 21:\n\nTempest test failed in Fujitsu iRMC CI environment.\n\nWe run following Tempest tests\n  - ironic_tempest_plugin.tests.scenario.test_baremetal_basic_ops.BaremetalBasicOps.test_baremetal_server_ops_partition_image\n  - ironic_tempest_plugin.tests.scenario.test_baremetal_basic_ops.BaremetalBasicOps.test_baremetal_server_ops_wholedisk_image\n\nIt\u0027ll take few minutes till log appears in following URL.\n\n\n- fujitsu-irmc-ci-job http://openstackci.jp.fujitsu.com/irmcci/39/797739/21/Fujitsu-iRMC-CI/fujitsu-irmc-ci-job/b500b62/ : FAILURE in 42m 58s\n\nAbout this CI test, see https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_iRMC_CI","accounts_in_message":[],"_revision_number":21},{"id":"7890033b7846c741399e2da7ffe00dccdf0f354a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-11-23 14:39:00.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"779dbf9b9a75db4fbbf1d7f5b14197cfd251cc87","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-11-23 14:39:20.000000000","message":"Patch Set 22:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/ironic on branch master.  The error was:\n\n  Unknown projects: opendev.org/openstack/ironic-inspector\n\nThe error appears in the following job stanza:\n\n  job:\n      name: ironic-inspector-tempest-partition-bios-redfish-vmedia\n      description: \"Inspect and deploy ironic node over Redfish virtual media using legacy BIOS boot mode\"\n      parent: ironic-tempest-partition-uefi-redfish-vmedia\n      required-projects:\n        - opendev.org/openstack/ironic-inspector\n      vars:\n        # NOTE(dtantsur): the inspector job includes booting an instance too.\n        # Excluding the abort tests since it hits \"node locked\" too often.\n        tempest_test_regex: test_baremetal_introspection\n        devstack_localrc:\n          IRONIC_BOOT_MODE: bios\n          IRONIC_INSPECTOR_MANAGED_BOOT: True\n          IRONIC_INSPECTOR_NODE_NOT_FOUND_HOOK: \u0027\u0027\n          IRONIC_AUTOMATED_CLEAN_ENABLED: False\n          IRONIC_VM_COUNT: 1\n        devstack_plugins:\n          ironic-inspector: https://opendev.org/openstack/ironic-inspector\n        devstack_services:\n          ironic-inspector: True\n          ironic-inspector-dhcp: True\n\n  in \"openstack/ironic/zuul.d/ironic-jobs.yaml@master\", line 273, column 3\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":22},{"id":"8db422cbe3a9cb7f059c01f6aa110b57cf7295d3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 15:14:08.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/749411ebee0f451eb70d5c1525a5bf55 : SUCCESS in 19m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ba945b4ceb4948b4b6c024bc23e599fc : SUCCESS in 18m 11s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"c546c66655579934b56295a75189999b3b83df7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 15:20:31.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/24c63f1f6a9544178d39770b785c02fc : SUCCESS in 11m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2dd71f322ec341edaba6a9a878f3e77a : SUCCESS in 11m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c22f5f4f87e44ebf98d7670400aac10a : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7730541c1339458ea5c2146669e140a8 : SUCCESS in 10m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4913c625d9d14283a2c04efbe3bc5a3f : SUCCESS in 12m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03a4bd09b60e477fbca790955070fbd8 : SUCCESS in 10m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce8c5d58004c41abbfee196d79ea8a0e : SUCCESS in 14m 44s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/cead60e6e6bd488592cfc6c58c41fdcd : FAILURE in 31m 19s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/9c8905e2b1c04d5089474b99710d5b12 : FAILURE in 35m 45s","accounts_in_message":[],"_revision_number":22},{"id":"cd4fb4ac9881ba2470536945ead05aa2eeeb6a9c","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-11-23 15:40:43.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/22/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/c2a02a6/c2a02a6.html : SUCCESS in 48m 05s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/22/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3/327d4f6/327d4f6.html : SUCCESS in 39m 37s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/22/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3/cb68fab/cb68fab.html : FAILURE in 36m 16s (non-voting)\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/22/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/5303864/5303864.html : SUCCESS in 41m 29s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/22/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/b3360f2/b3360f2.html : SUCCESS in 39m 24s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/22/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/dc5646d/dc5646d.html : SUCCESS in 40m 06s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"ec957aa722cefb2f0f044dbed21ec5109d0b5b70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-11-23 20:32:37.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"68a338721c6ec776a2c1c3056d04d3218a8315e0","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-11-23 20:32:55.000000000","message":"Patch Set 23:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/ironic on branch master.  The error was:\n\n  Unknown projects: opendev.org/openstack/ironic-inspector\n\nThe error appears in the following job stanza:\n\n  job:\n      name: ironic-inspector-tempest-partition-bios-redfish-vmedia\n      description: \"Inspect and deploy ironic node over Redfish virtual media using legacy BIOS boot mode\"\n      parent: ironic-tempest-partition-uefi-redfish-vmedia\n      required-projects:\n        - opendev.org/openstack/ironic-inspector\n      vars:\n        # NOTE(dtantsur): the inspector job includes booting an instance too.\n        # Excluding the abort tests since it hits \"node locked\" too often.\n        tempest_test_regex: test_baremetal_introspection\n        devstack_localrc:\n          IRONIC_BOOT_MODE: bios\n          IRONIC_INSPECTOR_MANAGED_BOOT: True\n          IRONIC_INSPECTOR_NODE_NOT_FOUND_HOOK: \u0027\u0027\n          IRONIC_AUTOMATED_CLEAN_ENABLED: False\n          IRONIC_VM_COUNT: 1\n        devstack_plugins:\n          ironic-inspector: https://opendev.org/openstack/ironic-inspector\n        devstack_services:\n          ironic-inspector: True\n          ironic-inspector-dhcp: True\n\n  in \"openstack/ironic/zuul.d/ironic-jobs.yaml@master\", line 273, column 3\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":23},{"id":"ae262370d77f1eeef8b53a0a891614149ec52566","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 20:53:52.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/260a00e30c18423f98f6a557602909bd : SUCCESS in 17m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2a1839a13c384591926004982c980e38 : SUCCESS in 13m 32s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"f65ebb9048f23a6e2cf35811306378be1f662792","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 21:18:58.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd87be1a6176433ca5c137f7969bc548 : SUCCESS in 9m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/967660fd544c4222854e501a5bf2e743 : SUCCESS in 12m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea9f8b9c729e486e89f560c4c840296a : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/afc7a9e5cce54092ae3f0e082bccf8ce : SUCCESS in 9m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/afc591187b364357803d33cf265bbd6f : SUCCESS in 8m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/115509b3ea5e4e3b95a5130f3f73d967 : SUCCESS in 10m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d28df6f316b40198b21c7f3b1c563cb : SUCCESS in 13m 25s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/8b0688cbbdbd462b9a1b68e5d5b12837 : FAILURE in 40m 16s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/912ea0e6dbbe44958f47cfb77a4085d8 : FAILURE in 43m 11s","accounts_in_message":[],"_revision_number":23},{"id":"977c85b1545e6803282b16497c7cdf34538e6162","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-11-23 23:06:59.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/23/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/96d1848/96d1848.html : SUCCESS in 49m 03s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/23/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3/3e34368/3e34368.html : SUCCESS in 39m 31s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/23/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3/ed1a885/ed1a885.html : FAILURE in 31m 09s (non-voting)\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/23/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/6b7afcd/6b7afcd.html : SUCCESS in 39m 34s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/23/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/0f56f9a/0f56f9a.html : SUCCESS in 38m 10s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/23/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/c73a61c/c73a61c.html : SUCCESS in 41m 11s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"34b5518cf5291cbcd9159227224d715955078da5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-11-23 23:45:33.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"2fc23ce6116a073b3d5fc0bf6ed492ec346c7416","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-11-23 23:45:49.000000000","message":"Patch Set 24:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/ironic on branch master.  The error was:\n\n  Unknown projects: opendev.org/openstack/ironic-inspector\n\nThe error appears in the following job stanza:\n\n  job:\n      name: ironic-inspector-tempest-partition-bios-redfish-vmedia\n      description: \"Inspect and deploy ironic node over Redfish virtual media using legacy BIOS boot mode\"\n      parent: ironic-tempest-partition-uefi-redfish-vmedia\n      required-projects:\n        - opendev.org/openstack/ironic-inspector\n      vars:\n        # NOTE(dtantsur): the inspector job includes booting an instance too.\n        # Excluding the abort tests since it hits \"node locked\" too often.\n        tempest_test_regex: test_baremetal_introspection\n        devstack_localrc:\n          IRONIC_BOOT_MODE: bios\n          IRONIC_INSPECTOR_MANAGED_BOOT: True\n          IRONIC_INSPECTOR_NODE_NOT_FOUND_HOOK: \u0027\u0027\n          IRONIC_AUTOMATED_CLEAN_ENABLED: False\n          IRONIC_VM_COUNT: 1\n        devstack_plugins:\n          ironic-inspector: https://opendev.org/openstack/ironic-inspector\n        devstack_services:\n          ironic-inspector: True\n          ironic-inspector-dhcp: True\n\n  in \"openstack/ironic/zuul.d/ironic-jobs.yaml@master\", line 273, column 3\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":24},{"id":"9756236083b98495af0aeb530d3c26e0110ffcba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 00:02:06.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9d8c7b7540ed4b57b9016ad327b3bc4b : SUCCESS in 14m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9237cf2d691649d584349c09bdccc70f : SUCCESS in 12m 41s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"60ae4d3c801b554d4f99e244f3abec050ce4d2ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 00:28:42.000000000","message":"Patch Set 24: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d04826ed11b49fda5606e53ac48c68a : SUCCESS in 9m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d149889b6ba46359deb2b6bf67d41cb : SUCCESS in 10m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c34f18b731b649b68c13326a04a5b93e : FAILURE in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ebdbff5fe3094fe5bc1b0e8493e12bd8 : SUCCESS in 10m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7717268b1f5468c85954bb7eb3b325d : SUCCESS in 9m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d04847c17b44c14a2852c3a888a4eb8 : SUCCESS in 8m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a0ee7c3f4f14975a0e487a0c142576d : SUCCESS in 14m 06s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/14f67ca8616041ce809ef630a25c4650 : FAILURE in 41m 39s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/66b01426aabf4b4e838047d0b9f64c88 : FAILURE in 29m 19s","accounts_in_message":[],"_revision_number":24},{"id":"5b5549a656d78d66c292aa21fe3dbd5da5244e58","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-11-24 00:39:24.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/24/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/256c0f6/256c0f6.html : SUCCESS in 49m 23s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/24/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3/eaec173/eaec173.html : SUCCESS in 38m 47s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/24/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3/f1ebd3e/f1ebd3e.html : FAILURE in 31m 22s (non-voting)\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/24/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/12f11d3/12f11d3.html : SUCCESS in 39m 59s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/24/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/2b6977c/2b6977c.html : SUCCESS in 37m 50s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/39/797739/24/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/8086129/8086129.html : SUCCESS in 38m 58s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"941b0021556a0c9fe0d30e7f4ef3b7bca5ecafba","author":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"date":"2021-11-24 11:21:26.000000000","message":"Patch Set 24:\n\nTempest test failed in Fujitsu iRMC CI environment.\n\nWe run following Tempest tests\n  - ironic_tempest_plugin.tests.scenario.test_baremetal_basic_ops.BaremetalBasicOps.test_baremetal_server_ops_partition_image\n  - ironic_tempest_plugin.tests.scenario.test_baremetal_basic_ops.BaremetalBasicOps.test_baremetal_server_ops_wholedisk_image\n\nIt\u0027ll take few minutes till log appears in following URL.\n\n\n- fujitsu-irmc-ci-job http://openstackci.jp.fujitsu.com/irmcci/39/797739/24/Fujitsu-iRMC-CI/fujitsu-irmc-ci-job/d55d82d/ : FAILURE in 35m 53s\n\nAbout this CI test, see https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_iRMC_CI","accounts_in_message":[],"_revision_number":24},{"id":"600147e646c310f39ec68c16f0d03b7830897e60","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-01-17 20:14:49.000000000","message":"Topic add_fips_job removed","accounts_in_message":[],"_revision_number":24},{"id":"e4eb5def5a4bebe54ea931e2f5290fd2c2cfa73b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-01-17 20:15:12.000000000","message":"Topic set to fips-compatibility","accounts_in_message":[],"_revision_number":24}],"current_revision_number":24,"current_revision":"02b79d7292ed1d59d3a192dc2b846eca02ab215f","revisions":{"13dbe334d25d6b1750e19d9667a813a9da5289b6":{"kind":"REWORK","_number":1,"created":"2021-06-23 18:55:07.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/39/797739/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/797739/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/797739/1"}}},"commit":{"parents":[{"commit":"eb18f8fce8fdfe5acf6d6525e9a18deebfc10f90","subject":"Merge \"Rename ilo_boot_iso -\u003e boot_iso\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/eb18f8fce8fdfe5acf6d6525e9a18deebfc10f90"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-06-23 18:51:50.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-06-23 18:55:06.000000000","tz":-240},"subject":"DNM/WIP - Add FIPS jobs","message":"DNM/WIP - Add FIPS jobs\n\nThis commit adds jobs to confirm ironic functionality when FIPS is\nenabled.  For simplicity, we have removed the other gate jobs until\nwe get the jobs working, and then the ironic team can decide which\njobs to keep.\n\nChange-Id: If026b9afd0356a4fd1ac0a2157a06b8d5b0b176c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/13dbe334d25d6b1750e19d9667a813a9da5289b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/13dbe334d25d6b1750e19d9667a813a9da5289b6"}]},"branch":"refs/heads/master"},"7a4a1001109df8390091412f5e0c63caeb1fb387":{"kind":"REWORK","_number":2,"created":"2021-07-29 21:31:33.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/39/797739/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/797739/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/797739/2"}}},"commit":{"parents":[{"commit":"b39989f50e0181625e45f81063f4893f3e3fc63d","subject":"Merge \"ci-workarounds/get_extra_logging.yaml : avoid grep failure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/b39989f50e0181625e45f81063f4893f3e3fc63d"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-06-23 18:51:50.000000000","tz":-240},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-07-29 21:31:25.000000000","tz":-420},"subject":"DNM/WIP - Add FIPS jobs","message":"DNM/WIP - Add FIPS jobs\n\nThis commit adds jobs to confirm ironic functionality when FIPS is\nenabled.  For simplicity, we have removed the other gate jobs until\nwe get the jobs working, and then the ironic team can decide which\njobs to keep.\n\nChange-Id: If026b9afd0356a4fd1ac0a2157a06b8d5b0b176c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/7a4a1001109df8390091412f5e0c63caeb1fb387"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/7a4a1001109df8390091412f5e0c63caeb1fb387"}]},"branch":"refs/heads/master"},"e4a8d3f9b864435bb9bc2e83f21fdc5fae03a15b":{"kind":"REWORK","_number":3,"created":"2021-08-01 23:47:54.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/39/797739/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/797739/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/797739/3"}}},"commit":{"parents":[{"commit":"3729682ce640fa325466a64e059dcf413720af30","subject":"Set postgress password encryption for FIPs compliance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/3729682ce640fa325466a64e059dcf413720af30"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2021-08-01 23:47:31.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2021-08-01 23:47:31.000000000","tz":720},"subject":"DNM/WIP - Add FIPS jobs","message":"DNM/WIP - Add FIPS jobs\n\nThis commit adds jobs to confirm ironic functionality when FIPS is\nenabled.  For simplicity, we have removed the other gate jobs until\nwe get the jobs working, and then the ironic team can decide which\njobs to keep.\n\nChange-Id: If026b9afd0356a4fd1ac0a2157a06b8d5b0b176c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/e4a8d3f9b864435bb9bc2e83f21fdc5fae03a15b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/e4a8d3f9b864435bb9bc2e83f21fdc5fae03a15b"}]},"branch":"refs/heads/master"},"eadb84bb3630ff9b448154865a805188e290758c":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-08-03 01:28:32.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/39/797739/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/797739/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/797739/4"}}},"commit":{"parents":[{"commit":"7b0c90cb5b80e7d3f04347db322fb65088de9914","subject":"Set postgress password encryption for FIPS compliance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/7b0c90cb5b80e7d3f04347db322fb65088de9914"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2021-08-01 23:47:31.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2021-08-03 01:17:19.000000000","tz":720},"subject":"DNM/WIP - Add FIPS jobs","message":"DNM/WIP - Add FIPS jobs\n\nThis commit adds jobs to confirm ironic functionality when FIPS is\nenabled.  For simplicity, we have removed the other gate jobs until\nwe get the jobs working, and then the ironic team can decide which\njobs to keep.\n\nChange-Id: If026b9afd0356a4fd1ac0a2157a06b8d5b0b176c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/eadb84bb3630ff9b448154865a805188e290758c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/eadb84bb3630ff9b448154865a805188e290758c"}]},"branch":"refs/heads/master"},"25e3ea6f2d8addc0393ce7402cbc74b1cb09c61e":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-08-03 04:20:25.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/39/797739/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/797739/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/797739/5"}}},"commit":{"parents":[{"commit":"729660b6a3c46d521d236e5d05224eecf1c07320","subject":"Set postgress password encryption for FIPS compliance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/729660b6a3c46d521d236e5d05224eecf1c07320"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2021-08-01 23:47:31.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2021-08-03 04:20:07.000000000","tz":720},"subject":"DNM/WIP - Add FIPS jobs","message":"DNM/WIP - Add FIPS jobs\n\nThis commit adds jobs to confirm ironic functionality when FIPS is\nenabled.  For simplicity, we have removed the other gate jobs until\nwe get the jobs working, and then the ironic team can decide which\njobs to keep.\n\nChange-Id: If026b9afd0356a4fd1ac0a2157a06b8d5b0b176c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/25e3ea6f2d8addc0393ce7402cbc74b1cb09c61e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/25e3ea6f2d8addc0393ce7402cbc74b1cb09c61e"}]},"branch":"refs/heads/master"},"813d89bf13ff78b9f28c507634b82c26ac5e1a67":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-08-03 22:55:36.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/39/797739/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/797739/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/797739/6"}}},"commit":{"parents":[{"commit":"d5c27fc2ad30207ed3c2d4b1e50ddfd9a154f1f1","subject":"Set postgress password encryption for FIPS compliance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/d5c27fc2ad30207ed3c2d4b1e50ddfd9a154f1f1"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2021-08-01 23:47:31.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2021-08-03 21:54:30.000000000","tz":720},"subject":"DNM/WIP - Add FIPS jobs","message":"DNM/WIP - Add FIPS jobs\n\nThis commit adds jobs to confirm ironic functionality when FIPS is\nenabled.  For simplicity, we have removed the other gate jobs until\nwe get the jobs working, and then the ironic team can decide which\njobs to keep.\n\nChange-Id: If026b9afd0356a4fd1ac0a2157a06b8d5b0b176c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/813d89bf13ff78b9f28c507634b82c26ac5e1a67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/813d89bf13ff78b9f28c507634b82c26ac5e1a67"}]},"branch":"refs/heads/master"},"a019e3794ae81862568cdec41d1d84e593d1d88f":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-08-04 01:18:29.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/39/797739/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/797739/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/797739/7"}}},"commit":{"parents":[{"commit":"95ae1c5259024019a97198afdab97fc5d28f257c","subject":"Set postgress password encryption for FIPS compliance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/95ae1c5259024019a97198afdab97fc5d28f257c"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2021-08-01 23:47:31.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2021-08-04 01:18:18.000000000","tz":720},"subject":"DNM/WIP - Add FIPS jobs","message":"DNM/WIP - Add FIPS jobs\n\nThis commit adds jobs to confirm ironic functionality when FIPS is\nenabled.  For simplicity, we have removed the other gate jobs until\nwe get the jobs working, and then the ironic team can decide which\njobs to keep.\n\nChange-Id: If026b9afd0356a4fd1ac0a2157a06b8d5b0b176c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/a019e3794ae81862568cdec41d1d84e593d1d88f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/a019e3794ae81862568cdec41d1d84e593d1d88f"}]},"branch":"refs/heads/master"},"350e7594e891dfc9e5172ea8eb14a06744f4d0e7":{"kind":"REWORK","_number":8,"created":"2021-08-04 20:52:11.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/39/797739/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/797739/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/797739/8"}}},"commit":{"parents":[{"commit":"bf88e7377d924707fb8b45e96090780d8e81da2a","subject":"Set postgress password encryption for FIPS compliance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/bf88e7377d924707fb8b45e96090780d8e81da2a"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2021-08-01 23:47:31.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2021-08-04 20:51:59.000000000","tz":720},"subject":"DNM/WIP - Add FIPS jobs","message":"DNM/WIP - Add FIPS jobs\n\nThis commit adds jobs to confirm ironic functionality when FIPS is\nenabled.  For simplicity, we have removed the other gate jobs until\nwe get the jobs working, and then the ironic team can decide which\njobs to keep.\n\nChange-Id: If026b9afd0356a4fd1ac0a2157a06b8d5b0b176c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/350e7594e891dfc9e5172ea8eb14a06744f4d0e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/350e7594e891dfc9e5172ea8eb14a06744f4d0e7"}]},"branch":"refs/heads/master"},"1e8d1e991e31d97301892c01949dd8deead58e87":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2021-08-04 21:28:45.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/39/797739/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/797739/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/797739/9"}}},"commit":{"parents":[{"commit":"844edec42c5889e9e66bc8a002af8ba72c2fe586","subject":"Set postgress password encryption for FIPS compliance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/844edec42c5889e9e66bc8a002af8ba72c2fe586"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2021-08-01 23:47:31.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2021-08-04 21:28:32.000000000","tz":720},"subject":"DNM/WIP - Add FIPS jobs","message":"DNM/WIP - Add FIPS jobs\n\nThis commit adds jobs to confirm ironic functionality when FIPS is\nenabled.  For simplicity, we have removed the other gate jobs until\nwe get the jobs working, and then the ironic team can decide which\njobs to keep.\n\nChange-Id: If026b9afd0356a4fd1ac0a2157a06b8d5b0b176c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1e8d1e991e31d97301892c01949dd8deead58e87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1e8d1e991e31d97301892c01949dd8deead58e87"}]},"branch":"refs/heads/master"},"c57267073d83c4255ae677b9181c842e45f6cf15":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2021-08-04 23:47:53.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/39/797739/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/797739/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/797739/10"}}},"commit":{"parents":[{"commit":"6af0eb374e5565520937e5b3d8fda086d336fe9d","subject":"Set postgresql password encryption for FIPS compliance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/6af0eb374e5565520937e5b3d8fda086d336fe9d"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2021-08-01 23:47:31.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2021-08-04 23:47:38.000000000","tz":720},"subject":"DNM/WIP - Add FIPS jobs","message":"DNM/WIP - Add FIPS jobs\n\nThis commit adds jobs to confirm ironic functionality when FIPS is\nenabled.  For simplicity, we have removed the other gate jobs until\nwe get the jobs working, and then the ironic team can decide which\njobs to keep.\n\nChange-Id: If026b9afd0356a4fd1ac0a2157a06b8d5b0b176c\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/803530\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/c57267073d83c4255ae677b9181c842e45f6cf15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/c57267073d83c4255ae677b9181c842e45f6cf15"}]},"branch":"refs/heads/master"},"ead4ac584681954793d1da7f22ce8f3dde741941":{"kind":"REWORK","_number":11,"created":"2021-08-05 02:26:31.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/39/797739/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/797739/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/797739/11"}}},"commit":{"parents":[{"commit":"6af0eb374e5565520937e5b3d8fda086d336fe9d","subject":"Set postgresql password encryption for FIPS compliance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/6af0eb374e5565520937e5b3d8fda086d336fe9d"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2021-08-01 23:47:31.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2021-08-05 02:26:18.000000000","tz":720},"subject":"DNM/WIP - Add FIPS jobs","message":"DNM/WIP - Add FIPS jobs\n\nThis commit adds jobs to confirm ironic functionality when FIPS is\nenabled.  For simplicity, we have removed the other gate jobs until\nwe get the jobs working, and then the ironic team can decide which\njobs to keep.\n\nChange-Id: If026b9afd0356a4fd1ac0a2157a06b8d5b0b176c\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/803530\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/ead4ac584681954793d1da7f22ce8f3dde741941"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/ead4ac584681954793d1da7f22ce8f3dde741941"}]},"branch":"refs/heads/master"},"0095eeb81fa63686bc01e83e0413376bce5d52a8":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2021-09-21 14:07:35.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/39/797739/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/797739/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/797739/12"}}},"commit":{"parents":[{"commit":"33f09ad4d634d3940e2285be89e32ef926a83d31","subject":"Merge \"API endpoints to get node history\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/33f09ad4d634d3940e2285be89e32ef926a83d31"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2021-08-01 23:47:31.000000000","tz":720},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-09-21 14:07:35.000000000","tz":0},"subject":"DNM/WIP - Add FIPS jobs","message":"DNM/WIP - Add FIPS jobs\n\nThis commit adds jobs to confirm ironic functionality when FIPS is\nenabled.  For simplicity, we have removed the other gate jobs until\nwe get the jobs working, and then the ironic team can decide which\njobs to keep.\n\nChange-Id: If026b9afd0356a4fd1ac0a2157a06b8d5b0b176c\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/803530\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/0095eeb81fa63686bc01e83e0413376bce5d52a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/0095eeb81fa63686bc01e83e0413376bce5d52a8"}]},"branch":"refs/heads/master","description":"Rebase"},"138e89be4e3b61deae17f3f3f09c6674252897f5":{"kind":"NO_CODE_CHANGE","_number":13,"created":"2021-09-21 19:39:47.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/39/797739/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/797739/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/797739/13"}}},"commit":{"parents":[{"commit":"33f09ad4d634d3940e2285be89e32ef926a83d31","subject":"Merge \"API endpoints to get node history\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/33f09ad4d634d3940e2285be89e32ef926a83d31"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2021-08-01 23:47:31.000000000","tz":720},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-09-21 19:39:40.000000000","tz":0},"subject":"DNM/WIP - Add FIPS jobs","message":"DNM/WIP - Add FIPS jobs\n\nThis commit adds jobs to confirm ironic functionality when FIPS is\nenabled.  For simplicity, we have removed the other gate jobs until\nwe get the jobs working, and then the ironic team can decide which\njobs to keep.\n\nChange-Id: If026b9afd0356a4fd1ac0a2157a06b8d5b0b176c\nDepends-On: https://review.opendev.org/c/openstack/ironic/+/810313\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/803530\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/138e89be4e3b61deae17f3f3f09c6674252897f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/138e89be4e3b61deae17f3f3f09c6674252897f5"}]},"branch":"refs/heads/master","description":"Edit commit message"},"ccde525280e53e71746f5624090ba69fb5fb41e8":{"kind":"REWORK","_number":14,"created":"2021-09-22 00:05:18.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/39/797739/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/797739/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/797739/14"}}},"commit":{"parents":[{"commit":"33f09ad4d634d3940e2285be89e32ef926a83d31","subject":"Merge \"API endpoints to get node history\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/33f09ad4d634d3940e2285be89e32ef926a83d31"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2021-08-01 23:47:31.000000000","tz":720},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-09-22 00:05:09.000000000","tz":-420},"subject":"DNM/WIP - Add FIPS jobs","message":"DNM/WIP - Add FIPS jobs\n\nThis commit adds jobs to confirm ironic functionality when FIPS is\nenabled.  For simplicity, we have removed the other gate jobs until\nwe get the jobs working, and then the ironic team can decide which\njobs to keep.\n\nChange-Id: If026b9afd0356a4fd1ac0a2157a06b8d5b0b176c\nDepends-On: https://review.opendev.org/c/openstack/ironic/+/810313\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/803530\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/ccde525280e53e71746f5624090ba69fb5fb41e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/ccde525280e53e71746f5624090ba69fb5fb41e8"}]},"branch":"refs/heads/master"},"99b3843a9f03e955838a11df6157499ef8917f2c":{"kind":"REWORK","_number":15,"created":"2021-09-22 13:52:34.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/39/797739/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/797739/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/797739/15"}}},"commit":{"parents":[{"commit":"33f09ad4d634d3940e2285be89e32ef926a83d31","subject":"Merge \"API endpoints to get node history\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/33f09ad4d634d3940e2285be89e32ef926a83d31"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2021-08-01 23:47:31.000000000","tz":720},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-09-22 13:51:39.000000000","tz":-420},"subject":"DNM/WIP - Add FIPS jobs","message":"DNM/WIP - Add FIPS jobs\n\nThis commit adds jobs to confirm ironic functionality when FIPS is\nenabled.  For simplicity, we have removed the other gate jobs until\nwe get the jobs working, and then the ironic team can decide which\njobs to keep.\n\nAlso fixes the IRONIC_VM_EMULATOR override to properly oparete on\ndifferent operating systems instead of the global default which was\nfocused on ubuntu/debian.\n\nChange-Id: If026b9afd0356a4fd1ac0a2157a06b8d5b0b176c\nDepends-On: https://review.opendev.org/c/openstack/ironic/+/810313\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/803530\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/99b3843a9f03e955838a11df6157499ef8917f2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/99b3843a9f03e955838a11df6157499ef8917f2c"}]},"branch":"refs/heads/master"},"e18bdc4efeb8c1b5701063ab2655c6fba58e778c":{"kind":"REWORK","_number":16,"created":"2021-09-22 15:52:13.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/39/797739/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/797739/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/797739/16"}}},"commit":{"parents":[{"commit":"fe78dd7687a61463e8bee8fd783d729709af2bf8","subject":"Remove legacy rpm install list and use bindep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/fe78dd7687a61463e8bee8fd783d729709af2bf8"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2021-08-01 23:47:31.000000000","tz":720},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-09-22 15:51:58.000000000","tz":-420},"subject":"DNM/WIP - Add FIPS jobs","message":"DNM/WIP - Add FIPS jobs\n\nThis commit adds jobs to confirm ironic functionality when FIPS is\nenabled.  For simplicity, we have removed the other gate jobs until\nwe get the jobs working, and then the ironic team can decide which\njobs to keep.\n\nAlso fixes the IRONIC_VM_EMULATOR override to properly oparete on\ndifferent operating systems instead of the global default which was\nfocused on ubuntu/debian.\n\nChange-Id: If026b9afd0356a4fd1ac0a2157a06b8d5b0b176c\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/803530\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/e18bdc4efeb8c1b5701063ab2655c6fba58e778c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/e18bdc4efeb8c1b5701063ab2655c6fba58e778c"}]},"branch":"refs/heads/master"},"a5b8d0f8af2d8735d45cfb389246138e3e591a01":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2021-09-24 16:45:12.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/39/797739/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/797739/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/797739/17"}}},"commit":{"parents":[{"commit":"c1e355011c96309fb6b44f1ee00a24232fe756ca","subject":"Remove legacy rpm install list and use bindep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/c1e355011c96309fb6b44f1ee00a24232fe756ca"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2021-08-01 23:47:31.000000000","tz":720},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-09-24 16:45:12.000000000","tz":0},"subject":"DNM/WIP - Add FIPS jobs","message":"DNM/WIP - Add FIPS jobs\n\nThis commit adds jobs to confirm ironic functionality when FIPS is\nenabled.  For simplicity, we have removed the other gate jobs until\nwe get the jobs working, and then the ironic team can decide which\njobs to keep.\n\nAlso fixes the IRONIC_VM_EMULATOR override to properly oparete on\ndifferent operating systems instead of the global default which was\nfocused on ubuntu/debian.\n\nChange-Id: If026b9afd0356a4fd1ac0a2157a06b8d5b0b176c\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/803530\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/a5b8d0f8af2d8735d45cfb389246138e3e591a01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/a5b8d0f8af2d8735d45cfb389246138e3e591a01"}]},"branch":"refs/heads/master","description":"Rebase"},"3bef95c83e2f0b73f7307b6e2e5a87cecce16a8f":{"kind":"REWORK","_number":18,"created":"2021-09-29 21:43:17.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/39/797739/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/797739/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/797739/18"}}},"commit":{"parents":[{"commit":"c1e355011c96309fb6b44f1ee00a24232fe756ca","subject":"Remove legacy rpm install list and use bindep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/c1e355011c96309fb6b44f1ee00a24232fe756ca"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2021-08-01 23:47:31.000000000","tz":720},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-09-29 21:43:11.000000000","tz":0},"subject":"DNM/WIP - Add FIPS jobs","message":"DNM/WIP - Add FIPS jobs\n\nThis commit adds jobs to confirm ironic functionality when FIPS is\nenabled.  For simplicity, we have removed the other gate jobs until\nwe get the jobs working, and then the ironic team can decide which\njobs to keep.\n\nAlso fixes the IRONIC_VM_EMULATOR override to properly oparete on\ndifferent operating systems instead of the global default which was\nfocused on ubuntu/debian.\n\nChange-Id: If026b9afd0356a4fd1ac0a2157a06b8d5b0b176c\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/803530\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/3bef95c83e2f0b73f7307b6e2e5a87cecce16a8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/3bef95c83e2f0b73f7307b6e2e5a87cecce16a8f"}]},"branch":"refs/heads/master"},"2575e5f92a75388fac6d2b631769b825c2897086":{"kind":"REWORK","_number":19,"created":"2021-10-06 18:34:49.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/39/797739/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/797739/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/797739/19"}}},"commit":{"parents":[{"commit":"e082a11174e69f0a731c1319f9dfd551b18ae85e","subject":"Merge \"Make iDRAC management steps verify steps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/e082a11174e69f0a731c1319f9dfd551b18ae85e"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2021-08-01 23:47:31.000000000","tz":720},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-10-06 18:33:20.000000000","tz":-240},"subject":"Add FIPS jobs","message":"Add FIPS jobs\n\nThis commit adds jobs to confirm ironic functionality when FIPS is\nenabled.  This adds the testing of FIPS functionality to two existing\njobs (ironic-standalone-redfish and\nironic-tempest-ipa-partition-uefi-pxe_ipmitool)\n\nAlso fixes the IRONIC_VM_EMULATOR override to properly oparete on\ndifferent operating systems instead of the global default which was\nfocused on ubuntu/debian.\n\nChange-Id: If026b9afd0356a4fd1ac0a2157a06b8d5b0b176c\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/803530\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/2575e5f92a75388fac6d2b631769b825c2897086"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/2575e5f92a75388fac6d2b631769b825c2897086"}]},"branch":"refs/heads/master"},"0a2cc8049d5d5344496919d46c205d34ad7f6957":{"kind":"NO_CODE_CHANGE","_number":20,"created":"2021-11-18 16:51:26.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/39/797739/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/797739/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/797739/20"}}},"commit":{"parents":[{"commit":"e082a11174e69f0a731c1319f9dfd551b18ae85e","subject":"Merge \"Make iDRAC management steps verify steps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/e082a11174e69f0a731c1319f9dfd551b18ae85e"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2021-08-01 23:47:31.000000000","tz":720},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-11-18 16:51:26.000000000","tz":0},"subject":"Add FIPS jobs","message":"Add FIPS jobs\n\nThis commit adds jobs to confirm ironic functionality when FIPS is\nenabled.  This adds the testing of FIPS functionality to two existing\njobs (ironic-standalone-redfish and\nironic-tempest-ipa-partition-uefi-pxe_ipmitool)\n\nAlso fixes the IRONIC_VM_EMULATOR override to properly oparete on\ndifferent operating systems instead of the global default which was\nfocused on ubuntu/debian.\n\nChange-Id: If026b9afd0356a4fd1ac0a2157a06b8d5b0b176c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/0a2cc8049d5d5344496919d46c205d34ad7f6957"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/0a2cc8049d5d5344496919d46c205d34ad7f6957"}]},"branch":"refs/heads/master","description":"Edit commit message"},"343205f05b6019e5b301e6597cf93b11ec9f9621":{"kind":"REWORK","_number":21,"created":"2021-11-22 23:03:54.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/39/797739/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/797739/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/797739/21"}}},"commit":{"parents":[{"commit":"e082a11174e69f0a731c1319f9dfd551b18ae85e","subject":"Merge \"Make iDRAC management steps verify steps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/e082a11174e69f0a731c1319f9dfd551b18ae85e"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2021-08-01 23:47:31.000000000","tz":720},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-11-22 23:03:21.000000000","tz":-480},"subject":"Add FIPS jobs","message":"Add FIPS jobs\n\nThis commit adds jobs to confirm ironic functionality when FIPS is\nenabled.  This adds the testing of FIPS functionality to two existing\njobs (ironic-standalone-redfish and\nironic-tempest-ipa-partition-uefi-pxe_ipmitool)\n\nAlso fixes the IRONIC_VM_EMULATOR override to properly oparete on\ndifferent operating systems instead of the global default which was\nfocused on ubuntu/debian.\n\nChange-Id: If026b9afd0356a4fd1ac0a2157a06b8d5b0b176c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/343205f05b6019e5b301e6597cf93b11ec9f9621"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/343205f05b6019e5b301e6597cf93b11ec9f9621"}]},"branch":"refs/heads/master"},"8844f894e069c38c18b94e960caccd5a7425b520":{"kind":"REWORK","_number":22,"created":"2021-11-23 14:39:00.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/39/797739/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/797739/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/797739/22"}}},"commit":{"parents":[{"commit":"e082a11174e69f0a731c1319f9dfd551b18ae85e","subject":"Merge \"Make iDRAC management steps verify steps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/e082a11174e69f0a731c1319f9dfd551b18ae85e"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2021-08-01 23:47:31.000000000","tz":720},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-11-23 14:38:26.000000000","tz":-480},"subject":"Add FIPS jobs","message":"Add FIPS jobs\n\nThis commit adds jobs to confirm ironic functionality when FIPS is\nenabled.  This adds the testing of FIPS functionality to two existing\njobs (ironic-standalone-redfish and\nironic-tempest-ipa-partition-uefi-pxe_ipmitool)\n\nAlso fixes the IRONIC_VM_EMULATOR override to properly oparete on\ndifferent operating systems instead of the global default which was\nfocused on ubuntu/debian.\n\nChange-Id: If026b9afd0356a4fd1ac0a2157a06b8d5b0b176c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/8844f894e069c38c18b94e960caccd5a7425b520"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/8844f894e069c38c18b94e960caccd5a7425b520"}]},"branch":"refs/heads/master"},"8b6748d47b41313a4926a89e4ed12d183605b2e4":{"kind":"REWORK","_number":23,"created":"2021-11-23 20:32:37.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/39/797739/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/797739/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/797739/23"}}},"commit":{"parents":[{"commit":"e082a11174e69f0a731c1319f9dfd551b18ae85e","subject":"Merge \"Make iDRAC management steps verify steps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/e082a11174e69f0a731c1319f9dfd551b18ae85e"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2021-08-01 23:47:31.000000000","tz":720},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-11-23 20:32:03.000000000","tz":-480},"subject":"Add FIPS jobs","message":"Add FIPS jobs\n\nThis commit adds jobs to confirm ironic functionality when FIPS is\nenabled.  This adds the testing of FIPS functionality to two existing\njobs (ironic-standalone-redfish and\nironic-tempest-ipa-partition-uefi-pxe_ipmitool)\n\nAlso fixes the IRONIC_VM_EMULATOR override to properly oparete on\ndifferent operating systems instead of the global default which was\nfocused on ubuntu/debian.\n\nChange-Id: If026b9afd0356a4fd1ac0a2157a06b8d5b0b176c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/8b6748d47b41313a4926a89e4ed12d183605b2e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/8b6748d47b41313a4926a89e4ed12d183605b2e4"}]},"branch":"refs/heads/master"},"02b79d7292ed1d59d3a192dc2b846eca02ab215f":{"kind":"REWORK","_number":24,"created":"2021-11-23 23:45:33.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/39/797739/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/797739/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/797739/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/797739/24"}}},"commit":{"parents":[{"commit":"e082a11174e69f0a731c1319f9dfd551b18ae85e","subject":"Merge \"Make iDRAC management steps verify steps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/e082a11174e69f0a731c1319f9dfd551b18ae85e"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2021-08-01 23:47:31.000000000","tz":720},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-11-23 23:44:56.000000000","tz":-480},"subject":"Add FIPS jobs","message":"Add FIPS jobs\n\nThis commit adds jobs to confirm ironic functionality when FIPS is\nenabled.  This adds the testing of FIPS functionality to two existing\njobs (ironic-standalone-redfish and\nironic-tempest-ipa-partition-uefi-pxe_ipmitool)\n\nAlso fixes the IRONIC_VM_EMULATOR override to properly oparete on\ndifferent operating systems instead of the global default which was\nfocused on ubuntu/debian.\n\nChange-Id: If026b9afd0356a4fd1ac0a2157a06b8d5b0b176c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/02b79d7292ed1d59d3a192dc2b846eca02ab215f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/02b79d7292ed1d59d3a192dc2b846eca02ab215f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
