)]}'
{"id":"openstack%2Fironic~798665","triplet_id":"openstack%2Fironic~stable%2Fwallaby~Ib4ab755c806f028d97379b80a8c27d6ade63cba1","project":"openstack/ironic","branch":"stable/wallaby","topic":"change-796432-5-stable/wallaby","hashtags":[],"change_id":"Ib4ab755c806f028d97379b80a8c27d6ade63cba1","subject":"Update the clear job id\u0027s constant","status":"MERGED","created":"2021-06-29 13:27:31.000000000","updated":"2021-07-05 15:07:02.000000000","submitted":"2021-07-05 15:04:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"798665-change-796432-5-stable/wallaby","cherry_pick_of_change":796432,"cherry_pick_of_patch_set":6,"meta_rev_id":"fff12594c056a0a62f5aac714d839f20c7da252f","_number":798665,"virtual_id_number":798665,"owner":{"_account_id":31871,"name":"Dhuldev Valekar","email":"dhuldev.valekar@gmail.com","username":"dhuldev30"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-05 15:04:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"recommended":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"all":[{"value":1,"date":"2021-07-02 15:35:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":2,"date":"2021-07-05 13:13:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-01 13:32:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":1,"date":"2021-07-05 13:13:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"}],"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-29 13:28:58.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2021-06-29 13:47:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-29 15:16:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-29 16:45:04.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2021-07-01 13:32:13.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2021-07-02 15:35:20.000000000","updated_by":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"reviewer":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"state":"REVIEWER"},{"updated":"2021-07-05 13:13:07.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"bcfb47509bd4db5029e2ce55371d9d31d12383e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31871,"name":"Dhuldev Valekar","email":"dhuldev.valekar@gmail.com","username":"dhuldev30"},"date":"2021-06-29 13:27:31.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"31246fd32172d7b80ac2b6471b1769ad40f6d4bb","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-06-29 13:28:58.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"6592cda2df03daf9a3762c20902688d223c48645","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 13:47:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8b1f8a60fb8d43999567dffec68d91f5 : SUCCESS in 16m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f84f4298e6e1494f8ba931cb5aaf50f0 : SUCCESS in 18m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"12f01563953b2e320919d7a9beea33c9c5bfa99b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 15:16:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80ffb086aa734d18970ae1f6780edc31 : SUCCESS in 17m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37d0bb242367441ebee0c43c2ddab2fa : SUCCESS in 8m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96571fa6b13945dab0c37b06fc5042a8 : SUCCESS in 12m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03def621ea2c47e89b25039d64e92a57 : SUCCESS in 12m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d755454a95264517af2062d8c15c4c34 : SUCCESS in 16m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/655cd623c79c496397482ab96430f9d5 : SUCCESS in 17m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8fe7f6e2178481bba9a306215aa49c2 : SUCCESS in 11m 30s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/18affe3a32bd4aab9772bd76a570e544 : SUCCESS in 16m 54s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/b293deb2c9664453a07fda2fa9432c18 : SUCCESS in 28m 57s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/3f0895a7f0ee433c9e72ea1829ba1b9b : SUCCESS in 1h 27m 10s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/9ac1967186b04316aaf24242f499ef64 : SUCCESS in 1h 28m 13s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/f7de5835830949c89294cc0d283bd3fe : SUCCESS in 1h 04m 36s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/834851e4166749d68cf2f3044bf2c311 : SUCCESS in 54m 45s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/c4653c291dcc438f9d03bcd27d03c102 : SUCCESS in 1h 08m 36s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/46fdcc08d96645dcb38f50ad97474513 : SUCCESS in 53m 31s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/e5c75034dbf4432483b4cb50cf209e74 : SUCCESS in 1h 13m 29s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/aa3d68fbca4c48a893bb9fe68a729ce1 : SUCCESS in 58m 24s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/c7dd698d553a40028549e55cb597e28b : SUCCESS in 1h 47m 46s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/00c9efef56064378b63eaeec0d06a3b6 : SUCCESS in 1h 29m 45s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/b4ec7784841944f7a3efbfee8bc2436b : SUCCESS in 38m 52s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/c25f371abee444efb3823f6767dc0183 : SUCCESS in 1h 00m 44s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/cd35606a4c4042569c51aaeaa135beb5 : SUCCESS in 1h 16m 44s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/ca094650e5834ec3a1eb4d6c181dd402 : SUCCESS in 9m 24s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/10385e7a605443b3bad9740b601f8084 : SUCCESS in 56m 15s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/4c07ca44ea06498da78ae487312450fb : SUCCESS in 56m 54s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/3a6d215318694dfb86be0a4535e30f93 : SUCCESS in 57m 13s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/2e9ca6fd6b19450f8f7547ef89b63142 : FAILURE in 1h 34m 54s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b948704707844650921dc236f87d0ae8 : SUCCESS in 29m 19s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/d24f6c0cc17847a59f3308a168691cbe : SUCCESS in 29m 10s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/5bd26733beb04255b3003ab6e3fec887 : SUCCESS in 1h 00m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ea8a0c42bc6edf6a079103e592408ead35e6e1f6","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-06-29 16:45:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).\n\n- 14G_bios_idrac_wallaby https://stash.dellemc-community.org/logs/14G_bios_idrac_wallaby/37 : SUCCESS in 3h 16m 51s\n- 14G_uefi_idrac_wallaby https://stash.dellemc-community.org/logs/14G_uefi_idrac_wallaby/37 : SUCCESS in 3h 11m 22s\n- 14G_bios_redfish_wallaby https://stash.dellemc-community.org/logs/14G_bios_redfish_wallaby/37 : SUCCESS in 3h 4m 33s\n- 13G_bios_ipmi_wallaby https://stash.dellemc-community.org/logs/13G_bios_ipmi_wallaby/37 : SUCCESS in 2h 53m 25s\n- 13G_bios_idrac_wallaby https://stash.dellemc-community.org/logs/13G_bios_idrac_wallaby/37 : SUCCESS in 1h 34m 46s\n- XE2420_uefi_redfish_wallaby https://stash.dellemc-community.org/logs/XE2420_uefi_redfish_wallaby/37 : SUCCESS in 1h 34m 17s\n- 14G_uefi_redfish_wallaby https://stash.dellemc-community.org/logs/14G_uefi_redfish_wallaby/37 : SUCCESS in 1h 33m 48s\n- 14G_uefi_ipmi_wallaby https://stash.dellemc-community.org/logs/14G_uefi_ipmi_wallaby/37 : SUCCESS in 1h 30m 13s\n- R7515_uefi_redfish_wallaby https://stash.dellemc-community.org/logs/R7515_uefi_redfish_wallaby/37 : SUCCESS in 1h 24m 6s\n- R6515_bios_idrac_wallaby https://stash.dellemc-community.org/logs/R6515_bios_idrac_wallaby/37 : FAILURE in 4m 38s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"7263341f597ae805ca83fa6fdc45d7b59c7d96f9","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-07-01 13:32:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"41206a4c99b5b39730b01e42f9aca6e110039edb","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2021-07-02 15:35:20.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"64a92ac720733358dd99eca79fd667b265a7755d","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-07-05 13:13:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"960a4fa2550fc461f7c9a43b099d298a7605a220","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 13:13:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e32aa8e3c18399c62372de46ca2e94fbf8da1ac6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 15:04:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12d1f772f1d04b80849c3920581becb8 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d00477960f14c24b1defd3bbd1f6aaf : SUCCESS in 8m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/047f7bd03efe48c48bb147b317ac2522 : SUCCESS in 11m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41255585fd7e4abaa17450ae28e64477 : SUCCESS in 17m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e1970c01d8249be97875685b87a9028 : SUCCESS in 6m 58s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/3fdcd6dacc044688bc1bc165d15c3ec9 : SUCCESS in 12m 23s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/6832b3f99f6142b9afb9b0bc9da2eb69 : SUCCESS in 28m 55s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/41f188349a9f467bad66e37b0e5fdc92 : SUCCESS in 1h 09m 23s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/7458efba5e19444681e0285e5754478f : SUCCESS in 1h 10m 35s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/ad7d202d3493490cacd4083ae5671abe : SUCCESS in 1h 16m 40s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/03591219ab174de89c41b98c666cdb33 : SUCCESS in 49m 27s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/1b16da89bc4442ab892a8a83be9a10c2 : SUCCESS in 52m 08s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/1863a13c1360417fa99450b1ebb3f9cf : SUCCESS in 53m 17s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/1bc3ca134aaa47aabe7f887f3a98877c : SUCCESS in 1h 21m 53s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/684f88478ca7434d9ca1dc717b3980bf : SUCCESS in 57m 27s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/6790b54790404d6fb3015e80ebf7fb1a : SUCCESS in 1h 12m 36s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/f7b2d61a4f67445e99f19c58bffd0186 : SUCCESS in 42m 45s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/a491f7c973c54fe58fad7175bd4a3a08 : SUCCESS in 50m 17s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/a4c0732681294697b6e41cbb07cdbeb5 : SUCCESS in 1h 37m 34s","accounts_in_message":[],"_revision_number":1},{"id":"244fb67a13cb3f0cf608870df34c3e0d169e030a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 15:04:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fff12594c056a0a62f5aac714d839f20c7da252f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 15:07:02.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dc3059d523c4430a9a0861ac33a7b18 : SUCCESS in 1m 17s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/09300a35f5cc4fb09f459e59449fc799 : FAILURE in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e1652c467949643afd7e0def8d617303d0087479","revisions":{"e1652c467949643afd7e0def8d617303d0087479":{"kind":"REWORK","_number":1,"created":"2021-06-29 13:27:31.000000000","uploader":{"_account_id":31871,"name":"Dhuldev Valekar","email":"dhuldev.valekar@gmail.com","username":"dhuldev30"},"ref":"refs/changes/65/798665/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/65/798665/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/65/798665/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/65/798665/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/65/798665/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/65/798665/1"}}},"commit":{"parents":[{"commit":"28d88acba4b1e0ede02a6840380050277b8e728c","subject":"Fix ramdisk boot option handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/28d88acba4b1e0ede02a6840380050277b8e728c"}]}],"author":{"name":"Dhuldev Valekar","email":"dhuldev.valekar@gmail.com","date":"2021-06-15 12:10:12.000000000","tz":-300},"committer":{"name":"Dhuldev Valekar","email":"dhuldev.valekar@gmail.com","date":"2021-06-29 13:27:31.000000000","tz":0},"subject":"Update the clear job id\u0027s constant","message":"Update the clear job id\u0027s constant\n\nFixes an issue of powering off with the ``idrac-wsman`` management\ninterface while the execution of a clear job queue cleaning step is\nproceeding.\nPrior to this fix, the clean step would fail when powering off a node.\n\nStory: 2008988\nTask: 42641\n\nChange-Id: Ib4ab755c806f028d97379b80a8c27d6ade63cba1\n(cherry picked from commit 741a4d8a88bbc51f5903fc00cc4f1decf8f4ea1a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/e1652c467949643afd7e0def8d617303d0087479"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/e1652c467949643afd7e0def8d617303d0087479"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
