)]}'
{"id":"openstack%2Fironic~798666","triplet_id":"openstack%2Fironic~stable%2Fvictoria~Ib4ab755c806f028d97379b80a8c27d6ade63cba1","project":"openstack/ironic","branch":"stable/victoria","topic":"change-796432-5-stable/victoria","hashtags":[],"change_id":"Ib4ab755c806f028d97379b80a8c27d6ade63cba1","subject":"Update the clear job id\u0027s constant","status":"MERGED","created":"2021-06-29 13:30:34.000000000","updated":"2021-07-05 15:07:04.000000000","submitted":"2021-07-05 15:04:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"798666-change-796432-5-stable/victoria","cherry_pick_of_change":796432,"cherry_pick_of_patch_set":6,"meta_rev_id":"a7701a39a938075d458aa5607df9b6bc1b4bbca6","_number":798666,"virtual_id_number":798666,"owner":{"_account_id":31871,"name":"Dhuldev Valekar","email":"dhuldev.valekar@gmail.com","username":"dhuldev30"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-05 15:04:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"recommended":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"all":[{"value":1,"date":"2021-07-02 15:34:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":2,"date":"2021-07-05 13:13:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-01 13:36:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":1,"date":"2021-07-05 13:13:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"}],"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-29 13:32:09.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2021-06-29 14:59:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-29 20:56:46.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2021-07-01 13:36:32.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2021-07-02 15:34:53.000000000","updated_by":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"reviewer":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"state":"REVIEWER"},{"updated":"2021-07-05 13:13:20.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"1039610c4131f28e67f46b77f5db7702eddf490c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31871,"name":"Dhuldev Valekar","email":"dhuldev.valekar@gmail.com","username":"dhuldev30"},"date":"2021-06-29 13:30:34.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c2503e40408d09b34f74a0ad98f718ab1f316195","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-06-29 13:32:09.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"81dc2cff6024e86202252e5cd175b8bee2142e61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 14:59:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e518659fd3c4479480de3e8b813481b4 : SUCCESS in 11m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0e0194bb013484fb45519e6d64b52f9 : SUCCESS in 8m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1dc617ae3a3749129829879666873ff1 : SUCCESS in 10m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8bbc96753834718980d3d4515f105ac : SUCCESS in 10m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/617184ce4e1b4ab8945a7d7634799048 : SUCCESS in 18m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2632e190b134e3080e72ce89273088e : SUCCESS in 9m 06s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/6eec1671b1f4415096491619cf5fbae4 : SUCCESS in 12m 37s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/bfa88df81f2c40efa1d7243b14e164d9 : SUCCESS in 1h 10m 35s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/83f96880d6644d9aa5f1966c42b2d1be : SUCCESS in 1h 20m 16s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/8926407859794ad0807b584ddf61bb86 : SUCCESS in 29m 54s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/abee7036b7ea4986b606785ef2e33e71 : SUCCESS in 1h 13m 23s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/3eb76022cc074d709dbca6ab1661592d : SUCCESS in 56m 32s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/ce441d5251b0420ba3ff055a03bb9433 : SUCCESS in 59m 39s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/34593c90197145e29a5a31b4f43045d0 : SUCCESS in 56m 29s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/3b634bb8a19245e594c0cee2f4007b49 : SUCCESS in 1h 26m 13s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/ab6838ad94c647a2abcd67b12c6e5843 : SUCCESS in 58m 03s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/1ca67a5015da44d18115113e64d1159e : SUCCESS in 1h 27m 35s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/627089adc6e048a588c03a2bc420cc9e : SUCCESS in 1h 16m 52s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/f1c8780fe7634a2d8d34627f8f24aeef : SUCCESS in 46m 06s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/b91864eabf4f407bb5835c2221733695 : FAILURE in 1h 15m 01s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/205b9d732851453cb2bb08b523fdb992 : SUCCESS in 1h 13m 19s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/d0f484d2ed80416b9e28fbc565f88e6c : SUCCESS in 8m 09s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/04d4642230394a4492e404529b537149 : SUCCESS in 58m 51s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/b9aa0b0f0877497fbf3ef50e1a681f78 : SUCCESS in 59m 20s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/37b930bc2e794a04a61d3cc519ff06bd : SUCCESS in 54m 28s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/8792398e97e6452eaa29ce70404a1f52 : SUCCESS in 57m 54s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/659ef083fe074b3d8819d4a07f7ff282 : SUCCESS in 27m 14s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/62fedd9aa632440b9b01e0d5ef414923 : SUCCESS in 56m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e6a1cb6ea51733b7a618a168d8b7ce6514284350","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-06-29 20:56:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).\n\n- 14G_bios_redfish_victoria https://stash.dellemc-community.org/logs/14G_bios_redfish_victoria/49 : SUCCESS in 7h 25m 27s\n- R7515_uefi_redfish_victoria https://stash.dellemc-community.org/logs/R7515_uefi_redfish_victoria/58 : SUCCESS in 7h 11m 38s\n- 13G_bios_ipmi_victoria https://stash.dellemc-community.org/logs/13G_bios_ipmi_victoria/49 : SUCCESS in 6h 53m 32s\n- R6515_bios_idrac_victoria https://stash.dellemc-community.org/logs/R6515_bios_idrac_victoria/49 : SUCCESS in 6h 19m 32s\n- 14G_uefi_idrac_victoria https://stash.dellemc-community.org/logs/14G_uefi_idrac_victoria/60 : SUCCESS in 5h 55m 21s\n- 14G_uefi_redfish_victoria https://stash.dellemc-community.org/logs/14G_uefi_redfish_victoria/62 : SUCCESS in 5h 22m 4s\n- 14G_bios_idrac_victoria https://stash.dellemc-community.org/logs/14G_bios_idrac_victoria/50 : SUCCESS in 4h 43m 5s\n- 13G_bios_idrac_victoria https://stash.dellemc-community.org/logs/13G_bios_idrac_victoria/50 : FAILURE in 4h 32m 45s\n- 14G_uefi_ipmi_victoria https://stash.dellemc-community.org/logs/14G_uefi_ipmi_victoria/59 : SUCCESS in 4h 30m 6s\n- XE2420_uefi_redfish_victoria https://stash.dellemc-community.org/logs/XE2420_uefi_redfish_victoria/58 : SUCCESS in 3h 2m 46s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"e5174c16aad1cfcf2cbedc997aee2c67005b6102","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-07-01 13:36:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4a5c23a57caeb10a94c1ce8d2dfae3b06983c03f","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2021-07-02 15:34:53.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c463363b93fe59c2d2ba55d3a9b5550ea1109f41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 17:33:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e21590ca3504659af897694f4cebbf2 : SUCCESS in 7m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd728eeefa654459b719eb531735c771 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/368f53cc6b12437086eae10b869514b8 : SUCCESS in 7m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45cfdde5f6684a73a35eb5aa1ab87b25 : SUCCESS in 8m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c82c40ba884483280ec3c009afc887b : SUCCESS in 14m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a671e7093c942c3b638375f8ea08d39 : SUCCESS in 6m 42s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/989180af0c9e4876bc9a5b110b969d7f : SUCCESS in 10m 16s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/3103de38961845cbb827514877316c43 : SUCCESS in 1h 07m 32s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/cddebd67baa64a139ac82136c747f7cf : SUCCESS in 1h 11m 50s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/68fd2e8297a64b309b78f0ce7f6689b5 : SUCCESS in 31m 10s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/3d23e6ccade44b319d5ba8938f958b95 : SUCCESS in 1h 14m 49s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/b01921847d564b109a36db1dbcd42a5f : SUCCESS in 49m 13s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/edc00eb276974c18b11d9edf174c3994 : SUCCESS in 56m 36s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/b01adc02121b4e2ab5a522bf5ac6c9c3 : SUCCESS in 54m 43s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/a42c916bfd094c4087a09cf20673c0c9 : SUCCESS in 1h 31m 40s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/2bbd0a06786b42a9819c43f34eeb0fea : SUCCESS in 1h 04m 52s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/abec36a8a76a4f5e89ab6249594199c3 : SUCCESS in 1h 52m 15s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/4a101aed823a4b1f95d328657d2c9184 : SUCCESS in 1h 16m 10s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/5a4eeeae6b7a4071ba3ff336a74eb7bd : SUCCESS in 51m 55s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/84df6fef9e1a426686b764707aadc2f2 : SUCCESS in 1h 04m 08s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/ea68a81d6de94deca61e4bea2ae2eb60 : SUCCESS in 1h 13m 37s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/fa24ef9ded6a40f28702bd1ac09690b8 : SUCCESS in 6m 41s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/79f7165ba709468695c05620fad54e46 : SUCCESS in 49m 00s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/289e4884b7de48f187a65672f99613da : SUCCESS in 55m 34s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/53d98e7ed03f40f7a8e3e1a2dd3a50ab : SUCCESS in 54m 27s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/d009ce2967fc4b75ad5471b717e4ed0d : SUCCESS in 58m 55s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/31e199b421524f14b4538a14e57afd0a : SUCCESS in 30m 05s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/61c22c0c6655449eaa0792c9daaf863c : SUCCESS in 58m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9186157626629bd52a01fb109b62ec4c43dbc264","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-07-05 13:13:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"42b9c6ba133fca14b75976abd3c624c4b970945f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 13:13:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ac64591183fca58b7bc2812641687832c9a1124f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 15:04:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42f32132c23844c68ce1cdd78fc5df64 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0f15dc434834c7ca539d72395237ddf : SUCCESS in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b9b2ea37afb4ad0804ce70a0c5f52ac : SUCCESS in 8m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb0c24f2d1744e1c9f8b52ab656fe437 : SUCCESS in 16m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83ff3dcdaefd464c9d3aaab96470c4d5 : SUCCESS in 7m 16s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/016bd97a8e904b06b85ec4d502e810b4 : SUCCESS in 9m 40s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/27cf3657312c4d17842a2e2623b48d80 : SUCCESS in 1h 20m 33s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/35d7b713c7ac4d47b69d6e4034673a17 : SUCCESS in 1h 20m 26s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/cbb95dc6496949c1bbaef5b83bf6b0fe : SUCCESS in 24m 59s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/5f46053e149c483dbcb2501c3211c7c7 : SUCCESS in 1h 18m 12s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/ebc5b0c5665745388ca73e26e062d251 : SUCCESS in 51m 48s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/c7b7b18923234acea6d44a64cf44527f : SUCCESS in 56m 59s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/60e3f272a8a642e8878645267215cacd : SUCCESS in 49m 03s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/e152a2a4968d4c538a10f2e176bf426d : SUCCESS in 1h 14m 59s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/00f78205829740f899565565348cd79e : SUCCESS in 56m 52s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/d432dc93488740a1878849759904ee97 : SUCCESS in 1h 27m 54s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/92bb7d9432844207b18546b52fc2f164 : SUCCESS in 1h 32m 13s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/1cbbe86aa7194cae9f673e8a77826fe9 : SUCCESS in 43m 03s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/8d15ce83abe745e7a51df97f686d9e1d : SUCCESS in 48m 34s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/3ca4d412453e4505873076843acb27de : SUCCESS in 1h 26m 49s","accounts_in_message":[],"_revision_number":1},{"id":"f925df7c61820dd9ada8d99d045841292e9bbe7d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 15:04:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a7701a39a938075d458aa5607df9b6bc1b4bbca6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 15:07:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f00ee608c7e44bf1a3b911f491a7ca89 : SUCCESS in 1m 20s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38d6a93eb01241b2b5f18f7f9796b976 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4ac6ad7317e20e1f2ebd99b1ce53d3ad2b1f09b0","revisions":{"4ac6ad7317e20e1f2ebd99b1ce53d3ad2b1f09b0":{"kind":"REWORK","_number":1,"created":"2021-06-29 13:30:34.000000000","uploader":{"_account_id":31871,"name":"Dhuldev Valekar","email":"dhuldev.valekar@gmail.com","username":"dhuldev30"},"ref":"refs/changes/66/798666/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/66/798666/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/66/798666/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/66/798666/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/66/798666/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/66/798666/1"}}},"commit":{"parents":[{"commit":"755c75e2e82ee6a7addfac3b7a30fafefb36ab56","subject":"Fix node detail instance_uuid request handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/755c75e2e82ee6a7addfac3b7a30fafefb36ab56"}]}],"author":{"name":"Dhuldev Valekar","email":"dhuldev.valekar@gmail.com","date":"2021-06-15 12:10:12.000000000","tz":-300},"committer":{"name":"Dhuldev Valekar","email":"dhuldev.valekar@gmail.com","date":"2021-06-29 13:30:34.000000000","tz":0},"subject":"Update the clear job id\u0027s constant","message":"Update the clear job id\u0027s constant\n\nFixes an issue of powering off with the ``idrac-wsman`` management\ninterface while the execution of a clear job queue cleaning step is\nproceeding.\nPrior to this fix, the clean step would fail when powering off a node.\n\nStory: 2008988\nTask: 42641\n\nChange-Id: Ib4ab755c806f028d97379b80a8c27d6ade63cba1\n(cherry picked from commit 741a4d8a88bbc51f5903fc00cc4f1decf8f4ea1a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4ac6ad7317e20e1f2ebd99b1ce53d3ad2b1f09b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4ac6ad7317e20e1f2ebd99b1ce53d3ad2b1f09b0"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
