)]}'
{"id":"openstack%2Fironic~798681","triplet_id":"openstack%2Fironic~master~I59996fac059dade0ef186598be1e8971e073eb04","project":"openstack/ironic","branch":"master","topic":"ramdisk-validate","hashtags":[],"change_id":"I59996fac059dade0ef186598be1e8971e073eb04","subject":"Ramdisk: do not require image_source","status":"MERGED","created":"2021-06-29 13:32:12.000000000","updated":"2021-07-02 16:26:09.000000000","submitted":"2021-07-02 16:24:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"798681-ramdisk-validate","meta_rev_id":"13574d6938966235a639c728cd18114824fbf682","_number":798681,"virtual_id_number":798681,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-02 16:24:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":2,"date":"2021-06-30 20:52:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":2,"date":"2021-07-02 14:52:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":1,"date":"2021-07-02 14:52:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-29 13:34:38.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2021-06-29 13:52:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-29 14:01:09.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2021-06-29 16:11:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-30 20:52:38.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"}],"messages":[{"id":"f42ecf93763b25286a3a9a36a8e35c0407150ad2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-06-29 13:32:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1cabd3f914f9ed88b6c0f981fe094f5b327bb1a9","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-06-29 13:34:38.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"f36aa41101e18135e47ce429f6cf78b7cdeeb5b1","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-06-29 13:39:17.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"24946e2233c339eda5b0c0830a3e7731dfc46dd2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 13:52:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/20f42bf895c949f3892c93c70d7b072d : FAILURE in 17m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a30b6d557a5a474991639e092d08cd58 : FAILURE in 18m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9a6470af0d36bfc6d48158648a60a1761311cf9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-06-29 13:53:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c82239f034263fdd511aa2398c875f347a7585ef","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-06-29 13:54:30.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"a65c44ddeccfbe60c0d06d3b936f24dbaadd8fb6","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-06-29 14:01:09.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"72a262e7361d0b2ae48438c7ee220cd3a3ed550e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 14:12:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cc974555a7c94b21b16a276d7d5a9bd6 : FAILURE in 15m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/679f7050ae1442529220bd05915ef7d4 : FAILURE in 17m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"da294b39990530ebfd91768654dafff4d28b68d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-06-29 14:31:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ce0f47ef28642201f418dafa63fb6f2719f628f3","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-06-29 14:33:13.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"c68015238ac5facac56965cf90b66900b02e8d9c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 14:52:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6c1ae66dd98a477e85dbe2e5c86336ff : SUCCESS in 17m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/90c5df719b4c4495ae426b3159aa355d : SUCCESS in 18m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a4381167808d41d355781c28164d6cc783b39281","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 16:11:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0571a01c57a4c0a90724b91f2497ee2 : SUCCESS in 18m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5016626cbeea44c28e20f4f53ffd09cb : SUCCESS in 7m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b933abd91094873b97b3d07852a6c63 : SUCCESS in 11m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f5a060ceeb84ed0a442480fe83d9fd2 : SUCCESS in 11m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e4a7bb707114bc9ba412a5386586cb0 : SUCCESS in 11m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5b29c10cc9d44b08a66f693fe643fad : SUCCESS in 18m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c69ed00dc6d45489c8b662c9897631a : SUCCESS in 9m 14s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/8a3ae6dc72624876b256ad458198f597 : SUCCESS in 14m 10s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/b771caa7c85f4a48b6adc8c393212d3d : SUCCESS in 29m 37s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/5770b9dac7ec45f68dc224860197abb4 : SUCCESS in 1h 14m 45s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/e1f1c3d268954f35a6b2e32cce9bef6a : SUCCESS in 50m 53s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/7b8aba379e0843508d8de179ca8a4f04 : SUCCESS in 51m 07s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/d4814843104c4449a57d90d34f697534 : SUCCESS in 52m 59s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/74186a09d0204ad2b77cc3f2010dce6f : SUCCESS in 55m 43s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/584ac5e77334439e8f6e279e24554019 : SUCCESS in 53m 57s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/7679f30303ff4f3cb529fec54a468ff2 : SUCCESS in 1h 18m 02s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/0cc578485e094c55a4e8df222fd78776 : SUCCESS in 54m 57s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/61d0544debe84ac1b8bec6919d23c2dc : SUCCESS in 1h 33m 44s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/97407feb8ef54176be7cc49ed420bd64 : SUCCESS in 1h 23m 02s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/6aaa1ad2b54b4a328396b984c25b65e1 : SUCCESS in 41m 01s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/f5b56d154f7d4d07bfbde92baa432d1d : SUCCESS in 48m 19s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/2fdea13dd03a49579ee691f5c7d667f8 : SUCCESS in 1h 28m 35s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/27560ca2ffbf4e9ba3b51d4cb9a703c1 : SUCCESS in 7m 12s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/b39abea54bbe44c6be6e9d487544e04b : SUCCESS in 57m 15s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/afbc26afc34440868c57010acd63a3da : SUCCESS in 1h 04m 42s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/ca47264ef9544962880442543d90cff3 : SUCCESS in 51m 22s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/b558c68f69a5411c9949936bde22762d : SUCCESS in 53m 01s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c16d38bde71842e1bfe1a1f23ac18670 : SUCCESS in 29m 30s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/2ed9728a9c6b4b698d326159faaf8873 : SUCCESS in 27m 54s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/8c284aea276749e48d6e165a855968ec : SUCCESS in 59m 39s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/56e036a0d1b34867b37d8aaecb9c00e0 : SUCCESS in 26m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"77fe15cdb2fd9ece7245e52e98b6c40de2962eea","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2021-06-30 20:52:38.000000000","message":"Patch Set 3: Code-Review+2\n\nThis LGTM; I\u0027m curious if we have CI coverage for the ramdisk deploy interface?","accounts_in_message":[],"_revision_number":3},{"id":"850f70850bb9abdaa1448fb392133627497fd430","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-07-01 12:11:12.000000000","message":"Patch Set 3:\n\n\u003e I\u0027m curious if we have CI coverage for the ramdisk deploy interface?\n\nWe do, in one of the standalone jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9b5bca89778c36a44266bd3cb86dcb9ef54585a9","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-07-02 14:52:54.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b80c8ff67ad2692c10cb414b42eec034ba679003","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 14:53:16.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9874ee70cd98b47bfc212ad512a127f17744201e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 16:24:14.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47119763ded84874b7bd022cf5ce8b0f : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e93500295feb411d82d3dcea28bc3c13 : SUCCESS in 10m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60107e5b3aca4f9c8ae634c545f1477d : SUCCESS in 10m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a96c47f05e814305a823d292a0f022ea : SUCCESS in 16m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49d968dcadff487b927c048c3a94197c : SUCCESS in 8m 40s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/0d5d3a6378db48ee8288b1b0fc926768 : SUCCESS in 29m 47s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/6a7d8247f0b14ec98411cd0bf845814c : SUCCESS in 24m 14s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/e7a38e3582df49c1bf827981117f5596 : SUCCESS in 1h 16m 35s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/46197a6f38b04b5e919c3e909dd758c7 : SUCCESS in 52m 55s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/c8277d763cd64b1c8921aeaed011137f : SUCCESS in 44m 10s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/3fb04b0b488c4373a900ebdb90e3e773 : SUCCESS in 52m 09s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/366b6906eb9e410ca6780cf710e148a7 : SUCCESS in 1h 02m 29s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/96caccc072614c518bff64f06fdb9afa : SUCCESS in 53m 32s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/69566cd24d5d4ee191d71b60e21008d4 : SUCCESS in 1h 09m 06s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/d4bdba74d17744feb5e0ef2017f920c3 : SUCCESS in 54m 55s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/24e63f481cc04b738ddfa1f7fa267f02 : SUCCESS in 1h 24m 19s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/64d00cdbaf774713b53e0b53c57251bc : SUCCESS in 46m 24s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/2dda1b87ebcf4946bdac21a34c831b27 : SUCCESS in 47m 39s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/3d22d718e7994538874299715dfca233 : SUCCESS in 1h 27m 35s","accounts_in_message":[],"_revision_number":3},{"id":"6a2dee97b0c24885f07eb103847541073b86a19f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 16:24:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"13574d6938966235a639c728cd18114824fbf682","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 16:26:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61f64edda3934bdea91dcf1ada7eb4bf : SUCCESS in 1m 30s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af759895002c4f2b8cc460fb3e9a3ccc : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d8ccfabda8649d82a28daa96fbd5a2891c0dea86","revisions":{"0a12a09a7baf51d4ab15c225dea4625db275a40b":{"kind":"REWORK","_number":1,"created":"2021-06-29 13:32:12.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/81/798681/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/81/798681/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/81/798681/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/81/798681/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/81/798681/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/81/798681/1"}}},"commit":{"parents":[{"commit":"1d6441cc347cfe984721f34ebb0cd64fd9e4d876","subject":"Refactor deploy_utils.validate_image_properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1d6441cc347cfe984721f34ebb0cd64fd9e4d876"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-06-23 14:37:18.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-06-29 13:30:51.000000000","tz":120},"subject":"Ramdisk: do not require image_source","message":"Ramdisk: do not require image_source\n\nBecause of the way validation works, the ramdisk deploy interface\ncurrently requires an image_source in addition to kernel/ramdisk.\nAfter 1d6441cc347cfe984721f34ebb0cd64fd9e4d876 it is no longer\nnecessary, and this change removes this requirement.\n\nChange-Id: I59996fac059dade0ef186598be1e8971e073eb04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/0a12a09a7baf51d4ab15c225dea4625db275a40b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/0a12a09a7baf51d4ab15c225dea4625db275a40b"}]},"branch":"refs/heads/master"},"8ae6ca4704efcdb1a1a77d0c291ff154c7fb4c76":{"kind":"REWORK","_number":2,"created":"2021-06-29 13:53:07.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/81/798681/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/81/798681/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/81/798681/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/81/798681/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/81/798681/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/81/798681/2"}}},"commit":{"parents":[{"commit":"1d6441cc347cfe984721f34ebb0cd64fd9e4d876","subject":"Refactor deploy_utils.validate_image_properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1d6441cc347cfe984721f34ebb0cd64fd9e4d876"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-06-23 14:37:18.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-06-29 13:52:27.000000000","tz":120},"subject":"Ramdisk: do not require image_source","message":"Ramdisk: do not require image_source\n\nBecause of the way validation works, the ramdisk deploy interface\ncurrently requires an image_source in addition to kernel/ramdisk.\nAfter 1d6441cc347cfe984721f34ebb0cd64fd9e4d876 it is no longer\nnecessary, and this change removes this requirement.\n\nChange-Id: I59996fac059dade0ef186598be1e8971e073eb04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/8ae6ca4704efcdb1a1a77d0c291ff154c7fb4c76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/8ae6ca4704efcdb1a1a77d0c291ff154c7fb4c76"}]},"branch":"refs/heads/master"},"d8ccfabda8649d82a28daa96fbd5a2891c0dea86":{"kind":"REWORK","_number":3,"created":"2021-06-29 14:31:50.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/81/798681/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/81/798681/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/81/798681/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/81/798681/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/81/798681/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/81/798681/3"}}},"commit":{"parents":[{"commit":"1d6441cc347cfe984721f34ebb0cd64fd9e4d876","subject":"Refactor deploy_utils.validate_image_properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1d6441cc347cfe984721f34ebb0cd64fd9e4d876"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-06-23 14:37:18.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-06-29 14:31:37.000000000","tz":120},"subject":"Ramdisk: do not require image_source","message":"Ramdisk: do not require image_source\n\nBecause of the way validation works, the ramdisk deploy interface\ncurrently requires an image_source in addition to kernel/ramdisk.\nAfter 1d6441cc347cfe984721f34ebb0cd64fd9e4d876 it is no longer\nnecessary, and this change removes this requirement.\n\nChange-Id: I59996fac059dade0ef186598be1e8971e073eb04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/d8ccfabda8649d82a28daa96fbd5a2891c0dea86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/d8ccfabda8649d82a28daa96fbd5a2891c0dea86"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
